)]}'
{"id":"openstack%2Fneutron~869012","triplet_id":"openstack%2Fneutron~master~Id85769f165684c9fcf67659dad485e1ecc12331f","project":"openstack/neutron","branch":"master","topic":"keepalived-no_track","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-01-12 23:07:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id85769f165684c9fcf67659dad485e1ecc12331f","subject":"Remove deprecated opt keepalived_use_no_track","status":"MERGED","created":"2023-01-02 19:25:25.000000000","updated":"2023-01-12 23:09:14.000000000","submitted":"2023-01-12 23:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869012-keepalived-no_track","meta_rev_id":"bade709c2f6d9542e5838c895afa59e160ba3ed4","_number":869012,"virtual_id_number":869012,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 23:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-01-03 16:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-01-12 18:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-01-12 18:54:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-02 20:19:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-02 21:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 10:07:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-01-03 10:07:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-01-05 14:21:25.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-12 18:54:27.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"ca8a49fadf85f3d0833689a58426dadec7177ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-02 19:25:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"138eacb1345a008dfe7baf10073c7a0ed887de17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 20:19:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c898ebf13941413d8ef9904c5971e74d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45f87afe86994bf1be5575ef4e16c780 : SUCCESS in 50m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f36621ad0f9f4535859b1c881712bbc7 : SUCCESS in 51m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5742001dce78d0eac7315a16124784b6cb1f6feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 21:27:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87ab529840749eb92f3986fff43b6a7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fd3f7e989e94427d8100d9199529c2e3 : SUCCESS in 1h 11m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/82937be6801343e0ba1a006540350bb7 : SUCCESS in 1h 09m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59ba9e1e61504a7b9eed1852b36a8964 : SUCCESS in 57m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a178c40776df408fa174ba60c90c69c6 : SUCCESS in 58m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e384eb8eb8074e27a3e604cb668010b0 : SUCCESS in 26m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c2e2e139b8649a1b985639412fafa27 : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67611a04af26425683af652bb400c33e : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d79b894fcac0459684523aee3e7de247 : SUCCESS in 30m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42ba0abc7c91402d821848c522dafb4d : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a92137a88c32496e9db100a2833f4c76 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da46a33893a1412db0c75bc694af2aa6 : SUCCESS in 12m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb85a1e140674c7aa277c235ae09a288 : SUCCESS in 47m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6777a8d266074b44adbdfd60d4abe9fd : SUCCESS in 1h 48m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/585968fd8ecb459db726c980965bd8f0 : SUCCESS in 59m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9041c87d3c9f477cb59b7725c0b9d9ec : SUCCESS in 56m 24s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/615a14712d5d49f8a57dae4ebf78fafb : SUCCESS in 1h 03m 46s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/88ddca02ed02457ea777c5d41aa22521 : SUCCESS in 55m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aff452cb36434ba8923b64e0bb44b639 : SUCCESS in 1h 17m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/32f24370ba6e4cc3abc3e7ecc988571a : SUCCESS in 1h 11m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d1f045229fe842c297bab47629fc5297 : SUCCESS in 55m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/35cf496d5bfa4d609658ebfb3ede7a97 : SUCCESS in 1h 33m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b55f916abc864b4da2bcb34d2c1dab63 : SUCCESS in 1h 56m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2abc55177fc404494244470e0034516 : SUCCESS in 1h 09m 21s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9959fb40817a44149602aa124dc67a81 : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8f51f55957b0b1df18931398bc3aed10b3f40901","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-03 10:07:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bc546e9e76c705c6c0cab97068af1b6d1b9e38ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-03 10:07:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac17f5c697d3043a00a9469e5258560249db47b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-03 16:20:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d91b27950cd9dcd75137cc6e24bdc2b820a252d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-03 16:22:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"34ef1fb7ccaa1510deb53d8f914b9ac62d8a515f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:29:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b6f53319324325a2643a295167a6f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9ae7e7a60f64af5b975090760994c3a : SUCCESS in 1h 03m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c833048e23bd4860910242cad3387814 : FAILURE in 1h 05m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"925c21ad41c3bca7d8c68e029909a7026e4546c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 18:26:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e1d687c4d04e68901f82be46bf4fbe\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/96071c62deb44eafa641d3b208f0551c : SUCCESS in 1h 07m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69884a532b0b48d3b6bf9d43cc2ca42a : SUCCESS in 1h 06m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4124c94375354d45967f895ea8b7c083 : SUCCESS in 54m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f06e07cffd9b463299624123e8c5f76f : SUCCESS in 1h 35m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/331175c371f446508b577ad136b07a50 : SUCCESS in 25m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d1fc3b4094245d1b896d45301f6a33d : SUCCESS in 42m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e510ec10fb4ec28fb85f6e8fad0458 : SUCCESS in 13m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9e573c8886a42da80551c9dd44e63b2 : SUCCESS in 24m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f63a415871784239b5cebdb97e7d49d8 : SUCCESS in 25m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46c31b3576746909d6d93072760e029 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b8dc355db2147488ebf7d556b9e858b : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/089b7a3949d94d80a7b5cc14f283702c : SUCCESS in 41m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5daaafc479524ea589dc239769da1146 : SUCCESS in 1h 43m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d687600096747328dd8ce031c6425d3 : SUCCESS in 1h 04m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/73e0b643b75e42cf9a23454daf441ebb : SUCCESS in 54m 16s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e1f4eb07917f4090b62a23f2d7a2c6a1 : SUCCESS in 1h 04m 18s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/806761cb4abe483ca22f0e0ba3afc3da : SUCCESS in 1h 15m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/befa607e7f86463ab0b96ae54590c102 : SUCCESS in 1h 12m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/358f2b0b909d4160ab0652ba4f78b0ac : SUCCESS in 1h 05m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ba16fac95ef4446791eaf2eb9fe31a2a : SUCCESS in 1h 01m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a01c2fcf1ef42c89e3f21db722e2a80 : SUCCESS in 1h 22m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/96c572f117144ddca1994af31610e23f : SUCCESS in 1h 59m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dd20d8a266194ea9aa8f5d77065e0b79 : SUCCESS in 1h 22m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/af982985336043148e18807b0f844e18 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":2},{"id":"30a880b6bd108d1fee27559c86387a3d92ac07eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-05 14:21:25.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/12/869012/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c4c23b3c41887fe5ddb0d55b3df547d6cdbb3c80","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-05 18:32:01.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/12/869012/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0e2ae156a81da02dca753a1b7510758072f81105","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-01-12 18:54:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c35a5047ebd9475c5f919f0863f543103aa6d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 18:55:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df3230f6a963561de476d15de719a9e1ad563bed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 23:07:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6977b8eb93ba45a3abaf24dd703a84da\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/bbb3d06304c54720afed556574e4751b : SUCCESS in 1h 30m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cf865fdf20be4f6fbd58653e739d4745 : SUCCESS in 1h 48m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f55060b8fc8b4a67a0cc134f3a3c70d6 : SUCCESS in 1h 42m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/caab83f91484408486eac521f3170779 : SUCCESS in 1h 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d70aae97f5b43c283b93f2b18e592bf : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cd855718a8f4688a72414d62dc5fb7f : SUCCESS in 26m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbedf902f3d642fd84462b23e12a4faa : SUCCESS in 30m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdad98760be34d8b96bcce03f120104a : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48b57576b5874155b6ce54531093d900 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d2c79f6599f4f01825aeed95f60db3f : SUCCESS in 44m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9b496930f5c4e808ab96ecdaffd9e78 : SUCCESS in 1h 56m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/073782b8f5054977bf313a7898aa2b67 : SUCCESS in 1h 24m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bf501d92cb64eb0acb2e4ffd31447d1 : SUCCESS in 1h 04m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/db0795d87e3440b288aac22cea0d1252 : SUCCESS in 1h 16m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ead6eaf00f7c457caae61e910c811727 : SUCCESS in 59m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ae0d97a9a70a4f23b955c18ecd1dd145 : SUCCESS in 1h 20m 20s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/334eb68eab50434a8a326faa3692f9e6 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"85d092e4b1a2c39e872798659be3104c024d0462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 23:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bade709c2f6d9542e5838c895afa59e160ba3ed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 23:09:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c6b2bc48164ae28597e5d746f1eadf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ff5356f69d4fc4a5f732e0746f500a : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a9df623a28749d6bc221ad32c6ba298 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26da863ca629f5974121a2fae4107ca4e98e7322","revisions":{"6c5a8efcf7073b5099043683c354f15f0de8d48b":{"kind":"REWORK","_number":1,"created":"2023-01-02 19:25:25.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/869012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/869012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/869012/1"}}},"commit":{"parents":[{"commit":"e444d543ed5864b77072aea720854b4859a77dda","subject":"Merge \"[Fullstack] use common assert for ncat connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e444d543ed5864b77072aea720854b4859a77dda"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-01-02 16:20:04.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-02 19:24:46.000000000","tz":0},"subject":"Remove deprecated opt keepalived_use_no_track","message":"Remove deprecated opt keepalived_use_no_track\n\nThe keepalived_use_no_track config option is\ndeprecated and it\u0027s determined by supported by\na runtime check instead.\n\nChange-Id: Id85769f165684c9fcf67659dad485e1ecc12331f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c5a8efcf7073b5099043683c354f15f0de8d48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c5a8efcf7073b5099043683c354f15f0de8d48b"}]},"branch":"refs/heads/master"},"26da863ca629f5974121a2fae4107ca4e98e7322":{"kind":"REWORK","_number":2,"created":"2023-01-03 16:20:43.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/869012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/869012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/869012/2"}}},"commit":{"parents":[{"commit":"e444d543ed5864b77072aea720854b4859a77dda","subject":"Merge \"[Fullstack] use common assert for ncat connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e444d543ed5864b77072aea720854b4859a77dda"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-01-02 16:20:04.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-03 16:20:36.000000000","tz":0},"subject":"Remove deprecated opt keepalived_use_no_track","message":"Remove deprecated opt keepalived_use_no_track\n\nThe keepalived_use_no_track config option is\ndeprecated and it\u0027s determined by supported by\na runtime check instead.\n\nChange-Id: Id85769f165684c9fcf67659dad485e1ecc12331f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26da863ca629f5974121a2fae4107ca4e98e7322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26da863ca629f5974121a2fae4107ca4e98e7322"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
