)]}'
{"id":"openstack%2Fneutron~869049","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ie9e2170c761c9a10f3daa991c3fb77f304bb07e2","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1742401-stable/ussuri","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-04 23:52:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9e2170c761c9a10f3daa991c3fb77f304bb07e2","subject":"[Fullstack] SG test - first wait until all vms can ping each other","status":"MERGED","created":"2023-01-03 15:25:44.000000000","updated":"2023-01-04 23:53:23.000000000","submitted":"2023-01-04 23:52:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869049-bug/1742401-stable/ussuri","cherry_pick_of_change":737741,"cherry_pick_of_patch_set":1,"meta_rev_id":"dc85c2deceea2de9a43050c5857a9527632c652b","_number":869049,"virtual_id_number":869049,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 23:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-01-04 14:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-01-04 05:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-01-03 19:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-03 21:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-01-04 14:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 17:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 19:10:26.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-01-03 21:14:28.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-01-04 05:47:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-01-04 14:32:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"8eb9675fac69f2ebee7f10f2ea23aff633629e14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-03 15:25:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fe00f4e7c8f2e6b223cb53f3954ca22c415c24ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:34:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/388f95d491f9448eb193caf36fa02d4a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91a1c1199c2f46268e3602c31e49ff77 : SUCCESS in 31m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b7daaea7f34611ae2bede7f93d55aa : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99e935a5154e48399c4ec55588680a4e : SUCCESS in 28m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03b7174974544140aab7ddd14e17ebbe : SUCCESS in 34m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11ed68ece4ea4c7888dc741150fb610f : SUCCESS in 27m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cd6ce4c79d44c2892aafd510212bab : SUCCESS in 10m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e4e745106144443387887bbaa56a9b11 : SUCCESS in 43m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4d2066d6704f4e2bb87e7376df39c78f : SUCCESS in 1h 19m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3da66e63188d47a7803049e900c5e05e : SUCCESS in 1h 22m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/92d9435084c9419e8a156dfb95be025a : SUCCESS in 1h 17m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7eed64ce6754a4b9110d723492d127c : SUCCESS in 1h 01m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1e28cb27e9064006b4a61f70e0d9f0bb : SUCCESS in 1h 10m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0bb641782fc1453f97d94fb7fb44cf2d : SUCCESS in 56m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc3efae4e4954cfe81fb30539018808a : SUCCESS in 1h 08m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/48b8e3d4195f4bd4a8f575b196393d25 : SUCCESS in 1h 05m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1090ce042dd64409b02549d9d94efafb : SUCCESS in 1h 20m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94d001ceec6a4a6bbf0ecfe5518c2550 : SUCCESS in 1h 31m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de690a7c78734dda8e7e8be5b37c3fa9 : SUCCESS in 1h 07m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/68819f7012324254940476a61dc59c87 : SUCCESS in 40m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4aba34349da54f3ab352dbfe96b81651 : SUCCESS in 1h 43m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdf50145d64a49778517e0853f11c8b6 : SUCCESS in 1h 01m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/521a7c1196604eefbf06e1f190f4d5ed : SUCCESS in 1h 16m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/14960dfef9d2495380d647ab6b5897ff : SUCCESS in 36m 12s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/68ce90c188ea42e8a5dc87324a9626ea : FAILURE in 23m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d94c668eaf59490c81f4a57ea9e0eded : SUCCESS in 1h 00m 24s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ace0c68bd2264b36964a4ebf46e5c1c5 : SUCCESS in 2h 03m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fc87ecc278a247038239538e6e3aa29a : FAILURE in 1h 09m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ad1ed355fdda2c1f0386c4f3ef7d8aea71ab850","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-03 19:10:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c13de7917347ea647d45d6c901c0b2d048b74b36","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-01-03 21:14:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"308f8b1198acdbc437646438b2fd4595f50a3b7c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-04 05:47:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c726a95af2eb33aec25bfc0a2b3e486269817b50","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-04 14:32:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f35890a5b7133f6416ee82a139637b571a2bf58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 14:33:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3455930ebf19040ba8e4f612fb9242cbb500d88c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 16:18:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d708f9ac94f44c958dbcceb9f65b668a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160d05b5b0e147a7aa513fe4c1529991 : SUCCESS in 13m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/270b135cd68b423caa202b72ce8bf832 : SUCCESS in 27m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6af6223198ad4d23965d65c3842f9752 : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d66a17e8bf6475d8866bf24a2efc984 : SUCCESS in 10m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/59c4632d32c8465ba641c48a83a93992 : FAILURE in 42m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ed15b1c0ea7d4974bcbccb1eb85722fc : FAILURE in 1h 08m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a86fd0622d94f65be307c9acfd2464c : SUCCESS in 57m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/71a717043c3c4e999a118f712be8d31a : SUCCESS in 1h 17m 44s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/83860f65d20e4486a9710de25c407414 : SUCCESS in 1h 11m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee9647134c594773ba94f653ff68f300 : SUCCESS in 1h 03m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/38521330759c4c1086c97c5ff238b752 : SUCCESS in 1h 09m 40s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7213ec8bc9714ed187978204afbd5ec9 : SUCCESS in 1h 34m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c9421ec74f04752981a391f649ecd1c : SUCCESS in 1h 02m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/94d08cebc2854e118c618874737f294d : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a171e28cf8dc9389ef2fb1485163b39a5a0691ff","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-04 20:31:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6305c943d215601b3e051a90836b00ebc35cd090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 22:22:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cc69168fc6d4ec58ea7ca413c6da115\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dab5685770fd45f183c7ff67d9a460af : SUCCESS in 34m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23236d6fc39d46179aa4cf90bc9f0133 : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29daa18ab61f41e7b0b12a4442e95462 : SUCCESS in 20m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/141825711e9941dcb5070af2078b79f2 : SUCCESS in 44m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24283bde25a0427c9dc893192075aa71 : SUCCESS in 22m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f46434867474cdba9c71ccb11bdd823 : SUCCESS in 9m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23e24c8f1d4c4111b8e6274e1020f586 : SUCCESS in 56m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c5b8b370d9054840b3f66f8c00050496 : SUCCESS in 1h 19m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69de0df460484cc9a5d45f42dff6ec12 : SUCCESS in 1h 11m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0aefb05a240b4e6382644547adf1e32e : SUCCESS in 1h 11m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fcc87134df8649fc9e0cb84875ee26c5 : SUCCESS in 1h 02m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b040b139870d4180ada32507c27cc87d : SUCCESS in 1h 00m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bea91e70607440f6a8bf59e5af1fc0fa : SUCCESS in 1h 04m 30s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/866e8437a545446d8e06f00adbb81dab : SUCCESS in 56m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7837b75ee61e44cfb6779bee6893a286 : SUCCESS in 1h 02m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1508e137acb24ff684945153f7f4ad31 : SUCCESS in 1h 23m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1217d79a98d4e19854af70b85a6ccf3 : SUCCESS in 1h 08m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98a30c7e7fc445339fc70bb0d5c65e8d : SUCCESS in 1h 19m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/24fe51ffa4ae4ac1b29e153cb8ddd08f : SUCCESS in 46m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aaf516d547b94129a18f24cd3a6c717b : SUCCESS in 1h 45m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47b4dac14abe49208651eabc66bb1204 : SUCCESS in 1h 00m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77b2f29cef1742d0b4158bc8e4a222e2 : SUCCESS in 56m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2c10e06922084b5a84cbba8e9ad613e0 : SUCCESS in 33m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/73bb30ef21d24975929f01d5f4d0b8c3 : FAILURE in 28m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd7e8b2602854f0bab8e12a9d7901136 : SUCCESS in 50m 50s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/373ad54449a74d42856c53ece8f2a98e : SUCCESS in 1h 28m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/528c9f06f0064634bc9159eb6ae6205f : FAILURE in 1h 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cba4ae50f57b0af46b4a5ccc85a2d14fc13b0437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 22:23:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e84883e127db6c2a4ad24f448d35cace7a43da83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 23:52:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54be73cd77de4c64bba99faabf323002\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f94603b713a4879b8bbb274e72da9b6 : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5983ed762fb4992a55408cc1d1fab10 : SUCCESS in 25m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39e474a058324b91a816235bb6fc50bd : SUCCESS in 26m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa88ce6f3abf4fb58d806f766f42e115 : SUCCESS in 9m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f762e2c593ff435199477fea13b7849e : SUCCESS in 56m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1524bb641f324fb98623612440aa2578 : SUCCESS in 1h 16m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bad7a2129b8c494b86cfd31f79fdb276 : SUCCESS in 1h 08m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f27f4daf15d540ee85984cb2850eb369 : SUCCESS in 1h 08m 18s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d59ccfd1e82b4bc78b5d10efb3692609 : SUCCESS in 1h 04m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eeca0fcbbeef4908b8742866a938c68f : SUCCESS in 1h 13m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e348174e562f461580bf2924386f8d84 : SUCCESS in 50m 00s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5573b7f90f0465ca5aeafd6c5699ff8 : SUCCESS in 1h 26m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cefa6f382064d7cbd49376180f8bb00 : SUCCESS in 1h 00m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/337fc1d319d641218a353eff845a03d4 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b60484ff184e322df876c1f8a3ee81857a11d241","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 23:52:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc85c2deceea2de9a43050c5857a9527632c652b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 23:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b335b9aa46442e7808f103c878932c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff93728623e45a7999e2cba3c31796d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4514641e9baa4b3f9856acb5847d6e639389822","revisions":{"a4514641e9baa4b3f9856acb5847d6e639389822":{"kind":"REWORK","_number":1,"created":"2023-01-03 15:25:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/869049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/869049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/869049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/869049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/869049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/869049/1"}}},"commit":{"parents":[{"commit":"e99d5b66ef00a6e16038cea5792d0b2ab78aa8f5","subject":"Limit tox version to \u003c4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e99d5b66ef00a6e16038cea5792d0b2ab78aa8f5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-24 10:32:14.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-01-03 15:25:44.000000000","tz":0},"subject":"[Fullstack] SG test - first wait until all vms can ping each other","message":"[Fullstack] SG test - first wait until all vms can ping each other\n\nIn fullstack security group test, after fake VMs are created there\ncheck if connectivity to some custom port is working fine with port\nsecurity disabled.\nAfter that there is called \"block_until_ping\" method for each vms.\n\nThis patch changes that to first wait if we can ping vms and later do\nnetcat tests.\nEven if that will not solve problems with failures of this test, we\nmay know more if the issue is caused by netcat or it\u0027s just no\nICMP connectivity between VMs at all.\n\nChange-Id: Ie9e2170c761c9a10f3daa991c3fb77f304bb07e2\nRelated-Bug: #1742401\n(cherry picked from commit 1e9a3bffd2cf565568ac479710104cd3a4cdae53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4514641e9baa4b3f9856acb5847d6e639389822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4514641e9baa4b3f9856acb5847d6e639389822"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
