)]}'
{"id":"openstack%2Fneutron~869112","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Id5f34ec858a940d2fc4a03b9290adc6161aa9db3","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-01-04 14:10:43.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-01-04 15:30:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5f34ec858a940d2fc4a03b9290adc6161aa9db3","subject":"[OVN] Fix availability zones changes check","status":"MERGED","created":"2023-01-03 15:54:51.000000000","updated":"2023-01-04 15:31:29.000000000","submitted":"2023-01-04 15:30:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869112","cherry_pick_of_change":868746,"cherry_pick_of_patch_set":6,"meta_rev_id":"995fdc6584f66adb5b205e5e3bfa50299522b2a2","_number":869112,"virtual_id_number":869112,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 15:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-01-04 08:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-01-04 14:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-01-04 14:10:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-03 17:01:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-03 17:09:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 18:00:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-01-04 11:43:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"0b14202df2efd3055c800181a3e8c22ea9e87c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-03 15:54:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a16f97ceea492a92e2334f35bd5cbb68f7f1b2a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:01:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a845a5632c9f4ef1ae98e6b003e8712f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbd1a64a2b7f4ac9a7f3b642dc0d21cc : SUCCESS in 1h 04m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00f6219d18484bde81120858842730a1 : FAILURE in 1h 03m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a28447826b829b4867d5c89bc46d5f5b5051534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 17:09:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0cff7ae6797496c8bda4a407ede1bb9\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/9748e95e8dfe405498df58bf6cd0a53a : SUCCESS in 43m 00s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/fe46793264244282b07a7fcffbe39b50 : SUCCESS in 1h 07m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce7e640b27d46fc8e34ceb4560cccb4 : SUCCESS in 42m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38671744e5f54bf69de5f9ceeabbfd65 : SUCCESS in 10m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b9ea2d9c2444e2cb3555c698528424c : SUCCESS in 24m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/740eadb30fc04412bcf337e609d0c6dd : SUCCESS in 29m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1868b9b4d294e629bf17582c43b78d1 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/57062b6c2476438a93869260bf2cd7fd : SUCCESS in 49m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a83bed306d8451092d1932135e9425c : SUCCESS in 53m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8bc171054f3545ae9f980617c675713f : SUCCESS in 1h 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8eb16a92faecec62bfd722bec2ac880c060b96f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-03 18:00:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0453afe8af161c3a0c6744253cb0c7535b1c078d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-03 18:02:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"552e438155f0ffa8fc04210319b5a08f6ca4e15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-04 08:40:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"04a54b54c34f58651b66c61fd98765aa571ee91d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-04 08:46:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dcbcc9fa2efeb85ab8811ed969b85dcf3156a58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 09:29:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f61eb6ab8840c79cc52075582adb65\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91a3f8fb292b4b899b2766e64fd5a9d8 : SUCCESS in 43m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab9c5384fd3348a3b61b7b31a7218060 : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"baecb9ee504f24b43395a420f7ff1a96004f674c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 10:09:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ff9fd5040049a3881f183b5e70ed39\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d0849068a090467c8d2dee7bc3ed26e7 : SUCCESS in 58m 10s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/2a3b1d303f5848d7bc3b7a913ea665ab : SUCCESS in 52m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/739ed6e1db644edb9ec136c9a5ddf8c2 : SUCCESS in 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1043209cd254f7582b334da812febf7 : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13236b03d43e4779b2fb3c465c6f6241 : SUCCESS in 27m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cabc62069a5428f8e9b847edb1a3033 : SUCCESS in 22m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8ab53d09aa4269ae8649a89dbada76 : SUCCESS in 8m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02ba0a824aba4b42b7e6468f16e86e1a : FAILURE in 39m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fefc6c529114435483b14e9ae9fcc48d : SUCCESS in 1h 22m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd5f38f951c8487191fb38c3c2abb6ed : SUCCESS in 1h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5628e4e842dc481a666ce2ae3d9641e04de4c63f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-04 10:19:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f30c32b995bab4a93b3defa83efd79f2572ee784","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 11:04:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ba61e92f7b04ed282c0e1d6891d3126\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fac592acaea402abf97e6bf8efdd5c3 : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/618875f3338a46e9a1b0c5ea30f2a262 : SUCCESS in 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca16a97dde01fbd5dd8680d4e9050fe4c481f757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 11:37:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8544dd25b494458be60beb77889653c\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/9f7e99f91eea4e5cb5ffb3dc90afb6c7 : SUCCESS in 53m 58s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/89b5626e63f449e2aba83d2f5f9b6f41 : SUCCESS in 58m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b4290ed19e240f5accc7a478d180d63 : SUCCESS in 48m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8979cb79c7aa422d9330e5beb346b0b0 : SUCCESS in 14m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8659a63c5ab644a78803bff2ca0b4f2f : SUCCESS in 34m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c845372af0f42a88f607a54d9acae0e : SUCCESS in 27m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28dea89d947846c1a3fa94dccf5fc248 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/119ee634109b4519aceb2fd2655580a7 : SUCCESS in 49m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bedb847ee6544508860454103f34522 : SUCCESS in 52m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/93d5d8555d4746b3885ad247efcbafd3 : SUCCESS in 1h 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b191dbb7d7dde6ceeb7e4ec4bd28bdc53b6b4730","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-04 14:10:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"09cef879d41ae9a68a4fc09290229880afba62d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 14:11:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"55dd4b7f92381846f720d646e7b96c3c00b8b488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 15:30:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2ed273d05545ab8bbf3a7b8db27236\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/5dea7546f91d429a92c0fa5b55a5bf69 : SUCCESS in 1h 04m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7406f691506c48fcadc8302c7e5783d1 : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf8f0ba3579c4666807e4e9debeeeac4 : SUCCESS in 31m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d2e085bb0d74f599f21a0b85429b4dd : SUCCESS in 22m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403bff45d07842a6ad2276bbcfd37131 : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a3ba1e950c74b7dae048fc98793a50c : SUCCESS in 58m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10758fa22fd245bd9c94fff08e2f671c : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ac0c0ec7c89f9759beb18071839131fa3a26f637","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 15:30:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"995fdc6584f66adb5b205e5e3bfa50299522b2a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 15:31:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b0920c5bfc4dd59c7cc94e2f244856\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/799043ea40c24017a71f177ef7019084 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2bf242a375933bc819493ee80f3aad98854f8e78","revisions":{"534c848dce60c507f5063acdc1ef69abb0a2d0ef":{"kind":"REWORK","_number":1,"created":"2023-01-03 15:54:51.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/12/869112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/869112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/869112/1"}}},"commit":{"parents":[{"commit":"0e18664513d579ac25c92240d8be52f513a85895","subject":"Limit tox version to \u003c4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e18664513d579ac25c92240d8be52f513a85895"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-28 13:47:21.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-03 15:54:51.000000000","tz":0},"subject":"[OVN] Fix availability zones changes check","message":"[OVN] Fix availability zones changes check\n\nCloses-Bug: #2000634\n\nChange-Id: Id5f34ec858a940d2fc4a03b9290adc6161aa9db3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/534c848dce60c507f5063acdc1ef69abb0a2d0ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/534c848dce60c507f5063acdc1ef69abb0a2d0ef"}]},"branch":"refs/heads/stable/yoga"},"2bf242a375933bc819493ee80f3aad98854f8e78":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-04 08:40:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/12/869112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/869112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/869112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/869112/2"}}},"commit":{"parents":[{"commit":"0e18664513d579ac25c92240d8be52f513a85895","subject":"Limit tox version to \u003c4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e18664513d579ac25c92240d8be52f513a85895"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-28 13:47:21.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-04 08:39:01.000000000","tz":60},"subject":"[OVN] Fix availability zones changes check","message":"[OVN] Fix availability zones changes check\n\nCloses-Bug: #2000634\n\nChange-Id: Id5f34ec858a940d2fc4a03b9290adc6161aa9db3\n(cherry picked from commit 5e19eaadbb298aaf58b5389fea74c9b8064d0f41)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf242a375933bc819493ee80f3aad98854f8e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bf242a375933bc819493ee80f3aad98854f8e78"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
