)]}'
{"id":"openstack%2Fneutron~869295","triplet_id":"openstack%2Fneutron~master~I647d00a30564ade246e704ff199b6aceafdc4c50","project":"openstack/neutron","branch":"master","topic":"bug/1580880","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-02-09 17:14:03.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-07 09:41:51.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-01-24 12:06:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-02-07 09:55:18.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-02-09 17:14:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I647d00a30564ade246e704ff199b6aceafdc4c50","subject":"allow manila ports to do multiple port binding for ML2","status":"MERGED","created":"2023-01-06 13:41:59.000000000","updated":"2023-02-09 17:15:43.000000000","submitted":"2023-02-09 17:14:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"869295-bug/1580880","meta_rev_id":"967af959bf432da967a9ec17fa94615ee286eeb4","_number":869295,"virtual_id_number":869295,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-02-07 02:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 17:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-07 09:41:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-02-07 09:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-02-09 14:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 14:19:37.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 15:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-06 15:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 15:51:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-01-17 09:40:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-01-24 12:06:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-24 12:27:33.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2023-02-04 12:31:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"90a492774c4c32a717c7c592705999e374219f1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-06 13:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c90c7a57f8cce58b1eb7c313a2cbea4faf585e32","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-06 14:19:37.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/95/869295/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"67ac00f89ead2c8bf9a63a1f8e5ac3416e209f66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 15:05:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18e064da9804404e87a63ac3a85c3215\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40218c9381424d04b8561c991ac3213b : FAILURE in 41m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/376e2139d7f849969ecca737ef43a749 : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"142b453d29e010dc32c8effd06920bc14a5455b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 15:46:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70cba1d2df4c47a884e0ee7732efb67f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f0ee74cce43b43b3bc9d9460797f1cea : SUCCESS in 1h 19m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be4f908ee19c4cbd8e562823ee58f79d : SUCCESS in 1h 13m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ccc949e6ea4436391a3a067c5e42c9b : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/474ffae884ee4588bd9ed5160dae191f : SUCCESS in 1h 02m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc98ea1fb2134754a247113779ebc0f9 : SUCCESS in 27m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f419e5e8d4df41f2a3529b594c5c4d60 : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aeb126a439d4cfabd270e9d2d9fe41a : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a828404fe844c9e8e568019efceb516 : FAILURE in 30m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/804ca568ae04435193fcc7d0ac89c179 : FAILURE in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c9c6f68ff04ca69c1bc41ec67f1c36 : FAILURE in 3m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3bfa854eea742579b446d599953fb09 : SUCCESS in 42m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f174fdda519469e9b7d37e2467e7f3b : SUCCESS in 1h 25m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/37a854a80b6a4dba960f9902e99b2976 : SUCCESS in 52m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a29c31ba0483408b8f2e67f42fecb0c0 : SUCCESS in 1h 03m 04s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb2dbd1573214854bcf700f4efa515d3 : SUCCESS in 1h 20m 01s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9d68e9e6040d406c8558b215c4fc931b : SUCCESS in 40m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/adf99641d2a54e4895e5cafdd9b45cde : FAILURE in 58m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1faded3b8770402eade6956d52a98e55 : FAILURE in 1h 05m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6505641758a84d808d40afb79a51bc2f : SUCCESS in 1h 05m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/622590a81b794beca5354ad6685475b0 : SUCCESS in 1h 51m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8eb228cfd9cf4558ae6ad2fd351f2a72 : SUCCESS in 1h 58m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/87b02a0b79ec4bd9a2c25c8b5c4461f5 : SUCCESS in 1h 23m 42s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bc13335fc147452ba355c6853757a160 : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6189995b4a7e382151b929332bcdfacee4d9067c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-06 15:51:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"930e7a6cfc9e8744147831d2351b0cd95dff1103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-06 16:11:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd1e83761deed75e2ad821b90bed461cfcb08dad","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-06 16:12:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bcf225a130609e523e518706e7edce6cc58dd38c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-06 16:48:13.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/95/869295/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8d99e1aeff0001064416367b5025ab36f5ac0ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 16:58:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6831a91ca4a14fa8acab596fb23a6350\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8631c4bd027d4139bb13d09f27da5270 : SUCCESS in 45m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd34ebe59db64bccbe93ddd5689f8413 : SUCCESS in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c1816a39dc620aa806ca2f096cf4834636075b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 18:11:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87be504641434d8794c6d33a667e13fc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d64e192f3c6a40fabaed3fc3819db9ec : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/635ed7b2213e4d7bbf92c2581222d14f : SUCCESS in 1h 33m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/033e1752e3164013bffa33a3bf6ef32c : FAILURE in 1h 36m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9fc0db4a1b44bfe95be72b5a6ab29e3 : SUCCESS in 1h 29m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/449525424fb64de497cf182e57fdde42 : SUCCESS in 52m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a023ca2f845b42d48cff631728ae4f59 : SUCCESS in 38m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1156e9c58640138ac840a2cb43f607 : SUCCESS in 14m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cc4cd6fb0e14e37ab8b96478344fa2c : SUCCESS in 23m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb14a22689c341b39bd6fe4da10996fc : SUCCESS in 28m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160d78c4202c47389c7ab98eab5db9a9 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e436381f30434e13b27183cf4617d73a : SUCCESS in 47m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4018c021b74407bbe02f789b44d3a5a : SUCCESS in 1h 53m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/313fed0b1cf946ca96ebb7954ce82638 : SUCCESS in 54m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/33441d0606d8428a93aa710a6e26dc2e : SUCCESS in 1h 10m 28s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/453139e1341e4c59a6758498bf44ac17 : SUCCESS in 1h 20m 51s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/df7db81eeba14fcdb2210ff11e8e9382 : SUCCESS in 52m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db2778ba95724414a90d17fb7a36f5cd : SUCCESS in 59m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c24a1fe874048e993feec73bfdfccd3 : FAILURE in 1h 08m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0171e1606e14a25b29a5d34543df1ef : SUCCESS in 1h 00m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f71a2acd2c14b6bb69c84b3553199a3 : SUCCESS in 1h 47m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0fc5fc435fa44b58a3f102e57a8f9528 : SUCCESS in 1h 40m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/23e518b464854cc79e349a95557561c4 : SUCCESS in 50m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/0100bafaac9b4fdfbc50a7e245c4ce28 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":2},{"id":"64424b7f730af7a3c1b2466ebc2bccc3a2513c7f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-06 19:15:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c587fc3ac08111a9f66b7ac7c9a6d8a8d288a95d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-17 09:40:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b4b7afd66aafc7d4f951cb2d2254241df543176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-17 18:09:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3deffdeac4a52bf47ddca2e80bdafdae0c5a2f7c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-17 18:09:28.000000000","message":"Topic bug/1580880 removed","accounts_in_message":[],"_revision_number":3},{"id":"d552fc2c140f948e46d897806a18cdbb8231c72b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-17 18:09:57.000000000","message":"Topic set to manila-multiple-port-binding","accounts_in_message":[],"_revision_number":3},{"id":"36697855586f21373ebb86fa2fa4c12b7f5637a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-17 19:25:28.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/95/869295/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b21de5c4a2f6da0a63a619fe5724791c17c0040e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 20:12:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff040b53f62b48dba997f5e77cde4c69\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/535c73e93a8449c1ad2fe27ac9ce628c : SUCCESS in 2h 01m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6b5410b2f0c4c94bbd1e2ecf266b9b2 : SUCCESS in 1h 19m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3945c99457744c7fb004fa2b6f3d7d70 : SUCCESS in 1h 31m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd4c661832bb4fe5b08c0b67531687b1 : SUCCESS in 1h 23m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6d9c37169a344c292095b72ed763836 : SUCCESS in 34m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b317da29e84c4b9db35302c0e67e5fd0 : SUCCESS in 31m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5785468fc867452984a7f86745a473f8 : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69dd8a8adb1c465f9fbfb5c990205ff8 : SUCCESS in 28m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bb7ce4b29ee4696be8d22d6596e2a3f : SUCCESS in 32m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917a6812a179455b849c5b58625c8845 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c69829075414ea4ba99a92fcf1709ad : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4105b428a92e4bd18421bcb5db696f98 : SUCCESS in 55m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd183ac085b243b5849d9e79963e4d1c : SUCCESS in 1h 50m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d9f81cce4be40ff9a775d55e84f8c37 : SUCCESS in 46m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0fcf39f5d57d4aa7bb21ab691ccbc85a : SUCCESS in 56m 01s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/21ddb35b410541368ff189780a2a3a2f : SUCCESS in 52m 52s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/36fd31ec04564195a23a84d01a1fed6e : SUCCESS in 53m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be0c6763bd56431b9e6c89ffe67d2970 : SUCCESS in 1h 17m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b5e038f08c8455e9f7e1f533cde6d25 : SUCCESS in 1h 23m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dcc09d1d14064e45b2f0c7f91da853af : SUCCESS in 1h 01m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f8908161e4a48e18625263540225593 : SUCCESS in 1h 31m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b67317c7abce4c3ba9431a578a325c49 : SUCCESS in 1h 40m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e145b40d8c0741398db851fcb343cbb7 : SUCCESS in 1h 49m 10s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e08f0e2b42b94050bbd3963728abe19d : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ed77fe72e25f0ff680242bf9c263fd8c3af9aab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:45:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09d13c1fef64180a4fbfc18662029ad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eddeb9fa34bd41c890bd26dee1a5134d : FAILURE in 1h 09m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e31365f4134b4b8789a7bc62f8ff98a4 : SUCCESS in 1h 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce5bd5ff50378ab3ce018a7af3f5b3159748c0d3","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-18 07:02:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b7f7e40e5e27b2508ada867d5e741c30828fce7f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-24 12:06:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d3480ae9b4783337c4dcf5c5a19d7ed46a01c5b4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-01-24 12:27:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7638e2f9a252b69dd2de4deae0078d9a2154ca6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-24 12:42:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c957ab88d6e8252c7f4fc6bdb11d858a00e60a9f","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-24 12:42:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"94e2ce56127cafb7247d3d73cc18701128f54d38","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-24 13:18:53.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/95/869295/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"7c655f3bd705d108b2d7af39b005fabd86ced166","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 13:33:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371f86f422744b94acd25763032fae54\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6f0ce90cfdf4c89aef401e67d0cfe19 : SUCCESS in 46m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/889059b19aee43cb8cbc8c54aaaf906b : SUCCESS in 48m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c6fe545ed5599f9ff16b30f9c7f07975a10310b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 15:00:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9a3e8277d04b2282b489185b8f36c6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d39cb42ce59b4438814eeca36133e78a : SUCCESS in 1h 27m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/97569524e6cf4a6587041163dd0c4afd : SUCCESS in 1h 06m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5436fdfa81f44c42bcb92e969e3a7747 : SUCCESS in 1h 23m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/176b950a50324751a5b0737d854d7b41 : SUCCESS in 1h 08m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b789bf47a55c488c9e180610ea3127fc : SUCCESS in 25m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e8eb9566f73427197808f0da0c0f31c : SUCCESS in 46m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e847ebc67a5742278fc2fccf58c98df4 : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfa1176a0d164a0db5cf0971f77bf01f : SUCCESS in 26m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2776a6d6aee4ffbb33d119aa07a7bf6 : SUCCESS in 24m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb368edec3c4e3f96fb1df136db9511 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac4b4d602ddd45349658d46a54fa672e : SUCCESS in 8m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdbec08ef026405abbd838866847ffec : SUCCESS in 47m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7fe36cd6d34640c79fe3dc0bcf771af0 : SUCCESS in 1h 21m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5663fc0b48948d9b4b4bc7e2009c59c : SUCCESS in 1h 03m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/debd2ab91621406e8eab82e0382207eb : SUCCESS in 57m 06s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cbfd520749ce420f99a401e6d8d5e458 : SUCCESS in 1h 07m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9a4dc994328041efad8691255c824652 : SUCCESS in 58m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d1239f681c04888b4b2565b35e35f71 : SUCCESS in 1h 18m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e930526eca024466bf85289aaf587711 : SUCCESS in 1h 42m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/383e3f9e8ee94d578c981b086cad47fe : SUCCESS in 1h 09m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a4cefac2af154628b006df446eab17e0 : SUCCESS in 1h 40m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1cc5e73e9b84334b5e359ededa2daa1 : SUCCESS in 1h 54m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93e194ff634c47d6b9037a4bddba84ce : SUCCESS in 1h 29m 25s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7b8b5b8fd66e4d55b1541d7d761d1247 : SUCCESS in 1h 09m 45s","accounts_in_message":[],"_revision_number":4},{"id":"7ad6d3a13cc88230f8aeff4a0fb1167e11ae895a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-04 12:31:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5c02407548b1515cc0d60c7f22cbff5ddbae2111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-06 08:21:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b34bafca9861eecd4da1ccbac8fe83c6473543a5","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-06 08:22:12.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5bff48cba95dca2f61870c435f15d8726a65f792","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 08:42:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d428d9117a949dd874ac4cce0e054cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68268979b1ea49e7abc8432d85fd3eb6 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7391210e2ec24798b3f16d5fec7a4f61 : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a8ff01eca31d5b6dbdea7e2a5ac0f92c00a380bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 11:27:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6700956e34804d6496042cb791a3a675\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/15c9e803baae465d8ceb69bfc1157c72 : TIMED_OUT in 2h 51m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c601e78b50144c368e09368138a748fa : SUCCESS in 2h 18m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab7b401e8df5473583aa7a5d35557b1d : TIMED_OUT in 2h 49m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0fa9a0df7881446d82e0859e4151e045 : TIMED_OUT in 3h 03m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/86fc7886ddc94b1e8167e3846bec441f : TIMED_OUT in 1h 02m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96437685326e4259b677dcae2c2c6b98 : SUCCESS in 34m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5685a595f64fc08dc77fa4b0e95035 : SUCCESS in 11m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c711d6f5c7354396a111c6b84062f31b : SUCCESS in 21m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89ac386fba284a2cb4e23be281c4e214 : SUCCESS in 24m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6323ccc55983421fb9b8f677b3a5f81b : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b4d62267f94923a2c3d76f60f6950a : SUCCESS in 6m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1de61a50b9c7449fa6cfd356d2a9f503 : SUCCESS in 22m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e286f0f9b9cd409e86c665f877a31306 : SUCCESS in 41m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/97faa4fa57e443e0850fc7f6e0f46c09 : SUCCESS in 1h 40m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/91eb50057cef48f6ab65dc770e9b2c4f : SUCCESS in 48m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/abc339fa851a4322b255c6bb691464b7 : SUCCESS in 1h 10m 21s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/38cb762fb2b64b80884efa098c1fab7d : SUCCESS in 1h 09m 08s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/73a44c5a63154fc1889f528dd751ea7e : FAILURE in 54m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8f9e30362d6141db9a58d5ab6cac720d : SUCCESS in 1h 32m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/747699f3e2624a20931562be488f4fbb : SUCCESS in 1h 22m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2ab8d5b295e9478aa8d98b661a8f92d5 : SUCCESS in 1h 18m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5551031fddf342b99db3cb11c5dac840 : SUCCESS in 1h 23m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef7f93039b204e23be519335550b147c : SUCCESS in 1h 29m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/361e463d8e254f08b20b12637618b7f1 : SUCCESS in 1h 06m 22s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/262ec388d7d84439bbc064fd265e4bb6 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2e672042eff1736cd52dcb8a95aca920e3f6fd68","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-06 11:32:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d85f0864834aa300844f8a55c590019e11693d17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 12:15:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f83793e48d9426ab0db0930310eb940\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbf093c98084401db394bcbad836994e : SUCCESS in 41m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b04050c452d546ee81d631d64a82d364 : SUCCESS in 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2202144ba4d62e8efa12d29efd47a5d14ebf81ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 13:52:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/604e877551c346f98bf4a0ba1840d797\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dfc0a3e792df4c7295c860382d0d4062 : FAILURE in 15m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3b57118af13242849f7ec58bf3a26eaa : FAILURE in 15m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36d58dd533c34f13886ba812c7a0d87d : FAILURE in 35m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d76a9d55569480b95d2611976d44a56 : SUCCESS in 1h 25m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5414d694002e4409ba97ae7142f12506 : FAILURE in 40m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ab882f1a00c4c4baabfa04387e5b5e1 : SUCCESS in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c3ad158c6a4e4298e6d3cdecdb62bd : SUCCESS in 14m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f9252cc54fd4c33a26391e68f9d08ba : SUCCESS in 24m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82c77b03937943f3a38274160879a03d : SUCCESS in 30m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff11d302024b4f589e6d0b3c49929fa7 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b83dc929cee4693a4c4d8f2f4d48405 : SUCCESS in 6m 30s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/30ac8cbc8a004592b9895c5bc27ac27a : SUCCESS in 22m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c9f5550eb83429291fcb57000c7d634 : SUCCESS in 1h 29m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae9ca47a43e942d1bd45c9e59fe4541d : SUCCESS in 1h 39m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72c5b254cd4a4318b817ec88f22d3f53 : SUCCESS in 57m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/189543807a704420869a015c360cd1c3 : SUCCESS in 1h 21m 51s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2a30bf6064264a28a5017fc87fb8d7bc : SUCCESS in 1h 07m 57s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d8872472d8fe4eba9d717d962e659ab8 : SUCCESS in 1h 01m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/03cd4a3c7d944ba98704672a01978e26 : SUCCESS in 1h 05m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0336c2533b0b471ab56df4c0af62e6dc : SUCCESS in 1h 11m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a4a7bebd8de41259642bf0e605cf305 : SUCCESS in 1h 02m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1b8d3a1b43b44a5c9dfa9a772b82fa9c : SUCCESS in 1h 46m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d42759118044cd6828d4a2ce237efe2 : SUCCESS in 1h 26m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c3313af3c2847908d1d76f13cfac897 : SUCCESS in 1h 21m 26s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5c2388fb36344517975e5b4e2319932e : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":5},{"id":"0bce665a990825ae6945d518f9eb5fda740b19b4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-07 02:58:53.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/95/869295/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"ee7215ec48cb34b53f065953f1c5a28d79b3ff92","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-07 07:39:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"741426cb1a508210f89c6e71b2175d319f7d9e97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:51:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d79f0cce30a4342939ce2d11107dea6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1deed6d96b2246988032b64f4cbb1098 : FAILURE in 1h 10m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e0ee32dfb65420796c033a7f19caa26 : SUCCESS in 1h 08m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6bce9e60b8ddc467b4575e8b7e6bcaaab7bb919f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-07 09:41:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"63a17891a9c6b27946ec975839a18a1b2f4e3a6f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-07 09:55:18.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a74595cef5d75a9bef9e622d260abe1e21b658b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 10:02:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f12f18a98ea4541b713b2364d4cedea\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4a94d93798b4406bb59cf3a656e10224 : SUCCESS in 1h 58m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/09337220b2f647708f1de86674842649 : SUCCESS in 1h 57m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d30b0a3b74bb4d3bb10bb15b2170028d : SUCCESS in 1h 54m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/795e128bef7d4daebced8c26d3db4bd3 : SUCCESS in 1h 53m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70752563fa2941a6a60a786fa917cde0 : SUCCESS in 41m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f4ff2ca21b641ef8851331815c882d5 : SUCCESS in 34m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea21f2c5d15e4e908fe689889ef93e39 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a9b44354cf44cefb9f50a4d9633a750 : SUCCESS in 26m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9ae20389bab4da59b4d712de832b34e : SUCCESS in 24m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd688578aea46e78d3c385e2ac03023 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/002e9920ca5544948ba5423622f5b9e0 : SUCCESS in 6m 36s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/65bd3c79523b4bb0b00766d7bdf83e66 : SUCCESS in 23m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f879b7b512ab4f5c97cc8e2eb8043145 : FAILURE in 50m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fdcb74e6db7423a851e247a585c2547 : SUCCESS in 1h 36m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b0eddf85d9e492fa99cf2ce76343a1f : SUCCESS in 58m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d55f0afb576b4ebd824386af7b267e5d : SUCCESS in 1h 03m 31s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb6af010813a46528cc4f24711ce68a8 : SUCCESS in 1h 06m 54s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2936185abfb04ea0841ba043bd838245 : SUCCESS in 58m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/594dc2d4bda0483e8921f8711d0065f0 : SUCCESS in 1h 24m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8957357bc930492b9879d1fe0fc3a604 : SUCCESS in 1h 18m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c17a70c387c24be2886e54620942f865 : SUCCESS in 1h 16m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0b855c3a44e441dc86de47cadcfad3ae : SUCCESS in 1h 49m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19d7624d45524cb594cb17185b1179d6 : SUCCESS in 1h 58m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5cfba4857119459882ee1dd3b3b950c6 : SUCCESS in 1h 27m 11s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e974f1a8187b4be6bc9da34fd90d1105 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":5},{"id":"d658b9f46894d4659665f6b6256bba28452d8c61","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-08 08:39:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c862c26fbc59a3c1648825fc39f522a0f66a7988","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:13:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba4563441a1143d3b4aca357dae513a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c04f70aec084695a9b71de6c5a337d5 : SUCCESS in 19m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07e5668e5deb458680a77db891eb945f : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a96d5c1509e54a9eab7a3f820baca908b227ef2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:13:15.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/214d6d4ed272457cbd59368c99237215\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/32683ee884434687a217c8533e73536d : SUCCESS in 1h 46m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/89c6046f1e9343a193b6bc99f4c6488e : SUCCESS in 1h 44m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dfea478177934f89b6bb0611ab0a27e0 : SUCCESS in 1h 59m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d5941487de7a4d9ead6d640ce468ec39 : SUCCESS in 1h 38m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6fc006edf0641839e26009b515fa376 : SUCCESS in 36m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50897e3436b14aacac9c51129ea6b09d : SUCCESS in 44m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b083f2f6e54eac915bd5cb97a2acfc : SUCCESS in 13m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f151246b08d4975946fdcd2762ac27a : SUCCESS in 32m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26c339e9e59547e1b28f368182689267 : SUCCESS in 30m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf24b31d67154327ba2d683de8cf0c43 : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29f75c19d531400c82f5b9afd3d6b2e6 : SUCCESS in 6m 44s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/77ed7369c8124fca85615eeb6ce1a0a6 : SUCCESS in 25m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b0fb21eac1b4595a4540b915c3d3c8c : SUCCESS in 42m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6978473d0f94d67b8359c7001716edf : SUCCESS in 1h 40m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c33dce9ce6049f283c07975654dce1d : SUCCESS in 57m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/185a07613c314bea8116bdfbd1509180 : SUCCESS in 56m 33s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/edcde3566b8f44e59b9ea080d7c84365 : SUCCESS in 1h 14m 15s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6284d222171b4da28193fdda6f9a92f2 : SUCCESS in 1h 00m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52a53a4fbeb043959a83d3b8eeac530e : SUCCESS in 1h 19m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de9cae00957c4aa0a4710015797029dc : SUCCESS in 1h 20m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/de938766f5674ba88611a1bf388ef720 : SUCCESS in 1h 09m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7a44e0dfe7574cd4b6c5c4957bd08bf3 : SUCCESS in 1h 40m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/001012e6afce4bf796f9988142845a3d : TIMED_OUT in 2h 13m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f8932ef98c94a9685c8a0f3cbfe578b : SUCCESS in 1h 17m 44s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/8a7b0050e3d74ecda78f3a6c26323463 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":5},{"id":"a55f4bebdff7d9bc6da2b9745acb02804de5fdd0","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-08 15:45:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"52f47e124bcad6223fc15581a36038f34f42325d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:07:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5ead7a969b4fff8ca88623f457f411\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33f40f3308e44eab83497d9990a0ef87 : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0012a48e3c342aeab5fadf96666768e : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cac24afe8533be231dfeb247dd9e08e46e4dd97e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 19:49:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dfa96fcc1904ff5ad20f29e9e9f57c3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/df1403b192b442cb8162c8a4096724f6 : SUCCESS in 1h 37m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7f6458c784ae4b1bb176431061d5d88e : SUCCESS in 1h 08m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c76085c35be94394a4cf11cd060aa541 : SUCCESS in 1h 11m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94ed7de99d5f41ad808c4a768e9a3c2e : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/268848cc10c74ee09172ff22e9c675eb : SUCCESS in 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74c7a1061d004fa2b5a649d73ab329d3 : SUCCESS in 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e532a371ed9e4cbe8afdc62babd0cbcf : SUCCESS in 12m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfed679b4ce44dc89d09aaef5774ef6b : SUCCESS in 28m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daf276b823aa48a4a9ab620237095416 : SUCCESS in 32m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f099e95f8bf34e9b8fc184565d104ebd : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3518128d7be04e0c8a9722cf468989f2 : SUCCESS in 6m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9296cb262fd94d3c84f64f8b7d26949d : SUCCESS in 23m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e90fa8c4957f4693b4b3f548b11ab200 : FAILURE in 42m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/065ed81c545947cd85a178c6e148d9c3 : SUCCESS in 1h 22m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ba07dcf8c8a46c8b1d2b5558e5af2ba : SUCCESS in 50m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e2cf960bad33413098d4ef97b5e6725b : SUCCESS in 56m 43s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d595cd9e5b8e4932a0716a09c53c7ac0 : SUCCESS in 1h 02m 23s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8815795afe7a47b6b9492f62062ae64b : SUCCESS in 59m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8ca2c208466743fcb5867e73a608e41a : SUCCESS in 1h 28m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6598a3bf50a84e1f9ff4598fb3813091 : SUCCESS in 1h 19m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f9074c5f3e58435d88891e91071ccd8b : SUCCESS in 1h 03m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d8e66d52364b46cba9d65b38b17b4c41 : SUCCESS in 1h 40m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c41acaf3711a42538107e133a241d01a : TIMED_OUT in 2h 13m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f8aebc06c2c84202a36efe66a0912c76 : SUCCESS in 1h 18m 34s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/260bbcf773ab4571b99aa0138a758f0f : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b60a47caba0f009aef214204571a1642cc614185","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-09 11:55:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"961dbd2aea328539db321baff5dfa2ed038de5c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 12:51:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9359dfcb97374cf4b4d7c7c0732b8952\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c91239dbde6459eac5a5256fb4050ad : SUCCESS in 54m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ed5aaa3ac22048b9af3cbc5076c55346 : SUCCESS in 53m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8bf2839788048107dc2d82c2eeacd46bd83ad6e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:00:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c68d4cd5dd4f539992d6481c347b1c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e4a3ab5f3b5a4cac9b8388397ad060b0 : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79e9828d888648d090f43e471a20ee84 : SUCCESS in 1h 37m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/930a387cbca240cd9db7d3dac15d9cec : SUCCESS in 1h 42m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a338bd486bfe48c8a44f7ee5490443c4 : SUCCESS in 1h 55m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7232b54c743d4504918e4703c99a1dc4 : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e41bbbceb4ff44208c1014c4e8d4ec67 : SUCCESS in 42m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb6430b6f724ebcbd194346e8dcf65c : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f14eae9c420141dc95704feab3379893 : SUCCESS in 29m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a26fe7de8a8342e99a9d36830e19f09a : SUCCESS in 27m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fb1e4784a74a9c9cfdbff9c3331284 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7ae7f57aa514e72ba30df97e956f390 : SUCCESS in 7m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/02c0599c49444a2e8b4930bdb1c4d2b1 : SUCCESS in 24m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3eeb7540a468423abe73147f9175f13a : SUCCESS in 48m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/41bafd7c651845ab9832821a1b82bce9 : SUCCESS in 2h 03m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c1fcedaf8554ed58046ae2686463712 : SUCCESS in 52m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dc4c89d044a7438fae1c63a9c632576c : SUCCESS in 55m 55s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71a9917dd1c345eaaf61e9d6a1fa1407 : SUCCESS in 1h 05m 02s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d36fa33d08f0446eacbcde14d59282f1 : SUCCESS in 1h 03m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c71396ff8f5a4164934af06a32a59208 : SUCCESS in 1h 32m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/84e7f920c8cf4b26b8d721e3c6ca0eed : SUCCESS in 1h 39m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8c5b68864e394c8abfebcc372d78b090 : SUCCESS in 1h 07m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8a95db7abd4541cabec1948abb0f5f7d : SUCCESS in 1h 31m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0599391805174855a24acf79ac465d44 : SUCCESS in 1h 31m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f38292648bf54d248cba855837761618 : SUCCESS in 1h 23m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fa88bf2140a94a43a8e70a7f56a43b76 : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":5},{"id":"0d4b179ad5d8dcdcd95a9bbc28150a6def103d4f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-09 14:41:35.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a971bbd0df1421fccd51e2f08212973a28d789b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:42:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b40a42ceaf3cf31bd684363cdbe32d17e0722f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 17:14:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebcabbfe44c34bf59da10e2fe934e715\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/908b70df868d42578c5b231b90090bae : SUCCESS in 1h 45m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b84abdbe1d6b4ace92f0f8a7d95dc131 : SUCCESS in 1h 22m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/34e00082c91a481780268dbc3479b558 : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/270f994f03024266ab5e51e8e85656ce : SUCCESS in 1h 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12825a8ac25b4a65b2948ebce19f0f2d : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a35e8eed3b6c43b598230f7fa4727763 : SUCCESS in 30m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/724406555b0740d88c66728fce81ff0d : SUCCESS in 29m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd09f4d70b043af9dd32712f01074f6 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8fc84c0760d4863a40fcb394a5f07ec : SUCCESS in 7m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c435aae3a08746acb659b61092db6e13 : SUCCESS in 54m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d7d4551e99848faaa7f9862d7b7a1ac : SUCCESS in 1h 31m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c03a0ec1c4014c0f9fa61149e9c85c3e : SUCCESS in 1h 26m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e3e381fc7f74df9a0e337d1c0f91366 : SUCCESS in 56m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/abf853c3067c41e2b4a41ed2b830b358 : SUCCESS in 1h 01m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6c6e0f31920e4f14a474ff709a9e311a : SUCCESS in 1h 03m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5d8fa0c003ad4de1815e769a8dabf770 : SUCCESS in 1h 23m 41s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ed89d22d7f644cb7a45707e3bcc20fcc : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3d34b048d6448e8a0a84afdea5062191d27aa08e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 17:14:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"967af959bf432da967a9ec17fa94615ee286eeb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 17:15:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce4476586f74e35992c0fa9e469ee0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e8e3b6696749379251f4dfa535f33f : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d00c9cfc81d240a78c51b994fff09021 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5c697b8d60571ef4a052586a73edd3d513d0d635","revisions":{"3afae8dece927734b513f83bb17d871780294699":{"kind":"REWORK","_number":1,"created":"2023-01-06 13:41:59.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/95/869295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/869295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/869295/1"}}},"commit":{"parents":[{"commit":"07fd0e101b62deda45206aff8bbdf151e23b169f","subject":"Revert \"Documentation for adding extra port conflict fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07fd0e101b62deda45206aff8bbdf151e23b169f"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-19 11:33:36.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-01-06 13:41:57.000000000","tz":60},"subject":"allow manila ports to do multiple port binding for ML2","message":"allow manila ports to do multiple port binding for ML2\n\nSimilar to Nova live migration\n(see https://review.opendev.org/c/openstack/neutron/+/414251/74/neutron/plugins/ml2/plugin.py#2005)\nManila wants to do share live migration, and needs to modify its ports in a\nsimilar way: issue port binding upfront to determine the segmentation id in\nthe target network segment.\n\ndepends on new constant \u0027DEVICE_OWNER_MANILA_PREFIX\u0027 in neutron-lib\nDepends-On: I9be781d99a08a6a015c3747eb15f20bc356cf08e\nPartial-Bug: #1580880\n\nChange-Id: I647d00a30564ade246e704ff199b6aceafdc4c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3afae8dece927734b513f83bb17d871780294699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3afae8dece927734b513f83bb17d871780294699"}]},"branch":"refs/heads/master"},"603e8af9468ef9c296aa359ee0fd7d8d843e652c":{"kind":"REWORK","_number":2,"created":"2023-01-06 16:11:10.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/95/869295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/869295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/869295/2"}}},"commit":{"parents":[{"commit":"07fd0e101b62deda45206aff8bbdf151e23b169f","subject":"Revert \"Documentation for adding extra port conflict fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07fd0e101b62deda45206aff8bbdf151e23b169f"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-19 11:33:36.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-01-06 16:10:44.000000000","tz":60},"subject":"allow manila ports to do multiple port binding for ML2","message":"allow manila ports to do multiple port binding for ML2\n\nSimilar to Nova live migration\n(see https://review.opendev.org/c/openstack/neutron/+/414251/74/neutron/plugins/ml2/plugin.py#2005)\nManila wants to do share live migration, and needs to modify its ports in a\nsimilar way: issue port binding upfront to determine the segmentation id in\nthe target network segment.\n\nPartial-Bug: #1580880\nChange-Id: I647d00a30564ade246e704ff199b6aceafdc4c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/603e8af9468ef9c296aa359ee0fd7d8d843e652c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/603e8af9468ef9c296aa359ee0fd7d8d843e652c"}]},"branch":"refs/heads/master"},"ab362bd2a4df9fe585b7d700429899ea8dd70ecf":{"kind":"REWORK","_number":3,"created":"2023-01-17 18:09:01.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/95/869295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/869295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/869295/3"}}},"commit":{"parents":[{"commit":"07fd0e101b62deda45206aff8bbdf151e23b169f","subject":"Revert \"Documentation for adding extra port conflict fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07fd0e101b62deda45206aff8bbdf151e23b169f"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-19 11:33:36.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-01-17 18:08:45.000000000","tz":60},"subject":"allow manila ports to do multiple port binding for ML2","message":"allow manila ports to do multiple port binding for ML2\n\nSimilar to Nova live migration\n(see https://review.opendev.org/c/openstack/neutron/+/414251/74/neutron/plugins/ml2/plugin.py#2005)\nManila wants to do share live migration, and needs to modify its ports in a\nsimilar way: issue port binding upfront to determine the segmentation id in\nthe target network segment.\n\nCloses-Bug: #2003095\nChange-Id: I647d00a30564ade246e704ff199b6aceafdc4c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab362bd2a4df9fe585b7d700429899ea8dd70ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab362bd2a4df9fe585b7d700429899ea8dd70ecf"}]},"branch":"refs/heads/master"},"08b0d3ec36bd563970419da893da239a768e39ba":{"kind":"REWORK","_number":4,"created":"2023-01-24 12:42:12.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/95/869295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/869295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/869295/4"}}},"commit":{"parents":[{"commit":"07fd0e101b62deda45206aff8bbdf151e23b169f","subject":"Revert \"Documentation for adding extra port conflict fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07fd0e101b62deda45206aff8bbdf151e23b169f"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-19 11:33:36.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-01-24 12:41:49.000000000","tz":60},"subject":"allow manila ports to do multiple port binding for ML2","message":"allow manila ports to do multiple port binding for ML2\n\nSimilar to Nova live migration\n(see https://review.opendev.org/c/openstack/neutron/+/414251/74/neutron/plugins/ml2/plugin.py#2005)\nManila wants to do share live migration, and needs to modify its ports in a\nsimilar way: issue port binding upfront to determine the segmentation id in\nthe target network segment.\n\nCloses-Bug: #2003095\nChange-Id: I647d00a30564ade246e704ff199b6aceafdc4c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08b0d3ec36bd563970419da893da239a768e39ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08b0d3ec36bd563970419da893da239a768e39ba"}]},"branch":"refs/heads/master"},"5c697b8d60571ef4a052586a73edd3d513d0d635":{"kind":"REWORK","_number":5,"created":"2023-02-06 08:21:29.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/95/869295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/869295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/869295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/869295/5"}}},"commit":{"parents":[{"commit":"07fd0e101b62deda45206aff8bbdf151e23b169f","subject":"Revert \"Documentation for adding extra port conflict fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07fd0e101b62deda45206aff8bbdf151e23b169f"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-19 11:33:36.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-02-06 08:20:55.000000000","tz":60},"subject":"allow manila ports to do multiple port binding for ML2","message":"allow manila ports to do multiple port binding for ML2\n\nSimilar to Nova live migration\n(see https://review.opendev.org/c/openstack/neutron/+/414251/74/neutron/plugins/ml2/plugin.py#2005)\nManila wants to do share live migration, and needs to modify its ports in a\nsimilar way: issue port binding upfront to determine the segmentation id in\nthe target network segment.\n\nCloses-Bug: #2003095\nChange-Id: I647d00a30564ade246e704ff199b6aceafdc4c50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c697b8d60571ef4a052586a73edd3d513d0d635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c697b8d60571ef4a052586a73edd3d513d0d635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
