)]}'
{"id":"openstack%2Fneutron~869348","triplet_id":"openstack%2Fneutron~master~I79bf7de9e972b75021e2fb2ebb0b247eaa9f85fc","project":"openstack/neutron","branch":"master","topic":"remove_from_self","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-13 17:31:34.000000000","reason":"Change was abandoned"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-02-13 17:31:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I79bf7de9e972b75021e2fb2ebb0b247eaa9f85fc","subject":"[WIP] Propose replacement of ORM from_self()","status":"ABANDONED","created":"2023-01-05 10:50:22.000000000","updated":"2023-02-13 17:31:34.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ddf1f1ceac7ef4f70da415634ae05bb637145837","_number":869348,"virtual_id_number":869348,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-05 12:20:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 10:51:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-01-05 11:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-05 12:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-06 05:54:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"32e2557a45deaaeba573ffc4e86d4ff8c62878cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-05 10:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0880ad81233ba142f018cee575f4d6afab65041","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-05 10:51:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a285df35862d140fa00114259403c8cfefcc307b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:58:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e161a42645134180b6076b513d08ff61\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46c0083ad4ed4c40ae8857264439a9fb : FAILURE in 37m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d32aab2a84ca4239a25e9fb3ab5274a0 : FAILURE in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b75e290be2b2f956394fb2a60d3a65378d2f040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:20:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0c560843e7f47cdb0765f1b48b38d2a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6f5f759dfe13448098456a424c4bf6b5 : FAILURE in 16m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f9c12ac391649f7a09cfe8733b1bb86 : FAILURE in 17m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e6b787aa4fe494e8738f6c13cd87dcd : FAILURE in 18m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d15eea321fcb4701b950f395c5a7f72b : FAILURE in 19m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55a4ff9987144c7081cccf21bfa7a424 : FAILURE in 15m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e32440de80b946ccb8cb447fb4247238 : FAILURE in 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc52debd25449af825a7400018a7e35 : FAILURE in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce56554232c8432d8b1d1fb0527bcae8 : FAILURE in 22m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd3e11f920c44127a0f432d5cd5f80fb : FAILURE in 20m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e634408df024f00b18ffce2a5db3b90 : SUCCESS in 12m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3a6bd1807a34a9ab2a202b50d33c229 : FAILURE in 37m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a39e879bc684a8b9a0c312138bf3fc2 : FAILURE in 1h 28m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ec601d93e85418dbe77a307a1baa1d6 : FAILURE in 53m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7d9e22d20b24befbe1fd209d60efaee : FAILURE in 58m 57s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8446f17444694e96babc5ca6d9199b83 : FAILURE in 59m 49s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bc4525445b604dcdb1173832d85d5339 : FAILURE in 54m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1d9c130b063d40058a3fc841120ba663 : FAILURE in 22m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8d3d4646d66f4916af82e4fff9c9e3dc : FAILURE in 22m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/529c827721e844c19acaa748800dfd22 : FAILURE in 21m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/feb68c4a86f84901a3a1d08df028a18f : FAILURE in 19m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c2cb70257cf64c40b8b00ad53f024fd7 : FAILURE in 27m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e20fc993fe3f46b9896d0cbcb07a8a53 : FAILURE in 22m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bc2d7b8f600b41d79fecefd4d6a00d13 : FAILURE in 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"23b1e13688e23c5558ff8a2b3087a9d7c570f7c4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-06 05:54:53.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/48/869348/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ddf1f1ceac7ef4f70da415634ae05bb637145837","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-13 17:31:34.000000000","message":"Abandoned\n\nPatch https://review.opendev.org/c/openstack/neutron/+/744512 is already merged and working","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c50d0b6e60b1050ef2289be3c0a2ca20a3ef0ae","revisions":{"9c50d0b6e60b1050ef2289be3c0a2ca20a3ef0ae":{"kind":"REWORK","_number":1,"created":"2023-01-05 10:50:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/48/869348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/869348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/869348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/869348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/869348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/869348/1"}}},"commit":{"parents":[{"commit":"c9708e692f66700d0aef3fa0bcac07a2c0ac3a46","subject":"Merge \"Fix incorrect mysql examples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9708e692f66700d0aef3fa0bcac07a2c0ac3a46"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-04 10:47:24.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-04 10:48:50.000000000","tz":60},"subject":"[WIP] Propose replacement of ORM from_self()","message":"[WIP] Propose replacement of ORM from_self()\n\nAlternative solution to [1], because [2] changes how the queries\nare called in Subnet class. Instead of creating a subquery, this\npatch adds an outerjoin to the subnet query to filter the service\nsubnets.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/744512\n[2]https://review.opendev.org/c/openstack/neutron/+/844959\n\nChange-Id: I79bf7de9e972b75021e2fb2ebb0b247eaa9f85fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c50d0b6e60b1050ef2289be3c0a2ca20a3ef0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c50d0b6e60b1050ef2289be3c0a2ca20a3ef0ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
