)]}'
{"id":"openstack%2Fneutron~869352","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Idf522a20f9735829ee568020bfed46345a95e294","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1977629","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-18 17:02:38.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-01-16 17:07:19.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-17 12:01:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Idf522a20f9735829ee568020bfed46345a95e294","subject":"[ovn]neutron agent show real heartbeat_timestamp","status":"MERGED","created":"2023-01-05 11:07:20.000000000","updated":"2023-01-18 23:44:23.000000000","submitted":"2023-01-18 21:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869352-bug/1977629","meta_rev_id":"014f081f21ddc0c08d708ae9226f5bf0e3fab490","_number":869352,"virtual_id_number":869352,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 21:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2023-01-18 09:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-01-16 17:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-01-17 12:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-01-17 12:01:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 11:07:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"CC"},{"updated":"2023-01-05 12:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-05 12:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-13 16:10:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-01-13 16:10:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-01-18 09:52:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"0445f415205b6829a82f678d57fbc9f7e4b927e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-05 11:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64b6d9a828e6d53e8cdd82e478f6071316a4b08d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:11:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31f056ad85b4f6f964eee986595ce88\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a74bece6aef46358f37cba0a0b9fa68 : SUCCESS in 51m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d65fd018e2a04f7b809aedce459b0ed1 : SUCCESS in 53m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7dafe5770bc851e6c1a0a8626661bdb4f105085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 12:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9164fd35baff485282964a4935da9968\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/503e35c64af4400aaa01b3aec29a9041 : SUCCESS in 42m 20s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/bcf86a47d826441aa34971be55e0fb84 : SUCCESS in 1h 13m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e525131f509349d6bc6d00dd2946cf19 : SUCCESS in 39m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511c1125b19e42fba8e1d7ff6830ccef : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/295ed3efdc32410d9282b8085241e143 : SUCCESS in 32m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9cee7348c24c0ca5e52696df902465 : SUCCESS in 31m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/131ae13241da4832987a87a69b76b1dc : SUCCESS in 27m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a125ffcf4b47b5a24e89fffd18c4d2 : SUCCESS in 11m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d97ed40b01c347068db0972e5330f978 : SUCCESS in 43m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6a21c1ae344d46979fd3b278973fc738 : SUCCESS in 56m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8ff0a63b842a4da1a9abbb6e20560d53 : SUCCESS in 27m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f723d2854d334074a10dabf477493bbe : SUCCESS in 1h 21m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/362231b346fa43a88ee05c785dc1826e : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9409321bb2e61f48b1bfbb4466286ce5c1cc2440","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-16 17:07:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90b168cf5f0bbd467ab120a423534c6bbc15edf9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-17 12:01:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3ec553905595b0f0f3007f6f5fb5ae85d7043cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 12:02:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a687cd954a50e673b74f25a834f8922ce24a1f7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:39:11.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/869349 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"1cbc9569f913e8a1a1b590fc8a5aa67ea2ee2e95","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-17 19:48:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a26b074bbe52e271ac59d465e87284a17525bb45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 21:32:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fae52b407cf461b8177236cc8586619\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/6de15749198040408461a89f887b1a69 : SUCCESS in 54m 42s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/cd2c4b1186f94036a380fdf9b7f74fa6 : SUCCESS in 1h 40m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/068644433a3d4890b32a0254af8cc024 : SUCCESS in 44m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84aa8c5d47741dcaff30adbff0b3710 : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb0c13f4cafd41b7929dfea52017553c : SUCCESS in 33m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff27411bc424e5995361796cecb5443 : SUCCESS in 31m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/867de648a91a44e0b764541ae791eeb0 : SUCCESS in 32m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5afe4d18d0b94f4c9afbe5d1a918e99a : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a62ecbe0d9bc4b77bd489a0ca358d065 : FAILURE in 41m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40392fd0fb8a4f7fad6221d5d3bab739 : SUCCESS in 1h 16m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b9eeccca83e24ddd9e7827f4b2a549b5 : SUCCESS in 27m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d88f5f8aed6f410094eb3908c443ed26 : SUCCESS in 1h 40m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1d6996419b5845189583b5dcd267d487 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"eac0f6c549935de23e651e4a8e0e6b969163f303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 22:51:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d62e6103b294164af6baad0af7ede76\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bda48a2c832a4e6c9c01c681b84b5b50 : SUCCESS in 1h 04m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e9843661f5c4659920aa5087c5d5d13 : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c05dcc58cb8f835df22e48eb09cf9f9e276d81e1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-18 09:52:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32df619c866462d6a7a7b0246233c23104711f6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:15:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ecc5dee1a34098be84b628fc345e22\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f38492f40f04d23888bf4c5a1461b7d : SUCCESS in 52m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00e8fbab08484cfd9a3e19ceb63723a1 : SUCCESS in 1h 04m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70cc433941da42cce26f8bf54dd37054ea0965f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 12:01:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2824321f82ec41ff94db41c8e8207ca4\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/cc997568df1e4faabbb818a165b25b50 : SUCCESS in 49m 53s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/e3aa21274cca4c46b770e03bd1db9cab : SUCCESS in 57m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f1c872e9000489bb3dd5ddc46ed0ff3 : SUCCESS in 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ea264e96ae4f90bfa46960b025c14c : SUCCESS in 10m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0fd25649d56406bbfab0b12c430a556 : SUCCESS in 35m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb76c03f1ff643ac88a261bb43b07daf : SUCCESS in 28m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b54b095ff1e48c3a8301bc2ac867980 : SUCCESS in 28m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72bb45bc7bb4a69adf3402b14dcc728 : SUCCESS in 9m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da8a0ef96a154b0d90f27c624dddd321 : FAILURE in 59m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/81ff7a7e8a61480299a826f11a723e06 : SUCCESS in 1h 13m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ea3710c3c81046d48b478e83c2914772 : SUCCESS in 41m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b5d70caf76c4e94ba6ea2ed2d3a6666 : SUCCESS in 1h 43m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/10b4a44d293048d2a402ec9cebc4d675 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"afddaafd9e510c9f509e200cc4e6725d423289c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-18 17:02:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"953704e845c02d503185152e06709b0dab21529b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be304c57ba3840a1bcfec031f7059447\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/0cf061b8f6ac48b9b567764b24e94b8a : SUCCESS in 58m 01s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/43c7fc1e894347e4915afad2e60076af : SUCCESS in 1h 06m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b3b211c2c034caaa9a8e0579a79974a : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4379b27665ca4c1a84f8e31e398eb56e : SUCCESS in 11m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06a512eebe2e452ab03aa0fdcd7a7d1e : SUCCESS in 30m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a04ef361f54e42d89fabdcdded890608 : SUCCESS in 25m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7967eaedf2d46e78199eee3d8427453 : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e98b7bb91c43eebd3ca936dccb12c9 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c39a479d85e4776ad888afee240b2cd : SUCCESS in 53m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/193e363fe8a3473ca2ab79455a5fe07b : SUCCESS in 1h 18m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/37ebf968ebbb4c85b4c58ead14b1884f : SUCCESS in 29m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7239801596c340c19ec9aa931c46644f : SUCCESS in 1h 24m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b9c68924d595450d8c90e3acfb9680f1 : SUCCESS in 1h 25m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1fbd55a677fcddb6eec0af7d9b9f9e7fa735d7d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:50:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d83599b3d3364d6bea5062aa60abd0f9b2aaf2ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:01:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d66695d94346c29b915af84cd41405\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/d02eaf9f34d7484b9a4fd1f8b1d8870b : SUCCESS in 53m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb8000b38224b86a9c75d264a3ff17b : SUCCESS in 16m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd30674c44364a9bb713dbe4513fcaf3 : SUCCESS in 33m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2927313810d54e7dabcc2c04c3aa18d8 : SUCCESS in 31m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2af9347bb834a0388a7b510affd9e41 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42ca314079644a8b8be13cebb20d2ac5 : SUCCESS in 53m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a3442f38c524d7c8db539be638a1952 : SUCCESS in 1h 12m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/47315241a7e44ee994f9e3b7e51fa471 : SUCCESS in 39m 10s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43005c4b7ace4b8faef7cc39db0aca08 : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e145a6ab8226b942a7c14ffcc068ce025297ddac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98b13f6eecf83d067c4eb4ff247c9d85688c8a98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:02:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10b061c684148128bc2ea66f5c9ca56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03eba5334f44a7bb25252b140380946 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"014f081f21ddc0c08d708ae9226f5bf0e3fab490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:44:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a9fe850df8944fca53690a91dc0a15a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffa983216d1548b4b04dfff101f6908a : FAILURE in 1h 16m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b377e3a97ab243a4aec90e5548da6dcf : FAILURE in 1h 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"938534ecefa941f5a1b83b5df2b0d2b9da5c04d7","revisions":{"938534ecefa941f5a1b83b5df2b0d2b9da5c04d7":{"kind":"REWORK","_number":1,"created":"2023-01-05 11:07:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/869352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/869352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/869352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/869352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/869352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/869352/1"}}},"commit":{"parents":[{"commit":"d34afc5c13b0251c51f8c493b331900a46737512","subject":"[ovn]support read chassis update time from nb_cfg_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d34afc5c13b0251c51f8c493b331900a46737512"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-04 11:01:57.000000000","tz":60},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n(cherry picked from commit 411ecc4865198e2bed5e3d0ab498e3a42408c17d)\n(cherry picked from commit 4303039e500e76e44aad0aa9f0c2c8caa8e6c4fc)\n(cherry picked from commit 4fb768fe9dcbb862aed8eb8de74d50dff5b8d881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/938534ecefa941f5a1b83b5df2b0d2b9da5c04d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/938534ecefa941f5a1b83b5df2b0d2b9da5c04d7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
