)]}'
{"id":"openstack%2Fneutron~869554","triplet_id":"openstack%2Fneutron~master~I3aafe1aba406a52bc2b57be5133dee15b8848796","project":"openstack/neutron","branch":"master","topic":"bug/1983053","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-01-27 10:47:12.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-27 10:47:12.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2023-01-27 10:47:12.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-01-27 10:47:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3aafe1aba406a52bc2b57be5133dee15b8848796","subject":"[API] Add API extension and definition for default SG rules","status":"MERGED","created":"2023-01-09 11:08:51.000000000","updated":"2023-01-27 10:48:38.000000000","submitted":"2023-01-27 10:47:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"869554-bug/1983053","meta_rev_id":"97b38121d0ee756402cd6d904e3f07a0201904fd","_number":869554,"virtual_id_number":869554,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-01-23 12:09:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 10:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-01-27 08:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-27 08:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-01-27 08:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-01-27 08:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-09 11:46:24.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 11:54:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-09 12:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 22:40:45.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-01-12 17:59:53.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-01-16 09:13:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-01-16 16:19:51.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-01-27 08:39:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-01-27 08:46:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-27 08:55:05.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"b254e831447c2edd08ed55ee551430163b7edb18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-09 11:08:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be5709c371531ef33cecbccf95d0c0f294734f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-09 11:46:24.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/54/869554/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"02fda8bc1b17fe460283e786148a1806fc4fa849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:54:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cb1f0f9872e468bbf1c179ce9f846e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9922254b5804472bb62ada80ba90b859 : SUCCESS in 43m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6bed8c2a596d4081969ca60f40812b85 : SUCCESS in 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f04f9ee60450f22e028044f9fc22774413b7943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 12:57:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c7a6b5169b4fc3857481311cb09627\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/91787705855e44969865f672ac3341d7 : SUCCESS in 1h 19m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/93645b5ddc4c407aabdda1f6547a410d : SUCCESS in 1h 14m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22eb4601095f418ca112da8ac89e3065 : SUCCESS in 1h 12m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4e8f8d2b68404545870cfc85c98e81fc : SUCCESS in 57m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b28ea673e28c4802bb49632232e4dd5b : SUCCESS in 35m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9372a70de71a4782ac5782ff3a68f3ae : SUCCESS in 39m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3081172de0497695ee482b3d0cc5df : SUCCESS in 15m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b044b031a4249778b58547ff40c0953 : SUCCESS in 34m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89a2e7f2d38b45d58c2f207ce5dbf3a0 : SUCCESS in 26m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0dc017c29b248fa9dfea4227da1b8d2 : SUCCESS in 11m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e51611b52439411d8ba1b9db06ef9760 : SUCCESS in 37m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfa1b9c72a624945b48b0f0647117932 : SUCCESS in 1h 43m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a15d3d31fb854ac7a7954c1d281e9a98 : SUCCESS in 45m 22s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/74483bee01874b7cb6afcf4af1c46032 : SUCCESS in 1h 07m 41s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b194562a423f45e69ffd80e03c1ed182 : SUCCESS in 57m 14s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d5606cac80634ed8807fdfbd9246ad7d : SUCCESS in 1h 10m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fcc876bba1c34f9680c4babedba65cda : SUCCESS in 1h 08m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb6c8b1986634731a514275b4ca116bc : SUCCESS in 1h 23m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2a3f34a543b04dee85d5cf32990f8024 : SUCCESS in 1h 17m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3b9fb9bf82f447faaf2308ceead5055d : SUCCESS in 1h 25m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b79304f7c33344a49847c090beacc4bf : SUCCESS in 1h 27m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bff6ab5ee33c407994ec8a5849c0dfe2 : SUCCESS in 1h 26m 05s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/096a4e1494d34e17bafba6f2bba8cb62 : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b9d36768c7e51cd99e6e5ec56fe87447f5e9d03b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-01-12 17:59:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5bd22e429eb0f4ee95688937de066685fb2f6d79","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-16 09:13:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ebd67f543f74df32060109e98892d4c6f822c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-23 11:35:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c245400fb3a4c17d65ce9c704c37942a99a9a8d0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-23 11:35:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb69662a9d920d2dc72e882c815998ccabc43b8c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-23 12:09:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/54/869554/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"dd44a2d08126b2d2945a4cba5a934786533a5f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 12:46:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93b27d7ee0b14c3298078f60204cc346\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d379a28076fd4158aeb8b256455af29e : SUCCESS in 1h 08m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ac7a6d010634031826f5942a68b24a3 : SUCCESS in 1h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfe3310d95f07b5d33d1a354429dc3e53e617bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 13:38:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29deff2bb996431db13c4ecaa7674fa6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f3f101cb73ea40cdb50fc661702b2e01 : SUCCESS in 1h 33m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/61f3b7a842914c5aa40b3cd7b5f37507 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6a0415a910c74521a327bca31248e043 : SUCCESS in 1h 25m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/58fa1c8c393441e080ca32777151c3ab : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d08d9a9f23124df28c824185ccfabda9 : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3337b38c2a414288f76b95367c317a : SUCCESS in 41m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f402b2b9ec45609441488705a53cb9 : SUCCESS in 13m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d447f9e3e82844ef98cb709fe344a4f1 : SUCCESS in 27m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86a307566bb844a2a5767bb73430c299 : SUCCESS in 33m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380eeb74666c4223a9e48772946d9dd2 : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1864b7bbabc43a88fdbe8aece4c4a55 : SUCCESS in 56m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a160f1e54c5044bd818771c864656adb : SUCCESS in 1h 47m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2adcc66b250d496381ee189e8c737194 : SUCCESS in 1h 03m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5b04967f11124355856f79cece0cf5db : SUCCESS in 1h 07m 08s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3acedbc7dd1d4c89821f0528a5faad3a : SUCCESS in 1h 13m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/957443a553404946a48d7e9ddc136550 : SUCCESS in 1h 07m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/58f0285e16c5452dbc465fc5fbee8043 : SUCCESS in 1h 31m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e256326e42154a9bad722c05ef6d7308 : SUCCESS in 1h 38m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a2229938a6a441338b8378108aa267ee : SUCCESS in 1h 06m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/361bff2a76da4ee185e8d3d48cdbf155 : SUCCESS in 1h 56m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7fac2cadd4d14834a40940321ebfa8cb : SUCCESS in 1h 55m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6876005abab4ac387fc86d62c173212 : SUCCESS in 1h 22m 50s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/69197300fd324f4687f4081b74f1b1e3 : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":2},{"id":"dc8c3ae70047ea1fb72cc3547e06234094ac9ac0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-27 08:39:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a48e46239085dad042f261a22e56b4e66e553b3e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-27 08:46:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72c7e78733cf8787bc07792d36e8575cf69a505d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 08:47:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0be10ebb89a7e8c3dece2648eb0454423b98d858","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-27 08:55:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0707c78349cc92f92a8f15ce8a1e06d766880803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 10:47:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e53e80c5d84d08a3645163aa6dcfaa\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/51c219d070034533b3ef9ff7b4a8dd9e : SUCCESS in 1h 16m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b89940a6e6cc41818ddb0a6e3ccc2d79 : SUCCESS in 1h 04m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5f67a0c3e1434b528c8e8876e77183e5 : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f869c32ab4304b74b86890198eca8233 : SUCCESS in 1h 24m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5042c96ac443beb7a7ca5c7bbf3e39 : SUCCESS in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ef42e522b184a708ef8fa0613d9838b : SUCCESS in 30m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8554dcee149f4d36a0d22783dde27839 : SUCCESS in 25m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a9f8cb4bd84a1e8acf0c98c0732caa : SUCCESS in 9m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cf3361431434a8185c1108337d1ec33 : SUCCESS in 47m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/91a6aa1f270b486b94d94e4c25695ee1 : SUCCESS in 1h 29m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b72c236cfacd4f7b892c530e7f444b80 : SUCCESS in 1h 30m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/324408e7454c47c1b8773b84f6fc8715 : SUCCESS in 50m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/453dfefdbd9541a9a4d64ddc33cc2dea : SUCCESS in 1h 04m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c438175aa9e24ae8afa1614502cd0a76 : SUCCESS in 1h 01m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3287fd53e43f4309add1e09d9455e455 : SUCCESS in 1h 38m 41s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/092b37b79da74c2cb8ab28f92a0b8b93 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":2},{"id":"0226384b91bc51a8f5ca3adfdba4fb31761d040d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 10:47:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"97b38121d0ee756402cd6d904e3f07a0201904fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 10:48:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57b22995c8d144a89582a18900f007f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9472450db594319a3bebb651f7f7388 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d73f75c551c36b15eed43f0b19ed796626cc8cb5","revisions":{"491202951d198e512649ae7da2fda985b03582d9":{"kind":"REWORK","_number":1,"created":"2023-01-09 11:08:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/54/869554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/869554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/869554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/869554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/869554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/869554/1"}}},"commit":{"parents":[{"commit":"1374b01cfb37c4f1ff745a1614403e6d913bd13c","subject":"Merge \"Discard port with ofport -1 in _get_ofport_moves\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1374b01cfb37c4f1ff745a1614403e6d913bd13c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-01-09 11:05:28.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-01-09 11:05:28.000000000","tz":60},"subject":"[API] Add API extension and definition for default SG rules","message":"[API] Add API extension and definition for default SG rules\n\nThis patch adds API definition and API extension class for\nsecurity group rules templates API described in the spec [1].\nAPI definition in this case is very similar to the securitygroup API\ndefinition and uses same converters and validators which are still in\nNeutron instead of neutron-lib repo. Because of that this new API\ndefinition is proposed to the neutron repo first and will be rehomed to\nneutron-lib together with security groups API definition later.\n\n[1] https://specs.openstack.org/openstack/neutron-specs/specs/2023.1/configurable-default-sg-rules.html\n\nRelated-bug: #1983053\nChange-Id: I3aafe1aba406a52bc2b57be5133dee15b8848796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/491202951d198e512649ae7da2fda985b03582d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/491202951d198e512649ae7da2fda985b03582d9"}]},"branch":"refs/heads/master"},"d73f75c551c36b15eed43f0b19ed796626cc8cb5":{"kind":"REWORK","_number":2,"created":"2023-01-23 11:35:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/54/869554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/869554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/869554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/869554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/869554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/869554/2"}}},"commit":{"parents":[{"commit":"1374b01cfb37c4f1ff745a1614403e6d913bd13c","subject":"Merge \"Discard port with ofport -1 in _get_ofport_moves\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1374b01cfb37c4f1ff745a1614403e6d913bd13c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-01-09 11:05:28.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-01-23 11:35:45.000000000","tz":0},"subject":"[API] Add API extension and definition for default SG rules","message":"[API] Add API extension and definition for default SG rules\n\nThis patch adds API definition and API extension class for\nsecurity group rules templates API described in the spec [1].\nAPI definition in this case is very similar to the securitygroup API\ndefinition and uses same converters and validators which are still in\nNeutron instead of neutron-lib repo. Because of that this new API\ndefinition is proposed to the neutron repo first and will be rehomed to\nneutron-lib together with security groups API definition later.\n\n[1] https://specs.openstack.org/openstack/neutron-specs/specs/2023.1/configurable-default-sg-rules.html\n\nRelated-bug: #1983053\nChange-Id: I3aafe1aba406a52bc2b57be5133dee15b8848796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d73f75c551c36b15eed43f0b19ed796626cc8cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d73f75c551c36b15eed43f0b19ed796626cc8cb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
