)]}'
{"id":"openstack%2Fneutron~869613","triplet_id":"openstack%2Fneutron~master~I160dfc4e893b93ac7a40e19b3dd6b89750dac57d","project":"openstack/neutron","branch":"master","topic":"migration_nfs_subnets","attention_set":{},"removed_from_attention_set":{"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2023-05-29 18:35:03.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-17 08:03:28.000000000","reason":"removed on reply"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-05-29 18:35:03.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-04-17 08:03:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I160dfc4e893b93ac7a40e19b3dd6b89750dac57d","subject":"[OVN][Migration] Enable settings backup subnet for NFS clients","status":"MERGED","created":"2023-01-09 22:41:14.000000000","updated":"2023-05-29 18:37:22.000000000","submitted":"2023-05-29 18:35:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869613-migration_nfs_subnets","meta_rev_id":"f0ba00fa30a4e4f52b17535faa134c0ff534cb74","_number":869613,"virtual_id_number":869613,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-29 18:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":2,"date":"2023-05-29 13:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-22 14:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":1,"date":"2023-05-29 13:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 00:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 02:07:35.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-01-10 02:07:35.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-01-10 02:07:35.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"state":"REVIEWER"}],"messages":[{"id":"9aacb7605c55f49047f0277d74709efe52ed0d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-01-09 22:41:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f57ac88b3d155655f2dcae8c3e4fc64d491f3647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 00:05:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c2d2f43e11346f886a05b5796a82474\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb0b32260254a68976ade2eae6d7559 : SUCCESS in 12m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6635752128b443b2a8919659493e93b1 : SUCCESS in 8m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fed87ee0358462ca98e587edf5ae172 : SUCCESS in 54m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc5e4669801e465c973a4e9d50a8449e : FAILURE in 1h 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"eb127b6d5f3de43044fe86a3beafb909cc2ecf41","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2023-01-10 09:44:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d896ed32b2c75893ab81474164cc42f736d5002","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:03:28.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"ed340d6e92a74c36e174fdc2dc1cae8ee6c3cf09","tag":"autogenerated:gerrit:restore","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-05-19 19:17:47.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9aeef2f15a2280841e57f62494e5f68bf08e5fe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-05-19 19:18:32.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e336290dc6614365ee93eb3f91e309b3dd9633b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 19:36:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b009af147d41f4ae36644e29ba1ea9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b54489ba86459bb15617683b0c9110 : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831575e8725047d5b858124caca3ede3 : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9ebd78b8f88e03387ba18022cad9e43ff13e6685","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-05-22 14:07:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ee120cb8b2b44621f21e17af4f66db8c2b0994e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-29 13:07:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d39366329280ab7a2b9eb9e5bfbe4284cd79172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 13:08:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72d58f035ea211e751eea78d1a5ebfb8b1114471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 18:35:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b126361fb1b847d4ae3913b71e05d2e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8567f930f3bc4f738df0be2eb1347a74 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57dff0bdab8a46488fb3710e94dfc87d : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b2b1b9de8ea51026e112b0152e9832804750ecd6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 18:35:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f0ba00fa30a4e4f52b17535faa134c0ff534cb74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 18:37:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c57aed62e57479291e4d193bf63e817\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f8bf0cab274e0c8957ba1e3eb92208 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b677d65b2d9cf9e3ff58677300712fe3fe40da04","revisions":{"92047373d0120c206901366abee458905ee45544":{"kind":"REWORK","_number":1,"created":"2023-01-09 22:41:14.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/13/869613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/869613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/869613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/869613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/869613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/869613/1"}}},"commit":{"parents":[{"commit":"d59bf5ff6c75c9f32129cd4ee49c2acb23e8bb9c","subject":"Temporary update envlist for tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d59bf5ff6c75c9f32129cd4ee49c2acb23e8bb9c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-01-09 22:21:59.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-01-09 22:21:59.000000000","tz":-360},"subject":"[OVN][Migration] Enable settings backup subnet for NFS clients","message":"[OVN][Migration] Enable settings backup subnet for NFS clients\n\nIf the BACKUP_MIGRATION_IP is set to a different IP outside of\nthe default nets[1] set in the “openstack overcloud backup”\nplaybook setup_nfs role[2]. Then the NFS will fail to mount\ndirectories during the backup, because they will not be\nreachable | permitted.\nThis change simply adds a new variable\nBACKUP_MIGRATION_CTL_PLANE_CIDRS into the ovn_migration script\nto allow the user to overwrite the extra-var used for\nOpenstack overcloud backup --setup-nfs command.\n\n[1] https://opendev.org/openstack/tripleo-ansible/src/branch/master/tripleo_ansible/roles/backup_and_restore/defaults/main.yml#L47\n[2] https://opendev.org/openstack/tripleo-ansible/src/branch/master/tripleo_ansible/roles/backup_and_restore/tasks/setup_nfs.yml#L127\n\nChange-Id: I160dfc4e893b93ac7a40e19b3dd6b89750dac57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92047373d0120c206901366abee458905ee45544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92047373d0120c206901366abee458905ee45544"}]},"branch":"refs/heads/master"},"b677d65b2d9cf9e3ff58677300712fe3fe40da04":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-19 19:18:32.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/13/869613/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/869613/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/869613/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/869613/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/869613/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/869613/2"}}},"commit":{"parents":[{"commit":"d59bf5ff6c75c9f32129cd4ee49c2acb23e8bb9c","subject":"Temporary update envlist for tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d59bf5ff6c75c9f32129cd4ee49c2acb23e8bb9c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-01-09 22:21:59.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-05-19 19:18:32.000000000","tz":0},"subject":"[OVN][Migration] Enable settings backup subnet for NFS clients","message":"[OVN][Migration] Enable settings backup subnet for NFS clients\n\nIf the BACKUP_MIGRATION_IP is set to a different IP outside of\nthe default nets[1] set in the “openstack overcloud backup”\nplaybook setup_nfs role[2]. Then the NFS will fail to mount\ndirectories during the backup, because they will not be\nreachable | permitted.\nThis change simply adds a new variable\nBACKUP_MIGRATION_CTL_PLANE_CIDRS into the ovn_migration script\nto allow the user to overwrite the extra-var used for\nOpenstack overcloud backup --setup-nfs command.\n\n[1] https://opendev.org/openstack/tripleo-ansible/src/commit/e281ae7624774d71f22fbb993af967ed1ec08780/tripleo_ansible/roles/backup_and_restore/defaults/main.yml#L47\n[2] https://opendev.org/openstack/tripleo-ansible/src/commit/e281ae7624774d71f22fbb993af967ed1ec08780/tripleo_ansible/roles/backup_and_restore/tasks/setup_nfs.yml#L127\n\nChange-Id: I160dfc4e893b93ac7a40e19b3dd6b89750dac57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b677d65b2d9cf9e3ff58677300712fe3fe40da04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b677d65b2d9cf9e3ff58677300712fe3fe40da04"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
