)]}'
{"id":"openstack%2Fneutron~869741","triplet_id":"openstack%2Fneutron~master~If714b0e290ce5a1966eca700677e81735a7f27c1","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-17 08:02:28.000000000","reason":"Change was abandoned"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-02-04 08:41:52.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-17 08:02:28.000000000","reason":"removed on reply"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2023-08-16 14:54:51.000000000","reason":"Change was abandoned"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-04-04 07:41:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If714b0e290ce5a1966eca700677e81735a7f27c1","subject":"Set DVR qr-xyz interfaces DOWN on backup node","status":"ABANDONED","created":"2023-01-10 15:55:50.000000000","updated":"2025-04-04 07:41:43.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbfa86ab6df75b26d8ee012b133f073cc81510c5","_number":869741,"virtual_id_number":869741,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2024-10-18 10:43:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-18 10:34:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 16:33:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 17:47:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-10 18:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 16:07:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-14 19:43:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-02-15 07:25:29.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2023-02-21 07:59:00.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-06-07 01:01:22.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2023-10-09 07:33:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-10-18 08:17:23.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2024-11-15 14:50:24.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"CC"}],"messages":[{"id":"0eff1da46ac561f3e9afec0a8b6f334d08628669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-10 15:55:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdb549f6fcadbfd9f69661599f48672f78681436","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-10 15:55:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d0b2b246be942291af27c36f35eba5b383534879","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-10 16:33:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/41/869741/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0f227e44f597b10982dbfc9825936434fe93abe5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 17:47:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b8673861b747968b49142de89ada72\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cb4b707084f4705a24faa4ac5c6cf43 : SUCCESS in 50m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1f1fce25845544b1986a78d8dbf471ad : SUCCESS in 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54ca2e4377b84f80e817439211a7fa1cbeec334d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 18:16:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbc0579f0ac84b9ba5e4ab797ee6b162\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b9a1c3e0b8204e8d898b77f532af28fc : SUCCESS in 1h 15m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fe502343ccc047a48b9ea25b421462c8 : SUCCESS in 1h 06m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/671de5ecf3144a99b5b16830ccd6012f : SUCCESS in 1h 02m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/48253a0176714b67be4946aa23a5256a : SUCCESS in 1h 12m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83780f08df9f4286be8b8ec6c1de1c1c : SUCCESS in 59m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964adff3f87c479183a21edfc8b6b5a6 : SUCCESS in 14m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/063b49046025420c8a00b2a1adc5fc2c : SUCCESS in 30m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/620259c89fd84e778727fb486ac062ad : SUCCESS in 26m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f26d565df60c4ee7a93a76e03d4a0b35 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9b0471a919145e99c3f3d5839bc87ca : FAILURE in 39m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/54d71349a1e04ee79ecb05befac41733 : SUCCESS in 2h 07m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eafaec1a644f4563b87bd7bbb7a4c1e6 : SUCCESS in 1h 24m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a2eeaa832b67444b9351bbf229b82c9c : SUCCESS in 55m 56s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2785a39937a04d1fa6296c67d5c8c3ef : SUCCESS in 1h 17m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/245e7f35e504482b9bc4221bf662066f : SUCCESS in 54m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2a7f11b1884e4d59a5563c6f21b7b084 : SUCCESS in 1h 46m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/457c2e98cf6d45d2a955706ecf3f6ca4 : FAILURE in 1h 19m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef67390bc0de452c877b4b825dc69cf0 : SUCCESS in 1h 06m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/af659744336c4d12a793cbc45ffe1801 : SUCCESS in 1h 28m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09eccf5c9693421cb5085c0f6107eadf : TIMED_OUT in 2h 13m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/88c8abc47629459abfea007f429534a1 : SUCCESS in 1h 40m 35s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4cce616211954b7f84140fc7f3863e42 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"95a8b53c838ce9da1b427bae2441d1e8e226fc69","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-24 14:31:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c562a347c3a11c4842c3c4b73b0549d5d6da15e5","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-01-24 14:32:05.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"d92144bf1f3b66465304dfb01f0a45ac0560b545","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 15:31:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd435fa085d455787b52293a4099280\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b659e4b164145cbbf120596ff2f55ad : FAILURE in 57m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8530ea98e754129be9329c1e7398cec : SUCCESS in 55m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28ddf8ed1241ce797da6c8216a5f846a086161e9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-24 15:39:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/41/869741/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"831bc80ca12b1f838675bdcde793a414a766a34e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:51:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ed45be2c15a48199082f255396abb09\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/19f041392a154eee9664f70bb7af17b6 : SUCCESS in 1h 11m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d1a503f7846442a68b4d4ab70819ae6b : SUCCESS in 1h 03m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a8cd79eeca64053a4d49174124398a0 : SUCCESS in 1h 13m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4cb301001ebd496c9ab8d4d708b7c56d : SUCCESS in 1h 07m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e98b359ebac84b028e4df44572c34c6f : SUCCESS in 38m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47ae14188da14775a7555c477f54356f : SUCCESS in 11m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eab6d9f763e9490aa0512e3128369919 : SUCCESS in 28m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3adf454a42af4042acf348af573d31f7 : SUCCESS in 30m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6929b4f6403842d4943f73325ba0b1cf : SUCCESS in 12m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0495e1fe1184cd99c80bad2633fddca : SUCCESS in 1h 03m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2948e66faa20403c8c09116947843392 : SUCCESS in 2h 17m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/afd603f924284202849bada08a2600b6 : SUCCESS in 57m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1026d4539b334c288f12a5b48cb535df : SUCCESS in 1h 01m 35s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1cf6fa1d4abf457180999c158b92eb4e : SUCCESS in 1h 06m 58s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bed95959db754d0fa264663cf7686bd5 : SUCCESS in 59m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5cb989ebf26e4f588a2097740e73996c : SUCCESS in 1h 43m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ad8862c0fdf4aafb4267eccc5e8ab32 : FAILURE in 2h 11m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cec7474e2b8f400bbf089238b763861e : SUCCESS in 1h 16m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/64d64d3520a04aae914b68bd3fdaf71d : SUCCESS in 2h 10m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/166da9e61ac841639e0758c559efc06d : SUCCESS in 1h 53m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a062b00407a4a01806c92c1f0359c30 : SUCCESS in 1h 50m 40s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/45b988319ae049c4979ea0c23ba7dc98 : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4b942057cbb3bd2dd61bbbd5b40e5bc6ace407f5","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-13 11:02:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e0d9fd60b0306e3776e0a782aa05f525ea3199c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-14 16:07:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83ba087f8a02dc4731e2ff1735bfddac646467f6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-21 07:59:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ec5fac00a330a71d0cbd522a7b5cf10b749b25a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-21 08:22:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"528e21e5680bad673dc6e486446a5c19f1b37ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-22 08:25:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4349312519095bcb53bcb5ff7f6220a3ed6f41ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-22 08:26:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"28f79b55dc9a06bb2bac357cb0460d898c3dfaf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-22 08:26:30.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a6babd46570064b2e97d9ca403fd126bcfc0db23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:49:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e0bcab6c81448e098a59f251c538443\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eef2d7dfd40141469c4517757bb8276a : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32cfa3fa543d484f9849956313e01989 : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a2d9675b3244755ef3c0c8b5f5e967086bf4dbd6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-22 09:03:31.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/41/869741/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"577a9214f342bca86f76ce37a0613649de9e11cd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-22 09:41:07.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/41/869741/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"cdc5a33f5e420cd4ae419b2c617339aa405d94f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-22 10:18:31.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/41/869741/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d3bc3fe862b15dd1282bdb3f362d02fc667500cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:45:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a1f98dc4d444f3e89a8a01d743fee17\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c55626f86bff47eea7b759185a7e186a : SUCCESS in 1h 48m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2f7632e54d1d49b19cbc3572446a6e62 : SUCCESS in 1h 12m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67b1cd9ff071480593d8559dc657eb1d : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e5f31726b6d44486b110376421875889 : SUCCESS in 1h 06m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8aa9c0e637f947808ac45d80c0c6fbcf : SUCCESS in 1h 21m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09dea0d0201145daaea235f66afcf006 : SUCCESS in 36m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2221b609ed4f69b68ab25f65a6af0c : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/744316346d98481fb403d6de17aba12e : SUCCESS in 24m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc05e8d61f104a8a9c62ab513f727764 : SUCCESS in 23m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d128dfffda7146b5b07e43cb8541bcc8 : SUCCESS in 10m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f215043cdc6b4ca4a444304eae3274cb : SUCCESS in 25m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/694ebe10418d445483de6a6d53a9a241 : SUCCESS in 58m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eee1b57ec1834a79943ae2f736574175 : SUCCESS in 1h 45m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bd4c7a1f79a417eb001f6591f2037f0 : SUCCESS in 59m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/80b699170d8d4d04849c42d94cfbc8bd : SUCCESS in 1h 02m 21s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9cceba5ad93a4a91be6b22e86bbadfd0 : SUCCESS in 1h 21m 38s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fb26df16e4e741b18c2ec2692aeb2796 : SUCCESS in 1h 13m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fac5db3ce6bb49ddadc8fe2d58a9973e : SUCCESS in 1h 26m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d28cd6021cfb41798736310a49d1e598 : SUCCESS in 1h 26m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2198517f35b44606aa09d8c8362e2f8f : SUCCESS in 1h 08m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1453a4adb8bf4bc39507d79030bcd68f : SUCCESS in 1h 55m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8b257e0777c048d992c571e5f38af516 : TIMED_OUT in 2h 13m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9618933a6ae14b07a78000563e3d583a : SUCCESS in 1h 31m 34s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/99a9e9a027b14200b57238dbd2682c45 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":5},{"id":"66d8bb372c6a68261edbfeadcd708fc90aaf497d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-28 17:48:07.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"11a893e8947c814227f96935cef949ebcf46b567","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:02:28.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5},{"id":"c49de125ed0fc08196aa28ef3a8a35ee19f77d41","tag":"autogenerated:gerrit:restore","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-05-04 09:52:23.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"e49c81aac688736f4c7ed0b7ae1507c6d8745713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-05-04 09:52:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3c1d72e717fa8e8c95c023db934cd73938eefebd","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-05-04 09:53:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a3e9f18c941952da5df054d31cb501b45f73cb2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:27:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae751fde234490aaffeeb9651a60e1a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24a2152f9b094bc7bf0a049289315596 : SUCCESS in 33m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/418f0347eb534740b94bca8dec486bd8 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1505a977ff2c4c178ca57b074f659a6d : SUCCESS in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"68e3834bb5980612845c362c1665961d03ace793","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-04 12:34:17.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/41/869741/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"c081e8ddcdf3ab7d9aa96d1226759ba44ddcdc2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:45:10.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8973a80bbc14b62b1e47d068df74579\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0038db8d74df4d76930b649bb523bb43 : TIMED_OUT in 2h 50m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c9a1ef162eba41fa841460a8a407a7e8 : TIMED_OUT in 2h 50m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/56bc9a11e3e04086bc033f716ef779fe : TIMED_OUT in 2h 50m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ebf88d97467465aa9e3e7ecee507d95 : TIMED_OUT in 2h 50m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3693ec9ac1c64b19aeb4d062c51c4d0c : SUCCESS in 1h 52m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a264d2298b8249de97ce49b6a481e5a3 : SUCCESS in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa7a41f2b28432280bc594bcb421053 : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/460e4459035b4d4d9f87c5f31a1591a5 : SUCCESS in 26m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25906288afc647e791a28228caa47338 : SUCCESS in 29m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de3159eed1bb445a8048edd7dcfe6762 : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50594f335dd44f0fb68f3c2cac79e3f1 : SUCCESS in 12m 19s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f992e1a9c58d40f48e03ff65471ad5a2 : SUCCESS in 25m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06251a3853174c1da3bf7068250bd190 : FAILURE in 58m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffceb60c10e4483f879d6fc5042ed95c : FAILURE in 1h 23m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/26531c6075bb454092ae4955776ed247 : SUCCESS in 55m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/874fbf25bd384a1884c2104515d8cafe : SUCCESS in 59m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4db472ef3b474005989b73b1bed96f6a : SUCCESS in 1h 45m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da0f845889034a79b4725732a093f3d0 : SUCCESS in 1h 12m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/079072b8e60a45b180bf9f74677bda5c : SUCCESS in 1h 08m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/59a4642146a940a58fc51d5879cacaf7 : SUCCESS in 2h 03m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b040d75db1c940e59d817c850bf553a0 : SUCCESS in 1h 56m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e10e29a9c26f4cbe9f786f9d14e71bf1 : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":6},{"id":"e702071c6550165cb6629c7ff3a005224475f91c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-06 15:14:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"25404b1ec73b2d3d208fa7d741334ab5094943fa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-06 15:54:46.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/41/869741/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"fdab0c9c4c8da1b22774206972c926cc070c09f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 16:05:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02aca2a6880e468c9c5d0f32a5c99dfe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5990438d8a7646d5bce141cabd81b353 : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75f8c12df945407fa05be6d8e6bda73f : SUCCESS in 46m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/caf158da53114e839f83075de467c6c2 : SUCCESS in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1d8d265288430c3ded45982437a706001763c722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 18:06:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf7f3df6f1ac4106899aa10a61ac1c07\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/573264581d94477ab98b4246b35434c8 : TIMED_OUT in 2h 49m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5ba2c63e55648a6b2f14368ac67d32f : TIMED_OUT in 2h 49m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/807cef8e746f4699b73fe7f86d2563d7 : TIMED_OUT in 2h 49m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f99b6cf0086d4fd9843e1c6c0e46444d : TIMED_OUT in 2h 49m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7435562055054249a8f41496bb9a4225 : SUCCESS in 1h 40m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5172bc9fb7274571ade10ff1fe8f7abf : SUCCESS in 36m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e0ddbb5f994630ae5ef330e6de2698 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa604e9b37e483993c31d20dc327828 : SUCCESS in 29m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8597c50b5f754a629c4f3d751bfe9e35 : SUCCESS in 32m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2303c6c6c98d4832aaf9febdbfeea723 : SUCCESS in 28m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b8dee9a1be4f608a4558bbfd5b46e2 : SUCCESS in 13m 13s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bde35364a969409d9e658368b47a839d : SUCCESS in 25m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02e3dca0d4ed4dd4bae8003e123a2ed2 : FAILURE in 59m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71ae590798b94d4598513a661a88fa69 : FAILURE in 1h 51m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/061e35d4e7c84deb8e873c86cd912f6c : SUCCESS in 43m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3afcb674481b482994d1bc9ba5042e08 : SUCCESS in 1h 00m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0042bb2a703a41a2a4f6174e5d5e935d : SUCCESS in 1h 26m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c2a0e363fac64265b6900d0452b295b7 : SUCCESS in 1h 43m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/707862b1f210472aa0087855dc2c5c5e : SUCCESS in 1h 10m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/687b83cbcfac45d0bcc924c00ed625a6 : SUCCESS in 2h 08m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99ea89cb8fb9478d9b9d8ac76f8b8d3a : SUCCESS in 2h 11m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bde69466203c451f905df186cdb2af9b : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":7},{"id":"1ac4ef0e97a4982b286a52ed5cffe917873da2de","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-06-07 01:01:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"331e8e23270206418a5bd8cb6dc92a67e5df38e8","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-09 14:08:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7c3d98cb80634b025971011c54e1a0812ed3dbc6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-09 14:18:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6be9f0159dbe70f8659aac9a5d9c42f65b22af48","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-16 14:54:51.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":7},{"id":"0177e611fbe26d3cbff587987632f7ef0e7010a6","tag":"autogenerated:gerrit:restore","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 13:59:25.000000000","message":"Restored","accounts_in_message":[],"_revision_number":7},{"id":"866283800b6b4bb788b64ab6165d7d3d2d1b50ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 14:01:31.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"e6ed2a045a45fd1377d2267e5f4f2b320ad8832e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:05:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38377426293e426c8732dd6b0b153433\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c21ce84bb6aa4df7b3fbf6c42325a1b8 : SUCCESS in 59m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a63b98eccb204781afc2374f485fb728 : SUCCESS in 1h 02m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f4defd86f7f4289951b5181cb3595e9 : SUCCESS in 1h 00m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7538415ae10b4948aec11aa6e3d0bf18 : SUCCESS in 52m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"09cb16820fec9711e5c1d31934656b1cddb5e481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 16:52:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6f7609ba4314b9ca5b4ba1a5912d7c9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a7aeef2105b242d8a6cd054ba6f3a1dc : FAILURE in 16m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e50ca4947504d99800d77a9aa665b28 : TIMED_OUT in 2h 49m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ebc92ae9d4f84b42a17011d78b7e5a96 : TIMED_OUT in 2h 49m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87b531b91d23476ba5b0cf91638add5b : SUCCESS in 1h 45m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15f0dcbcbf2e4db2afeb089dd8e85684 : SUCCESS in 50m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f700d058bbea4b11a065746e3262db4d : SUCCESS in 10m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e283fc19fd1425b8d5fee961268e64a : SUCCESS in 24m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/376565f27e9f4eeba315ff7827f51934 : SUCCESS in 24m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e124eeecfca4a46ad38d36969d02521 : SUCCESS in 27m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e2f35d278884a5c94c59197c9a3cf63 : SUCCESS in 24m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e7e3145d734026b612b59de76953a4 : SUCCESS in 13m 03s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7e938d5ea6a344d7b49cd7897b40bfc7 : SUCCESS in 27m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8dbcfacae19449ca185df9ec9d666fe : FAILURE in 1h 01m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f93bb20fd4ba44958dfa1df6091e4f19 : FAILURE in 1h 51m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b98401906b47411c908d398cdf45053f : SUCCESS in 1h 07m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66e2329e5da14f3b93d68e456e9ce8be : SUCCESS in 57m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db1a3d07043c4362bbb3b161313b62c5 : SUCCESS in 1h 33m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2c30a2a4219247be8d86c539ace24e29 : SUCCESS in 1h 22m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4798494011df4ab4ba4a45343be32638 : SUCCESS in 1h 12m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e8cbba6e367a4dde887a45ad8f0b9fd7 : SUCCESS in 1h 50m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0fd71d4da0b487fbb044980b596a9d4 : SUCCESS in 1h 34m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b3596cf9e29490ab3b0e14e38ea2fb8 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":8},{"id":"4efdfdf58c3185eb1c56e654bf5a36506f51fdf8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-07 22:58:13.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/41/869741/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"d43246209018fcfa114cd0f01616d5564a43ee1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-26 13:03:51.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"df82178aef7fa1776f160a5da9121d142693c162","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-26 13:52:30.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/41/869741/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"90000315677cfaae5af6b54c42ebf62a283a4a85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 13:57:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e49d62615bfb4e25b50d57ae0e97f765\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8b2ab484ec04f6cad20ff1bfdde86a9 : SUCCESS in 51m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d136721570004c47bcd341d3d0e438e1 : SUCCESS in 52m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a0e7eb338ad458ea98dc77ba65db9ea : SUCCESS in 50m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/679b934e3169424e9fa8950451c79757 : SUCCESS in 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b66dc487c5d20e01417d3f5d142aba44d9cba0ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:56:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90a608daa05142f7838a9791ad1ace89\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e1902fa40214747b15a2327cafea202 : TIMED_OUT in 2h 50m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a75cbd797abd4471b6129fdd33a81d5d : TIMED_OUT in 2h 49m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f6dd869890034d24a9d187e7175b9b10 : TIMED_OUT in 2h 50m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3406654d055e4195b48db6bad0789dfe : SUCCESS in 2h 12m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd3a92f6184e4a0cb65c43e998d8a121 : SUCCESS in 48m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91db9dc84c5c45afb1501675f2ed528e : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7494d2030a08482aacd2caba8d34ee0a : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fccf44c127ca4b99a3a8f4aef4890266 : SUCCESS in 27m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a652f752ac54288832c8b4cc803c55e : SUCCESS in 20m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cf995a267e84241b41d20c73d11a434 : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78e7002e4a54cd2a1c6199e318d4af8 : SUCCESS in 11m 30s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e2826ea8910047dbb4474d21aa18a9c1 : SUCCESS in 29m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae6b23ebbad242de892237180af1da91 : FAILURE in 1h 05m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d23c295f7ee4d3ba6374ffa08faea91 : FAILURE in 1h 56m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6651cd6e9d1447e8b5581ff94f7eea8d : SUCCESS in 48m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c23af1a7458a4670ad19824b4fd5ed09 : SUCCESS in 52m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c932b3266d354894949c0d8431bd4423 : SUCCESS in 1h 33m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0323a98cb0de42d781af0930236c7982 : SUCCESS in 1h 15m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a1c401765b534e44a1943ccd3f74c173 : SUCCESS in 47m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0e0edd69ae9d42669c9cb82499e153d1 : SUCCESS in 1h 25m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc934abf9dc645408a4818c233c447f9 : SUCCESS in 1h 58m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1970d92036f245ccba74896fdd3f6624 : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":9},{"id":"e1ff1a9ebfad337399b06ee97e653398f566f6d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-28 13:57:41.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"354d67cc12b5900f0731ac2e29eb12303b740c22","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-28 14:48:54.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/41/869741/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"eb333a5467009a606e6c07341e5e1653888fc93b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 16:09:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d49776c33114aaba17975821d74eacd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9e5ac8f9c2f41d6ae1749664528e526 : SUCCESS in 41m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43c2b7be85fe427abd1d2518ae58f4fa : SUCCESS in 49m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ce4642f07a84dc5957c125d83189753 : SUCCESS in 41m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c6f2bdc028c44aaaf81045a8b8d30e5 : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d8e91354351e7b0176b952df3384e55265a36827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 16:37:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0db76c741014bf891d147d01bc8501c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/374c7d0aa6e44468b9f6ca7724ca9065 : FAILURE in 1h 44m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92854e2fe02e4e388af6a93a3fc8e487 : SUCCESS in 1h 55m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d311c005c2da48adae86b9e22581657a : FAILURE in 1h 45m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/83b12214db964c90bd31cb69f68cf399 : FAILURE in 2h 04m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88f86ed58f734f09a7940cf0df6f9d67 : SUCCESS in 44m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eacfe416e54fdb87a4264057831f3d : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef666509a224d9984adefd8764b7fb7 : SUCCESS in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9eb44f5c04b49ccbe2b6895c21d580b : SUCCESS in 25m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c14058e589a49e5a057dc8725cdb896 : SUCCESS in 24m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1706d5cff184123858658a691563fe3 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4575ad32c6b0493ab8d3216ffa1241e2 : SUCCESS in 12m 47s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c0f44c38e8b44fb0938328af75b4ca51 : SUCCESS in 28m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4692a5a4f8a433b9ba9d4d691cd44de : FAILURE in 58m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06588b88420745bfbb146b45a64215d5 : SUCCESS in 2h 31m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ebf02ba2cb045628a69bdbd433d1dbf : SUCCESS in 41m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c62ec731884e4da59a283e1427d08ffa : SUCCESS in 41m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b9bd50e1666452e85044cb4401de97f : SUCCESS in 1h 24m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2d6b8d563184f8bb761bbab0614200f : SUCCESS in 1h 21m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4c59c9f2918442b48e554763756214ac : SUCCESS in 1h 04m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3ce61b4ce1854bccbbf66557a035f647 : SUCCESS in 1h 32m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa68139ab97a4e76a316ce70ca3a6248 : SUCCESS in 1h 27m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f387ab6c97c4022bac6367cf54df341 : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":10},{"id":"c6d298709794c95c596cf31a737380d7c81d9e4b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-09 07:33:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"32349b533101c47164430f46ecdfe6f0d5b756cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:03:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc9f99ac3cf4f05a8811cdce54c8a05\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f848075b2616435fb67b25fd3c94822a : SUCCESS in 39m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b45dda7c52645c790b13598ab0abd78 : SUCCESS in 1h 08m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e76dc68accbc4080a28f316b0ff913ee : SUCCESS in 1h 14m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8a424392146f4afd90059afb31163778 : SUCCESS in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ce7da0b074fa50741950d568d746b59ca75bbae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:20:06.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0df7680c302a4ecfb889e4f3338e635e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5dbd6c1827f94ba0b47fb51654c8a962 : SUCCESS in 1h 24m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b965708786f4c339d267a3afa23d612 : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da771953881e41ddb2cbc11f98f4a1b2 : SUCCESS in 1h 52m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d327af93acfa42fa8945131f474084de : SUCCESS in 2h 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fc3c719321a4858b4976a729f15f7f5 : SUCCESS in 35m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76d93263b334b22ae7c18036bb33426 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57aec29673ec435dacb1f1fb237c7120 : SUCCESS in 21m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3586bc2dfab349de9ab76dc2084d0f8d : SUCCESS in 24m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46fdaa4589084d1db4f143a726054c5a : SUCCESS in 24m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/345a609558d5402ba4e79652962dc74c : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1a88f8219742558043d76129b41c00 : SUCCESS in 12m 48s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6367f3a229f143939663a8199b484f17 : SUCCESS in 29m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f613e1adb21a471f84291aadb98b319d : FAILURE in 52m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b67c298698048ddaaa97faa9bfde0f7 : SUCCESS in 2h 39m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b12642f4b224f23ae1864be0b233228 : SUCCESS in 52m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a02d03f9a1514cf187a309fa206dc44a : SUCCESS in 1h 03m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6656414fc5b84610a42031fc45a755e7 : SUCCESS in 1h 12m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b0696354dbb441219c2f4a6e9a621827 : SUCCESS in 1h 28m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/07ceb9f0f8614a38a91235a2990d660e : SUCCESS in 1h 08m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dfe02d4d8eb5406fbc654e89f7bbaf20 : SUCCESS in 1h 37m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f13b933c9a54eddabd8558cf0e453b3 : SUCCESS in 2h 04m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/04e984555ba14cc89a7d8da1a6959fae : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":10},{"id":"9662fb79b8ef60cae0a83e013b83b44685c10eed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-04 11:03:26.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":10},{"id":"859a27d07e3ed26dd30009f2a23fe7f248ff7b8c","tag":"autogenerated:gerrit:restore","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-11 10:46:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":10},{"id":"a7d2ee79ca762ebb00f50bf2bcae07c90bffbcbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-11 10:46:21.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"ad860654ef1390ed8d890787172f16ed09ca3f58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 11:07:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ca81a2cdb414ff383c352c61adb6e21\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6fbd41a838b4a7280bcdea3ab80b35b : SUCCESS in 18m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c50a1f8fdb84e978406693cc6d55244 : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cae76c345428432e966d1103b05a5ddd : SUCCESS in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b004ea60fec69201b8711223c59343eb1db24daf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-11 11:16:37.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/41/869741/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"10eb7557614c1a83f9b05f8b9e3d10a04bc249fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 13:00:15.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afad73717c5c4dab947e9f7af7081a3c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/884fdff096d3455aa89be4a7b19bb2ca : FAILURE in 1h 40m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3defa8e9f47f46e9a17ef7e72e892caa : SUCCESS in 1h 52m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/bf362c540c684ee89c2b8d6e009cab19 : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d046276834ae408f96b47276b263b8b1 : SUCCESS in 1h 47m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e120badaa0d047afa118a83537faa785 : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/630195743a2b40639fe0319b8f1f6c5c : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b17279362914313996ac1aac76a0c39 : SUCCESS in 19m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16861c412e2040d194cdaeedeea8b937 : SUCCESS in 18m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892920e7d7444e77bf9e6da27813070d : SUCCESS in 9m 56s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/89377fb0fde74778b9a178bbf2669bf8 : SUCCESS in 22m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/977ca2821b0447318cb85f81ce10a15b : SUCCESS in 44m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9fe46d8056847ab9db5f22529bbd605 : SUCCESS in 1h 06m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38753340e00d45759c9f80e7ac8b1f19 : FAILURE in 54m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee6e4f917b1241e4a163abeeb5083b56 : SUCCESS in 2h 06m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2be540f4056545e697f6052849347279 : SUCCESS in 1h 03m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b8b640896c9b49468ab1ef8a912adb99 : SUCCESS in 1h 11m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/95fe44cdaf7c4a2ba24eb5c55778f416 : SUCCESS in 1h 35m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eba26e6ec1864dd69b20064a52aa3b93 : SUCCESS in 1h 30m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f9d417511c66473f87045cb2c1f9a7c9 : SUCCESS in 1h 06m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2a96a3a2c8544046905ef5738aa265c3 : SUCCESS in 2h 06m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e757c424364463aaba3700d42a7488e : SUCCESS in 2h 03m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9ef247c4b81948298db1c31de6b00a65 : SUCCESS in 1h 21m 55s","accounts_in_message":[],"_revision_number":11},{"id":"6d5e3876ea9f02301b823936ffa3e189974984f6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-05 12:02:28.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":11},{"id":"248d7d06389fcadb931220a1d3b2dd7f843bde71","tag":"autogenerated:gerrit:restore","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 14:13:48.000000000","message":"Restored","accounts_in_message":[],"_revision_number":11},{"id":"832cd5febe09c71db47a47a2557693d3c217d64b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:43:12.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3175d816d6b34c0dbb1b3542af30eb90\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6acf37b6c2e478187bd64f0b5d75b81 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5056456aa2bf4a69b93714cc4eb63ae7 : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b22180673bc94477b1ca9d46d9ff7e0d : SUCCESS in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6cbd5cdbda80d07308d4621e68e390e437efb88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:17:22.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cf101b7ba144231a2865a4caaa83b7b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e06fba92d564c5b92da1435a54402aa : SUCCESS in 1h 29m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eddf4d78a27644d2ae9ef9e59becc024 : SUCCESS in 2h 02m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9d761cb150f54e57838e1a4844a471da : SUCCESS in 1h 58m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/643fd8b315c94eac839a25626ca3ca93 : SUCCESS in 2h 19m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7852b2ce75940eabbb1839f463390a8 : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e7142ed0294d5a960f326949e1571f : SUCCESS in 11m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7659d46803ac4d7aa72e4cafe1942f36 : SUCCESS in 25m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81d6c7879cb049339312ab75cdac43d7 : SUCCESS in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6eeb49ea2c040e8b4559f22954ea953 : SUCCESS in 12m 40s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fbbc9bd5466347fb9d945572bcd29bf5 : SUCCESS in 29m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6e211aa5f2af43cf9b3082f843872648 : SUCCESS in 1h 00m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/450ea22cb0de434fbc2777c832319eb4 : SUCCESS in 52m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3064ebcbede742c59d3fef6921677f3a : FAILURE in 40m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fa4a536ebce49fcbfc95491620f5ca7 : TIMED_OUT in 3h 02m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3037389978c545e0aa824958eacde89b : FAILURE in 49m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1ecb06491a724e0aa0f855d63b007c01 : SUCCESS in 1h 06m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/761627ca5e784ae2ab1422fca1fd9e6d : SUCCESS in 1h 27m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/844ac9ea588346bd971e59082c735876 : SUCCESS in 1h 37m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a353aea3d0c345dba49c24b014b190a8 : SUCCESS in 49m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2591103722494b83b315958009ad2110 : SUCCESS in 1h 16m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f94b2ffa609f48a1aa34b767db3eac4b : SUCCESS in 1h 21m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4baa2ed006f4b52a745541451d18297 : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":11},{"id":"a5bcdd876846bbdac923b9731d87662f9e36dc38","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-04 09:01:08.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":11},{"id":"65e0a2e105e5c0a95d5db0805b795e4a002e9e69","tag":"autogenerated:gerrit:restore","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-18 08:16:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":11},{"id":"0983d6b74d5c61c1f478efc882c4d8dca26bf420","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-10-18 08:17:23.000000000","message":"Patch Set 11: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 869741,11","accounts_in_message":[],"_revision_number":11},{"id":"4d8aa8535c078bdeb66f1b84c5bb76ddf4e75dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 08:17:25.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 869741,11","accounts_in_message":[],"_revision_number":11},{"id":"06c8725cec8772b77e5d5beb5e5f77ce0cc93f98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 08:17:25.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 869741,11","accounts_in_message":[],"_revision_number":11},{"id":"d395cbc980265aa7f893e5c8a1fccf7d4bb8eb10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-18 08:21:15.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a14e333bb0c32ec63028d00fd4a6cb86af4d2928","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 08:46:17.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea4ec97386184e0482be3a970cbfd0d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db1237f32a2c443b8ec548c694dd0413 : FAILURE in 23m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5ce18981114741e9aef999f28515f2bb : FAILURE in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6a3b3f39244f5cb64cc0b4c765113d01e916a521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 10:34:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06a51b1599324f3e8fe85481e0214f17\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/afddca28286f49d19187ae69ffad24f0 : SUCCESS in 1h 32m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6634d21e11c942788800ff096eecb7f8 : SUCCESS in 1h 57m 02s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6bfeddcade6a457ab17dfc2dc1ce07fe : SUCCESS in 2h 08m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/541d717327a24b02a49b261aecda01d2 : SUCCESS in 2h 09m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bf9af8fd5734fd0abc24715ea85460b : FAILURE in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca12b46ad36243d680886bcba552bcc2 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1640c5b5a1cc43f39007c53491893309 : FAILURE in 21m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/320325ecddb14179a5cf47167cab0723 : FAILURE in 22m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf1df9577164728a2d138e97402288e : SUCCESS in 11m 02s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/50d752003ccb44049e509efcd55b201f : SUCCESS in 56m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e6736e0c7ea145ad9cbf978b74f77c08 : SUCCESS in 45m 54s\n- neutron-tox-mypy https://zuul.opendev.org/t/openstack/build/36897ce4a1e3422eb16640c1f5dc5871 : SUCCESS in 4m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a49b88a10f604abbb1b8091ed827052a : FAILURE in 55m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e433036eb3f4ec28d149f0b9a8f3f0e : SUCCESS in 1h 25m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99577171595549e8926b6f3add14b39c : SUCCESS in 59m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5137dcec891542a6bd54f70e5285aa61 : SUCCESS in 1h 18m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/076fb02f37da4461afd95efb980ceba6 : SUCCESS in 1h 34m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c52e226937284eb7ae2ac0a27430b746 : SUCCESS in 1h 30m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8a589db01ca45f18adb3fe7baef4541 : SUCCESS in 1h 00m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5d1082c95d2f4c948d25a9cca071265f : SUCCESS in 57m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f410b98c24f24958baf92b28be1b1df7 : SUCCESS in 46m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d032c222c03c44e28b9b7b384b081eb9 : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":12},{"id":"99c1cd8089e8e31ae0bb419ca0549b29e0e35dc1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-10-18 10:43:45.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2024/Arista_Tempest_Testrefs/changes/41/869741/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"4e98b519189ac82662db6a9038fd26f4a9f1ddc5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-10-28 13:04:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"759e03c080b322bed2b5d7101c5fb6812311ee81","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-29 19:10:08.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b8ac3d2691770aa8c93977babc250156b7e48929","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-04 08:41:52.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":12},{"id":"227357f98cc2d621bff23c1f53c67d23a9c1ff0b","tag":"autogenerated:gerrit:restore","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-04 10:27:32.000000000","message":"Restored","accounts_in_message":[],"_revision_number":12},{"id":"c1faf7dc0e242f6e90e7c2ce7803304383b8b5c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 11:07:15.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c68bc0708dcf49039e3e00dfd14c3661\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffaea91ee09540448961c16eeac4d5db : FAILURE in 37m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f0ff781cb824e3fba5b827e2f222675 : FAILURE in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3f1ae22367bf1a34836dac505db6bd47fbae6937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 12:21:19.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f37c62daff4a4cb29f3acd1cb6d3d704\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5a05e12dcbe4378a4cf89330301e0e1 : SUCCESS in 1h 19m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c222814663ca488ab9c89d51799ca019 : SUCCESS in 1h 48m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/779f0872605c4f09b2be43bef3436543 : SUCCESS in 1h 52m 31s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/560b3c0232b34e12824af4efd1eccb2f : SUCCESS in 1h 24m 56s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac6491a5a6ba43bb8540ffa96c3274cf : SUCCESS in 1h 42m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/882e8ea66bd6405a905da1d47ba4b544 : FAILURE in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11d65fd38af4f848f705f9c7a712edc : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7a9f5a0019848739ce6677f408fdb81 : FAILURE in 20m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f06f74f9d36040188b5006904d035d28 : TIMED_OUT in 1h 00m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1127bbe781d45c099a4337c5ca09ad3 : SUCCESS in 9m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/744651935b014f4f94cbb0de1d6391bb : SUCCESS in 48m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aad3868f96e746ecb2e245ff37e1884d : SUCCESS in 1h 05m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d1c2480933bf43739f374c4cc096eef8 : FAILURE in 1h 16m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9132ea329f07461981b132d7c835263e : SUCCESS in 1h 41m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/85557e907c8f4d60be89ec337860c2b0 : SUCCESS in 1h 04m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9812ce62984449b6ab5f59d8a39006e9 : SUCCESS in 1h 15m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/31cd401d02624a5e8d2191c7fd6960ba : SUCCESS in 1h 32m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc357e8a072249cda59af7a1b90d11d3 : SUCCESS in 1h 32m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f825e9294053480999e1af6487d5ad67 : SUCCESS in 1h 10m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/86a969baed4643c6bf05a69fb72d7671 : SUCCESS in 58m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ee06fbdba53a46bcb44c0771a7efc0f2 : SUCCESS in 1h 18m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6fb64747ba3547c78422b956c82ee385 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":12},{"id":"fbfa86ab6df75b26d8ee012b133f073cc81510c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-04-04 07:41:43.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"42018ba570626b5d4ca93b69d2f464679206f7ac","revisions":{"eee359000b97f79218f5e97a4b8265d4022dee13":{"kind":"REWORK","_number":1,"created":"2023-01-10 15:55:50.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/1"}}},"commit":{"parents":[{"commit":"1374b01cfb37c4f1ff745a1614403e6d913bd13c","subject":"Merge \"Discard port with ofport -1 in _get_ofport_moves\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1374b01cfb37c4f1ff745a1614403e6d913bd13c"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 15:54:18.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 15:54:18.000000000","tz":60},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz).\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP.\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eee359000b97f79218f5e97a4b8265d4022dee13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eee359000b97f79218f5e97a4b8265d4022dee13"}]},"branch":"refs/heads/master"},"ee8e85a3be4b125f28be71321408647c4570e1aa":{"kind":"REWORK","_number":2,"created":"2023-01-24 14:31:39.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/2"}}},"commit":{"parents":[{"commit":"21927e79075ce0f3e521e56fca0bed8f1de61066","subject":"Merge \"[OVN] Introduce the new OVN Neutron Agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21927e79075ce0f3e521e56fca0bed8f1de61066"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-24 14:26:22.000000000","tz":60},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee8e85a3be4b125f28be71321408647c4570e1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee8e85a3be4b125f28be71321408647c4570e1aa"}]},"branch":"refs/heads/master"},"b7e8dc168668978a5f18fc12b904e632fca81d1e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-22 08:25:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/869741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/3"}}},"commit":{"parents":[{"commit":"21927e79075ce0f3e521e56fca0bed8f1de61066","subject":"Merge \"[OVN] Introduce the new OVN Neutron Agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21927e79075ce0f3e521e56fca0bed8f1de61066"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-22 08:25:28.000000000","tz":0},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/874536\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7e8dc168668978a5f18fc12b904e632fca81d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7e8dc168668978a5f18fc12b904e632fca81d1e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8e06d86efc5a4e4a14a4a1f4941e90af85b58e60":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-22 08:26:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/869741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/4"}}},"commit":{"parents":[{"commit":"169ef0558926656fe9ef38fea42dcada2786ae0d","subject":"Change neutron-ovs-tempest-dvr-ha-multinode-full job\u0027s config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/169ef0558926656fe9ef38fea42dcada2786ae0d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-22 08:26:20.000000000","tz":0},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/874536\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e06d86efc5a4e4a14a4a1f4941e90af85b58e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e06d86efc5a4e4a14a4a1f4941e90af85b58e60"}]},"branch":"refs/heads/master","description":"Rebase"},"4ef6043085b68d352fa83df09b0fc3756b2ef85f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-02-22 08:26:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/41/869741/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/5"}}},"commit":{"parents":[{"commit":"169ef0558926656fe9ef38fea42dcada2786ae0d","subject":"Change neutron-ovs-tempest-dvr-ha-multinode-full job\u0027s config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/169ef0558926656fe9ef38fea42dcada2786ae0d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-22 08:26:29.000000000","tz":0},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ef6043085b68d352fa83df09b0fc3756b2ef85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ef6043085b68d352fa83df09b0fc3756b2ef85f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e860d33661e944ffe693166e7e96f2be8b4d4b3c":{"kind":"REWORK","_number":6,"created":"2023-05-04 09:52:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/6"}}},"commit":{"parents":[{"commit":"7ebd753bbfadb18be07ac88606083805caacbc6d","subject":"Merge \"Add a method to retrieve router gateway ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ebd753bbfadb18be07ac88606083805caacbc6d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-05-04 09:51:25.000000000","tz":120},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e860d33661e944ffe693166e7e96f2be8b4d4b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e860d33661e944ffe693166e7e96f2be8b4d4b3c"}]},"branch":"refs/heads/master"},"0aee99ac10311074a07f748011430865b69b270d":{"kind":"REWORK","_number":7,"created":"2023-06-06 15:14:45.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/7"}}},"commit":{"parents":[{"commit":"7ebd753bbfadb18be07ac88606083805caacbc6d","subject":"Merge \"Add a method to retrieve router gateway ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ebd753bbfadb18be07ac88606083805caacbc6d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-06 15:14:25.000000000","tz":120},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aee99ac10311074a07f748011430865b69b270d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aee99ac10311074a07f748011430865b69b270d"}]},"branch":"refs/heads/master"},"ec5f9a2b41ef900d70893bbb27e476e37a83a623":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-09-06 14:01:31.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/8"}}},"commit":{"parents":[{"commit":"37c49986994abce3afa9b28cc30b7810dc281f04","subject":"Merge \"[CI] Bump OVS_BRANCH in ovs/ovn source deploy jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c49986994abce3afa9b28cc30b7810dc281f04"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2023-09-06 14:01:31.000000000","tz":0},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec5f9a2b41ef900d70893bbb27e476e37a83a623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec5f9a2b41ef900d70893bbb27e476e37a83a623"}]},"branch":"refs/heads/master","description":"Rebase"},"2285941c0618fa9e69a12d5e90310244a8cc0922":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-09-26 13:03:51.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/9"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2023-09-26 13:03:51.000000000","tz":0},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2285941c0618fa9e69a12d5e90310244a8cc0922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2285941c0618fa9e69a12d5e90310244a8cc0922"}]},"branch":"refs/heads/master","description":"Rebase"},"bcf2f652cb9de84f050964fa615e3c47639b5bd7":{"kind":"REWORK","_number":10,"created":"2023-09-28 13:57:41.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/10"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-28 13:57:20.000000000","tz":120},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcf2f652cb9de84f050964fa615e3c47639b5bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcf2f652cb9de84f050964fa615e3c47639b5bd7"}]},"branch":"refs/heads/master"},"975d92fb752af331aad8602d5190226aa87a37d1":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-12-11 10:46:21.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/11"}}},"commit":{"parents":[{"commit":"a0ef04d3b7b38608842289d352018c3fdfc4e505","subject":"Merge \"[OVN] Remove backwards compatibility with OVN \u003c v20.09\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0ef04d3b7b38608842289d352018c3fdfc4e505"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2023-12-11 10:46:21.000000000","tz":0},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/975d92fb752af331aad8602d5190226aa87a37d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/975d92fb752af331aad8602d5190226aa87a37d1"}]},"branch":"refs/heads/master","description":"Rebase"},"42018ba570626b5d4ca93b69d2f464679206f7ac":{"kind":"REWORK","_number":12,"created":"2024-10-18 08:21:15.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/41/869741/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/869741/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/869741/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/869741/12"}}},"commit":{"parents":[{"commit":"6d05519e93540227d228defd2495d81e6e405ff9","subject":"Merge \"ut: Remove unused method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d05519e93540227d228defd2495d81e6e405ff9"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-01-10 14:11:41.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-10-18 08:21:06.000000000","tz":120},"subject":"Set DVR qr-xyz interfaces DOWN on backup node","message":"Set DVR qr-xyz interfaces DOWN on backup node\n\nWhen using DVR in HA mode, only one SNAT node is ACTIVE.\nEven if the BACKUP node is not managing the external gateway, it was\nstill answering on the internal interfaces (qr-xyz)\n\nThis patch is setting the qr-xyz interfaces down in qrouter-abc\nnamespaces for DVR+HA backups routers.\nAs soon as the router is elected to be MASTER again, then the qr-xyz\nwill be put UP\n\nCloses-Bug: #2002417\n\nChange-Id: If714b0e290ce5a1966eca700677e81735a7f27c1\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42018ba570626b5d4ca93b69d2f464679206f7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42018ba570626b5d4ca93b69d2f464679206f7ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
