)]}'
{"id":"openstack%2Fneutron~869909","triplet_id":"openstack%2Fneutron~master~I57de801473fc30f06acf1bc8a65cb2ff76b2954a","project":"openstack/neutron","branch":"master","topic":"bug/1998608","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-01-24 02:16:49.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-01-24 02:16:49.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-01-24 02:16:49.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-01-24 02:16:49.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-01-24 02:16:49.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-01-24 02:16:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57de801473fc30f06acf1bc8a65cb2ff76b2954a","subject":"[OVN] Introduce the new OVN Neutron Agent","status":"MERGED","created":"2023-01-12 09:14:59.000000000","updated":"2023-01-24 02:21:30.000000000","submitted":"2023-01-24 02:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869909-bug/1998608","meta_rev_id":"a7dc869c7c41966c60b0639001da7a4976d25162","_number":869909,"virtual_id_number":869909,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-01-19 16:20:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 02:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-23 12:38:56.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-01-23 12:16:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-01-24 00:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-01-23 15:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-01-24 00:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 09:52:49.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 10:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-12 11:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 13:10:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-01-17 13:10:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-17 13:10:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-01-17 13:10:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-01-20 04:12:17.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2023-01-23 12:16:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-01-23 12:38:56.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-01-23 15:28:00.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-01-24 00:26:42.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"66431bc96342de3ce0ff1f5115dbc369384118e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 09:14:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91cb55c66a9e77a4ed5c12a9dd9ee70ce7048657","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-12 09:52:49.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/09/869909/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"69b842ea8b66bb2ecd40bffab0298acd3337a51f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:07:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c19d345c4f3c48ce988a85f75e7d8b70\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2996a62a90654937b10ae542d79c944b : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b23acc37e61a49b88b0c17c734daeecb : SUCCESS in 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34f2ecbdb396942052615a8ae9a12d71ae30f5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:34:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9916cf6e833b4baf9095385c55e38f12\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/057e5be5576b42438d92c7b1d053dab7 : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/33299db448844d758daf22d2e725ce2b : SUCCESS in 36m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/770108208943496a922379675ddbbca0 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a99ba3056d435b991a8c59e21908b8 : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/996339b32138491ea0881bcd56f3215a : SUCCESS in 27m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9444d59eb29a4f538317edb51a2d27b6 : SUCCESS in 26m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188a5e25749a42f0b9f642651e6853eb : SUCCESS in 14m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c34c621656ca468abd07ecb5b408cc37 : SUCCESS in 55m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/40cc4ae000b94b62ae0a306831039a1f : SUCCESS in 1h 48m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/72f417e3ea6444c19db52d50479c7632 : SUCCESS in 1h 16m 27s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e10ff5e6ad744c9fae7a541ab3557f50 : SUCCESS in 1h 02m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce3f3f5750eb41f28e6f4deedf541e3f : TIMED_OUT in 2h 13m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b9a226ecbc6d4bf8b842b8040fccb4c2 : SUCCESS in 1h 24m 39s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b1ddcd9e767c4968a80f1c74c6599faa : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2df09c2984449d563997409773b9065323286cef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-12 14:13:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77e88243b9b902689c5c3b9227f7fa23e00ff078","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:04:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aa3eb74ad7e4b6298d0ae2cd305a99b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a390aa48f184a9683d8f713f7da5a78 : SUCCESS in 50m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b7afa75b2b74ac79dd7c12e0b595d29 : SUCCESS in 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1825065ecb0ae801a897e835aaee8edbfe6329bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:28:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcfc88ea57fe48c9b5616995ee67ce6f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/31be3a3021454f429b715887f93ff5a1 : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf4a89d68c784749a2df3d1d2bba8218 : SUCCESS in 31m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3afff730b0b04e27938b28afc875919a : SUCCESS in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506690c936a04ed8988666ad78a067fc : SUCCESS in 14m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eee4a7fc6f6949e6b26fb784bd92a662 : SUCCESS in 49m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63e986432ac544f9bd564ed7142f8980 : SUCCESS in 38m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ea579147a84bc3893f1e45f819709f : SUCCESS in 12m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7568ea469628483891bdbe05f122f650 : FAILURE in 42m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fd2e495117345a889947cab73688690 : TIMED_OUT in 2h 12m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/584a84deaeb545d59f8f12f771f9d218 : SUCCESS in 1h 10m 18s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e40087a6fde94b2b87bcacdb6679558c : SUCCESS in 1h 03m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8cd409c229a14f90b4d2e9d465b39d08 : SUCCESS in 1h 47m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c02a58a4edb2461180b8437ff9562548 : SUCCESS in 1h 30m 04s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/07a9136bf9334a2097fb40692092d575 : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4a64effc62a23d19ee6be48254ba6ff250cc1167","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-13 09:21:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a9715710d4fb898ece1c7aacef5e76495de15d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 10:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f891790fd84096ba0c099a94b6ec58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da5e8abdc1f4421aa19a10a12e17369c : SUCCESS in 51m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/862440a4e5e24f25aa61275aed370ba0 : SUCCESS in 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d93faf3b55559597ba4b9ef9cd2f3728fad35748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 12:20:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf8ac7bb28d94fe38f72ba9911b5a211\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eb3242bb7527435699a3b7d1604ccdfd : SUCCESS in 2h 56m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fd45f1009ed14dddb03f2e5bfe1833eb : TIMED_OUT in 1h 04m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c37fa8d9cb044e3e8ed6a9a21efc3055 : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7516be2104c7407a985aaf183b742a92 : SUCCESS in 14m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab93515a2c624fb799b163c1ef6ced96 : SUCCESS in 31m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b8e61ed22704982920e4fc3ad5c6833 : SUCCESS in 27m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fc533487b64296b33d278be5315d71 : SUCCESS in 11m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af7bbc0ea46442c887658693fdc547cf : SUCCESS in 59m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ecb02e8093e4aa5b539867fd76ee451 : TIMED_OUT in 2h 12m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/275ed46f07864367a7d8e03457e272f4 : FAILURE in 59m 53s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c2521bb3dfc54f25a2c85a76fd30831d : SUCCESS in 1h 20m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1b43347252544778d20f5ef030ecc3c : SUCCESS in 1h 44m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9904daa65fd54c86ada61bc261e1b465 : SUCCESS in 1h 24m 30s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/6e42092b0da24e3b853e901df210c267 : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8c690b056f51a5001ff5fa498dbabc9325a4d32c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-13 16:27:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0289bad72e224c3a0e490c17334411282fa13864","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 17:49:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c6e9e3ed6f4e7bbfb660a73f8be8b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b35bc0230a140d5a0b0805fad88e455 : FAILURE in 1h 08m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d09a0318a16e497b985e643924bb4eba : FAILURE in 1h 09m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afaa48ad4eeb097a80a9a0c3567c0d532a095e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 18:00:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794bbb6918a8479bb64bb90a42abbec1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cd349f06a66d499a87c04868fa5f4973 : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce51022f2775480d9d8910fcf2c1d39f : SUCCESS in 26m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18ee89960eff4ecbba3b134ca3e1ca6d : SUCCESS in 32m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d76b26ae6b347cdbf4ed652ba8b2d7e : SUCCESS in 13m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7010a4f3dd714950bbd010c386b99ce6 : SUCCESS in 27m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdc45a4e16434bbab0fb63d9835a9745 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d65f1287d942afa1736dbf281ca646 : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5b6a2df8320433683a899ae51581907 : SUCCESS in 51m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/148079e1d7cb4d5d9d8a03a2d8c5bf7c : SUCCESS in 1h 24m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b00c0fc2d71d4d5eb259f25bee1cc408 : SUCCESS in 1h 20m 05s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/195a00b7f7a1468f8ba1cadfea3e22fe : SUCCESS in 1h 03m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/080fa669190944ca833b95ffb061d4c5 : SUCCESS in 1h 29m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a860750ee6b44b6690e3493f89616c37 : SUCCESS in 1h 13m 12s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/48891e315fc64024ae982513bbe31d6c : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"38f8839300a9dd5fa4a45fdbff1c977e28533365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-17 17:32:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0690b7d89c4209842d9ecaacbc9707dcfc291fa5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-17 18:10:44.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/09/869909/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"401a8c52fe3911a4881ed1f6aba100ea1a232f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 19:38:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a34c8fc08eff480ba1c3300b5348435f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/75ca985eb09e4189b30f139a8e887eae : SUCCESS in 1h 10m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/907b61d67e5347aab4b508531cc90ad3 : SUCCESS in 33m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8fe9d67a25d40a8b33fbbeec9f045e5 : FAILURE in 37m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8143b17b23cd4edf8c2a332928a7eec2 : SUCCESS in 12m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26b1f9b41faa4da9aa051303d4d61361 : FAILURE in 25m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d197fe1a40da498a81de29e4507c9441 : FAILURE in 22m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acafc299b09c459b92ab1d076fe00dd8 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87f3741002c44f09b2fbe4a31815909b : SUCCESS in 49m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5829d907a42640b89af34ba89a988353 : SUCCESS in 1h 42m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7d21e17194fc4ac69b547c6a974a7b60 : SUCCESS in 1h 05m 48s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7282b21086c749a7ae68f9725bd923f6 : SUCCESS in 1h 01m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1bd72f601e248349c77a6841b92dde9 : SUCCESS in 2h 00m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d372efd05f974a2b974917c7bae0cc91 : SUCCESS in 1h 04m 47s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/36d55dcdf424408c97b98c27b23e8286 : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b8c155101540585bbfe1d990ec7c541ac548c2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-17 19:57:35.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fb4cd35bd2faa5ca11b9452046eaaa5207ecc491","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-17 21:12:12.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/09/869909/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7a70c9ae2b5c0d8a19bd8a4208bd4eff4a7afdb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 22:16:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56d09b1ba2744a5fa6ec52d937d76f21\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc9cbb768cee4eaab2b9c35aa2e5106f : SUCCESS in 1h 02m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/709a2acd34144695b9a82f68f916e9ae : SUCCESS in 27m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1307a881ca9a437cab36149a68386fc6 : FAILURE in 38m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9642e953846444b5a25c10c65f5baaa2 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e8cb8997c334f07aa8496575d9c3139 : FAILURE in 26m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da245a700fd545f5a8c884c08efa2156 : FAILURE in 24m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc11b1d2bbeb48cc9a7270fec6ff01f3 : SUCCESS in 11m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20e97d22a4d340deb5cbf2cd5c722377 : FAILURE in 40m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/129cfbed7bee412ca758d4f963553365 : SUCCESS in 1h 31m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/78182da57b454a63817d740bdf994532 : SUCCESS in 1h 11m 52s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16f8c2cd1d3245efb1c227dae199ab60 : SUCCESS in 59m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7d2876912054bb59266e0e998f642e6 : TIMED_OUT in 2h 13m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5a8409a9d34e4adfa5849c953958e467 : SUCCESS in 1h 16m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2efa70bafcd542389d68fc9d570c2622 : SUCCESS in 1h 00m 43s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py","accounts_in_message":[],"_revision_number":3},{"id":"89c1d355540f66987dbbc4d69b49c54ac6946a16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 23:49:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/966b3619f51a4dc0b8f9500004bae16e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33d8d3a74fb44fd8aa33b8c7ee8b4d9d : FAILURE in 1h 19m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49fd65944855427d825c214c5615e305 : FAILURE in 1h 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"daaf0f39f912ccb163dedae996096d5b05c51983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-18 12:47:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5e6cec3175f5ddd68f8a80f15e152917c522c642","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-18 14:18:42.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/09/869909/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d552709862588688bea13ae814f4b98e24b1bb32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:12:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd69b0390ec04afd8f6efc0ad80174e3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/30c41a4468334b548557645dfd22d0c0 : SUCCESS in 1h 16m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea8720905d2a472c92952af498223e36 : SUCCESS in 27m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5810f5bf55294d07886ada0a9e697438 : FAILURE in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260324661a54494d9625aa1321fcb75e : SUCCESS in 14m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee3182adc25341a7a340ee2714e57817 : FAILURE in 23m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d34e8771514d4fd08277b5ded5e69282 : FAILURE in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d3374cf3ea4c448f56b7bb3b44cf82 : SUCCESS in 13m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/37b8673507494144be0f5028e853fd16 : SUCCESS in 42m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cff4431e2e84565ab51dcb29d69a9a6 : SUCCESS in 1h 36m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7fb94aaeb271485390e164c9ca727a04 : SUCCESS in 1h 19m 22s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/153726fdb6f642299c8ee87c1d65822a : SUCCESS in 1h 12m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5337c61f74ba4358b6f4ecbce8958f21 : SUCCESS in 1h 46m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/74b8890cdb4d4b03a46d85849fbda279 : SUCCESS in 1h 28m 01s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e2598faccf144d2b8fd40a47a3ce3345 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":4},{"id":"35022581bf8d36c77b8e2668d9282d4f80ae9ef0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:33:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5067e840fcb4bc986e725deb2512ee7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7662af44b7e46919c5524dadc83ec7c : FAILURE in 1h 19m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f860c1a5bd194efdaed3ea4c179e147f : FAILURE in 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"27cadc2f02a6b46b28c166bd19229387e14c4ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-19 14:36:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2dd042fddf329523b107dbfa93a11464e7499bcc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-19 16:20:34.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/09/869909/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d75932d7c0925603206544b460a0dd4b31481f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 16:34:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d11a875a44f4572ae05ec02a38a9ced\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85271664f6f4471b98fadb4e44bb09a5 : SUCCESS in 1h 19m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a21bfb5024a7491cb1a92679fe097953 : SUCCESS in 25m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b20b8e380aeb4a4aaa957c7f79a72fda : SUCCESS in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f741245f3ade459ea1bb23efab86433c : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d871b01050946028db1019179736b48 : SUCCESS in 25m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e54666bc6c245e3b43ec04ff17b8450 : SUCCESS in 27m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8f0067ee72457caf1cd21d31506125 : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5807a5adb894ad49efec8a70ee03cd5 : FAILURE in 51m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88d9b61a7b9d4e9883d55887a81a57b5 : SUCCESS in 1h 42m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f3c264350b0548d595930ff914bdd5a9 : SUCCESS in 1h 11m 24s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4d8202b67e0b4ea4be7cc9d20ed88405 : SUCCESS in 1h 12m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ebc60a2413d040f8845636390f9dab38 : SUCCESS in 1h 35m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1377e071e7644a5cb3c11b596f382ef1 : SUCCESS in 1h 22m 44s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/550da2012d054581811883fbaf3b918f : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a1cee315e00af267d3d937fa5a679cac202a9eed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-19 16:47:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a6f3428fc93fd14a3ff4901cdd8573f130350de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:50:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb439bc0a87b49d288ed14fc30bf0bc3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/072ac60690c9449eb506fadca16db8a9 : TIMED_OUT in 1h 21m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae2abaae2f294ac79d718ab0946e0454 : TIMED_OUT in 1h 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c5f5b9dd305b80ab37a283b9ded1246edeca3089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:59:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d24f8cbb276406fabea0c59cb8f21c0\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47370e1ef7954469a9c6066da8e6e4f9 : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e99ecbac6a98473f8f223f8b4bcaf014 : SUCCESS in 27m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07eab98a00b14bfa9c7feeae2733b542 : SUCCESS in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f2c1d755f9e46f9bd7c31c91170ca6d : SUCCESS in 15m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d307e952472420c81037353f6c77438 : SUCCESS in 24m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93a9078a31654b76828c79d679ccb910 : SUCCESS in 25m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec87dfff55104667b852ad95bb10012e : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0719af3931fe468b84eea55c417ecb15 : SUCCESS in 49m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1dd66b9d2014bd7b502f56c53d91238 : SUCCESS in 1h 42m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da9255f6b22f42168878359fec5f6003 : SUCCESS in 1h 04m 30s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/65185a19e8d94869ba651b09fa647cfd : SUCCESS in 1h 04m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7f62dfb95ed4b91bb832f6f8ad2e8f0 : SUCCESS in 2h 06m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20adb105eff144dc91fd458407e5df92 : SUCCESS in 1h 22m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/62b8f2a6fd824b63bcedfd98c7fbe4c4 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a17541feb5769f557d785f987bdd2968b8c21407","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-23 12:16:03.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6664589dcf9f041a1d06ad6b8ee3a11f0a24c44","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-01-23 12:38:56.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8242485a92d5b99e0430d5e7da36a48961fab3f1","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-01-23 12:39:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"27713dd812e34634a3dacd2c78057dccdc76eb34","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-01-23 15:28:00.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a235a8243e1670c5a0213cbb670f2a92afb991e6","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-01-24 00:26:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e51517234be08df8a6855c77443632cb77f3d79d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 00:27:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"25cc871cc134b134bdf1d2aef1e17141448f5fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 02:16:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642a0e880da442a8a619e04f5624d42d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b90f52cb27b54d7e9c79743d539fae7b : SUCCESS in 1h 33m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e6c816423f43c3a627ba98d603b344 : SUCCESS in 12m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/969c3bd9b59b4476ab2f0d284342fc0c : SUCCESS in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bff12ec7b4e443cb2e9833ead304267 : SUCCESS in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beda245acc27450f82c1ed2e4d5af6eb : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bfd438f65d044d8b31d5339493b1cd4 : SUCCESS in 44m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3cec46fd5ae47298b8e609c69b34b85 : SUCCESS in 1h 26m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fe135444a30748c0964a534ccc4de550 : SUCCESS in 1h 00m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/50c4635fc9304f3bad0f325e9b02ae47 : SUCCESS in 1h 20m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/6a992bf20a144a47982340a9b74b931e : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":5},{"id":"f9f703918e18778293fdee33280d86dbd3b622e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 02:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a7dc869c7c41966c60b0639001da7a4976d25162","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 02:21:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3d5b31f4ca4288a664d4073fb2fb51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55020a91e05948959491424266d6044e : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"78c3d52bb4f81dc338370240fd3da587dbb3a5e2","revisions":{"4e056cc307cded2ba509b9984b70c65e2a09408f":{"kind":"REWORK","_number":1,"created":"2023-01-12 09:14:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/869909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/869909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/869909/1"}}},"commit":{"parents":[{"commit":"fc10c16c66db03890b582bfeab9b8c7ed3befd6c","subject":"[OVN] Add vnic_type and binding profile capabilities to LSP info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc10c16c66db03890b582bfeab9b8c7ed3befd6c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-05 19:27:01.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-05 19:28:12.000000000","tz":60},"subject":"[OVN] Introduce the new OVN Neutron Agent","message":"[OVN] Introduce the new OVN Neutron Agent\n\nThis patch introduces the new OVN Neutron Agent definition in the\nOVN agent list and creates a new class ``OVNNeutronAgent``.\n\nRelated-Bug: #1998608\nChange-Id: I57de801473fc30f06acf1bc8a65cb2ff76b2954a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e056cc307cded2ba509b9984b70c65e2a09408f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e056cc307cded2ba509b9984b70c65e2a09408f"}]},"branch":"refs/heads/master"},"ef26f3c50365b668eab08c1668d77c61b731b8ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-17 17:32:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/869909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/869909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/869909/2"}}},"commit":{"parents":[{"commit":"37b12a4566d6f6aa63c027598de0c7afc91679e4","subject":"[OVN] Add vnic_type and binding profile capabilities to LSP info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37b12a4566d6f6aa63c027598de0c7afc91679e4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-05 19:27:01.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-17 13:25:32.000000000","tz":60},"subject":"[OVN] Introduce the new OVN Neutron Agent","message":"[OVN] Introduce the new OVN Neutron Agent\n\nThis patch introduces the new OVN Neutron Agent definition in the\nOVN agent list and creates a new class ``OVNNeutronAgent``.\n\nRelated-Bug: #1998608\nChange-Id: I57de801473fc30f06acf1bc8a65cb2ff76b2954a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef26f3c50365b668eab08c1668d77c61b731b8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef26f3c50365b668eab08c1668d77c61b731b8ce"}]},"branch":"refs/heads/master"},"d843c38c8e932b0176645a8fc0561d7b3544cdf6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-17 19:57:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/869909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/869909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/869909/3"}}},"commit":{"parents":[{"commit":"e2f4ae2837772f0dc0ee52cc798ab51d3641dc5e","subject":"[OVN] Add vnic_type and binding profile capabilities to LSP info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f4ae2837772f0dc0ee52cc798ab51d3641dc5e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-05 19:27:01.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-17 19:56:24.000000000","tz":60},"subject":"[OVN] Introduce the new OVN Neutron Agent","message":"[OVN] Introduce the new OVN Neutron Agent\n\nThis patch introduces the new OVN Neutron Agent definition in the\nOVN agent list and creates a new class ``OVNNeutronAgent``.\n\nRelated-Bug: #1998608\nChange-Id: I57de801473fc30f06acf1bc8a65cb2ff76b2954a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d843c38c8e932b0176645a8fc0561d7b3544cdf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d843c38c8e932b0176645a8fc0561d7b3544cdf6"}]},"branch":"refs/heads/master"},"ca31ac8cf1d16baa19447bcfb57a0d1eb2ab53c2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-01-18 12:47:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/869909/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/869909/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/869909/4"}}},"commit":{"parents":[{"commit":"505f34440c3800638a9300e7f49e55e435e46fad","subject":"[OVN] Add vnic_type and binding profile capabilities to LSP info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/505f34440c3800638a9300e7f49e55e435e46fad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-05 19:27:01.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-18 12:47:07.000000000","tz":0},"subject":"[OVN] Introduce the new OVN Neutron Agent","message":"[OVN] Introduce the new OVN Neutron Agent\n\nThis patch introduces the new OVN Neutron Agent definition in the\nOVN agent list and creates a new class ``OVNNeutronAgent``.\n\nRelated-Bug: #1998608\nChange-Id: I57de801473fc30f06acf1bc8a65cb2ff76b2954a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca31ac8cf1d16baa19447bcfb57a0d1eb2ab53c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca31ac8cf1d16baa19447bcfb57a0d1eb2ab53c2"}]},"branch":"refs/heads/master","description":"Rebase"},"78c3d52bb4f81dc338370240fd3da587dbb3a5e2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-19 14:36:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/869909/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/869909/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/869909/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/869909/5"}}},"commit":{"parents":[{"commit":"b6750fb2b80453f4efb2706ed8a8ef1e1e5db37e","subject":"[OVN] Add vnic_type and binding profile capabilities to LSP info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6750fb2b80453f4efb2706ed8a8ef1e1e5db37e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-05 19:27:01.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-19 14:36:39.000000000","tz":0},"subject":"[OVN] Introduce the new OVN Neutron Agent","message":"[OVN] Introduce the new OVN Neutron Agent\n\nThis patch introduces the new OVN Neutron Agent definition in the\nOVN agent list and creates a new class ``OVNNeutronAgent``.\n\nRelated-Bug: #1998608\nChange-Id: I57de801473fc30f06acf1bc8a65cb2ff76b2954a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78c3d52bb4f81dc338370240fd3da587dbb3a5e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78c3d52bb4f81dc338370240fd3da587dbb3a5e2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
