)]}'
{"id":"openstack%2Fneutron~870979","triplet_id":"openstack%2Fneutron~master~I68f79af5b02ff271d5e82052940368de23da54dc","project":"openstack/neutron","branch":"master","topic":"improve-ci","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-17 08:02:08.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-17 08:02:08.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I68f79af5b02ff271d5e82052940368de23da54dc","subject":"DNM - TESTING PATCH: functional and fullstack CI jobs","status":"ABANDONED","created":"2023-01-18 17:14:37.000000000","updated":"2023-04-17 08:02:08.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c907fb0b0a04ecf5f064e6e10181f3d2bbc2bd78","_number":870979,"virtual_id_number":870979,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-09 16:55:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 20:05:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-03 04:59:30.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-08 14:46:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 08:02:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"53f1ce466dd1da94ed30e3d80a9f0caa54e27681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-18 17:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5660121debda5fc2803d7ca4274de646d70c839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 20:05:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f652bd948247c1872b9d0f76a74db2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/50b6518a36824aa9b12b3a419ad4d997 : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/046461f8b4a44893aaeb09ce0234d99f : SUCCESS in 2h 02m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20ec15ea6a6f404aa2ebb017fee6c05a : SUCCESS in 1h 33m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9cc10aa6cedf4820ad4be9f0c9cf46ce : SUCCESS in 1h 13m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/517c6051bc274a439cff2fdca205416b : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a338ebe3cd436faa7f6ba7329e3379 : SUCCESS in 12m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c533c04bdd0443ab9cebe23d0a71a5ab : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/aa34f517353f4a91aaac7a4e84d40e90 : SUCCESS in 53m 56s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/3cce601db6c746a8b297b8201f8aec02 : SUCCESS in 54m 33s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/92c0efb555c74a9a8c84bd35bfd10fa8 : SUCCESS in 54m 38s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/60ff33d8e04248fca38738b1d22e9908 : SUCCESS in 1h 07m 27s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/f28eeb6171644359a277502cc78c7b3d : SUCCESS in 51m 08s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/28db036ae3774de688d0d63ca4cc7278 : SUCCESS in 48m 40s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/cb24917e144e4a6084b1248a5c10d93a : SUCCESS in 48m 54s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/b01c374878aa496398b4e58048c4ff9d : SUCCESS in 59m 12s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/70b1050ea8f847c8b5fbca5b4f246f2b : SUCCESS in 52m 24s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/fd3e6baad99143549116cc277aff65e0 : TIMED_OUT in 2h 12m 43s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/873b98262cc745588a79f715ac210261 : SUCCESS in 1h 04m 30s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/6c50cddcedca46acb7c2e0dd72a8bccb : SUCCESS in 56m 34s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/c51cf5552ba741e186be4425a93b1c80 : SUCCESS in 51m 23s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/a2d93d8e918948899a2c641ee419fad5 : SUCCESS in 52m 57s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/d7023ed271a64319810af83058891859 : SUCCESS in 44m 34s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/a0a2ecec799b4642bf15668e846f3bf0 : FAILURE in 43m 11s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/1bed4afd12b742bcbce2926fa2c06e1b : FAILURE in 42m 23s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/3424ea176d7d468585b105f13bafd3c2 : FAILURE in 47m 18s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/f454706ed1574e7795761d68c20c0124 : SUCCESS in 51m 13s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/59139be6f7b2432c90062e297e4f95e5 : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3a883a615763cc66a245a6d5536ab315bdde96d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-19 16:55:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"adddb05c0c73871511ea7b33d3d72018546b5b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:40:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7d93922e4a2447194c8e89b2ad59d37\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c38f00966746443e8dc7dfe219e14fc9 : SUCCESS in 1h 39m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4594e7650fe8462e85867ffa6c400b58 : SUCCESS in 1h 25m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab5144ceec6542589fa48a862f760d1a : SUCCESS in 1h 41m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/32c43a43b6af40039a9191a142e73bae : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f7fb65394ea74d5f86bf553681d927b6 : SUCCESS in 34m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca319467f3c04a3792b3402739519927 : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bdd69e4ba6a449ca8e0c87f52db7ca8 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/8e8609129a304fd9a2db8205b9207396 : FAILURE in 41m 21s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/7b8a284f4ba543e396a421c677c1f9d8 : SUCCESS in 50m 25s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/d0ed5d02fa5f4de4ac8ef59202c5f5e8 : SUCCESS in 49m 03s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/48c373e2c9b44759a729774768d2c783 : SUCCESS in 53m 53s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/8920f12b50c34342b0c9a5e68a03e23f : SUCCESS in 49m 49s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/751fe6396f04412ab81df5c3e4c9cf5c : SUCCESS in 49m 53s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/497c5e2d0f1943d3994d89e32076fe3f : SUCCESS in 49m 50s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/59ab8c4172b34ad7b0f1953a045bbad2 : SUCCESS in 50m 08s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/4e827e830db74687a5d431f9f244c614 : SUCCESS in 51m 15s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/d7154d3f6c18447d9cfd2d2068e11bd9 : SUCCESS in 59m 11s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/05d2af5d92ed4d3091b396465c974463 : SUCCESS in 51m 58s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/e9f21ba89c604b00bb632d4b03819eb2 : SUCCESS in 53m 29s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/c8fa40cbeaa5486c94dd7abf1e357b34 : SUCCESS in 58m 23s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/55fc9705981049c683eab620d095d961 : SUCCESS in 50m 17s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/c4fb0898572a4b6c978e6f6b277bb5ae : FAILURE in 50m 11s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/9c7fa8eaa9cb4287952acaf977d3ca17 : SUCCESS in 59m 59s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/c8adfe5faf004f6ea9678861b0a33e8d : SUCCESS in 58m 08s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/3323efa69b0d48bab1eba94dadfabbc2 : SUCCESS in 42m 26s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/38cfb641ce9f4a088661ea38ddde2abf : SUCCESS in 54m 11s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/b6d9870cc2234f56baa38297a3ba2111 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":2},{"id":"08a3823b3c84f2b573eb2797fd2e36c1c33759c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-23 08:24:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d755964e16f00d2f662212d4db5078d5a5b9d684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-25 18:19:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f8c0f11ca760737dfaa5f659dd0cc7ab1c6019ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 19:37:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/955826926f78495e92da992727966a67\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dbb757c362c643a1a31dd9053f7837b1 : SUCCESS in 1h 12m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/884e2f32fcef463e99764ab3d79e1325 : SUCCESS in 1h 16m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be84750a65de47b5bf1e2b33b9b38cbb : SUCCESS in 1h 10m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/23cf40799e364dd5b9fa3a6e46184e9f : SUCCESS in 1h 14m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b3895e5f5d84807b40975ad008313ae : SUCCESS in 28m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/913aa9a67df845e9989a1d24b7b28e93 : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c351c77e1d854e82b51644a0423790ac : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/eef0b19c46c244fd8d71c2fadad08c1d : SUCCESS in 48m 15s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/51f6f85ee14a401387e5941e6a5b204d : SUCCESS in 46m 30s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/06c76f3651d945e585dfe7a2d8c5a39a : SUCCESS in 47m 01s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/9a86e590c4744d189f0efb6320e9f567 : SUCCESS in 43m 40s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/edd120c3207344589ff3a499aa6d2e36 : SUCCESS in 45m 37s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/d054f8d6dc4d427f8468c7165ced29b8 : SUCCESS in 41m 20s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/46994fc64dbe4dceafe3a8407c35e0a8 : SUCCESS in 47m 24s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/56b9320a074c4aa98263b853e49db66e : SUCCESS in 45m 17s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/58127107a96346d1a339a0d6b335cd36 : SUCCESS in 47m 22s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/0aca0e107b954eb98e72423e1d5b94a4 : SUCCESS in 51m 11s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/bc1dc8324ff4438e88624a67252c0634 : SUCCESS in 41m 22s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/109f2378f8ea46b58e4fa175ce80672c : SUCCESS in 50m 00s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/6f0c96a1650c4e00a6a85e1e834de456 : SUCCESS in 44m 59s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/f7aa16e9b3154c48b8fa0ec3ec57450b : SUCCESS in 56m 59s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/14f94fb4119345cfb2a4b2476260d2aa : SUCCESS in 46m 46s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/dce0ac8db0f340e1aee51c23be2073b7 : SUCCESS in 50m 14s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/5fe8ce3c04b24c1cb7e8c0fd0ac20eff : SUCCESS in 48m 15s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/08620dce1a31495195dff997686f36d9 : SUCCESS in 56m 48s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/50df38be54274b3d9bc62c76cf37c852 : SUCCESS in 47m 21s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/0ef68c8112ab481c80cd99d85ccc1b7a : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e08f645b05ee82ff1d7a31825d0ef16ffaaf0765","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-02-03 04:59:30.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"b68a6e5d574e5afb60673f2f51f164202adc9d26","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-03 10:24:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"2d9408613f1f77b0a75bcb72557a202f9086878c","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-08 14:45:40.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"05ffdfcaf4e6fdcecc635fe684d344306ac0c98e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-08 14:45:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc493902507f979c9af86f493ed257383962f983","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-08 14:46:02.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 870979,3","accounts_in_message":[],"_revision_number":3},{"id":"ce53f7b2af92b13429da9a84045bc3e23058af01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 18:23:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/929eb7c046324c20a538b21309bcdeb1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4538896628ae49f991e09dda1d4aba14 : SUCCESS in 1h 11m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ad51cfb9964d45b18f799eda5c89bec8 : SUCCESS in 1h 19m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b64b28ccabbd4047aa3af10bf43a219d : SUCCESS in 1h 11m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f798048ec2154392892b0774333de507 : SUCCESS in 1h 04m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bba0c4e7a04e43cfaf44c0ee7be554b1 : SUCCESS in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be7ade60c6b41359a8acdd6251c5c0d : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5a5fa483c84f47955da355444912a9 : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/43706688669c4cda9e1a11a66f1cbb7a : SUCCESS in 47m 57s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/27a1ead43e7f43eab4885c117c317812 : SUCCESS in 48m 01s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/30a51e8168ee4bf1a735c5e6e1445824 : SUCCESS in 49m 32s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/eb22482df0fa472fac66ffb373581978 : FAILURE in 47m 31s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/1bd108393bc24dcfac5c0301895f3430 : SUCCESS in 58m 08s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/4975c6dd92c7433598526a93e06442a8 : FAILURE in 36m 30s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/1cafac77606f4a87a1b92124f8b91cc7 : SUCCESS in 59m 33s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/86852c47f76d4eabb15166dfad5538a5 : FAILURE in 46m 19s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/6560d0bc1c6b423c999c3f91e9404163 : SUCCESS in 1h 01m 45s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/668aaf9f34c74a79abbe537c5dacc009 : SUCCESS in 55m 43s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/ec14f5a2a622473691c2c00a3ecf68ba : SUCCESS in 51m 58s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/dce26ff84500427dab04ece6e6f2d6d0 : SUCCESS in 54m 29s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/d4d0fee0c1624db3a4c6d2504bcb8916 : SUCCESS in 54m 36s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/c743c652d63d4b5282bfa18fd428ecd3 : SUCCESS in 47m 22s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/a1b6f9fd733744e59cc81954d8c4a680 : SUCCESS in 53m 08s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/39e604ee67df4cb68c8b5add4985185b : SUCCESS in 54m 42s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/5057bb1126bb4e78a466048da6fa3e3e : SUCCESS in 49m 18s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/57635f6032a34c6d958bd8ee92846e6a : SUCCESS in 50m 17s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/4e8822a095df42e3ab49d213e7f61d19 : FAILURE in 51m 57s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/ae45d56075e0409ea88850ea72983d5c : FAILURE in 37m 42s","accounts_in_message":[],"_revision_number":4},{"id":"9d2450361257dc54b85890e6475d7ec1bd3e6ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-09 15:52:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a46c616efb4c234a44a8d396588d18245eb7278e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-09 17:58:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"92b555a038886a25db1a1ccc41cdae03b2411317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 20:21:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a43cf44b11ef430996193f82cec8d74c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e0eea369eb1c46389182058e70e9a09c : SUCCESS in 1h 22m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6996f765ec64ee88bf276201410c345 : SUCCESS in 1h 18m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6ae93bde14742aa83758a22c0181672 : SUCCESS in 1h 05m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6996fe4a38f446191fa4932bb2faee6 : SUCCESS in 1h 18m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e298702d7884d9985f10740ba948347 : SUCCESS in 25m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeae4ea22f1f406d99337888a5c476c3 : SUCCESS in 14m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34a06d653ce6460fa48baac925885742 : SUCCESS in 11m 35s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/e5b66d7c7267404798866588d36c6bc0 : SUCCESS in 48m 28s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/9cdef8bf16d54a699642936823ec0f7c : SUCCESS in 47m 10s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/04002277b6824f28bbdf22b1ffd01831 : SUCCESS in 44m 20s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/2829590d067346a8b4b451309ea278cb : SUCCESS in 51m 40s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/579c35b634a84010a8f89fbdccde4811 : SUCCESS in 48m 36s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/b5e2e8fa71f0409e93d27da24b847a8e : SUCCESS in 40m 01s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/4b36882c43594c52948321c08b88cf5d : SUCCESS in 50m 38s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/5c788d926bde4f21b4d33646d93f0be4 : FAILURE in 47m 20s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/068f8d4768284e6d81eed76d7f04728e : SUCCESS in 39m 36s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/6b2ad09169a3470b9a760ac1560f2d2d : SUCCESS in 36m 09s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/5268b16280354ea5bdfc68e87f659e34 : SUCCESS in 49m 04s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/3d1440ee806340f3a96f20a1391de74a : SUCCESS in 1h 24m 07s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/4c83212160154a54a8098a8fe3b0dae8 : SUCCESS in 47m 34s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/82fda488d6e143b9bf6ea5482651d7f1 : SUCCESS in 54m 46s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/c62e870737544b18be8886d832527ce8 : SUCCESS in 56m 32s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/ba7f3260e25e47fb99eafce9fbfb0e46 : SUCCESS in 44m 53s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/ef28e5de5298483bb39f6c82df7680a4 : SUCCESS in 1h 15m 43s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/e08997c287c84709a24d163e390c54ed : SUCCESS in 45m 45s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/af728fd030e24d16aa9ad1e6b8731797 : SUCCESS in 48m 49s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/d634e2a7f17d422ea7d83be78a9d3f1d : FAILURE in 44m 52s","accounts_in_message":[],"_revision_number":6},{"id":"4032a620715c6b880ca5fb80f190252fa5a59a31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-13 11:37:27.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0a83c498a07ac148df6d80ed88dbcf1d4d432c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:51:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e9727039c2741589be302a66dd0ec82\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5c7e9c1ac876444197bc9020ebfb6122 : SUCCESS in 1h 40m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/33bc1562cf5549b2bc693bb9046cce65 : SUCCESS in 1h 43m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/191e76d689644b6199e711e1a7ac35ef : SUCCESS in 2h 11m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3bec54e4af98422587828a32bf56b99d : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c60050bb40d94dd4b9cfae9bd5a84b9b : SUCCESS in 34m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8f6603ff2b4cddb588673224bcd78e : FAILURE in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86d3355f982b4a5897dd1eb86c168c11 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/2531cc25548142fbbd9f98b9f96531ac : FAILURE in 45m 47s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/b877d06516ff4e0e9b9eda39084a6ff2 : FAILURE in 46m 09s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/f3eaa801ae3a491d9fc4ece0962c41aa : FAILURE in 41m 39s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/8f1e03d3622947c5b8114a275ba374af : FAILURE in 43m 59s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/8acee868e986494da1782d670e86125c : FAILURE in 38m 57s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/bf699d5c340b4c2aa84097becfce9835 : FAILURE in 43m 26s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/aa1b828be23648e3a4315eb02f19400f : FAILURE in 52m 12s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/4f064b2cd33d4776902c8177e79949c9 : FAILURE in 37m 25s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/6b01937d047e4c4fa5b0fa5d461b40d4 : FAILURE in 43m 33s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/1e0b4f5598af49939cc87e9e6854f505 : FAILURE in 35m 53s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/0402693ca4d547b7ac9421fd5a25db50 : FAILURE in 44m 52s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/1013dabcb6d14bc688877c5b99aaa24c : FAILURE in 38m 00s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/ce26c6110d9e4b23b5877bbe070f3ad5 : FAILURE in 36m 35s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/a33e4dc42bd2494a86aef488434ce44f : FAILURE in 37m 17s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/ec6c8c7471e7468d98649afb7ea36327 : FAILURE in 39m 27s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/0c3df47b5dfd44149f530c5abc71e42a : FAILURE in 34m 51s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/afb2d8b55dc74124b32a287b39d32f4f : FAILURE in 45m 02s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/dc2ffa7461974206a345f011d1371cf4 : FAILURE in 41m 32s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/3e872b6a1d6e4ff1b5c6235d95ec0364 : FAILURE in 51m 20s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/6da3d09ddb0b4b2b9f572206dec38625 : FAILURE in 42m 42s\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/agent/ovn/extensions/test_qos_hwol.py","accounts_in_message":[],"_revision_number":7},{"id":"e148a9ce6f8eee01061a739588ef37f0db87cf02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-13 16:27:48.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"99d77299fc9a5e4780f8c3fb0d57902d79d53c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:48:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/366e912d9d6b4781b135abdf97e7e961\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c4bf861753194329834d21adbb6d87b8 : SUCCESS in 1h 18m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be3d4641ab6745df8fc8dd58cb7abae6 : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8fc9588aeb4c44a984a2827ee7bc2ccb : SUCCESS in 1h 14m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0fcca28ca86040a787121ec91dc3af70 : SUCCESS in 1h 14m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1cb689859efe4246b5f4dd962da66f75 : SUCCESS in 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5874c767d2354ac7a3380a694f79e805 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e3fbd994f14c26b8510140a230fe84 : SUCCESS in 11m 54s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/7eb535ce077f4e51b4dc83b25bb7b93b : FAILURE in 42m 49s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/23cff5a20a05453495f60e35bbfcab82 : FAILURE in 37m 35s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/07ed9049b1f04f17a3d74049bba56193 : FAILURE in 50m 56s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/41d9cad60812411ab4e8ccb7e637bad1 : FAILURE in 43m 09s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/665e35eaeca24a528c495c173fb570d6 : FAILURE in 34m 50s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/a443ca2d12bd43e590cc463756c7baeb : FAILURE in 47m 57s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/b36c0b52bc904711a5da08e78575fe36 : FAILURE in 39m 08s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/c086bdd7a4a341d1ad0f81c18a53490b : FAILURE in 47m 24s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/ccc9b50d08da49fb95eaf37b50177309 : FAILURE in 57m 36s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/dbf1feb95eca49b3b2057fe94a831c20 : FAILURE in 39m 33s\n- neutron-functional-with-uwsgi-11 https://zuul.opendev.org/t/openstack/build/cc54daec620f4457aafd90266b172e45 : FAILURE in 49m 14s\n- neutron-functional-with-uwsgi-12 https://zuul.opendev.org/t/openstack/build/4ff04886f9ce436a8d3b0edaa2a37a00 : FAILURE in 51m 04s\n- neutron-functional-with-uwsgi-13 https://zuul.opendev.org/t/openstack/build/5be84b985fb34cc894625c8f4c808f02 : FAILURE in 43m 22s\n- neutron-functional-with-uwsgi-14 https://zuul.opendev.org/t/openstack/build/6a84414c3c0e45198a858bc583a0c927 : FAILURE in 45m 57s\n- neutron-functional-with-uwsgi-15 https://zuul.opendev.org/t/openstack/build/0923cba8df9d4876a4eedcdab9a52239 : FAILURE in 40m 03s\n- neutron-functional-with-uwsgi-16 https://zuul.opendev.org/t/openstack/build/49dfd294501a4d269b22e712a94d4003 : FAILURE in 43m 29s\n- neutron-functional-with-uwsgi-17 https://zuul.opendev.org/t/openstack/build/1227126f95ca4b75b0edcd8a7f178a21 : FAILURE in 52m 49s\n- neutron-functional-with-uwsgi-18 https://zuul.opendev.org/t/openstack/build/4e479b5943394bcdbb75ffc0a95f4db4 : FAILURE in 41m 48s\n- neutron-functional-with-uwsgi-19 https://zuul.opendev.org/t/openstack/build/a9cae02804274801a442cb866344ad5b : FAILURE in 44m 27s\n- neutron-functional-with-uwsgi-20 https://zuul.opendev.org/t/openstack/build/7d11d19b5a014df9805ef2e3179a37de : FAILURE in 54m 08s","accounts_in_message":[],"_revision_number":8},{"id":"b9d81cd0eee280b887d57abff025e0ef75faa69d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 10:32:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cd70d2e70937f8b9cda59453cd80743bf7335406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:12:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87503d8c587643d09eaeee3953b414c8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a35e9fb1b5e74d2597357bfe7395bea7 : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/32be951b7788437eac90ed04e01467b1 : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/214ebb1812ed4575a3ec666af9efaff2 : SUCCESS in 1h 44m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2762a7df220a4f1f82bbf99a9869e9a7 : SUCCESS in 1h 23m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f6cfbefdae5c411faf679cab2d0f797b : SUCCESS in 1h 25m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e928068091c492ba61be98d5e4f4c65 : SUCCESS in 26m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1683a853f28a4db58bffdc7c630cbb1d : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ac2b06d9364b98a79a678082f98276 : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/c3a893cd37e94d66a46a44f46955c8f9 : SUCCESS in 48m 34s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/0790a82fbb0140ffae56ebad87f0bd99 : SUCCESS in 56m 34s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/2b52d2c84c6046ca94eb1b12aedc8eb3 : SUCCESS in 1h 12m 06s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/8e34b2871a874bf892d2767832278f75 : SUCCESS in 1h 02m 10s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/653de744bc6b4546b2e435a1d18021aa : SUCCESS in 59m 44s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/c616a934dac343cf89b7edc88e997093 : SUCCESS in 41m 13s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/79c45f1a46964cbe9541627e52fa561a : SUCCESS in 46m 18s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/b5c58eebd3a747419a8f139a19661072 : SUCCESS in 44m 45s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/7daf35e71ab146629cf4b0cc399bd747 : SUCCESS in 53m 35s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/71f49a1e88024f45a02b4b24609b94e0 : SUCCESS in 48m 25s\n- neutron-fullstack-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/0940a601ac234c69a113d285df37dda6 : SUCCESS in 1h 46m 48s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/790d4521b43c488b86dd5d43609b3f6a : SUCCESS in 2h 20m 26s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/841535552eeb4c148313d40fc506d3e6 : SUCCESS in 1h 27m 19s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/b5db13c47c2f4c5a9d3e43417a4a20e5 : SUCCESS in 2h 16m 21s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/bb8bd2eff07e43aca48a7994bfe40f02 : FAILURE in 1h 25m 18s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/55d0ae2835654502be1c9faabcff7e80 : FAILURE in 1h 16m 35s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/e85173f1e5d1471f8444a1782d5ef8bc : SUCCESS in 1h 41m 48s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/1335857d4309498a889b5aa8df680d5c : SUCCESS in 1h 29m 06s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/3284ce4eef72429799f993a7417897ed : SUCCESS in 1h 45m 23s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/b8caf2fb0163461684d37f2a98ea4ed9 : SUCCESS in 1h 51m 08s","accounts_in_message":[],"_revision_number":9},{"id":"86b39493bf9165459234b01706c129346608ba96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-09 14:08:33.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"5cc6c1c6f37cfd4f404b61ddedd992539b523b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 16:55:09.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7de9fe3eefd4e40a66e05c4814924d6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e6170177fd4b44349ed9c558f775c59c : FAILURE in 1h 13m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/de981e0bc96e442b8eb79453b8aa9814 : SUCCESS in 1h 13m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da3d8f76a7344b7d86d0d8380eb6730e : SUCCESS in 1h 09m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/745c5fc2639b4aa982d06fb7a66f6a62 : SUCCESS in 1h 10m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/28ea23d2ab294a7b8239d20e9ec1e21b : SUCCESS in 1h 14m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65596dadc72c4373b84903a31301d5ca : SUCCESS in 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe5d273eae14c4bae568733798c7db3 : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48c6f31438b4998bf788cd7a93c291e : SUCCESS in 12m 01s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/ce68fbe328ef40bb8a706f88e8e92938 : SUCCESS in 40m 58s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/a4ab9a9777f3455bb3b8ba0c8d3cd41a : SUCCESS in 43m 18s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/1e03324b3815408fbd965cd00e64b44a : SUCCESS in 1h 04m 56s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/e95d1e06d584487ea14615182001b571 : SUCCESS in 1h 01m 19s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/10a50a4c272643ee8f00e6d6a04a3f03 : SUCCESS in 1h 03m 26s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/8c1fbe5c0620471884474a7d700177bb : SUCCESS in 1h 00m 24s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/6b34a3c6abb6460489ed695692d29ceb : SUCCESS in 1h 02m 26s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/38b9fb9d0f174997bb05ac0642da0a4d : SUCCESS in 55m 41s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/060adf2e145144c181322b05924904f5 : SUCCESS in 53m 29s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/9a72bb253d784112b05cf37eb4f48283 : SUCCESS in 55m 18s\n- neutron-fullstack-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/c7d8f74eaba548c199434f7922bc7782 : SUCCESS in 1h 48m 22s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/3d03a7aeccf74238b6cf6f4c58e32340 : TIMED_OUT in 2h 32m 23s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/79f83528bbf349298fd2b9a5ef036472 : SUCCESS in 1h 53m 30s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/5a4972fa731247d68ca3c1d718b673f6 : SUCCESS in 1h 54m 51s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/db78934d13c24fc98f5feb0d54362c2e : SUCCESS in 1h 29m 06s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/5d8660d9dbb54325be7a973760cc640a : TIMED_OUT in 2h 32m 48s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/95a47e07f3bb40b7a9af860c3a0aca61 : TIMED_OUT in 2h 32m 35s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/e3e03700e97045b88dbcae73de23d771 : FAILURE in 1h 10m 32s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/f3d738166f214a609e03b885c752983f : FAILURE in 2h 30m 57s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/b806a1bc7ff44f0680aea3d5fc654a8e : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":10},{"id":"c907fb0b0a04ecf5f064e6e10181f3d2bbc2bd78","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:02:08.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f1bd92f7169b60938bbd33b51fd5330d091a7ca3","revisions":{"fc4cb3c8a504ba34b86b863cc7ee2779d0897c65":{"kind":"REWORK","_number":1,"created":"2023-01-18 17:14:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/1"}}},"commit":{"parents":[{"commit":"2c3d1d315dbaced49fe726caed1279c2ad880700","subject":"[FT] Change the logic to check the IPv6 address state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c3d1d315dbaced49fe726caed1279c2ad880700"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested: https://review.opendev.org/c/openstack/neutron/+/870967/\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc4cb3c8a504ba34b86b863cc7ee2779d0897c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc4cb3c8a504ba34b86b863cc7ee2779d0897c65"}]},"branch":"refs/heads/master"},"575684ac57683ca414857b36ad92586d0e9c86d8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-19 16:55:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/2"}}},"commit":{"parents":[{"commit":"ef4b1c7fa72723fdc940565327533ce4b16223e0","subject":"Add new debugging logs for ``ARPSpoofTestCase``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef4b1c7fa72723fdc940565327533ce4b16223e0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-19 16:55:57.000000000","tz":0},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested: https://review.opendev.org/c/openstack/neutron/+/870967/\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/575684ac57683ca414857b36ad92586d0e9c86d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/575684ac57683ca414857b36ad92586d0e9c86d8"}]},"branch":"refs/heads/master","description":"Rebase"},"6ecf69217f12ac2f4f9787efce1cebec93ee354e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-25 18:19:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/3"}}},"commit":{"parents":[{"commit":"f235ce8791ab31bac598a3ebb40fb48cf2e4d1ab","subject":"Add new debugging logs for ``ARPSpoofTestCase``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f235ce8791ab31bac598a3ebb40fb48cf2e4d1ab"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-25 18:19:32.000000000","tz":0},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested: https://review.opendev.org/c/openstack/neutron/+/870967/\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ecf69217f12ac2f4f9787efce1cebec93ee354e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ecf69217f12ac2f4f9787efce1cebec93ee354e"}]},"branch":"refs/heads/master","description":"Rebase"},"bb33d27ed2c3c180f33583e4f04480f5d0fd8fc7":{"kind":"REWORK","_number":4,"created":"2023-02-08 14:45:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/4"}}},"commit":{"parents":[{"commit":"d67fcb112cfed9ad34058b02d6df332c94c1e68d","subject":"Increase port name size to avoid collisions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d67fcb112cfed9ad34058b02d6df332c94c1e68d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-08 14:45:04.000000000","tz":60},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/873118\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb33d27ed2c3c180f33583e4f04480f5d0fd8fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb33d27ed2c3c180f33583e4f04480f5d0fd8fc7"}]},"branch":"refs/heads/master"},"abb424b1fea848fe2fa1f0bdb0d2f3b9876acc6a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-09 15:52:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/5"}}},"commit":{"parents":[{"commit":"b3ed9cee14e84d636ff018473df14c8c085c1369","subject":"Increase port name size and type to internal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3ed9cee14e84d636ff018473df14c8c085c1369"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-09 15:52:17.000000000","tz":0},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/873118\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abb424b1fea848fe2fa1f0bdb0d2f3b9876acc6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abb424b1fea848fe2fa1f0bdb0d2f3b9876acc6a"}]},"branch":"refs/heads/master","description":"Rebase"},"d64f1a937cacbb4a6ccc59f7ff1028d9e98b0c3e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-09 17:58:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/6"}}},"commit":{"parents":[{"commit":"58b84be238675127494330faf602ed8bae65a13d","subject":"Increase port name size and type to internal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58b84be238675127494330faf602ed8bae65a13d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-09 17:58:43.000000000","tz":60},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/873118\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d64f1a937cacbb4a6ccc59f7ff1028d9e98b0c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d64f1a937cacbb4a6ccc59f7ff1028d9e98b0c3e"}]},"branch":"refs/heads/master"},"1a761d79c6ffe7130a0dc50379a63a118f09aa95":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-13 11:37:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/7"}}},"commit":{"parents":[{"commit":"de2b5e635d5d913fc84fb113f51a3f4c1c2f3cb3","subject":"Increase port name size and type to internal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de2b5e635d5d913fc84fb113f51a3f4c1c2f3cb3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-13 11:37:27.000000000","tz":0},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/873118\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a761d79c6ffe7130a0dc50379a63a118f09aa95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a761d79c6ffe7130a0dc50379a63a118f09aa95"}]},"branch":"refs/heads/master","description":"Rebase"},"e236dce834612267cf396a0ea8783d1313713338":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-02-13 16:27:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/8"}}},"commit":{"parents":[{"commit":"d5cdc9a34855e4e1c0ae199c62c473f8f2f691ea","subject":"Increase port name size and type to internal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5cdc9a34855e4e1c0ae199c62c473f8f2f691ea"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-13 16:27:48.000000000","tz":0},"subject":"DNM - TESTING PATCH: neutron-functional-with-uwsgi","message":"DNM - TESTING PATCH: neutron-functional-with-uwsgi\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/873118\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e236dce834612267cf396a0ea8783d1313713338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e236dce834612267cf396a0ea8783d1313713338"}]},"branch":"refs/heads/master","description":"Rebase"},"edadd286342e823209b8bfe6a3e6c39400d2c820":{"kind":"REWORK","_number":9,"created":"2023-03-09 10:32:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/79/870979/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/9"}}},"commit":{"parents":[{"commit":"745c415d5c8ba06c97ce423b9dfc7cb0ce8d76cf","subject":"Try to optimize Mysql server mem usage on some CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/745c415d5c8ba06c97ce423b9dfc7cb0ce8d76cf"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-08 10:13:15.000000000","tz":60},"subject":"DNM - TESTING PATCH: functional and fullstack CI jobs","message":"DNM - TESTING PATCH: functional and fullstack CI jobs\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/876556\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edadd286342e823209b8bfe6a3e6c39400d2c820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edadd286342e823209b8bfe6a3e6c39400d2c820"}]},"branch":"refs/heads/master"},"f1bd92f7169b60938bbd33b51fd5330d091a7ca3":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-03-09 14:08:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/79/870979/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/870979/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/870979/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/870979/10"}}},"commit":{"parents":[{"commit":"efab60c0bf22dec3190b46d63568dfe2a4892ebd","subject":"Try to optimize Mysql server mem usage on some CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efab60c0bf22dec3190b46d63568dfe2a4892ebd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-18 04:11:14.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-03-09 14:08:33.000000000","tz":0},"subject":"DNM - TESTING PATCH: functional and fullstack CI jobs","message":"DNM - TESTING PATCH: functional and fullstack CI jobs\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/876556\n\nChange-Id: I68f79af5b02ff271d5e82052940368de23da54dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1bd92f7169b60938bbd33b51fd5330d091a7ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1bd92f7169b60938bbd33b51fd5330d091a7ca3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
