)]}'
{"id":"openstack%2Fneutron~872008","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I9d79b123bb20ded327208d84a14d4f8d2e505087","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1956958-stable/yoga","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-01-31 15:10:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d79b123bb20ded327208d84a14d4f8d2e505087","subject":"Use common wait_until_ha_router_has_state method everywhere","status":"MERGED","created":"2023-01-30 05:55:21.000000000","updated":"2023-01-31 15:12:06.000000000","submitted":"2023-01-31 15:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872008-bug/1956958-stable/yoga","cherry_pick_of_change":848585,"cherry_pick_of_patch_set":1,"meta_rev_id":"234b38914fe2598f1279fa2ad2e4c327e187cda7","_number":872008,"virtual_id_number":872008,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 15:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-01-31 14:11:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-01-30 08:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-01-31 14:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 05:55:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-01-30 06:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 08:18:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-01-31 14:11:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"61f7aa48b5536f7a765b05adbd3f7aa86536920a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-30 05:55:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"633b47a21614532470462df14ac882bd5a968669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 06:40:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b9f0e34775491f94be5c91530aa15d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7ec0053a54422f9585d00f9f8332b7 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b8b36e1f2994deb9c699f24f529ab0c : SUCCESS in 7m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8937f6d70ea549e79d14a5102a9499d2 : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7c65498ebdc0eb27f0740e1d3762822dea9c56d0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-30 08:18:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6908cfc57773b0e0511ef0f48f7250f73d86fd2f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-31 14:11:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09c11a386bedf9133a24b2b9d9f013508e5f1fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 14:11:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3139497dad78c5e0d3112d1a43f455bb1c1267ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:10:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/072d5051a6b44be68f866ec29ccd1163\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c7b4d70c73f493aab56260e43d424e9 : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464c88fb966a4d1e8a21bd6b8aa20065 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3665d44862fb42e6bddc94df666e6680 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":1},{"id":"208924a2dfb78460f6f268c31d5210476e386851","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:10:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"234b38914fe2598f1279fa2ad2e4c327e187cda7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aef42b658ba4aa8a41621468672086d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e13053aeb2f4abdb47872e0352d6f40 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ef9d235d2f9ba764f96d32741fffc28c80076f2","revisions":{"6ef9d235d2f9ba764f96d32741fffc28c80076f2":{"kind":"REWORK","_number":1,"created":"2023-01-30 05:55:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/08/872008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/872008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/872008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/872008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/872008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/872008/1"}}},"commit":{"parents":[{"commit":"5cb3428d770ccd5a95bd3a5fcf799ad07f30f64d","subject":"Never raise an exception in notify()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb3428d770ccd5a95bd3a5fcf799ad07f30f64d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-04 09:09:23.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-01-30 05:55:21.000000000","tz":0},"subject":"Use common wait_until_ha_router_has_state method everywhere","message":"Use common wait_until_ha_router_has_state method everywhere\n\nIn the L3 functional tests framework module there is already helper\nmethod called wait_until_ha_router_has_state which should be used to\nwait for desired HA router\u0027s state.\nThis method has proper debug logging added so debugging issues in CI is\neasier when it\u0027s used.\nIt is also decorated with unstable_test decorator to skip tests when\nrouter will fail to transition to desired state (see related bug for\ndetails).\n\nIn some tests this method wasn\u0027t used so we couldn\u0027t benefit from the\nlogging and unstable_test decorator there. Now it should be unifed and\nused everywhere in the same way.\n\nRelated-Bug: #1956958\nChange-Id: I9d79b123bb20ded327208d84a14d4f8d2e505087\n(cherry picked from commit e39011c73396af68fe5002cdbe4c6b3fe8b7cf23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ef9d235d2f9ba764f96d32741fffc28c80076f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ef9d235d2f9ba764f96d32741fffc28c80076f2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
