)]}'
{"id":"openstack%2Fneutron~872023","triplet_id":"openstack%2Fneutron~master~I936855214c635de0e89d5d13a86562f5b282633c","project":"openstack/neutron","branch":"master","topic":"bug/1995078","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-30 06:34:42.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-05-25 08:41:27.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-29 13:21:01.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"I936855214c635de0e89d5d13a86562f5b282633c","subject":"Improve \"sync_ha_chassis_group\" method","status":"MERGED","created":"2023-01-27 15:19:06.000000000","updated":"2023-05-30 10:08:51.000000000","submitted":"2023-05-30 10:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872023-bug/1995078","meta_rev_id":"14ad35385a19e5cd8dff2ce43230cb0af0f126b2","_number":872023,"virtual_id_number":872023,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-05-24 15:34:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 10:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-25 08:41:27.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-05-29 13:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-05-29 13:21:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 15:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 15:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 16:00:02.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 16:27:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-01-31 13:37:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-03-02 02:52:50.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2023-04-26 14:26:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-04-26 14:26:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-05-06 02:00:04.000000000","updated_by":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"reviewer":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"state":"CC"}],"messages":[{"id":"04a274a62fd31e93a9572cdf1855d540d524bc8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-27 15:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d8cf8082891e8d4ee8869326482d598988a0c9d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-27 15:19:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8c7f25c919967b19bf09353d9d1130ecc3ec89f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:29:31.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871836 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"5ff7374426ba75b9f37cbc1894b914bd92d21620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:29:33.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871836 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"da40357cb9c737fc802b308a68258032670ecfb5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-27 16:00:02.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/23/872023/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"17dc38964fed75eb1cea88e67fc7c5476c08dc34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-27 17:04:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"162a71b8e18e5439db3448b0d4d21af4c424d74f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-27 17:13:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6dd0e12ebd1c527213e80f3c433f1b8717706013","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 17:35:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0aa36e6f86d4da2a4aa64dc134b81be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7525aafed4e043e7ad511ef141f7be45 : FAILURE in 20m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8416471300f841a68da5c209c44ee6b6 : FAILURE in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"00e8b28ca66deac4e2ca470a15b3ad5e562901e7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-27 17:45:33.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/23/872023/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"bcff2bd0d03aa6c6a401721ce3807ef5b8f0a4e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-27 19:08:07.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/23/872023/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f64a73515c2e5dd4c1438447fbb938c5de246bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 19:28:06.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d04afbd325e44ba5be0eaed9682d6535\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a8ed3ff3ac154b1caabb69f4215631bd : FAILURE in 18m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/929a3f6738ae41259893f09748facbdd : FAILURE in 19m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a33ea47a9c646f88c93da4a16f1c218 : FAILURE in 56m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f53bb25f7580432481b418bf8ea374e5 : FAILURE in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7f1eb770371488e9ccf83fb8474be28 : FAILURE in 28m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae0250228ee6420e9c197eba765882cc : FAILURE in 44m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d231db9aae24c7cb81e336ccb4257e2 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6a8ea1942014f3a977ddba48897c4e9 : FAILURE in 37m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e6222da93fc4f73bdafe0c95420cc23 : SUCCESS in 2h 12m 19s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0b0596d6aaae4cc5916c19a23a941f07 : FAILURE in 50m 21s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0279cb6bcb7b4322839bb43b0f0d50c3 : FAILURE in 19m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/be60090ad3744336885ec9979fa71f1a : FAILURE in 39m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/89f09bb9c6814efc97bb3a6b522f34e9 : FAILURE in 15m 03s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/170d08f2e2bb47e687470fda3d0e6e29 : FAILURE in 21m 04s","accounts_in_message":[],"_revision_number":3},{"id":"d7a8226686662454f28337769feb3aa5d5e3e711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-01 12:02:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f0dff14ac41c42ffbc23d0ca88e3458855e7b333","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:12:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a0e781e4854d1eb3f8ef34e6f2baa4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d592be43185469095b0d6a3ec35cea8 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d118637e5fb44a1fa20590878e817caf : FAILURE in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"399db0e85d2908685f3ec9067d289b534521a8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 13:52:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5083c97835b5470ea7832dcba636e941\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/98b533c8957b4432885191c4b65e9456 : FAILURE in 15m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9934eb3bb10447ea9dfa9ccfab27be3 : FAILURE in 16m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c3c0dde69384ef2bf1e48a529e84799 : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3eb246102d84a8d9c08613d7edf72eb : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6b31087a2624deeb85673fbb1557bee : FAILURE in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c828a71aaf4b8b931fee6663587186 : FAILURE in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62667df3fc5f49ce8d46ab6d9023965d : FAILURE in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aad29103c2cf4b01bc69d0279cdd2dd8 : SUCCESS in 3m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b557ffe13de4e5d94894a1e82cd99fb : FAILURE in 23m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a54cef9eeeef40abb073c87f6421ae25 : FAILURE in 24m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cf32b2b5b784866baa6c2ad16ba9a12 : SUCCESS in 1h 42m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a60736237ff246b1803eaa4227a15f35 : SUCCESS in 1h 18m 15s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4b0cdbc1812e44e383dfedbfe87171c6 : FAILURE in 1h 05m 02s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cb1d50f2980f430793ce4d2c29e1e0c6 : SUCCESS in 1h 04m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23053c3ce8c4496b9643280adf873149 : FAILURE in 20m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6103a331815c4218a949c32a4bea686d : FAILURE in 21m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b6b1cc5e4e14ccf9d41a91458fe1ab7 : FAILURE in 21m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6aca05823bf4405f96cffd15edb0931c : FAILURE in 16m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1830e3e7e654e72b5c5934e6818f500 : FAILURE in 19m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e592bd2e937467692dfa0de22070c6b : FAILURE in 21m 08s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d1c9321c91f945ab9ec0340a49c6cbee : FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":4},{"id":"065aad50327971f36bd2e66d1ee547972dbbec78","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-01 14:07:21.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6eb08853500c76e5060a6c97904e458422201923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-03 17:33:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0f155bfdecb9bc4a057583c89ff30703e2f5ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 17:37:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9287b65712f4f31b8e5a2a709928ea6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57399fd54c8e44f7af80054a7f0a115e : FAILURE in 3m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3bfdec5fb2ac426da9d658091edbe5b6 : FAILURE in 2m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"65ab08c4d08ffbadffe024151d2e365a1a472786","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-03 17:46:31.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b33dce9ee4f385d11c26e6d33c4de6eced37d6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 19:01:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ce6fc9550c7460c9ca41bdb7db0df88\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8b45990c9684a8fae7c8ee266fda7df : FAILURE in 16m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b26db0fee5f44b4b04f0cfa2f415a4c : FAILURE in 17m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/499ff9980dd84f3f9f8d86bb4f2dd46b : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a990743a493242e2bc5477b4e061ee19 : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e01c6ce1954487baa9383aa3b191cda : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36c59a33118841ca8b25d4d4dcca6fb1 : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca711214f124f77bd7f0b60616a273f : FAILURE in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70813511d08b4bc4b754199a4780c3f7 : SUCCESS in 3m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c207a3bdafa8405fa7bf06062b79bb44 : FAILURE in 3m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6073d1a524cc4d03915011df870f5a73 : FAILURE in 22m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b8fa02224d446ae8fd7b159508fa1d8 : FAILURE in 8m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49a832e674a6449b94848fcbc17e695c : SUCCESS in 51m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19b58652474b4fdda989421979e1d948 : SUCCESS in 1h 03m 56s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a0b1c99ec8614d73847b68a388f5914c : FAILURE in 56m 39s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/874e74c3961543b1afe9064ae69f967a : SUCCESS in 1h 21m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d4925f9325584924be52ebece1197bd0 : FAILURE in 13m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c2cfa04adac849c78430ebc928f2c009 : FAILURE in 11m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8fb4e821a1904362a9a9ed9bf4a36758 : FAILURE in 8m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2c71cdd214c64418b22eab02f560668c : FAILURE in 9m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c9b96de827b450c895392d1ff48d54b : FAILURE in 20m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3cae2aed460c4715a0dbe9da9f44f714 : FAILURE in 11m 30s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/77dc75dd542b4ea590fb1493a36abb7f : FAILURE in 18m 37s","accounts_in_message":[],"_revision_number":5},{"id":"5669c96e3df9985595ed274c9c194c74b6372f42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 08:46:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6e07c10547d162b415e82934b48f4e9279fdca60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:05:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0414850a8bb941dcb538985566adbc0e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d5e077fbb714cbdb576f124aff6e932 : FAILURE in 18m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e4a516ed37d46f09fb4dfaabe090796 : FAILURE in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96a5891627e61262c5a2e005742e1f47a9b317cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:38:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f613af624f634653b776b396d56f7f30\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/97800b06b87648a0883ec73d39069c7c : FAILURE in 15m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e78d44b79bc4d55b1447aee28c8bb8c : FAILURE in 20m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/528d5476c8c54ed381911d2fc847dd19 : FAILURE in 35m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780fb2da7f10465f9cb2e76c8deed8fd : SUCCESS in 13m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c6a8e31bda479982c856d43884f3fd : FAILURE in 27m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ba2c4e25a4b4afba82f635369d5e43d : FAILURE in 22m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4317cb0aa93e40f48df6677194fbd31b : SUCCESS in 10m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/72e2aa359a1e456195c590069373dd4f : SUCCESS in 2m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8d26604879bf44dd9f90e6a2bab78b9c : FAILURE in 25m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28094152f0644ed3ac0ec887a0020abf : FAILURE in 49m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d92cc9a481fc43a6a94e6846ebb59529 : SUCCESS in 1h 32m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b3199cbfd294cefa38baaf92eaf1a51 : SUCCESS in 1h 01m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/26a24cdf2a2546cda01bfbc7bb040041 : SUCCESS in 55m 14s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a6f5aa4554e14520b8b5f8af8c9e72c5 : FAILURE in 1h 06m 38s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0313603938734a21b81173a11feb769a : SUCCESS in 55m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b84b30de955f4cb69fb41f64bda10aa7 : SUCCESS in 1h 34m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b7eda64944f749f2bc892d58ad959054 : SUCCESS in 1h 24m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0ea5606a4e44aacb9942fafd9ae72b1 : FAILURE in 18m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2abe0f4d91204a7ea583447740b8e507 : SUCCESS in 1h 34m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e5901b38f4d402cb6420c6494a82fe4 : FAILURE in 21m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f20d2a41639c4464934a7faef0e8c147 : FAILURE in 19m 50s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c1f0d2bf2d0e4bbf891ce034c70ebf4c : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":6},{"id":"a3c8ac119702137ec78ee42f70fe1f86f4ad76b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-17 10:59:38.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"ceb28a66b9d6e6b349ad6a3a8ff1ec36375d8582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 11:34:50.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"fa13f1c5a2806cc8c4da4b7a3211ffc2f4f7b46a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 11:35:10.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"e0f34190fcb6dfc8318ae324333bf049bdcf87dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:00:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7964155fcfa54ed5897c42a3663c7f03\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/949611f17b824a6284e42e576fe6d24a : FAILURE in 20m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40007224ddbe40689b7332dc479eddb2 : FAILURE in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f94eeed7f508e5b0e43474da102bb73950e7bd57","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-17 12:11:40.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"6b6c9524249dd43b0caa2375eb4ef278d1701960","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-17 12:48:43.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"0fa43e9ac51e646debeae12e7fa3ad50146194cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:41:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f969bc2d54774bc1b539158e08b05912\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1786528f4b39430dbf0e850171d95bda : SUCCESS in 1h 40m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09a7598f8f9c4ba6a171edc4eaba3861 : SUCCESS in 35m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97e9e9f701024d0a8372e2afa5ef0f31 : FAILURE in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6965380d8e54441a951b98964b4344cc : SUCCESS in 15m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/540e594d229e4442be65aeb875d4fdfe : FAILURE in 22m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/026ba37eeee044feb77c7bc3a4a7ed44 : FAILURE in 25m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662595cb18b94dae937da72e9e2a486c : SUCCESS in 11m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b41a71258852459d806cbb8475e057b9 : SUCCESS in 2m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5d31e2bc61af48f0bae46a66170c89ea : FAILURE in 21m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1414900b1444039bff4b82b9e5d0bb5 : FAILURE in 1h 09m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f6c09a82928460ba4ed9c35773f52e8 : SUCCESS in 1h 41m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d7f093875ed43709f4baabade8c0519 : SUCCESS in 57m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4a8a1896e02d461a90a2b7554b84b6c1 : SUCCESS in 1h 11m 41s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d053af053fd4dfe831ed7b51ec9bc71 : SUCCESS in 1h 15m 21s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/747c6eb722f8445ab814823466210517 : SUCCESS in 1h 06m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dc3670ee7d76417889acb093516cddc3 : SUCCESS in 1h 35m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e213ff9f2c7442ca8ebd0e2180180fc5 : SUCCESS in 1h 17m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/37e215bee85c43b7a1730fccac26c8eb : FAILURE in 1h 09m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3dd1ea6ae3ea41558c776632f874aca1 : SUCCESS in 1h 40m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65c8db99ed5544059deeec9cb7531a18 : SUCCESS in 1h 46m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d54f53149164b1b946edb9f65629212 : SUCCESS in 1h 29m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b3c599bbcff24894a5cb9a200d82a153 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":8},{"id":"4e2e02d110d7412e2ef0c0b6650a3b4db54b07e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 15:07:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"03c6a16290ec0d275fc524b0cd1c9402de399d47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:53:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6421e3cf084e218df41c7c6adbced8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/935fdba17c2845e38f54c30f1853bf01 : FAILURE in 25m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc6ab46d70b14abebb3f0dba654239bf : FAILURE in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e7045a2b67eb94ad4e52e55b064f250d917abde6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-17 16:29:54.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"f3e46fbc11be24816865d214675ff099e8c7a098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 17:15:06.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/342336f1feae4fb88e32b8e3d5d206f7\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b741f3347d3a47df942e5d0a8c293316 : SUCCESS in 1h 13m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/69c757d103c14be087c9378e6dc9750c : SUCCESS in 28m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/751baa8571694c5d9cebaac6e8c7e10f : FAILURE in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b913ea502e643b7a69c5fadd4d0b5fb : SUCCESS in 12m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bc26d658cee4ccb8c507bc924936923 : FAILURE in 34m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c59b02f956ee42b7bb93bb5c766a80ac : FAILURE in 33m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85aacbf495d64bb2856ee0b60cbac75b : SUCCESS in 10m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/33128d3117ef4bd4a2f0637b7ec0e248 : SUCCESS in 3m 20s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6f9b45e4108a4c158d77a5a6413c939f : FAILURE in 22m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad1a2a1983b141fea8f9687c88355c93 : SUCCESS in 46m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec55cb4aaf154b7a9209f716a8fece9e : SUCCESS in 1h 38m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c93dff8c1e647feb6609ab553bed1e6 : SUCCESS in 58m 32s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6426acc7cec74384aaae3e95abc133a8 : SUCCESS in 1h 06m 56s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f0ab677f5f7d44f780ba06e2181e2fcf : SUCCESS in 1h 25m 33s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/afbe3df5f32a48d6b49b467ef08985c4 : SUCCESS in 1h 09m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2c7af531749141338b9c7cec2ce9e6f4 : SUCCESS in 1h 19m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/26b96d1125534de8ae240bbe36a41f81 : SUCCESS in 1h 39m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/767576de70a74a2abbe1093926865ebf : SUCCESS in 1h 02m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7ba99de1efc84419bc2075f9221aec3c : SUCCESS in 2h 01m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70cdec7c253449479d368f5bab7626b9 : SUCCESS in 1h 51m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b025f1437224e0f9cef21942fc59eee : SUCCESS in 1h 41m 20s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f252f7e27a2d4393a37d89259cad784a : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":9},{"id":"77455e91ae76d4f1824f574d176b05983ad763c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 14:23:13.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"95b9857fdb43e248e741d43d34e6f17242f5c575","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 14:56:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e24acb0478c42019b692024188104a7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7aec8639d2be4e0ebe6e9ae6f4d8e0c7 : SUCCESS in 31m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/87d667f921ec436fa5d44ffd16e0bad0 : SUCCESS in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"62773844ce4e30b76d421daf009817050eae5915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 17:27:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7770c81ea3b44e58a6451a7d8a8e1bbe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/022a354ff304478693ac862a7f4552a1 : SUCCESS in 1h 15m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48283c7ba9374af58ed55b8aa1a348dc : SUCCESS in 28m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0e99f1fd36045a5a93c6767cc06a6b1 : SUCCESS in 39m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5935817f52b42339789b5a1777c0089 : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d16d44f6cc748f99ace392c0379ef82 : SUCCESS in 27m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9058b5198a6146d99cb048d5da6865bd : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8ade8834d74f20a91c5d9ffe549681 : SUCCESS in 9m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4dfca7f56d2f45d4b6eb1cdb94566e9c : SUCCESS in 3m 00s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/2dae0a8f757f4ffe9810b12bc75888e3 : SUCCESS in 28m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/231aaa7c3a734c37a5b3194f1554e65e : SUCCESS in 36m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/607a2ddcfe77423c996fea1d1194b15d : SUCCESS in 1h 28m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2b5e02207cb4e259c2a7477641845c0 : SUCCESS in 53m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fbea6e7a66054873ad0f8baa431aa634 : SUCCESS in 56m 25s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1612513bd7294a3bbca3cb28bb238252 : SUCCESS in 1h 03m 34s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6bc57acc75b4f6a8a0832bb15c69afe : SUCCESS in 1h 04m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/89f5d55fbe5442299b724da8b4aa5c01 : SUCCESS in 1h 44m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dcd4e314121048988f59eafe751860df : SUCCESS in 1h 27m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c1a5fb89c4164d7daecc44a0fe16b5fc : SUCCESS in 1h 01m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df69eae0266645f8a855db968efef691 : SUCCESS in 1h 37m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44286bdeb94f4ff2b490f33c530068f2 : SUCCESS in 1h 38m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/458b55dcc77c4814af21a6025960c62b : SUCCESS in 1h 15m 56s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/166521718ba84c74872f695eb34579b8 : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":10},{"id":"d5fc1df0c6992a177cc5eb664aaee6f0b877c2e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 18:11:40.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7c92e491f7198b24f673bc21d2eff410afd14204","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-20 18:13:19.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"46ceb7ca3aaa41c53855d1d2f86a1cfd9a40ec04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 19:00:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d51326246f8d4e5baf149f42311d196b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2e2b70e2fb9421f88abaa19a302f4a9 : SUCCESS in 44m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69087274f76944fe84b1c76cabf5452a : SUCCESS in 45m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"64db65210d9e69682efc6e749d6c034c477afd56","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-20 20:04:25.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"4a7fe747e4b89601e6da2ac7508272dea673e8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 20:18:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d8d89b65eb3475ea7c1ff60cd9bf3e3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/feeaccf96827421da759c43af18d5df5 : SUCCESS in 1h 26m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6cb68f051ed4142b0b3efde610f27a2 : SUCCESS in 28m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15d52b3b069b4c8aa6f6a92fcfe066c5 : SUCCESS in 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c980f1fe1f4a22983bf0dd256ede85 : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0597b0eedce458e8f77adf494b0fa71 : SUCCESS in 28m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ded52f37a4e46ebb422a32f306069e3 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c2615b87264379bfd3edc57d9d0fbe : SUCCESS in 11m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a021a3e8f12b492eb844f48130c602a6 : SUCCESS in 3m 35s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5e39d83abbf841f8a59040a55125db73 : SUCCESS in 21m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eff3f7fc6e0445a99639579bf960c794 : SUCCESS in 57m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/279fba244c5c427b8a7983476fe88cbf : SUCCESS in 1h 43m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/54df137925584dfe89d9e972083f660e : SUCCESS in 56m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0be00919cd1045ff8eb075f69ca315a8 : SUCCESS in 1h 01m 09s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8e73b1252a324e3a8389439d2515af8f : SUCCESS in 1h 13m 13s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ab73d44ba9c74d27bd432f1cc914eac6 : SUCCESS in 56m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/121d7a9e770e439190a6a497e600dada : SUCCESS in 1h 34m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/28507d30283c41ed80a43e2b69cfd190 : SUCCESS in 1h 26m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/89cf6ad9c1294706899487a97340a775 : SUCCESS in 1h 05m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c92ba04edceb4c1e9f23ecf7c59c951e : SUCCESS in 1h 46m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5ee7267c17e45b592dd49747dbc4884 : SUCCESS in 1h 58m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f072ab9fc60144a2b0a11c9f600ab1cd : SUCCESS in 1h 06m 40s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/578142e93e5142daad9906e8dd49e08c : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":11},{"id":"34e47d1f7e12010ad1f5b7e1d0a531ff6f85907d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-21 14:50:55.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"44431599789cc2246a48dc861059d45503292ebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:14:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31da7ab269964afaa9d0deb5f723bb37\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b390fac73fcb4ca59974026c9f47f839 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5227ad2445bd4fb2995389199b6ca543 : SUCCESS in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6aed8b13add145ac356ec662e8158d5ba79dbde7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-21 15:27:58.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/23/872023/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"aa4bdd482fc8681d74113a772f5c8783b375e05e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:34:33.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f31319fcab446982b4084e200209ed\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/890d4e3b5da742c9aaded9353a5809b4 : SUCCESS in 1h 08m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/161dfc95e75c4326b77578075ed40c0e : SUCCESS in 26m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a29f170dcbbc48be8fc4be2d9a116a55 : SUCCESS in 41m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c86a5c40e4349dda0a45166f070cb4f : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac0c1084e5aa47c3b8f66865609fa8e5 : SUCCESS in 27m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/110c6e21f6c043efb737d0e53a665a07 : SUCCESS in 24m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a753fffd302846dd9ef918fa0c9b338a : SUCCESS in 11m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bdef7ab2101847929f56696e7fe75ceb : SUCCESS in 3m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/62abf50cca244656bb7bc462fc88f6c3 : SUCCESS in 21m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef819ed77eee45b6b0fab63e4e133302 : SUCCESS in 42m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/426d8b27a9be43a2aa96fa586c9b1514 : SUCCESS in 1h 35m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0983316732b40dbaf18714e29706e95 : SUCCESS in 1h 06m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/73cef766884a43bb928931f4ae2b44e2 : SUCCESS in 58m 06s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8ee8c95b8cd04fb881ae1595dc8c3226 : SUCCESS in 1h 06m 57s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/abe9733a4c314f7dbd6f46d39c657a6f : SUCCESS in 55m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f55bf4e1e26d400eb1ab8e27d35bc99e : SUCCESS in 1h 36m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9be9377c8d645e5be146271f1a29fc3 : SUCCESS in 1h 20m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/479daf131a8d4f168fcd8dfc7a8466ca : SUCCESS in 1h 00m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f7ddb612d38446778ca192b9087d8efa : SUCCESS in 1h 37m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb141d9f2aa046a5ab21ecf8424a1fa6 : SUCCESS in 1h 38m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93bd607ac9e940a7be8f2a508c4c20c7 : SUCCESS in 1h 19m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d3d60bd1597b4ff1bc6b4334f98b03b1 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":12},{"id":"4541e915ed897c92337463881ad3f244efc6f2a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 15:15:14.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"f8ad89009defc1594617e0d22400f6c8987837bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:43:03.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03744b4c626a4d1daba2b1aca6c71b9f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fe302cc9c9348bcb8419fefe6cfb11e : SUCCESS in 26m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65a5a9d518c34bcd90c54aadaedc4e96 : SUCCESS in 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5cc8512d52f3347eeca0c3f144f55c0a8c35c0c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-01 15:53:52.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/23/872023/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"226c428124ce58526eab8c6e493917812feda7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:32:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0af44137730b4ecaa75413cb0be32b0c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/35968152da0f495f82c1c993d2c4fd46 : SUCCESS in 1h 14m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa61dc3b800e4ac9b4c44874c9b6c62b : SUCCESS in 26m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08903763cae24d6a80b595502b6bd077 : SUCCESS in 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1a8e1ebdd645d9a4d00ce3cd8f0cea : SUCCESS in 14m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45bc5ea03ac7465692c899d74af7cb5a : SUCCESS in 23m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d23faed907f4e31bf9dc33938021474 : SUCCESS in 23m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4130938a643e49a283b707f66cc94246 : SUCCESS in 11m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/88feb89d097e40b580472e72df2d72c6 : SUCCESS in 3m 05s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/462a08e189b5408c9090d1a7b77f6371 : SUCCESS in 24m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7231b15679ed417f971fdc584e2b4b74 : SUCCESS in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4742c88757d44c4932dcb4fc4b635ed : SUCCESS in 1h 41m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/64c6a22556f04741b4409a0bffccd882 : SUCCESS in 59m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/831faeda72b841e1bb30d799b9ddde38 : SUCCESS in 1h 03m 54s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3963826ab25746809708cd694dcbc478 : SUCCESS in 1h 17m 28s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cbd3331af81d494691094125e5bc747a : SUCCESS in 58m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecd9c5aee90040f886c64440f8001353 : SUCCESS in 1h 21m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/534ce44066bb4eff93a7923aa48a09e2 : SUCCESS in 1h 16m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5583f07bf604491381cbf75433f57b72 : SUCCESS in 1h 06m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b709a745fca94adc98715a2581211522 : SUCCESS in 1h 49m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/720800531c6b480daec78d4d23dc60f0 : SUCCESS in 1h 50m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/16f0a9c241124c6999c7701ac32c6b4d : SUCCESS in 1h 05m 40s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/22eb4d737e9f40f0a1cc1fb725b01a55 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":13},{"id":"46c3a7af24f5ffbbff2be437378a2f52578e4c9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-24 14:18:07.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"f08f0db9191fa36f80879c75a92db836ffdc042f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 14:46:23.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1724610cc52245fd9d536c79fbb427de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8abbe64879ab49b0b4cfe1f1aaed0706 : SUCCESS in 20m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a817f3b7c504af69ed5f8c5af232ce3 : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b601535f05c745b693e607eca64e1b58 : SUCCESS in 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"34018ef65fc62510416095c80bc43d8588b4f347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:27:50.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4a1af7fe764fc8824433301386684c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/077b10473f78426d8bef00918ba9b8d1 : SUCCESS in 1h 21m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06f9ed290b4c4286892eed6d7fcd4a2b : SUCCESS in 28m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405e9a0a30c24d6a8148f6382053a164 : SUCCESS in 43m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9756ba65c704c5ea74f205b8eefa2a1 : SUCCESS in 16m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1489552b08054330a142c26588e7abe1 : SUCCESS in 23m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54912716f3094837895662ab2f3469dd : SUCCESS in 26m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59946562b8c041bbb00bc6f092b1dc4e : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71775c3c4199496cb510c79030cc8d10 : SUCCESS in 10m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee71a0b0868e4ccc9c59a5522288ad25 : SUCCESS in 2m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/dbd5b8b1309f4638b0424d20786baf2d : SUCCESS in 22m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d747962557034993a810e123326bb252 : SUCCESS in 1h 18m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/674da0d818394a6c9870c19b28bbc729 : SUCCESS in 1h 42m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/de48a364438344168660a3795309b394 : SUCCESS in 1h 23m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4c9da371a159488d9b4bc73562531e0a : FAILURE in 20m 37s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0c2537c528324ab1a7d73dfd74134df2 : FAILURE in 25m 27s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5a2f6be25b1f422796a4f2be593d34c4 : SUCCESS in 58m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b58a2673acc45be93e5d7477c52aa93 : SUCCESS in 1h 55m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a4c25c001a6443089d311f5e269f9988 : SUCCESS in 1h 19m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d81eb2e08198426caaf2755091bb1543 : SUCCESS in 1h 11m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5a2efc225dfe4bf782da591cfad6f740 : SUCCESS in 1h 42m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/837a7ee1f29b44749dbdecb23a6d855f : SUCCESS in 2h 03m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00683a3218c74f7c899b59d5bb59ee20 : SUCCESS in 1h 10m 10s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4b452b60711448e483f7c08acf8c0066 : SUCCESS in 1h 39m 34s","accounts_in_message":[],"_revision_number":14},{"id":"b0fe730828efa9d0c92e34bc14622b3a8465eff3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-24 22:09:16.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/23/872023/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"9d6dc7e5f1aa74260372379c255aafd723b7eb91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-19 07:44:47.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"310b16b6cb0636d0a6378f2bdab61e0624c0e640","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 08:08:56.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd22228773443cf83f772b485307c39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81211a5894a242718a60d7be56a8c1d1 : SUCCESS in 22m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9b9c3acb3a847cb995485b9740bb667 : SUCCESS in 21m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/15ca37adecc94adba1afad93bf09c9fe : SUCCESS in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6e5f48fd0b0537fa781bec6d68f157f94aec3dcb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-19 08:21:34.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/23/872023/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"93c9081cc05db4669b6763ab131e4af3e4432c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 10:38:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b07749ec360464aa044b55e59185a30\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca8acf9d574e48f78ad07a652bddb468 : SUCCESS in 2h 51m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdd73537fbbe4482a8a17ff34c2ccc3d : SUCCESS in 45m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71a4dc1ab9fd4b56b3bee6699634ad61 : SUCCESS in 36m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a53fd26c184524b841402945846d96 : SUCCESS in 12m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbcf4cdcfa94dc0bf82d2c2ceeea792 : SUCCESS in 25m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/607e86cae26642b9819e1528bf4babda : SUCCESS in 27m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b1bb422567e47609145054a3a1d2eb9 : SUCCESS in 24m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/015695bdb2a347c6997c744c4b72bfc4 : SUCCESS in 11m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f97ea926c30493ba70a6bacbeac4ce1 : SUCCESS in 2m 39s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5dd41c55fb5a4f44a5ab181c841f9706 : SUCCESS in 22m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f895691b4c44f33b02f225f5f5cfd51 : SUCCESS in 1h 00m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e931b88a6ba241e58f7ff3e4e02bb3f1 : SUCCESS in 1h 49m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/93feca47bdfb414cac635b2309bf0f9e : SUCCESS in 59m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0393539d54714d6b9b9c8f8757372aad : SUCCESS in 1h 01m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b4176d2e7f674df3bfa1f9aec3ceba9b : SUCCESS in 1h 27m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4e2560552343474d828e492d2eaa8b7c : SUCCESS in 1h 25m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0e5cf7c71604213b26167d94d307428 : SUCCESS in 1h 11m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/84494b2eb5b8437e8fb6e09064c57d91 : SUCCESS in 1h 46m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8e656ec0a4f4d0f8c3c819f04a806c2 : SUCCESS in 1h 54m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/342a009c1f604d049537e73cdba6d49b : SUCCESS in 1h 16m 27s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/280e4e469d674adbbf25d5e2882c82cd : SUCCESS in 1h 14m 36s","accounts_in_message":[],"_revision_number":15},{"id":"107a672cd956439cf5fa37dba25e70c66b5c193a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-28 13:14:48.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"65c43f80e459a10c59ae826c5c70f0ff5c8e1e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-09 09:51:44.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"86c35829a77aef664f8dbe595cb74828931c2288","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 10:22:04.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4e0ff8e91b4897845b83a4ec5fb2a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d74791054ef48d482b153f1f169afd7 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bc1a1b2a57b467fabc66437bb196a17 : SUCCESS in 28m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2f87a6f49a14b7c84d24df341475266 : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"3842081b8739583bc6c6ef7da95cd5e34a07aa4d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-09 10:42:03.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"19119ee8774156c6bbe64165dc892952484ad0d3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-09 10:54:27.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"7f3d0a9433f1267730e2ecd3da7fe5978058bb27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-09 10:54:55.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"2df80007d850127bc35fb630d6f7e1ac6020ba07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 11:04:30.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dbc0a81c4c24010a7ae9da28df3f265\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c26cacde34b949bfb23ffac658e43d95 : NODE_FAILURE Node request 300-0021172477 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/406fdfd7d0c44a248db95f0bf60b6d04 : NODE_FAILURE Node request 300-0021172478 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f69e85f8c8a04b09b25dd10faaa2a842 : NODE_FAILURE Node request 300-0021172479 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"86cf49ede5036d9180607dfea3ce5360f15534b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 13:00:01.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d173a641182c43ab935fa029f01f58dd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53db342fc4234fb1a7703beb4e36b0cf : SUCCESS in 1h 51m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae773b3a05db4bf7b1a01cb564f6a65a : SUCCESS in 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aabfc558337848d1b2b73963be07ee37 : SUCCESS in 37m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bf3ffbb5d34669bf9cda00960c5fff : SUCCESS in 13m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57c2d82753144406801faf4e05cd6e2f : SUCCESS in 28m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75902547cede4caabe90f7c8b9114e40 : SUCCESS in 29m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd4ff7938cdd483cadc543e89c147704 : SUCCESS in 31m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb71f49d52a4f97abb3f2876b704ee6 : SUCCESS in 10m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c17c362441e845cea58216096186789a : SUCCESS in 3m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/859abd55f511461daa85c98d07432c07 : SUCCESS in 26m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df9fa44077a849309f6b438fbcab0fed : SUCCESS in 48m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfd43f57e07d4e5ea42bdc17be1b029c : SUCCESS in 35m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/929f8a40b4294a54b418cf010d565cd5 : SUCCESS in 1h 00m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/098b0cb0889d4f16b8d56e7199ff85c4 : FAILURE in 42m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/276ed4c284c34ed4b0c8b50a5e3f984d : SUCCESS in 1h 24m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/33f282ee0f5d43e4a80d9aab08f73667 : SUCCESS in 1h 20m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d2a3eb1da4c64596bb32d672deaae11e : SUCCESS in 56m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/89ffd27bf9444b3d88bd28033259ea43 : SUCCESS in 1h 34m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53e1ae81012c47278d30380c61051e68 : SUCCESS in 1h 56m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8c3921a21dc74a05a160675f85cec035 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":17},{"id":"00237e4b1b61417f9723f22602eb00be06c3f623","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-09 13:09:03.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/23/872023/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"35b5d11ad03318f0fba608deb8efb935bd220df9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-09 15:24:20.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b175a8e56db502ff9811d347f7d2bec92c82a150","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 16:12:15.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eeae1b5a5ff40918896395135c1a6c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/facf581d14a84fa68917856a9bd8edf5 : SUCCESS in 46m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c825a61627e4dc0befa6fe8ec63d5a4 : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ed1903095aac440ba45aa1e8f47b5f23 : SUCCESS in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3c88fba39e1482fc2978566b055e3aadcc51e7f1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-09 18:03:06.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/23/872023/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"6f613ea108daba6dc38f81ad62b7264d0829b2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 18:22:47.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b9638af90a4a98be412864ea26ed50\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cae86d9b9ef6450d9f4bc05b08ce1d37 : SUCCESS in 2h 11m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70433a40f2c74ad29bee34ca57a920e0 : SUCCESS in 39m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da6c3f7eaa14cb48794669edce4b7e6 : SUCCESS in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c662560d745442698ac1b5267d9bc610 : SUCCESS in 14m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40a75e5cd4754315908a2fc18db3accb : SUCCESS in 27m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/322648c05bd14566a5a440fb09572a30 : SUCCESS in 26m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c775189b97e48be955225cda3290577 : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7047ad78cb4d43ae37ae1c9375ef59 : SUCCESS in 10m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9ba76b6e02884ce9a605707c1cc1ae49 : SUCCESS in 2m 35s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/45195e51eb514257803154690948b004 : SUCCESS in 25m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a298fab182eb4660b8fd6145106baf78 : SUCCESS in 58m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c45eb8601b7e431298461e011195c8ac : SUCCESS in 29m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f804567384e5412aa524060851273f09 : SUCCESS in 54m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eaa3ca56c1254eb2bf7a573eeaa857bf : SUCCESS in 59m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a87038d6394d43df8f270e79bf779fb2 : SUCCESS in 1h 25m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f95d3a5c6ac44a5afd14c78c223d1dc : SUCCESS in 1h 28m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/25bfe050aa67419bacca4d82ab27f0c7 : SUCCESS in 1h 04m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c4ff5e452f804b398e6bd9696525fc1b : SUCCESS in 1h 44m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8d34cbc459f42aead58372895573911 : SUCCESS in 2h 09m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c282c7f8fc5e435396ee370a93ed275b : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":17},{"id":"8d9768ba1d2c1fe44b645dfd96bb377e3221ca42","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-11 15:05:18.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"a8eca4870ce868252e8e0c9dd5973ea5abbdbf36","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-24 08:19:21.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5c9dee3a61adc23b35aa230346d8c7fa1fb4bb88","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 08:25:50.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"33a1514b628164fa6805d584352898df793b94be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 08:39:41.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"70e75782ff87e8ff0f6efe89d52b69c16d0ef888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 08:56:50.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"711476e0ecac431d0c7c01a2d5124627f26672bb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-24 08:57:25.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"4e1ccb86ce28be5297f69d512c13721b0962e77e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-24 08:58:11.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"04851f442e8c46848a3f964ac239fb9d36d20874","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:53:44.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d92c12827e4f2bb3116a07a1c1c939\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99bcaf7df424d1988b018420963f99c : FAILURE in 21m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a829be1c7bc749efa30348dab379765b : FAILURE in 49m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b78c86f187064c46a72593ff48d06549 : FAILURE in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5525109f951788dd69af46c9774e758c3e6a3509","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-24 11:03:48.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/23/872023/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"ef97c92d504a08fde0c885393092b6d5b63562f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:04:18.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/043680960160476db4f6e49c20ebb745\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fff4086959074231b69468aa47252ab1 : SUCCESS in 1h 25m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ac4ddddd1ae4b83b99438ea421fd8b1 : SUCCESS in 31m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f5c4f70dfff4236a6e7a2db3fc063ac : FAILURE in 34m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acfe0eab84194fd8bb75a1113b1541b8 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/211d98f6643147d09207d621d737c0fb : FAILURE in 28m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67937e6867a7409391b36d20e5e05b64 : FAILURE in 26m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/710f3b16dcd740f1b68eb008bf5dec66 : FAILURE in 28m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4377f819fe0d4a04a5f27351584cc089 : SUCCESS in 10m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9448380ad4044f918e19056cca74cb10 : SUCCESS in 2m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/447aeaf87cf244a3aed9f96f9c04d569 : FAILURE in 24m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d32ec018c77e42b78872be6bce765ef6 : SUCCESS in 59m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0da19f3b882e4f5a9f24460c53c65361 : SUCCESS in 1h 44m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed66626daf114768acbf683b34ee350d : SUCCESS in 59m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e84709c696724a75b012b02ef4b241fa : SUCCESS in 56m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c16ec3002fbc4b7191bf8089886aa5a5 : SUCCESS in 1h 21m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e381f1d951bc474bba50e24f969f2a6f : SUCCESS in 1h 24m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/46d73ae45a3d47abaf44ef7fb1db233a : SUCCESS in 1h 06m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a0058c84407945bead2fd4cb7a8f0dbc : SUCCESS in 1h 46m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b0a141cb49924ebbb40054bce4894364 : SUCCESS in 1h 52m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a44378f1f8b42eea8771581d7a42528 : SUCCESS in 1h 26m 31s","accounts_in_message":[],"_revision_number":19},{"id":"ceec46948c95a4e146bbd95f5d26c26181f12d0b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-24 11:47:00.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/23/872023/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"fce08a1aa008ce58426d8bac94c781b102046e63","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-24 14:18:17.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5c464f338988d84309ff446f915f7c89eca3033a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-24 14:19:43.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f86806ded761385ac75c8c909f7463d5a64e8ce3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-24 14:23:27.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d48859360cf34ca47619b25d947425595e378ed7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 14:36:49.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"316d051fa5cb1475a15c4308260df9fe7d4be670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 14:37:31.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"9619dc79686394b6f5c1becab46b812cbf6f4e6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:09:39.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c7bce3ecada40efb70cd1b7cfd6d868\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3b11f900a134fa7b79b9a7e2c8bc2dd : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c92175d5ff1436394705887582205a6 : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb51e4fb2f734ce4a267704d88692605 : SUCCESS in 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"079a87eb5dba9a5c548e6bd9fafee0a22bbdc95b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-24 15:34:21.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/23/872023/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"c20d0c503016fe9a4a1e2bd9991768e06d986d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:32:34.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ae602ed1f3480cb7bf7d4e2fafc00f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/413a01df639041ccabe33f3421c5deb4 : SUCCESS in 1h 33m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/34cad7111be54b12b8affd5e374835c1 : SUCCESS in 39m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e46121e508f54edc962727f1ff566acd : SUCCESS in 34m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704468d0c4e84d7597d76666d47e80d2 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c9d8664c15449e3b948112c44548d5c : SUCCESS in 23m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4264c27c18e4abb960bbfcce7d65c40 : SUCCESS in 29m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8923f7d02f334694a759171879ffbcf8 : SUCCESS in 29m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78515de514d4cb2bc24097ad1884911 : SUCCESS in 9m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bea75d3430714c6e919f27cc5df663e0 : SUCCESS in 2m 59s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3baf46ac3a9240f0883e47db45f5e2c8 : SUCCESS in 23m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/421d02d6dde648b59fe124072bf4ada8 : SUCCESS in 46m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ead519c795d14d5a902dbad2c9fdc06b : SUCCESS in 1h 47m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0330c0d50cba48d69bfd1f44347d7456 : SUCCESS in 55m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0be24a231fe4d528a6e92b8822f9e76 : SUCCESS in 58m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e95d3b184b849cfa04668ccbe741efe : SUCCESS in 1h 14m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/596c04128aa8445183d804d5fe2926bc : SUCCESS in 1h 31m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/937ca9bbd1c74871b1a1e85dc037af4c : SUCCESS in 55m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/53291ff52e014e3a9468ece34a6549cb : SUCCESS in 1h 28m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3dc4e0c5181c475098115fb3201c7201 : SUCCESS in 1h 32m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/32f433d908d34e2dbede32c07c747113 : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":20},{"id":"86eba21f5f36b54405783d20affb497f37a459b3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-25 08:41:27.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"c80073ce6387a74d4a99a25ad6c356a3beb00282","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-29 13:21:01.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"61074540a75269bb1fbaa7783bb7d4d25b839592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 13:22:12.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"a851580d6bfc2df3373d0c9432a3fad556be8a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 19:23:14.000000000","message":"Patch Set 20: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee37bb53ca1945e9865ff427ed936cb4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de9283375c27409583b0992acbc55fec : SUCCESS in 1h 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86fd9fe7c6984f159bba439087e81b27 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8d4fad012884f3abad1ca4d89279598 : SUCCESS in 30m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f85563d627444fdfa8337fd4b0c28864 : SUCCESS in 27m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27cbf509b3e3423f93c5b9576d1218f3 : SUCCESS in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0c1f2c319e48e48119392172fc8014 : SUCCESS in 11m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/68ecb451ba1944d388c93bb6814fb895 : SUCCESS in 3m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/360be0db8e1f4cc4970ca6a71ddcbf63 : SUCCESS in 1h 09m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f8d140b014c45709adf73ed491207f2 : SUCCESS in 1h 34m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3507a61b4c0b4640a29f303a425e82c2 : SUCCESS in 1h 31m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/744edcdd5fc8496ea711866d2c975499 : FAILURE in 42m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a803d6ac62674cc08ec4a5d7da0d87e6 : SUCCESS in 37m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a21e3937676243ceaf3249d6838452e4 : SUCCESS in 1h 15m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/932910b4d33347ee945462ae6614bc6a : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":20},{"id":"8a2e6523b4563142a65d021d3d5bf4f0075cd154","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-30 06:34:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"50645e548abe7a19809c88d4dfbd611ba68488c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 07:31:45.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6e23b463c74d7f9db048a1124da42a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa8f4f3504f846e788d9d7254ff2c562 : SUCCESS in 53m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f640458af91148dc9447a18d072df01e : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e0dc01837cc4ce5bceea4b5a0155a36 : SUCCESS in 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4c92c5523fd41ce458bf9f56096b35115f3a0e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 08:22:41.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d97c81c106d546e890e2f5a3c3143b17\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/98b199dde4ad4413a451a7d9ce8a3e2f : SUCCESS in 1h 46m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/94703b4c395e40e986cbcbf50a5800ca : SUCCESS in 30m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/073f30f711964fa5a6418ee171f3eab6 : SUCCESS in 47m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b28a4a6bc55422b9bda2e0d3b4147d7 : SUCCESS in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89fdccf11e144219824e1f03b26e4449 : SUCCESS in 29m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25051ffd795d45b99a55a256c3538f3e : SUCCESS in 28m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0dc091f2cc8744fda6d6cc6e4cc12c21 : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ae17c808364de2ad559fd1882ac6b3 : SUCCESS in 10m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3663d07004c42e2b36e7661f86a6d72 : SUCCESS in 3m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a58a5159bd84492b8904a74eb9b87c2e : SUCCESS in 27m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/14283a7034284ccab2df4ad6b37d3cb6 : SUCCESS in 57m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/45aa95d5faae4e8cb707c93473e81e4e : SUCCESS in 1h 24m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2f0e4b905764930b0acec2f86357c6c : SUCCESS in 47m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9715d7c261be4f71842e52ef9e373ba8 : SUCCESS in 48m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/91ee6a67af924d33a116669691afb42b : SUCCESS in 1h 21m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/be5603070aed425a93b0ca5f8fbe5eb7 : SUCCESS in 1h 14m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3d74f33ba07e4142816a9749d0c9f690 : SUCCESS in 1h 02m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/508b0c6674974a808868b7b87bf4e348 : SUCCESS in 1h 31m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/500162ac047c43be9eda778c652a68d9 : SUCCESS in 1h 41m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c73a021b5ffa4ce0bd8e76077b1c5aa4 : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":20},{"id":"eefc2679610a43e513b817496ac7197a28dd7e66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 08:23:09.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"e1fd84e8ff258ea7a6478579f7b7c69b78da4232","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:07:41.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd72e5c1a08f4b5a96ae6721945d2d79\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c05f3ebebe9f4de980b9858f0019b3a1 : SUCCESS in 1h 43m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef9021b7d234c1680b883b4106bd0ec : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a430acb6e55645b09dbb47a5dbd8e9eb : SUCCESS in 26m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a04aa5b08daa4d2c8ada24bad0b7dcc6 : SUCCESS in 28m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bc5e69edfab441cb5c4ab17978536f1 : SUCCESS in 32m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be95e738cc794f01aba083bcc5874f71 : SUCCESS in 10m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4df0623ed4324998900a8d977f7d10f8 : SUCCESS in 2m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4606c86a829e42788aaab3944f7af170 : SUCCESS in 45m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/014b8c0109dc47848bc3fdd658f4d131 : SUCCESS in 1h 24m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd75d833a14b4bb3aac56973a564faf1 : SUCCESS in 1h 33m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/046ad91af799454ebce2a17828e887df : SUCCESS in 1h 10m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5439e7c5ff874226a6882c2c9fb47c2f : SUCCESS in 1h 02m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5a7919cdccb5416cbaa99a7e87f3fd83 : SUCCESS in 1h 03m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cd047bcc3fb74e4783ef88ecad5159bc : SUCCESS in 1h 23m 55s","accounts_in_message":[],"_revision_number":20},{"id":"1ed922a5c632a4f4c6f26133fe90d7d2688d08d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:07:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"14ad35385a19e5cd8dff2ce43230cb0af0f126b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:08:51.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff44377329514681af55f079b937d92d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8c95b15c084894a83355117e6fa118 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"ac231c817473c018dde8fa31594b1c9a78a36c13","revisions":{"de39eea00b9c885112742341f347c75ac1c60634":{"kind":"REWORK","_number":1,"created":"2023-01-27 15:19:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/1"}}},"commit":{"parents":[{"commit":"30aac23a5eb2ae89d91f5f375ceaea7633784e8d","subject":"Merge \"Add doc note on nf_conntrack module requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30aac23a5eb2ae89d91f5f375ceaea7633784e8d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 22:02:56.000000000","tz":60},"subject":"WIP - Improve sync_ha_chassis_group","message":"WIP - Improve sync_ha_chassis_group\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de39eea00b9c885112742341f347c75ac1c60634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de39eea00b9c885112742341f347c75ac1c60634"}]},"branch":"refs/heads/master"},"93f94e83e25bcfbd66ef158a5b75f66cff026b89":{"kind":"REWORK","_number":2,"created":"2023-01-27 17:04:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/2"}}},"commit":{"parents":[{"commit":"30aac23a5eb2ae89d91f5f375ceaea7633784e8d","subject":"Merge \"Add doc note on nf_conntrack module requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30aac23a5eb2ae89d91f5f375ceaea7633784e8d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 23:46:37.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93f94e83e25bcfbd66ef158a5b75f66cff026b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93f94e83e25bcfbd66ef158a5b75f66cff026b89"}]},"branch":"refs/heads/master"},"b1b601dbf898660382dc49dd478c75c5a9fa0611":{"kind":"REWORK","_number":3,"created":"2023-01-27 17:13:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/3"}}},"commit":{"parents":[{"commit":"30aac23a5eb2ae89d91f5f375ceaea7633784e8d","subject":"Merge \"Add doc note on nf_conntrack module requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30aac23a5eb2ae89d91f5f375ceaea7633784e8d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-01-27 17:13:19.000000000","tz":0},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1b601dbf898660382dc49dd478c75c5a9fa0611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1b601dbf898660382dc49dd478c75c5a9fa0611"}]},"branch":"refs/heads/master"},"c4856a256de77d83253e8f3c868c41c5f801641d":{"kind":"REWORK","_number":4,"created":"2023-02-01 12:02:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/4"}}},"commit":{"parents":[{"commit":"65bd4a24f5543053ccfe9e9ab37fd516a11dc14b","subject":"Merge \"Use SLQAlchemy ORM \"relationship\" instead of \"relation\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65bd4a24f5543053ccfe9e9ab37fd516a11dc14b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-26 04:06:40.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4856a256de77d83253e8f3c868c41c5f801641d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4856a256de77d83253e8f3c868c41c5f801641d"}]},"branch":"refs/heads/master"},"56fc2b9e77a0a90287142bfaba848eed1e9d0a24":{"kind":"REWORK","_number":5,"created":"2023-02-03 17:33:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/5"}}},"commit":{"parents":[{"commit":"8aea97157450af8dcf1b36b20c67dc1b439f2023","subject":"Merge \"Add debug logs to OVSInterfaceDriver.plug_new\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8aea97157450af8dcf1b36b20c67dc1b439f2023"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-03 18:32:09.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56fc2b9e77a0a90287142bfaba848eed1e9d0a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56fc2b9e77a0a90287142bfaba848eed1e9d0a24"}]},"branch":"refs/heads/master"},"dbfc636c70d0acded5108836ec6efdadead1b9b8":{"kind":"REWORK","_number":6,"created":"2023-02-17 08:46:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/6"}}},"commit":{"parents":[{"commit":"aface4c9bc4c4463a6ae4b9cffda5f1f47cdf6eb","subject":"Merge \"Improve scheduling L3/DHCP agents, missing lower binding indexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aface4c9bc4c4463a6ae4b9cffda5f1f47cdf6eb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-15 17:05:07.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbfc636c70d0acded5108836ec6efdadead1b9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbfc636c70d0acded5108836ec6efdadead1b9b8"}]},"branch":"refs/heads/master"},"beee06a9d3f501ed07b258160c11fb082081b507":{"kind":"REWORK","_number":7,"created":"2023-02-17 11:34:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/7"}}},"commit":{"parents":[{"commit":"aface4c9bc4c4463a6ae4b9cffda5f1f47cdf6eb","subject":"Merge \"Improve scheduling L3/DHCP agents, missing lower binding indexes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aface4c9bc4c4463a6ae4b9cffda5f1f47cdf6eb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-17 11:34:50.000000000","tz":0},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/beee06a9d3f501ed07b258160c11fb082081b507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/beee06a9d3f501ed07b258160c11fb082081b507"}]},"branch":"refs/heads/master"},"ab11e3c7ab4c83a928d06bd44fa255c0d5e09bad":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-02-17 11:35:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/8"}}},"commit":{"parents":[{"commit":"265ce3de7ba2afd3b9cc2260d7d51fb197efbf02","subject":"Merge \"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/265ce3de7ba2afd3b9cc2260d7d51fb197efbf02"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-17 11:35:10.000000000","tz":0},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab11e3c7ab4c83a928d06bd44fa255c0d5e09bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab11e3c7ab4c83a928d06bd44fa255c0d5e09bad"}]},"branch":"refs/heads/master","description":"Rebase"},"5e91928c5b98cde43fc54af76c4f2b9c3d0b1594":{"kind":"REWORK","_number":9,"created":"2023-02-17 15:07:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/9"}}},"commit":{"parents":[{"commit":"265ce3de7ba2afd3b9cc2260d7d51fb197efbf02","subject":"Merge \"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/265ce3de7ba2afd3b9cc2260d7d51fb197efbf02"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-15 23:26:35.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e91928c5b98cde43fc54af76c4f2b9c3d0b1594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e91928c5b98cde43fc54af76c4f2b9c3d0b1594"}]},"branch":"refs/heads/master"},"c79e3800f4ed2f5f0a53421a0b19b1af91ea7747":{"kind":"REWORK","_number":10,"created":"2023-02-20 14:23:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/10"}}},"commit":{"parents":[{"commit":"2e1616ca97e16df8f29f717e2685fe46a3d6a9b0","subject":"Merge \"Move tox-py311 job override into job-templates.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e1616ca97e16df8f29f717e2685fe46a3d6a9b0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-16 07:09:49.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c79e3800f4ed2f5f0a53421a0b19b1af91ea7747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c79e3800f4ed2f5f0a53421a0b19b1af91ea7747"}]},"branch":"refs/heads/master"},"43e8b83ee4c94bfb656bcbc8a55415acdf7e0c90":{"kind":"REWORK","_number":11,"created":"2023-02-20 18:11:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/11"}}},"commit":{"parents":[{"commit":"2e1616ca97e16df8f29f717e2685fe46a3d6a9b0","subject":"Merge \"Move tox-py311 job override into job-templates.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e1616ca97e16df8f29f717e2685fe46a3d6a9b0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-16 10:58:13.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43e8b83ee4c94bfb656bcbc8a55415acdf7e0c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43e8b83ee4c94bfb656bcbc8a55415acdf7e0c90"}]},"branch":"refs/heads/master"},"0bde2fe45f68c3a7d74b562d89e07d7a12f2cc53":{"kind":"REWORK","_number":12,"created":"2023-02-21 14:50:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/12"}}},"commit":{"parents":[{"commit":"c178c28fb8f21c1ea145f696cf724aa9038012b2","subject":"Merge \"[OVN] Bump the port revision number in trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c178c28fb8f21c1ea145f696cf724aa9038012b2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-17 06:46:26.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bde2fe45f68c3a7d74b562d89e07d7a12f2cc53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bde2fe45f68c3a7d74b562d89e07d7a12f2cc53"}]},"branch":"refs/heads/master"},"b834b2ee17fdada085a78586696649f18adc9cbf":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-03-01 15:15:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/13"}}},"commit":{"parents":[{"commit":"fb08c34561bebf537a728c8043f643b0fffad771","subject":"Merge \"Check port.tag is not DEAD_VLAN_TAG in ``DHCPAgentOVSTestFramework``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb08c34561bebf537a728c8043f643b0fffad771"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 10:34:45.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b834b2ee17fdada085a78586696649f18adc9cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b834b2ee17fdada085a78586696649f18adc9cbf"}]},"branch":"refs/heads/master"},"33ec32063692c6628e4a39ce9e77c5531bceb1a9":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-03-24 14:18:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/14"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-23 08:16:25.000000000","tz":60},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33ec32063692c6628e4a39ce9e77c5531bceb1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33ec32063692c6628e4a39ce9e77c5531bceb1a9"}]},"branch":"refs/heads/master"},"64cddc8b00361b2155797e2880ca198211f43b5e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-04-19 07:44:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/15"}}},"commit":{"parents":[{"commit":"18bc5b82a1d313a27411a9a88488f1cc306f0fee","subject":"Merge \"OVN: Always try and create a metadata port on subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18bc5b82a1d313a27411a9a88488f1cc306f0fee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-04-19 07:44:47.000000000","tz":0},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64cddc8b00361b2155797e2880ca198211f43b5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64cddc8b00361b2155797e2880ca198211f43b5e"}]},"branch":"refs/heads/master","description":"Rebase"},"a3226351269dc47f527e8637c9a01e7f22bf76a0":{"kind":"REWORK","_number":16,"created":"2023-05-09 09:51:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/16"}}},"commit":{"parents":[{"commit":"9319ba00a986e693eeb4707055aa53111601e7c8","subject":"Merge \"rbacs: clean-up to use defined constants ACCESS_*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9319ba00a986e693eeb4707055aa53111601e7c8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 09:51:07.000000000","tz":120},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3226351269dc47f527e8637c9a01e7f22bf76a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3226351269dc47f527e8637c9a01e7f22bf76a0"}]},"branch":"refs/heads/master"},"627b9493ca6ccf61900eedd6e12d0415be35bf87":{"kind":"REWORK","_number":17,"created":"2023-05-09 10:54:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/17"}}},"commit":{"parents":[{"commit":"9319ba00a986e693eeb4707055aa53111601e7c8","subject":"Merge \"rbacs: clean-up to use defined constants ACCESS_*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9319ba00a986e693eeb4707055aa53111601e7c8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 10:54:37.000000000","tz":120},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/627b9493ca6ccf61900eedd6e12d0415be35bf87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/627b9493ca6ccf61900eedd6e12d0415be35bf87"}]},"branch":"refs/heads/master"},"96b2518f9b6d345d49ba73579ffa62efb1ba87a7":{"kind":"REWORK","_number":18,"created":"2023-05-24 08:39:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/18"}}},"commit":{"parents":[{"commit":"0360eb8e127a9f162a987fe2fc8cfefdafdb9c32","subject":"Merge \"Revert \"Delete sg rule which remote is the deleted sg\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0360eb8e127a9f162a987fe2fc8cfefdafdb9c32"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-23 19:16:23.000000000","tz":120},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96b2518f9b6d345d49ba73579ffa62efb1ba87a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96b2518f9b6d345d49ba73579ffa62efb1ba87a7"}]},"branch":"refs/heads/master"},"dcffdbf927f03ca841baebb0b74a63fe19a546d7":{"kind":"REWORK","_number":19,"created":"2023-05-24 08:56:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/19"}}},"commit":{"parents":[{"commit":"0360eb8e127a9f162a987fe2fc8cfefdafdb9c32","subject":"Merge \"Revert \"Delete sg rule which remote is the deleted sg\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0360eb8e127a9f162a987fe2fc8cfefdafdb9c32"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-23 19:33:33.000000000","tz":120},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcffdbf927f03ca841baebb0b74a63fe19a546d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcffdbf927f03ca841baebb0b74a63fe19a546d7"}]},"branch":"refs/heads/master"},"ac231c817473c018dde8fa31594b1c9a78a36c13":{"kind":"REWORK","_number":20,"created":"2023-05-24 14:37:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/23/872023/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/872023/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/872023/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/872023/20"}}},"commit":{"parents":[{"commit":"0360eb8e127a9f162a987fe2fc8cfefdafdb9c32","subject":"Merge \"Revert \"Delete sg rule which remote is the deleted sg\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0360eb8e127a9f162a987fe2fc8cfefdafdb9c32"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-23 15:17:46.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-05-24 14:37:30.000000000","tz":0},"subject":"Improve \"sync_ha_chassis_group\" method","message":"Improve \"sync_ha_chassis_group\" method\n\nThe method \"sync_ha_chassis_group\" now creates (or retrieves) a\nHA Chassis Group register and updates the needed HA Chassis registers\nin a single transaction. That is possible using the new ovsdbapp\nrelease 2.2.1 (check the depends-on patch).\n\nDepends-On: https://review.opendev.org/c/openstack/ovsdbapp/+/871836\n\nRelated-Bug: #1995078\nChange-Id: I936855214c635de0e89d5d13a86562f5b282633c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac231c817473c018dde8fa31594b1c9a78a36c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac231c817473c018dde8fa31594b1c9a78a36c13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
