)]}'
{"id":"openstack%2Fneutron~872099","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I9d79b123bb20ded327208d84a14d4f8d2e505087","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1956958","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-02 07:16:19.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-02-02 05:22:21.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-02 07:16:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d79b123bb20ded327208d84a14d4f8d2e505087","subject":"Use common wait_until_ha_router_has_state method everywhere","status":"MERGED","created":"2023-01-30 05:58:02.000000000","updated":"2023-02-02 07:18:11.000000000","submitted":"2023-02-02 07:16:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872099-bug/1956958","meta_rev_id":"2ba6d0115022f3baa1637cad6fd9ccbb922f036a","_number":872099,"virtual_id_number":872099,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 07:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-01-31 14:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-02-01 14:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-02-01 14:10:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 05:58:02.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-01-30 08:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 14:11:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-02-01 06:07:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-01 06:07:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-01 14:10:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"434ab546bc44dfc150de67aad81c6a37d12f9127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-30 05:58:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4ecfbbe449fd845804ee3ea0ae6be5480915691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 08:21:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5657f7a3d1b441b905d71003dd1f57a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1eb5ee1dd744e4c93eb228279657d8e : SUCCESS in 13m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48d1e1e8940f4b18844277f03a97bd94 : TIMED_OUT in 40m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bf129bc67e24f5285d03557662b51f5 : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/735a268807af4b41aac47d449dafd1ff : TIMED_OUT in 2h 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b3125210b883486700c0c0e3b2308488d63cf9a9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-30 11:34:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f255fd4debed20d1c8a376a3b8930c29ec71b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 12:43:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a6cec0b3e94b16ba7a46fc6ad6b8f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8466776d25bd4582a5ca5ec9f5369eb4 : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e557d436da94183bb194ac53a3a8bbc : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9122700917804543bdc4d0e0112f6b60 : SUCCESS in 10m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4d15f4dc5c649b383950397dd9a6c20 : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"12cdc21a6c8c61f127402ab836ba49faa4e0fd4a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-01-31 14:11:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d58ff6683e4d211787d5b472958dc963dd18837","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-01 14:10:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3940e1ee71041586ab59b2329f7ad8d618e363d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 14:11:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17e9dbca0afba37b06be95b5cf5ccdbd5b50eb59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:43:19.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/872098, https://review.opendev.org/872007 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"927e73639673f1c2c8951a0266d716b71cb809b5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-02 05:22:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ed22ca0ab5371ad28d417f2607399a53c8f9744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 06:00:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/420ec34ec89847ae9055de269f756463\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d9a35cbd444c40a06a6d706b0b6593 : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0680fe4b388b4433a6edd5e6d8162fae : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408b24f512194191a51aacab3494f8a8 : SUCCESS in 8m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fe090b16f654a77bf7f9bb90d0cc2ac : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"39355ca09e2d191d190b835a29e250de2464c7a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 06:01:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4205be65973abeeccc8179d4979aefbb2dcc2f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 07:16:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bf18c703c7c43519e33225a990d3ddf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4fd8ad8cd449b1a2fa86c13e104383 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c06dcdf86b54ac49e7f759c54decaff : SUCCESS in 7m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c90fe7a950a45d9ba48e1259c5a6853 : SUCCESS in 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ce5d147a89a1514597141224cb693803be176d0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 07:16:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ba6d0115022f3baa1637cad6fd9ccbb922f036a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 07:18:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94fb9549f4e847bb85e01e329f0632e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b394d31524184a29b50a8cdede165654 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4213c8da94a6acd32eb492c98d8c145c44f0bf8f","revisions":{"4213c8da94a6acd32eb492c98d8c145c44f0bf8f":{"kind":"REWORK","_number":1,"created":"2023-01-30 05:58:02.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/872099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/872099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/872099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/872099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/872099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/872099/1"}}},"commit":{"parents":[{"commit":"858caf578ad13f359f46c29d6b48e42b2ddaab79","subject":"Mark functional L3ha tests as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/858caf578ad13f359f46c29d6b48e42b2ddaab79"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-04 09:09:23.000000000","tz":120},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-01-30 05:57:36.000000000","tz":330},"subject":"Use common wait_until_ha_router_has_state method everywhere","message":"Use common wait_until_ha_router_has_state method everywhere\n\nIn the L3 functional tests framework module there is already helper\nmethod called wait_until_ha_router_has_state which should be used to\nwait for desired HA router\u0027s state.\nThis method has proper debug logging added so debugging issues in CI is\neasier when it\u0027s used.\nIt is also decorated with unstable_test decorator to skip tests when\nrouter will fail to transition to desired state (see related bug for\ndetails).\n\nIn some tests this method wasn\u0027t used so we couldn\u0027t benefit from the\nlogging and unstable_test decorator there. Now it should be unifed and\nused everywhere in the same way.\n\nRelated-Bug: #1956958\nChange-Id: I9d79b123bb20ded327208d84a14d4f8d2e505087\n(cherry picked from commit e39011c73396af68fe5002cdbe4c6b3fe8b7cf23)\n(cherry picked from commit fc62d1ea8ef5024cd1b851dab2f765ba425279c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4213c8da94a6acd32eb492c98d8c145c44f0bf8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4213c8da94a6acd32eb492c98d8c145c44f0bf8f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
