)]}'
{"id":"openstack%2Fneutron~872219","triplet_id":"openstack%2Fneutron~master~Ibf09254b5b2fee6efd3de5e5dc6f013424831db9","project":"openstack/neutron","branch":"master","topic":"fix-gw-subnet-error-messaging","attention_set":{},"removed_from_attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2023-02-01 21:44:00.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-01-31 15:52:19.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ibf09254b5b2fee6efd3de5e5dc6f013424831db9","subject":"Improve message for subnet gateway out of host IP addresses range","status":"MERGED","created":"2023-01-30 21:29:08.000000000","updated":"2023-02-01 21:45:11.000000000","submitted":"2023-02-01 21:44:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872219-fix-gw-subnet-error-messaging","meta_rev_id":"e2b7a7b18d2bb7c43f0d4c7d8fb8f26f2d40d63e","_number":872219,"virtual_id_number":872219,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-01-30 22:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 21:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-01 14:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-01-31 15:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 06:52:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-01 14:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 21:40:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-01-30 22:06:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 22:32:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-30 23:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 06:52:08.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-02-01 14:12:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"bf94b862d738afbb3df8379f14c76f6e3aa24e25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-01-30 21:29:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"955a79acd23527516be28c6bd017b1f1feb73419","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-30 21:40:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c20d3e528d393475c1e237cb2755c7f73dff87c1","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-01-30 21:43:37.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d0146d80563af28ebea80288d0330fff7eec41a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-01-30 21:43:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e1fed4010ab727b78eb39ccb3c6ee3bffef031d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-30 22:06:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/19/872219/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a91102d5b51f788040611a622deb3d4237f5f619","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 22:32:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26e21896e6841daa26a35bb0136099d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d2c96eea7d5452d8ad187309df37ef4 : SUCCESS in 24m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0b126eea29647c9acb778c4531e59da : SUCCESS in 46m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b1063183be688a65fbba14cfff61eddbcaec43d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-01-30 22:46:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2023/Arista_Tempest_Testrefs/changes/19/872219/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"260ca039170bb6e740b46716ff54ec86e156c468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 23:50:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d12e70ee17d4cc29abda972cac3c9e8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/77db7b346e9141e8b828af12c72eb684 : SUCCESS in 1h 15m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/211d4971db394fef975b1d36283e7626 : SUCCESS in 1h 22m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2487ce503a524fbaa9095910873376e1 : SUCCESS in 1h 00m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6df9494a49c4d2583d7e5991408ee7d : SUCCESS in 1h 12m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31ea04e3745943f9860edb74f04f5a93 : SUCCESS in 26m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d49c7af5b634afb8ed3eb87d35621a1 : FAILURE in 49m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc440a1187f45ffbca6803a06f71cf3 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b690a525647542b097ad5748de853077 : SUCCESS in 27m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16194fe1368e412492aee621e7c72c9b : SUCCESS in 26m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31988c9fbca24b43a2fa83a9051f2784 : SUCCESS in 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48d5d89333d04b49b4c5f351325ad51c : SUCCESS in 41m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/19c281a1754e4de3b0943d8f0e0c984a : SUCCESS in 1h 25m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b965ab4ce4d44048bb54213b86c5afe0 : SUCCESS in 47m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/be6e0b968302495591b5d1ed023bb3e2 : SUCCESS in 48m 10s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d04c7125d5d34ae19881d8620253418a : SUCCESS in 1h 03m 27s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/43fd3e0bb44c4d12b221ef208cfe97d9 : SUCCESS in 48m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/347b22830616465d81ed4ee01eb92cc4 : SUCCESS in 1h 12m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/443ddf7482024a66a806b2bf78676a14 : SUCCESS in 1h 04m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/930bce903ee6462e9197d66139258526 : SUCCESS in 1h 02m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/98017356d16f46d4834e043d2584727b : SUCCESS in 1h 58m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/26ccdd6add6142f2981d54cb974c1351 : SUCCESS in 1h 29m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac143e77064345fda66a07dbd2e301af : SUCCESS in 1h 18m 14s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3b8e008f59ec468f9eb69b0ca761d5d6 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1a81ea7dd9d2069fb387e978bc8cb4c57cd14d33","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-01-31 01:22:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34d1333e3a6e2f03d4f33634ac805aea306bbff2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 01:57:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49bb4a7c630347aab2b24b584ada0013\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28aae51c683147d58e096b8e91fbc78b : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01705747542b44c5a8504a44614c81b4 : SUCCESS in 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c3a06f8ef5bfaf66e11248d8d29949f52c7f6f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 03:18:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0b7a62e8bfd48c789f42082602e0d85\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c58ed38fcb97422d9dd6f9d2973b98c7 : SUCCESS in 1h 27m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/05bcdbce36fb4f6f9e4196124637a2b4 : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1dbe36d9d8b14b1cbeb811253c1dbf7f : SUCCESS in 1h 25m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3a2e08619797471e9233174747682814 : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c020077041a45cfa50b67cc2bb909fe : SUCCESS in 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0dab01179604dbcbdb1ae768bf96f13 : SUCCESS in 35m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861fe54b85ba4fc7aa4d61151012e748 : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ee5d1f352d7492fbb74b3b3008c18be : SUCCESS in 25m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d9a35a722fe491ab876d15d4252edba : SUCCESS in 24m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9161f1d910b4471b9e103041068744b8 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0805cdc4765742398fc565d27eec26ba : SUCCESS in 53m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ebfb5a25ef14e79b36694a7a9903f8a : SUCCESS in 1h 49m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6902476d6bc54fca8a45fde3dd68f217 : SUCCESS in 1h 03m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1312eedf05d04a61a01ea57087630b44 : SUCCESS in 1h 01m 17s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/01680dbbf16049b290477172fec252cc : SUCCESS in 1h 16m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae06e726e5784814b49866df9271c8c7 : SUCCESS in 49m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3142eb0cc7fe4965badc4a1de912243f : SUCCESS in 1h 20m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/30351712624140689458a8eafba9eddb : SUCCESS in 1h 10m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfc0c99116804fd299635e877f251ee8 : SUCCESS in 1h 03m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e7b11f62745c4ca3b46565bb96afcc63 : SUCCESS in 1h 32m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/85cb73ef1f6a4c029ffcadfcc336cc7b : SUCCESS in 1h 25m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a41409ff12e84ed58c56f367d8f0a013 : SUCCESS in 1h 17m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/852975607d844671a40aa30399348672 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":2},{"id":"88104651c1f7c0797e256f18e8e39f43e77b13aa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-01-31 06:52:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"795e95d1871b72d2f2c6f51145d3e2aa207a19a7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-31 15:52:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c2f2e47fe7775367e69416e63ceb64148182ba8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-01 14:12:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b394683ff38bbf40e28df9878253c1a122569644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 14:13:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"05e57a5a5c0abfed6ba19f7480a50441660b2beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:44:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6118acf1d958484bb2dd31c7640cf8e4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/25f26d7878d0495f973ef193dd465a87 : SUCCESS in 1h 48m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c1198a3aa084a7b9aeaba370379edee : SUCCESS in 1h 36m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/60c6e3d70b7240ac8f5ea5157a69b20d : SUCCESS in 1h 35m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb84a753e0b4452b9e307cd2904da230 : SUCCESS in 1h 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d17376da19344667b59ce71b2c98a83f : SUCCESS in 12m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02f6da3bc71e45c49928c98bf94b85bf : SUCCESS in 28m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf44b88d850e4feea0e9e86473d501f8 : SUCCESS in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977274a9ecac412fbaaf6f718cceba93 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7aad2773717540faa9361b6c60edde36 : SUCCESS in 41m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/423c2a466827491e9ea92636ceb8fc89 : SUCCESS in 1h 27m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f5678a08bc640b78a2866a43a761a83 : SUCCESS in 1h 39m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/99085533f0984625a6db4d2263bf63a9 : SUCCESS in 48m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/61682bdb11c24c0cb6fef3499a1eb19c : SUCCESS in 1h 02m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/62bfef815e6949238c4ac3e458b8393a : SUCCESS in 1h 04m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3636f673872c4ee9928174408cc9747e : SUCCESS in 1h 54m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d4b5091862214e8ca9c7de57d1c61528 : SUCCESS in 1h 36m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6f5a3fdfd1b38cec5c74e2ae3d7f478cc510beb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:44:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e2b7a7b18d2bb7c43f0d4c7d8fb8f26f2d40d63e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:45:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b898465030447185637e85e55afa0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be0676c1a214c258d027691a7df89b6 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83da1e6d79c4d971718d31cbd3aa038353adbae8","revisions":{"2e36113c78abc94ad7b1bb7a98ce86a4c0d5d96d":{"kind":"REWORK","_number":1,"created":"2023-01-30 21:29:08.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/19/872219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/872219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/872219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/872219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/872219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/872219/1"}}},"commit":{"parents":[{"commit":"e6eeb700c88499281d3924de6d8fd009d5dc09b7","subject":"Merge \"Honor debug mode in keepalived-state-change script logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6eeb700c88499281d3924de6d8fd009d5dc09b7"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-01-30 19:10:26.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-01-30 21:28:48.000000000","tz":-180},"subject":"Improve message for subnet gateway out of host IP addresses range","message":"Improve message for subnet gateway out of host IP addresses range\n\nFollowing the discussions on [1], we notice that there is an error\nmessage in the Neutron code that is misleading.\n\nThis patch intends to improve the exception message, and provide a\nlittle bit of documentation regarding the gateway IP address\nvalidation for subnets.\n\n[1] https://bugs.launchpad.net/neutron/+bug/2004004\n\nChange-Id: Ibf09254b5b2fee6efd3de5e5dc6f013424831db9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e36113c78abc94ad7b1bb7a98ce86a4c0d5d96d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e36113c78abc94ad7b1bb7a98ce86a4c0d5d96d"}]},"branch":"refs/heads/master"},"83da1e6d79c4d971718d31cbd3aa038353adbae8":{"kind":"REWORK","_number":2,"created":"2023-01-30 21:43:41.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/19/872219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/872219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/872219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/872219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/872219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/872219/2"}}},"commit":{"parents":[{"commit":"e6eeb700c88499281d3924de6d8fd009d5dc09b7","subject":"Merge \"Honor debug mode in keepalived-state-change script logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6eeb700c88499281d3924de6d8fd009d5dc09b7"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-01-30 19:10:26.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-01-30 21:43:24.000000000","tz":-180},"subject":"Improve message for subnet gateway out of host IP addresses range","message":"Improve message for subnet gateway out of host IP addresses range\n\nFollowing the discussions on #2004004, we notice that there is an error\nmessage in the Neutron code that is misleading.\n\nThis patch intends to improve the exception message, and provide a\nlittle bit of documentation regarding the gateway IP address\nvalidation for subnets.\n\nRelated-bug: #2004004\n\nChange-Id: Ibf09254b5b2fee6efd3de5e5dc6f013424831db9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83da1e6d79c4d971718d31cbd3aa038353adbae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83da1e6d79c4d971718d31cbd3aa038353adbae8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
