)]}'
{"id":"openstack%2Fneutron~872273","triplet_id":"openstack%2Fneutron~master~I1b813c5b3248ab5d3e321f1b31dd095d82e2feca","project":"openstack/neutron","branch":"master","topic":"bug/2004265","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-03 17:38:05.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-03 18:31:16.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-02-02 19:46:35.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I1b813c5b3248ab5d3e321f1b31dd095d82e2feca","subject":"Add CI jobs using SQLAlchemy master branch","status":"MERGED","created":"2023-01-31 11:16:45.000000000","updated":"2023-02-03 18:33:22.000000000","submitted":"2023-02-03 18:31:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"872273-bug/2004265","meta_rev_id":"754c6b629d66323ba8da79fde54cb21025d052cb","_number":872273,"virtual_id_number":872273,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-03 18:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-02-03 17:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-02-02 19:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2023-02-03 17:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 11:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-31 16:23:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-02-02 04:50:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-02-02 09:37:53.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-02-02 11:36:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-02-02 17:33:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-02-03 17:38:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"},{"updated":"2023-02-03 17:44:16.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"49c499916921f60b4d56b2c8e32928cb00845e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-31 11:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9af6f51e24f5cf1189b087a9b4c454160c163bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 11:17:14.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":1},{"id":"bdecac2c69e9c3d151cf52ffb5a1f4a23a821b8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 11:17:19.000000000","message":"Patch Set 1:\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ec20bbff796dff201271e93efee10ef4161e8ace","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-01-31 11:18:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"285e9f42fe3e05726ebbe1f477ee6247631dd040","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 11:19:22.000000000","message":"Patch Set 1:\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":1},{"id":"cbb56ede0efc8475b03b59e241ee93607e652b81","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-01-31 16:23:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5e7ee72150fb1b373edf4a19f641c27a26d3ab50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-01 16:31:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"67fd1f2e73fbac00da7be2cc3a7ef3e51d0f00a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:32:26.000000000","message":"Patch Set 2:\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":2},{"id":"999ba12ccdd77e26a848e3646e86f195adc8879b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:32:34.000000000","message":"Patch Set 2: Verified-1\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":2},{"id":"6eadda64ce9e7127c88b5626e7e665c98e11ee33","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-01 16:32:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ac8834ca6b865f7ea906c899e8d98f6781085aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-01 16:43:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b80202d2047b2217f494861d3b377bf1c2704cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:43:56.000000000","message":"Patch Set 2:\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f0e339042959ecfc6fa329d733329ef97f5f9c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:44:12.000000000","message":"Patch Set 2:\n\n(13 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Unknown projects: github.com/sqlalchemy/sqlalchemy\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-functional-with-sqlalchemy-master\n      parent: neutron-functional\n      required-projects:\n        - name: github.com/sqlalchemy/sqlalchemy\n\n  in \"openstack/neutron/zuul.d/base.yaml@master\", line 209, column 3","accounts_in_message":[],"_revision_number":2},{"id":"3f665810211869a78f1aebe1d0c7760e7144045e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-02 03:24:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5d2c80f72b81ab06a053efe655d37a1d430a4ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 04:01:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2299c7228f134e96b9539f449847a3e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/758ccc2263b6416592c720d13d7f0aff : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edcc82e07b894c4f9ebde7e7ec25deba : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f282e364fda643728ecc8c236bf5beb3 : SUCCESS in 25m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc4e3ceb387b4f2eafaed64ee437ea6f : SUCCESS in 25m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4541836b7f484bada1c777f447981f93 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":2},{"id":"64135a46deae19ced55ef8a9fd7a145b5d732dd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 04:19:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a94189e01d406f8faef0647b6c47db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abc8cc6147d3452ab1520c651af3a3c2 : SUCCESS in 52m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01035a4377ae46b591e29b7b9252329e : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ea44e89c611978ca06df0bacabe1ef8a2c2e0b6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-02 04:50:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"39dd76e5bb5ecf2cdf77cdc01d9b42cc2552b376","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-02 09:37:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61d0af5c4ba75ce8319de0ea1453569609ef06ec","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-02 11:36:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2db29c2a46a59c695f762cf2cb4b2700e2d4602a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 12:29:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f4005b852339e092939f6a36bb45c971c2ae245d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 12:30:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f6bb27d6e1e43f86db872f05e259e0274b409fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 13:17:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d3d32dcbc14f6aa34c161c3d336e50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7cbd1df84ec41f087d91c573197dd8b : SUCCESS in 42m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6aba1a6bb44183b63d5b7b1c64884f : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7237ffc47a74ab7918374732ba65807 : SUCCESS in 33m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/469bcc5aa60443fbb44510abd1e6301c : SUCCESS in 31m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d5f907c1af4945b5fe733babcf60c9 : SUCCESS in 14m 02s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/aa4389bcaa40457583396992ae50dc07 : SUCCESS in 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8feb095867bd7f58ced86c75244d739289061301","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 13:53:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a640b9047c4ee6a9eed0aa20f69e59\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ae4ae06aaa548c3bd3f51cb7dee3437 : FAILURE in 1h 20m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c94db544e44d4628b100330b18728bc1 : FAILURE in 1h 21m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb6a52a606303fa3c070c3d4ebeede3786f4bbbf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 16:27:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51e3577ad56201ad7d5da02b4b0afcdc725471a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 17:33:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8fe2e53865abe05983a2da8ecb1be8ef07316259","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 17:33:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8acc222b0d57faa1db353b0d9ef6f0d51cf3409d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 19:01:44.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d3fa191c8c4fc6bba717d50b343e34\n\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/95b136b122f34d1f8a28daa400d5e500 : FAILURE in 14m 32s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/07d5f8ab5f0c4918adb97d49caaa134d : SUCCESS in 1h 04m 33s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/626d0f870b20433c90b440a5dc0c1833 : SUCCESS in 58m 40s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ec579b99e05a4d068f8b508d3b04b1df : SUCCESS in 54m 17s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/97cb275aa90b4813afcb4e2001afb62c : SUCCESS in 41m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8979a5c2af4e49acb3e53f8e0f01be3a : SUCCESS in 1h 00m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/feae63f3b3b4418595ef38840b134090 : SUCCESS in 1h 24m 04s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/864a1a81aaed468dbc03fafbd125d2c4 : FAILURE in 51m 46s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/bb9e47d486a8494097ab61bb49d59897 : SUCCESS in 26m 22s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/6325f7e617ab4089af915e790a7c6fb5 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":4},{"id":"b60b0299a60925cd193df09ea1dbb9d070b5d5ec","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-02 19:12:22.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4db4a080a162cd7fc43d59a1f0328f2f06ed6bf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 19:12:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32b63253df42435596c6e76f7581e789\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac1679c457974148bf40ee1471dc1e52 : SUCCESS in 35m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b98e1ee9d92b41e4b0ac3ba12e075ab9 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f59a2ab7c37f457c95edb8a060fba8aa : SUCCESS in 28m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d1aecbd18124ac5b95388ea6053087b : SUCCESS in 26m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e186a70d6243efb957e0a333d2f6a1 : SUCCESS in 9m 43s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e5aa5f17b5c147809aa906923fa2d53c : SUCCESS in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6bd45e237b7eb92e9dbbe4ca6998f1d4ba2ebccc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-02 19:37:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"743cc411a8fe43701ab1462926633912dd607bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 19:44:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83446cf5e9e495a9a6a4946bfaa5daa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a761d2b6b7f4459fbf2dbfd4d098a4c7 : SUCCESS in 53m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/013434af2df54113b81e9c4c4cd87e7f : SUCCESS in 48m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aa2511d210fce73a4db07557dce656501bf68c7c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-02 19:46:35.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78d19b08c4e90cd6e6316ade40b3787cd618fc1a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-02-03 17:44:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5226fce1feb386176c77554ca9c5bfd54d2f8c72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 17:44:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cdd647e4d7cc0a0a3ed70ba8cf09bf1325a31220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 18:31:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b8d8eb30b78477089723794d1522a93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113b763d2aba403793414b2ef221bad9 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf78716914d74b429fdc54aa926fa607 : SUCCESS in 23m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62c19b554ce64e3ca1740a4d6f6b9efe : SUCCESS in 22m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeecea92cd084ee19452a57253e7e643 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ea8fb1abbe9e50f7c5421e2d5179f42ab918578a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 18:31:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"754c6b629d66323ba8da79fde54cb21025d052cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 18:33:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8532d5ad0e4d2a858a27147ae63376\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20f2517c5cf4531953206af37dcf984 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3c2af924d972b175ddd7f030822d2e37b2b4085c","revisions":{"306f7391bd01d41f705c80a2358028ef0765f4c0":{"kind":"REWORK","_number":1,"created":"2023-01-31 11:16:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/872273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/872273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/872273/1"}}},"commit":{"parents":[{"commit":"6a6cee932b2f11a771693148aebc12d447e4dce7","subject":"Propose replacement of ORM from_self()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a6cee932b2f11a771693148aebc12d447e4dce7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-25 03:16:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-25 03:20:30.000000000","tz":60},"subject":"Add CI jobs using SQLAlchemy master branch","message":"Add CI jobs using SQLAlchemy master branch\n\nThis patch add the following CI job definitions:\n* neutron-ovn-tempest-with-sqlalchemy-master\n* neutron-ovs-tempest-with-sqlalchemy-master\n* neutron-functional-with-sqlalchemy-master\n\nCloses-Bug: #2004265\nChange-Id: I1b813c5b3248ab5d3e321f1b31dd095d82e2feca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/306f7391bd01d41f705c80a2358028ef0765f4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/306f7391bd01d41f705c80a2358028ef0765f4c0"}]},"branch":"refs/heads/master"},"945c4591e7490e9507c07c936dafad106edb4125":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-01 16:31:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/872273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/872273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/872273/2"}}},"commit":{"parents":[{"commit":"e98a268de4316c4f889925a3da6266e1c70306aa","subject":"Propose replacement of ORM from_self()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e98a268de4316c4f889925a3da6266e1c70306aa"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-25 03:16:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-01 16:31:55.000000000","tz":0},"subject":"Add CI jobs using SQLAlchemy master branch","message":"Add CI jobs using SQLAlchemy master branch\n\nThis patch add the following CI job definitions:\n* neutron-ovn-tempest-with-sqlalchemy-master\n* neutron-ovs-tempest-with-sqlalchemy-master\n* neutron-functional-with-sqlalchemy-master\n\nCloses-Bug: #2004265\nChange-Id: I1b813c5b3248ab5d3e321f1b31dd095d82e2feca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/945c4591e7490e9507c07c936dafad106edb4125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/945c4591e7490e9507c07c936dafad106edb4125"}]},"branch":"refs/heads/master","description":"Rebase"},"fbdf2b3a24b1c2f82a6941b551dace355c7bb366":{"kind":"REWORK","_number":3,"created":"2023-02-02 12:30:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/872273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/872273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/872273/3"}}},"commit":{"parents":[{"commit":"e98a268de4316c4f889925a3da6266e1c70306aa","subject":"Propose replacement of ORM from_self()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e98a268de4316c4f889925a3da6266e1c70306aa"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-25 03:16:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-02-02 12:30:00.000000000","tz":0},"subject":"Add CI jobs using SQLAlchemy master branch","message":"Add CI jobs using SQLAlchemy master branch\n\nThis patch add the following CI job definitions:\n* neutron-ovn-tempest-with-sqlalchemy-master\n* neutron-ovs-tempest-with-sqlalchemy-master\n* neutron-functional-with-sqlalchemy-master\n\nCloses-Bug: #2004265\nChange-Id: I1b813c5b3248ab5d3e321f1b31dd095d82e2feca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbdf2b3a24b1c2f82a6941b551dace355c7bb366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbdf2b3a24b1c2f82a6941b551dace355c7bb366"}]},"branch":"refs/heads/master"},"3c2af924d972b175ddd7f030822d2e37b2b4085c":{"kind":"REWORK","_number":4,"created":"2023-02-02 17:33:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/872273/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/872273/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/872273/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/872273/4"}}},"commit":{"parents":[{"commit":"8234250e011faee880e3cf13c98ed6e3b5d093f8","subject":"Merge \"[OVN] New OVN Neutron Agent extension: QoS for HWOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8234250e011faee880e3cf13c98ed6e3b5d093f8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-01-25 03:16:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-02 18:32:03.000000000","tz":60},"subject":"Add CI jobs using SQLAlchemy master branch","message":"Add CI jobs using SQLAlchemy master branch\n\nThis patch add the following CI job definitions:\n* neutron-ovn-tempest-with-sqlalchemy-master\n* neutron-ovs-tempest-with-sqlalchemy-master\n* neutron-functional-with-sqlalchemy-master\n\nCloses-Bug: #2004265\nChange-Id: I1b813c5b3248ab5d3e321f1b31dd095d82e2feca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c2af924d972b175ddd7f030822d2e37b2b4085c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c2af924d972b175ddd7f030822d2e37b2b4085c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
