)]}'
{"id":"openstack%2Fneutron~872745","triplet_id":"openstack%2Fneutron~master~I715029c311c4516f3212054c5c72533b12fd0986","project":"openstack/neutron","branch":"master","topic":"bug/2004581","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-07 12:48:53.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-02-06 21:57:38.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I715029c311c4516f3212054c5c72533b12fd0986","subject":"Filter subnets by \"enable_dhcp\" flag using the correct type","status":"MERGED","created":"2023-02-06 10:18:27.000000000","updated":"2023-02-07 12:50:14.000000000","submitted":"2023-02-07 12:48:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872745-bug/2004581","meta_rev_id":"39e07a1b1d6dd3995e9d4cd42195fdc48ebfcf48","_number":872745,"virtual_id_number":872745,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 12:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-02-07 10:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-02-06 21:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2023-02-07 02:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-02-07 10:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-06 10:42:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-06 12:28:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-06 13:28:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-02-07 02:28:51.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"REVIEWER"},{"updated":"2023-02-07 04:31:41.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-07 10:57:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"5c6b42b846654a87074e3d114dda5bc5d28c5329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-06 10:18:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d98e3487f5a751d559b1145a1b0ca3002124ec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 10:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd195c92c324c23a3e29c39a459961f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a32c34192f35441c886f2d1200336eea : SUCCESS in 22m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59f77540e61142b4a8dc56a630c671d5 : SUCCESS in 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6c9a112a424838f301310bd1bce7e7d52901ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 12:28:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/026e8245fdb84e109508048aa93f6d2c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/02b8b0f30f03473cb0b3ee011bdedd01 : FAILURE in 41m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22cec19982694aeb9b9a0f1d5991f169 : SUCCESS in 35m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ff57e8906b425b83cde561d8180dcb : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/826719371a954646a1e23a56da3b81a5 : SUCCESS in 23m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9c3d5f8eacc44d18a45055a6128bba5 : SUCCESS in 27m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3038387ca7a243bbb5a259ae50a83baf : SUCCESS in 14m 36s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/dc281baba3524698a596889cde79cf23 : SUCCESS in 27m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f007836e39b49f6b96f4d9cb4c72dda : SUCCESS in 56m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b0efb630a3ec4b7f8a0746749d39c642 : SUCCESS in 53m 30s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f52e7c4fa95947cf9da51d920e9c7963 : SUCCESS in 57m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3679b5bdab5420ebf4c6cdb76c3dd8b : SUCCESS in 1h 53m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8ceae1f162f44c68d1e6a75eb2f442d : SUCCESS in 1h 30m 26s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e365bc4324b54d6397387f8efadd7e57 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8b9fd84483d8865645916e993798dbb98d8c72a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-06 17:19:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a82e58e200fcd8cdcd0b6a7bcdf477f0dce23d39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 17:41:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/446b2cd7e42043df9fdb383e89931c7b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f24e5e02627142faaafa6ed08b66cd71 : SUCCESS in 20m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78b9b9c72f6742bba2cca68206949dc8 : SUCCESS in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6da35b1e678eee90672b25afff484f80bc18d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 19:19:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876f50a2b3a54b34bd009587e53d4b0b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/752fa4f1c2a44844a48916707be8f9c6 : SUCCESS in 1h 51m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36bd632cdd24476b8b52f269ed4ca8a : SUCCESS in 34m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39d4e9dd0a04217ad4405a3efb95e29 : SUCCESS in 24m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6a0f82863b241f6a2bfd76eab341c6c : SUCCESS in 50m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f05954e3c264cd69bf345ac78e5b532 : SUCCESS in 28m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a93c8ad8f5e49e68079682e7c1e5d73 : SUCCESS in 9m 36s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3550a694ae3541c0acc6f1c52c14fe6e : SUCCESS in 25m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1618a5775ca44e3199db3cd08079413a : SUCCESS in 42m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/52a823fa33ff414098f56ca1815c7de8 : SUCCESS in 1h 04m 33s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6a973d929c704b2b9fb153a65563dc3e : SUCCESS in 1h 52m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da18e3864dc748e98a3989f1a7c433df : SUCCESS in 1h 35m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05de1fe8ad354bbaa07d9c05123951d5 : SUCCESS in 1h 08m 02s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/271b1161147848f2a760de4aba3361ae : SUCCESS in 52m 16s","accounts_in_message":[],"_revision_number":1},{"id":"44e5013c63f73e640eaed1c8aa9953267e94f723","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-06 21:57:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e909b4a1667d4ff8b139864b211527c50bf09bf9","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2023-02-07 02:28:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dee2b69e1e917a26fbf4be39fc9d9bca2a12705d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-07 04:31:41.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/45/872745/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"457c8477b683942fa63bfd28e7fb79a00dc50fe4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-07 10:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de26eee61f24edfeea4bdd3e75541a8630fe5e36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 10:58:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0be5e08341fcc9f68b0ed0c93f57dfdeff4b160b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 12:48:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18501a1c13c84599b7c3a7148e33d663\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/efb30561056f4948ab3b00ba3966ecb5 : SUCCESS in 1h 45m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71de5fefd244e1d863366370de6fb16 : SUCCESS in 14m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1634c58e4e954930a69efbf8c8f544c4 : SUCCESS in 26m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9186199ca0ad4ade9caeaeb7b523e32a : SUCCESS in 28m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6a7aa575d64b748cff65f3d4b37418 : SUCCESS in 12m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d7631ffabf84efca6d5963f939d920d : SUCCESS in 54m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a6a9440251ba4ff99e07ec2856060c3f : SUCCESS in 1h 17m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e238d87b70074a16867999d430d97408 : SUCCESS in 1h 19m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/eec0295a713447b4aee3460f745dfc42 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d15d80919991f1a6596e6a1f770b20807c13cac0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 12:48:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"39e07a1b1d6dd3995e9d4cd42195fdc48ebfcf48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 12:50:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1709a021b8eb4505a140d59b2d3c01df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02b32bad57349819afe8386c2bbc176 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61b2917a3e8596cbb4074024cee0844e319ed551","revisions":{"61b2917a3e8596cbb4074024cee0844e319ed551":{"kind":"REWORK","_number":1,"created":"2023-02-06 10:18:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/872745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/872745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/872745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/872745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/872745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/872745/1"}}},"commit":{"parents":[{"commit":"b260b7cba2d03ee426506fc89dd6a58a9d1a8ffa","subject":"Merge \"Removed duplicated job call in experimental queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b260b7cba2d03ee426506fc89dd6a58a9d1a8ffa"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-03 20:06:00.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-03 20:06:00.000000000","tz":60},"subject":"Filter subnets by \"enable_dhcp\" flag using the correct type","message":"Filter subnets by \"enable_dhcp\" flag using the correct type\n\nWhile other SQL engines can compare interger and boolean types,\nPostgreSQL needs explicit casting to compare variables. Method\n\"_sync_subnet_dhcp_options\" is currently raising the following\nerror:\n  operator does not exist: boolean \u003d integer\n\nCloses-Bug: #2004581\nChange-Id: I715029c311c4516f3212054c5c72533b12fd0986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61b2917a3e8596cbb4074024cee0844e319ed551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61b2917a3e8596cbb4074024cee0844e319ed551"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
