)]}'
{"id":"openstack%2Fneutron~873280","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I23540114120f6ea52754116cfaaeac35e09543b4","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1986428-stable/yoga","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-10 15:21:43.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-10 14:01:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-02-10 15:21:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23540114120f6ea52754116cfaaeac35e09543b4","subject":"Fix deprecation warnings in pkg_resources","status":"MERGED","created":"2023-02-10 10:11:18.000000000","updated":"2023-02-10 15:23:04.000000000","submitted":"2023-02-10 15:21:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873280-bug/1986428-stable/yoga","cherry_pick_of_change":853053,"cherry_pick_of_patch_set":1,"meta_rev_id":"9f2834ec75b4879766295c80393c27dfbc062f4c","_number":873280,"virtual_id_number":873280,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-10 15:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-02-10 14:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-02-10 12:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-02-10 14:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-02-10 10:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 10:11:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"CC"},{"updated":"2023-02-10 10:17:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-02-10 10:17:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-10 10:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-10 12:01:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a8eda6284123d656f197fca23df96c1ec0e7467f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-10 10:11:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2803fc2387d1617bea3779c13843911625bfaafa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-10 10:17:44.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4e87680248774cb6594bf45341565c9edea0076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 10:39:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d83a2ad76554fb58b47f19ce71ab2aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cea2da9038c43c59b1e6e948a1ddb6e : SUCCESS in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ae21e1b2359448d831d68f10f0f0092 : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f229264976140908326c6fc2e168124 : SUCCESS in 22m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50a163933b94f1e85a0433741193c3d : SUCCESS in 25m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ebbae43599a4d6ab3efc25db69e72da : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"35b24f2b13627cfb062076328dc92db62003ed9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 10:58:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fbed887fbab4174aed0a79e707574ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e24d34c696dc4834a6016a2d24204a1b : SUCCESS in 44m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c08727fda4643559c72b0bea7a2da27 : SUCCESS in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78e9b84736f46849ada96ddfe65a1b93424e348b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-10 12:01:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e524a6640e858e89914c08a34c146cf380a3866","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-10 14:01:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d931ceda6ca1c50274b42ef51a61d5902b6e5a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 14:01:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41fd1c791b3039056e3d6e9310c8ece392f376c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 15:21:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cddb0d95b0c74b2c8b802694fe55946b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76d9b90f31142d699fd13f5a8f26471 : SUCCESS in 11m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa9bc0fcd8c540c0a2aceeb6f4bef579 : SUCCESS in 26m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b99257a1c21049eaa70163d5d5125258 : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8532d7e94b462a9a4bfcaf38143d1e : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"07acec4569512a20e5f54ee7a7b4f617d929e253","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 15:21:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f2834ec75b4879766295c80393c27dfbc062f4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 15:23:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9226d1c4c874403c896d8845dc453ef8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7341b8f10302441b91d8deaec8922c21 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aaafcbef3367a8b937a0b3283170031384d5504f","revisions":{"aaafcbef3367a8b937a0b3283170031384d5504f":{"kind":"REWORK","_number":1,"created":"2023-02-10 10:11:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/80/873280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/873280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/873280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/873280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/873280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/873280/1"}}},"commit":{"parents":[{"commit":"a70cfffef35f4ad90754b6a9c73766fc3584b2d2","subject":"[OVN] Allow logging all traffic related to an ACL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a70cfffef35f4ad90754b6a9c73766fc3584b2d2"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 13:30:51.000000000","tz":540},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-02-10 10:11:18.000000000","tz":0},"subject":"Fix deprecation warnings in pkg_resources","message":"Fix deprecation warnings in pkg_resources\n\nFix the following deprecation warnings.\n\n  PkgResourcesDeprecationWarning:\n  \u003cMagicMock name\u003d\u0027execute().split().__getitem__().__getitem__()\u0027\n  id\u003d\u0027140417024565696\u0027\u003e is an invalid version and will not be\n  supported in a future release\n\n  DeprecationWarning: Creating a LegacyVersion has been\n  deprecated and will be removed in the next major release\n\nChange-Id: I23540114120f6ea52754116cfaaeac35e09543b4\nCloses-Bug: 1986428\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n(cherry picked from commit 76cf6b4a9e9a8c8f7b01f44d787b66b9b894331b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaafcbef3367a8b937a0b3283170031384d5504f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaafcbef3367a8b937a0b3283170031384d5504f"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[]}
