)]}'
{"id":"openstack%2Fneutron~873296","triplet_id":"openstack%2Fneutron~master~Icda34910ea7fe308814e0cc60999b465d3540b67","project":"openstack/neutron","branch":"master","topic":"bug/2006735","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-21 13:28:28.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-21 07:56:25.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Icda34910ea7fe308814e0cc60999b465d3540b67","subject":"[OVN] Bump the port revision number in trunk driver","status":"MERGED","created":"2023-02-09 15:39:11.000000000","updated":"2023-02-21 13:29:59.000000000","submitted":"2023-02-21 13:28:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873296-bug/2006735","meta_rev_id":"afb295748f4dd4fc8c44228fda23637f4f453ec4","_number":873296,"virtual_id_number":873296,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 13:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2023-02-13 11:52:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2023-02-13 12:00:32.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-02-13 11:51:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":1,"date":"2023-02-13 11:53:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-02-13 09:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-02-13 12:02:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2023-02-13 12:00:32.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-09 16:56:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-09 18:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 09:29:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-02-13 11:51:16.000000000","updated_by":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"reviewer":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"state":"REVIEWER"},{"updated":"2023-02-13 11:52:40.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2023-02-13 11:53:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-02-13 12:00:32.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-02-13 12:02:26.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-02-14 16:27:02.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ed65af097efec8991004e3f0b4818fc98f1bf48c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-09 15:39:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12f75a7e3d894d16a20c0666b497b4109d58701c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 16:56:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c3d19323f0f407f8619eb54fc906450\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76334bf022ee4132928420dff6ee0a3a : FAILURE in 1h 13m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a86e7d0eae94e76aa9592cb69d01784 : FAILURE in 1h 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"390031152ff1c2047d879596d755c69b6a7bdf5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 18:07:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78a7ad84ef8b47c8852e84772ddb7e1f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b3dee6915de14b2787b1f4f38fd858dd : SUCCESS in 1h 08m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6818275153c44c7d9e99cbb646b2cbf4 : FAILURE in 42m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4019159ca9ae4ff58449facc84b5a215 : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd23c9acdd0346599c871c2a6cce522c : FAILURE in 45m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c879e157919a419ebc427b977347a2b9 : FAILURE in 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e472a98a5c84adaa9a15009bf8a406f : SUCCESS in 12m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/44e6912caf3f41469391a5cd07e0af43 : FAILURE in 23m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ee414d0edef4c4d9e8368856472dacd : SUCCESS in 47m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2e501a0015e4791a20c880324f03107 : POST_FAILURE in 1h 33m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/76253c03baec4119bff4e7749debba75 : SUCCESS in 1h 17m 12s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ac713e3a3f4e476bafce015c2bfb5118 : SUCCESS in 1h 21m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7fce948ea424decb74e32861cac8e5d : SUCCESS in 1h 45m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7c1064b87e9464393b29c5f06eee724 : SUCCESS in 1h 21m 36s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/632e1fb9ced24c87a9129962636970b5 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d411bd425d9de3504ef7cd03c540144a69507e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-10 15:09:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f220a6c20d7b7eb237b49048ff197df7bd828a85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 15:30:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db15dcecdf9b4bedb3c8d747014b4a42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7983cfaf93134beb82531f401bc98f39 : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6921dfa4934e4ced96272705087daaa8 : SUCCESS in 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d88963aa8c1a0b74d54b240dd02920d703e73e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 17:58:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840af49f4c69424db41a509e120609cf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a85c97ce561046c0a16af93a2b348810 : SUCCESS in 1h 11m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ff8b3801c6f48e3b283699d80939761 : SUCCESS in 59m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d193c39537844a6bf67bba0099dd379 : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/934eba41b2ed4fa09079f0f7b9ef689a : SUCCESS in 25m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/516b0b88a53b43d68ee2704286fdf46a : SUCCESS in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9459efce5dc34ff8b801311b7de84f3f : SUCCESS in 10m 19s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fe0ddc0bfdc04967a9ba52fb2429bbab : SUCCESS in 22m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/954eb82f102b4056b71a599a80d92d3c : SUCCESS in 49m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f049a62b9b2049c1803c8b6ea6ec40bf : SUCCESS in 1h 31m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3b06bab1c2fc4234bb27fccdd2e99aed : SUCCESS in 1h 08m 15s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5970bfd0d8324a4bacd519d6192f8abb : SUCCESS in 1h 10m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09cce3fbbb9d46b681b1e406920336e1 : SUCCESS in 1h 48m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ceaae418f52c431dab52b5c7ee955150 : SUCCESS in 1h 17m 01s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b45ac093d3594c558afe88054c8696fe : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2bf58b4a261a3cc79dffabfaaabcb2a599ff94cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-13 09:29:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"54dda91a78d18a99bdc08a37f7ab68ce8ae855b8","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2023-02-13 11:51:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2448f1817de307a5a53745c0b5edfa7956606d81","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-02-13 11:52:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1dfff7c973b48ce36a4a7937617c1de5db4fab92","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-02-13 11:53:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8aaad9276a8ab60cc5a6917b458f1ed3118dc87d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-02-13 12:00:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dacd3c07570d32e4991d84a2a0539bdf7216e358","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-02-13 12:02:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2cdbe27571746abd76b0fc46c74e5f619c1fd427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 01:04:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b1c1cd6feecfe27ca4e5e8edcee1dffe9a2041b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 02:37:06.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/873247 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"0fb63024333682134dfe2b1292bb5c8ca86c15bb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-14 16:27:02.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/96/873296/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d8faf9e810c8da3f74cabdec112ac0dbdd5f6248","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-15 01:17:54.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/96/873296/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7ae629d491878345cbd70b836fee657e5c804b87","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-21 07:56:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"caf2ee257072a72114577b28000b19d86437ad8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 08:15:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0875770605e744ffb4038f5e702ee8f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba41b22651dc44d292020fc550d0d43c : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42b1db665bdb4aef850a97e0b6b0d6bf : SUCCESS in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8dbcf10dce5a79b80367d77673877b568bb1a839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 09:58:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/686af296184346d2bb7a01a378c7aab5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/91de60b018cc49a2a15ee49a77f03df1 : SUCCESS in 1h 59m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb1e754003b24359a718cb55c3cf110f : SUCCESS in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8cc6b2c01fa478097b013013c508e60 : SUCCESS in 14m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38105bbe61a143b79b8fe96cbe74e85c : SUCCESS in 31m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a249f5d87814c1ebddc3797c6db1c60 : SUCCESS in 31m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1e2602910d4105b89d3662e6ff0e5c : SUCCESS in 11m 17s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/4b3586a050824a75a086acb0b3eb308e : SUCCESS in 24m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/815f4470881e43a18b822745de12adf3 : SUCCESS in 45m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a2c68efd9884f27b73e93cf805ff2f9 : SUCCESS in 1h 45m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0b2797cc084e4950a2ed10fb372ff7cb : FAILURE in 55m 57s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/de257523496b4c2ca1508f1f126060b7 : SUCCESS in 1h 17m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/171182831f3245a8addb5c9aca7aba5e : SUCCESS in 1h 45m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3cc3ed294ecc4579aefbb5849457e877 : SUCCESS in 1h 25m 49s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9f2e43bf0c8a43c9ac5fadf80025e5e6 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"35f5ab355a8e18a8fa5837f6a9af049e8998c4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 09:59:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a851acedcbe3fdc1fd5b2155be141b0322460a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:28:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88e90ec1d6c459697b71dd97dd0dfed\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cc92896c59c04db5b77989b5a57dc3ad : SUCCESS in 1h 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457b9840cbb74b3e8a60a7b2012aedfc : SUCCESS in 12m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6139ce7a9fd74248804525e66b5a489d : SUCCESS in 27m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1efae4bb73724e3eb5bb8a8f02dc8d77 : SUCCESS in 23m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664ca288f8e4440d84738fc647ee54fe : SUCCESS in 13m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/43ee28397d8d49738819ddaffd2868ca : SUCCESS in 58m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdb6a780a0b1450eb9c60970d89c8f0f : SUCCESS in 2h 22m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0095b5cd3b3a4e75b130f4a9a054db49 : SUCCESS in 1h 11m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8591a86e8564b39b3deaface21d12aa : SUCCESS in 1h 31m 59s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5a5abb724e694c61b8eb36bba06a9714 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":2},{"id":"01ab5516146c197022d3f54eb534e367cf2e8af5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:28:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"afb295748f4dd4fc8c44228fda23637f4f453ec4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:29:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92dd856472954be7a2f27af6f306941c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e7cb56eceb4018a972d6caac48aacf : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fce516e3461054716ff2d4623c240d3e006acbeb","revisions":{"a2782ef5a54dd61ec3a69d8e3120efa6d6679059":{"kind":"REWORK","_number":1,"created":"2023-02-09 15:39:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/873296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/873296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/873296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/873296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/873296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/873296/1"}}},"commit":{"parents":[{"commit":"cce9e739f14190fafe7549649a16645c04c87a7a","subject":"[OVN] Add missing LSP device_owner info in trunk driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cce9e739f14190fafe7549649a16645c04c87a7a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-09 15:36:14.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-09 15:36:14.000000000","tz":60},"subject":"[OVN] Bump the port revision number in trunk driver","message":"[OVN] Bump the port revision number in trunk driver\n\nWhen the subport binding information is updated in the OVN trunk\ndriver, the OVN revision number should be updated too, same as\nin other ``ovn_client`` method, for example.\n\nCloses-Bug: #2006734\nChange-Id: Icda34910ea7fe308814e0cc60999b465d3540b67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2782ef5a54dd61ec3a69d8e3120efa6d6679059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2782ef5a54dd61ec3a69d8e3120efa6d6679059"}]},"branch":"refs/heads/master"},"fce516e3461054716ff2d4623c240d3e006acbeb":{"kind":"REWORK","_number":2,"created":"2023-02-10 15:09:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/96/873296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/873296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/873296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/873296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/873296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/873296/2"}}},"commit":{"parents":[{"commit":"d83e243c6d54f29ded7a0bea6e5538772fff3a8e","subject":"[OVN] Add missing LSP device_owner info in trunk driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d83e243c6d54f29ded7a0bea6e5538772fff3a8e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-09 15:36:14.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-10 15:08:56.000000000","tz":60},"subject":"[OVN] Bump the port revision number in trunk driver","message":"[OVN] Bump the port revision number in trunk driver\n\nWhen the subport binding information is updated in the OVN trunk\ndriver, the OVN revision number should be updated too, same as\nin other ``ovn_client`` method, for example.\n\nCloses-Bug: #2006734\nChange-Id: Icda34910ea7fe308814e0cc60999b465d3540b67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fce516e3461054716ff2d4623c240d3e006acbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fce516e3461054716ff2d4623c240d3e006acbeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
