)]}'
{"id":"openstack%2Fneutron~873553","triplet_id":"openstack%2Fneutron~master~I544ef84a015d1a90969dd41713b6248c6b84424d","project":"openstack/neutron","branch":"master","topic":"bug/2006264","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-15 15:33:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I544ef84a015d1a90969dd41713b6248c6b84424d","subject":"Increase tox-py311 timeout to 3600 seconds","status":"MERGED","created":"2023-02-13 11:58:12.000000000","updated":"2023-02-15 19:54:29.000000000","submitted":"2023-02-15 19:52:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873553-bug/2006264","meta_rev_id":"7c4c9fbfae06ed20004a5634fa65c6ba00a92663","_number":873553,"virtual_id_number":873553,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 19:52:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2023-02-13 15:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2023-02-14 13:27:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2023-02-14 13:27:04.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 12:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 15:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 15:44:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-02-14 13:27:04.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"7f94635cad6583374c197e8be4b0a9f24e39142e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-13 11:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4c021d8ea4cdcffc4fa489501d2856e0c17f566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 12:41:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab9be7911b54e11bdf1aa2a5c343f7a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0648fb5226b54fc2ad56ae298b658ab7 : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/719d08dc4bbc4f5aa268712854f5d43d : SUCCESS in 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de411b165cf21cf9785f3d8bcc16a842ca96b981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:26:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e63bbdcdb7c4388a04a6a6215cf01fd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/589d183d5f514d82b26f4782cc92f97e : SUCCESS in 1h 32m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2cab2856358c41c7bf792119835a1f58 : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9b9e7dbaf0d7411eb2f0d84c576709a2 : SUCCESS in 2h 06m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dfb5aedefc09471dab5bcc46977139c6 : SUCCESS in 1h 20m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c928f18bd0c74da797ddf3153bc9c7b4 : SUCCESS in 37m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0397cc4f03542e9be376de52ece4492 : SUCCESS in 36m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2736022fc84c17ad63fd17c2909fdb : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afa0c424769f4c4ab2c52ee6698004b7 : SUCCESS in 24m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e20adef4e8ca45c6b3d834ea25264b15 : SUCCESS in 20m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4802e1f59a364b9b8dc65d6fa7968b42 : SUCCESS in 9m 42s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7ee2e9381780462e9dc1fe70a1d23bb6 : SUCCESS in 23m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/afaa204851364f299be52435f0550724 : SUCCESS in 50m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a31c5e5528b4919820e065d852f1512 : FAILURE in 2h 29m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc9d0b4424a64b0eabeb129f251f1fcd : SUCCESS in 44m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9be90920e81d40629bf8021a2c0a826b : SUCCESS in 57m 26s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6519e2d76aaf4a0c9e7e6c59c720a78a : SUCCESS in 51m 18s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b9bd6bfe991b4d149d2230ca8cbc5fa6 : SUCCESS in 57m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3e5ef9f34e1745a9a539062d67ada6a5 : SUCCESS in 1h 28m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9397942bcd3d40058747188caf410002 : SUCCESS in 1h 16m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d94229fee8a14f809bae81f7c96417f6 : FAILURE in 1h 05m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/13aece25dcd34efe80793d47c10fa9d9 : SUCCESS in 1h 37m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a1855f3b41947aaa072772ed355d800 : SUCCESS in 1h 45m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a2330a6cffdd4a60ae5f9f638d0019a4 : SUCCESS in 1h 14m 38s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ae7540e631134a3684797cfd93021e93 : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2253e38540a5d798f953e5cb0121b2903ef75938","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-13 15:44:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"071d3e9502214f150c17b6ab17eb1f94b53519e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 16:32:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b9030b2f09a43c089a50dd24aab1ad8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da936d06da2a4aad9b70c7ec13a51347 : SUCCESS in 20m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b3c52e0dc1c45e78cd47343e16a8b82 : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4746e577283d274328011c6da8bc324258a33a0a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-13 16:37:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1be0e76f13be39b3429ce16b91da9b81044f0434","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:01:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c831ab3aa7b4c70bc12105037f5666b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4e1ecf0b14f421a85e8d1a346f3dd34 : TIMED_OUT in 1h 22m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6bba5f77757b4280a7fe18223f294a1d : TIMED_OUT in 1h 22m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a23626422e7d2e2dd23b7032c0d80a6fa17a22e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:50:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bddaddda9afa4d4fba8b26d9c7d1015e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/36a115d08d7b4a0baf2d2b2c85b0fd0e : SUCCESS in 1h 52m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73c2901112c8429981e43cba4b6d70e3 : SUCCESS in 1h 55m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b031d89b2714f9a98114cf707d0600c : SUCCESS in 1h 34m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a1191dcd0c594d6898baf95d27e50a89 : SUCCESS in 1h 31m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fde50f84cb0742b7aa3f52f95d5bdca8 : SUCCESS in 26m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bac42fdbb7174bf3a3088147b3f0aa02 : SUCCESS in 40m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e00458622674a30a3212f00a344a65b : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762c7a20a4ea460ab3d4529066502a23 : SUCCESS in 27m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27548da2ee744ad1bfe5b9233555d4ff : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b87c92b031345448c4a5ab76141f853 : SUCCESS in 13m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0490a24494484216b3b55dc409ff67dc : SUCCESS in 24m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5528fbd4b3e34268ac756e9f0dd75711 : SUCCESS in 42m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/625ec2b697124714b9bf9480a9a507dc : SUCCESS in 1h 21m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdc7ccfdadf34a5781cd35ae0dde713f : TIMED_OUT in 3h 03m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d65394468ec44352ac1d9740dbf958fb : SUCCESS in 58m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/82e457b5e2804b6d9d2d2dad6086b625 : SUCCESS in 1h 14m 17s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/754c0294ddef4c71a60c930adaf7d2c6 : SUCCESS in 1h 07m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/931c1e4cd79f4e5cb6f9d6eaf3c818af : SUCCESS in 1h 25m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88961b9e6a3846f898e1a1e752961bbb : SUCCESS in 1h 44m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6800220d59024e11bdf14fef52674b4b : FAILURE in 1h 06m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f122288ff716431890fd5ee49147ea9b : SUCCESS in 1h 24m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8a9172ad6aba45aca8870104025ae2d8 : SUCCESS in 1h 30m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d4b4ffa14454a2c8e1e7224d4633e7a : SUCCESS in 1h 02m 27s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ca89608c841e4d2a8bc4945a93038222 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"96b769e87da0df1ac12e4255a770654f186dc37f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-14 08:37:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1df7614bde5f204f018c6dc8ea00126d52bc2a1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 10:01:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94cbe31fe69945e3906d91737bbb5f8c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52f6182ff0224f458df3cac86f733db9 : TIMED_OUT in 1h 21m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e53566cb05144fcf8a51e908d4383b9c : SUCCESS in 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cbb463a650c54dfe3d1de9961eaaa31c2421763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 11:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c821148c81c34f6ea5a0ef673af500d1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1d5f18d96959473eb2fb00e2244b0dde : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/65efb69f4fe74ac79b580a3ecb0e8422 : SUCCESS in 1h 16m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78f5aed958a64e65bfc21020ef3404a4 : SUCCESS in 1h 24m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0e6a0cb4a0fb43db914dbfb8bbb8e98b : SUCCESS in 1h 10m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76ddfa7132024031ab2bd2717382474b : SUCCESS in 28m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c7ec19bd1284e3fb8632914e2244e14 : SUCCESS in 44m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f8630c15c9432aae0b71354cbe2653 : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581bc374fd7941878934649ca73a9216 : SUCCESS in 30m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c29f6d45c88b4105980837523a7069ff : SUCCESS in 29m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dabf292de2d4b069245c26f881b9b6d : SUCCESS in 13m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ab562f8dcd394a809b6dc5936df3699c : SUCCESS in 27m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/328fc55ee3f348378939ee6bdfb37f5f : SUCCESS in 46m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1978671755b14bbd877a7e93bb6cc6f1 : SUCCESS in 1h 31m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8d96b905639457c8c287525d96a7f76 : SUCCESS in 1h 00m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c42bc20ee7234a93bc779920d23531d6 : SUCCESS in 1h 07m 17s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/faa7993333a24ffcafe404e58a3ba7be : SUCCESS in 1h 20m 07s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f420f73027644d3bafa8df45683235cb : SUCCESS in 57m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a188b17cb0b246cfb394b9d5addee479 : SUCCESS in 1h 34m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/481214e29eb84023a7c7b68be5729eeb : SUCCESS in 1h 19m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fccf3abd061e4905a74a7eaab1d87740 : SUCCESS in 1h 09m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/341064b530b442cebf0565ee23d77e77 : SUCCESS in 2h 19m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4890c84c86fe4e06858850a11141fee7 : SUCCESS in 1h 42m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f96e27d1552c4dd39f75360cf04f5aec : SUCCESS in 1h 11m 39s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/909c6b57527445e9aa559bb619358054 : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a6a23859f17949742e6e85175f8bca4c717710fa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-14 13:27:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1173f354183d2f5bc3d40efa87d79373a473234d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"012cf7779bae34b2f88c15c2d8772bad0bbbfb59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:33:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12e00f57304a48e8ae3f43c98585ba76\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/81500da824884e1281dfe2bb282e9c4c : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a01ad2d101e5432bb8a5017db08d21a5 : SUCCESS in 1h 14m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5cf0fa7bebdb40838b6a321c9a4206cf : SUCCESS in 1h 11m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f706a2f3673d46f49b05e4a620208c1f : SUCCESS in 1h 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2beabf0b524f41a5bb551459319f7f7c : SUCCESS in 13m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01caf8d4746b40b197e365f529777c1e : SUCCESS in 28m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b244232a6185489f8e39b923a9c55478 : SUCCESS in 27m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0869c600ff064f59923391c2445dc5d9 : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8160efb7fd154277a754c331451db095 : SUCCESS in 41m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd3bc6cb588842f79f0100ee645203be : SUCCESS in 1h 33m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2c588eec4ee9491481811c3d81c0c221 : SUCCESS in 1h 25m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7e9a8c6b6714a0084085926965de0f6 : SUCCESS in 1h 08m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/62d3d6d8093047039773047d7f61ccc2 : SUCCESS in 58m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8764d29d8a3441588cab2b781d6aa00f : FAILURE in 1h 07m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/faffdf34c1464f2cbcf55151a1ed7710 : SUCCESS in 1h 22m 28s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bd88940638184f25b75f4201ae0064ee : SUCCESS in 1h 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0a40ed63fe40dc32d2d8b591eed69ba459f21824","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-14 18:03:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e93070ab9b92c786e64f78fc9965b81086679a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 18:58:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9d078ac0a2407d904cf739eae4ff6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/045620e6d31e47b3a29e7944c4dbef05 : SUCCESS in 52m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a661d344d064ed49dad004ffa062d6a : SUCCESS in 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be04108533007811e1b450340f490f8f75be5690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:29:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98a49c496063439d8ec9146813ecf2c3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/adb3b7c86a4b453a88b0926cae751a70 : SUCCESS in 1h 28m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e71331b55ac4adcb362f42ce2d4bbd0 : SUCCESS in 1h 18m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/664134b4b6e349b2bcfb8165be30739d : SUCCESS in 1h 12m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2226395913743e29926eb1c454601a8 : SUCCESS in 1h 05m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/380c920e251444948cebee502f8c3262 : SUCCESS in 30m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/117c8c05136d45c9b26cd3ef06a17a1f : SUCCESS in 35m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b117b422d6024630b5bf8e29bbe4f1dc : SUCCESS in 13m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/388947aef31d4e0a82ace7040ec4890d : SUCCESS in 36m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/555edb9d56ea4b9cafc8d64820bfc930 : SUCCESS in 28m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9c722253de42e096c154f8573d67f6 : SUCCESS in 10m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ffff9990395a4118b066d6c58ec1d847 : SUCCESS in 24m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6e7902a2fe44989b54b324501a7865e : SUCCESS in 1h 20m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df1ebc4c161043ad9a465e3d2141818e : SUCCESS in 1h 41m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/432674ecc26d44c1b9d75c49e0ae6bf1 : SUCCESS in 1h 04m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3578efe3a2a247deb125f988f2b75be6 : SUCCESS in 1h 07m 37s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1d68ce5e924247058f22d97508630b66 : SUCCESS in 1h 07m 14s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/153208dc7bdb4ac98c88251418e879b1 : SUCCESS in 1h 07m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/55a4cb71cf7c4939800235d7ba180f0e : SUCCESS in 1h 23m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f3083b91dee4362bcd341c0201903a1 : SUCCESS in 1h 46m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b073297118ce4fc7b813d16af3224709 : SUCCESS in 1h 04m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6df125f1874a4948b9efecf6608eee66 : SUCCESS in 1h 35m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dbb752abc2ef49bca25c4415e3eb8e50 : SUCCESS in 1h 46m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/95dacbd3165f4e4aaba9a8f06cde914b : SUCCESS in 1h 15m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4d7b8492fb364ac5bd2f25e6e7cf2f2b : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bc049adf10bc01814626b8337f0ebe58bf0c9eda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:30:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19fb963145ddfc0f73aa30594e81d601c1b5584b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:14:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d07d3f001d2433dbacda2e1d2a773b8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ab802d70d2414655819d395968c23cfe : SUCCESS in 2h 26m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6c9b40708067440bbddfac52e0c868c0 : SUCCESS in 2h 01m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/385e3bb09a714baebe20cfa9c4413cc0 : SUCCESS in 1h 27m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7aa1e05bfc3b4cbeade7bcc85f0a96f7 : SUCCESS in 1h 36m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2efee715ef42b28e2d61ebbfd957ec : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71f8b79781a048d79727607d5d0b748e : SUCCESS in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/486319920c3949b7848986e36806e219 : SUCCESS in 26m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce396fa1705475e8c69194d1b5243c7 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0843f74cc354bdeac4cb6055a686722 : SUCCESS in 50m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc87adfa7e224898af50ebb5d586c61d : SUCCESS in 1h 43m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/491256e47d2a47f8b5171ddf438e530e : SUCCESS in 1h 54m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4676da174b1545fa932e2f73a394248a : SUCCESS in 59m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9a1a1abcf713427f902d2af66e3139b3 : SUCCESS in 1h 06m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d63672d40afd44b597e34769479b047e : FAILURE in 1h 06m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11e1bc0c0e35477b9b0e2bea0a89373c : SUCCESS in 1h 20m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/28859a54c7af4023984323987cb6cbef : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7ae044e262b100ac15ec11cd8a14e769f968638d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 09:25:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"debb82cb004799715fe4816f80a29e6015aaac49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 09:53:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b940867167654af3971216799e8c16c6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94cb7525492c48e3a8eb5ebbe4f5565b : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2c86c6e660244dfb11ee8f1ac6e2f15 : SUCCESS in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51241d7e7dc0498f863fbe16a3a2701763dd94f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 12:22:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7dc7a858c2c409cb1d1da7b1d62fa70\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/62e6100848b74a97bc992cbfaf877823 : SUCCESS in 2h 03m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/604134170bba450999b1ca5227d5e267 : SUCCESS in 2h 51m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/362ccad72e984b33bd1157df3bb66fc1 : SUCCESS in 1h 17m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0a7b37efb33415aa3b0bca1f9f3f79d : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07dd3eebf800497da681415c3ed4f651 : SUCCESS in 28m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdccebea5c624cb7a3f09a6d328f6c32 : SUCCESS in 52m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a47a63ebbb1467da716c00a1361f751 : SUCCESS in 13m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac2f802c467b4c8a9f3e575eee90429a : SUCCESS in 28m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/635be13d57e54a7485a60119025ad509 : SUCCESS in 26m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d6126f0f1544a7b11047e44c3ae809 : SUCCESS in 9m 37s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/4c0f186666d340bdbbeaf43f603799ef : SUCCESS in 30m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfb7f40e8a9449da959fb4dff1837016 : SUCCESS in 58m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a9ca3b7a5004fc0a23903f0f146a81b : SUCCESS in 2h 30m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b939064d2174411a4768e32c1458916 : SUCCESS in 1h 05m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3f2dd42071f142659fdb4eccc17d7bbd : SUCCESS in 1h 05m 12s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b990792902ec407fad1b0b8c691a8a6c : SUCCESS in 1h 44m 11s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e85a223cf07842fba6e16dbc3faca8b6 : SUCCESS in 59m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/145a96d2ca3d470a8c5613071155e243 : SUCCESS in 1h 34m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/af2b4ec1547343f893e751bd37804e20 : FAILURE in 1h 03m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ff67c62ea99d4a888b751b9a7da85c8a : SUCCESS in 1h 11m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2a3f09ac46be459d86b14c1511b885fa : SUCCESS in 1h 43m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14c161ca67a34506aa86b11907c9465c : TIMED_OUT in 2h 13m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/200c0338d626480aa4fcf57c28525e16 : SUCCESS in 59m 17s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1aa5f02b51ef4d9286c6cd09b257d3b2 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d6d00e98f1731a89623bcc864737b77842243a43","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 15:33:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa90088bfac6e70c3cc5cb679a74a8b1e720421c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 16:44:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f87623d8c421456d81f82564eee7cd6a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a9869342de846f583c0d52b7707ae6f : SUCCESS in 1h 07m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f41af8d6d81479295ed4401a692731c : SUCCESS in 1h 03m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f255a965a63b3a1ab93406dc5505a87a08f37390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:29:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24bf456c08f84d1f922c73ae79634ae6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d6cda934847a4fd49f9dca388f7a42d5 : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3b68bb129394412590d8c96a3a4fdc46 : SUCCESS in 1h 44m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5b08ae3db7c4ee9ac7064d64f5921ec : SUCCESS in 1h 16m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/30aa879a6b2e49acaf460c0da6bbeee6 : SUCCESS in 1h 33m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37941687cf3b4cd6ad3d68fd790a966b : SUCCESS in 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f232a5c388843289a4bb532b3813f33 : SUCCESS in 41m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d123e58658402087c68db21d18e4dd : SUCCESS in 14m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e943c3d6f93498b888b2687acc849aa : SUCCESS in 25m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/515d8754cd7b4b19bd27c69b2251a32f : SUCCESS in 23m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced835c125c84ee4bee8eba34bbacacc : SUCCESS in 10m 20s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d115352f409645bf9472b98dc649d4c3 : SUCCESS in 21m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/78eaed4c639c47c6b087a79a088f962d : SUCCESS in 44m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddff5e378eaf4e788f68726ab175e1b2 : SUCCESS in 1h 32m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b2171e3b75e48ac85773d1f7f531b89 : SUCCESS in 56m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/29d66799251b4115be7ec620c71c29b6 : SUCCESS in 58m 36s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/415fdd8f869849febb8fea515f2536b6 : SUCCESS in 1h 03m 49s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/51f261c7e91a4cdf83439b02fe93885a : SUCCESS in 1h 22m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aecb6cea3d90433e8fdccbf47337f023 : SUCCESS in 1h 19m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/93c3dd817b2b4251869e25740005f83b : SUCCESS in 1h 25m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/39e192f913bc41539f5d57927899d674 : SUCCESS in 1h 01m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f87b763cf924ec4aebe8d1c5d82ae73 : SUCCESS in 1h 23m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ecc2ddee5f6e45d0a85b6498b578c0de : SUCCESS in 1h 41m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/03a99055526b43958cbb26b655f5044f : SUCCESS in 1h 09m 38s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/38cd75e09fa04247aaa8438cfa5b3bac : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b4ec2832e9fa997d09eca2d5b54916ba92d84478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:30:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c94c9f9fc189fe8da0fe0b693773cabfac2be65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:52:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18051b819f5547c182bdcc52dc65e208\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/94e65b4bdf2a48488fd9e7000743fd38 : SUCCESS in 2h 04m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f359464b67d7433c997a20a85cde907f : SUCCESS in 1h 42m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/45bab8f1209f413f87cb918cc67bbb64 : SUCCESS in 1h 58m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d1b8da7bef14e17a3835e6343f02548 : SUCCESS in 1h 35m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/360e93e670fe424181c3907b885ff187 : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b742b73415644bbb9bdb0695ed14802 : SUCCESS in 25m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9be28bd93ba34a84b5675cf2850f80cc : SUCCESS in 27m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e828a75b9440608d1607a93ddfa2d1 : SUCCESS in 12m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b734d2a697c04d05863f5800b03e5975 : SUCCESS in 1h 14m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df01849565454fe8ac7d0f8b50829ce7 : SUCCESS in 1h 31m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a2a107329c33412f96fc5a1c9f147b6e : SUCCESS in 1h 42m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7eeea3cd0e0f4137bc41ed0ee08636a4 : SUCCESS in 58m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3c67be332a1544418cfe9a20a8959768 : SUCCESS in 1h 03m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/70c6eeb8fc0040a980d47c6306823a1c : SUCCESS in 59m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/07e60049a87245c7af6a99a1681e531d : SUCCESS in 1h 07m 04s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ea38b4589f3f4ca7a4fd4fa562c4bdeb : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c3eec16c5ab388760848cb182166f8e1ac834a24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:52:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c4c9fbfae06ed20004a5634fa65c6ba00a92663","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 19:54:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56299348555e4908b5479e2c0b1bbf77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f06f98c8c2e465d930ab084f99da411 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9880c20133f1cce76ffcf695d3e153881d64e53","revisions":{"f9880c20133f1cce76ffcf695d3e153881d64e53":{"kind":"REWORK","_number":1,"created":"2023-02-13 11:58:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/53/873553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/873553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/873553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/873553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/873553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/873553/1"}}},"commit":{"parents":[{"commit":"f8121ee981888179e785ba509263273f75b5ed6d","subject":"Merge \"allow manila ports to do multiple port binding for ML2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8121ee981888179e785ba509263273f75b5ed6d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-13 11:55:37.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-13 11:57:46.000000000","tz":60},"subject":"Increase tox-py311 timeout to 3600 seconds","message":"Increase tox-py311 timeout to 3600 seconds\n\nThis is default configured timeout for Neutron \"openstack-tox-pyXX\"\njobs.\n\nCloses-Bug: #2006264\nChange-Id: I544ef84a015d1a90969dd41713b6248c6b84424d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9880c20133f1cce76ffcf695d3e153881d64e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9880c20133f1cce76ffcf695d3e153881d64e53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
