)]}'
{"id":"openstack%2Fneutron~873585","triplet_id":"openstack%2Fneutron~stable%2Fxena~I23540114120f6ea52754116cfaaeac35e09543b4","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1986428-stable/xena","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-15 08:58:15.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-02-14 21:09:25.000000000","reason":"Reviewer/Cc was removed"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-15 17:46:52.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-15 09:23:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-02-15 13:48:24.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I23540114120f6ea52754116cfaaeac35e09543b4","subject":"Fix deprecation warnings in pkg_resources","status":"MERGED","created":"2023-02-14 19:18:59.000000000","updated":"2023-02-15 17:49:38.000000000","submitted":"2023-02-15 17:46:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873585-bug/1986428-stable/xena","cherry_pick_of_change":853053,"cherry_pick_of_patch_set":1,"meta_rev_id":"3e77fee6c9133b8c5a407f63804950cdb411cd15","_number":873585,"virtual_id_number":873585,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 17:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-02-15 08:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-15 11:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-02-15 09:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-02-15 11:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-14 19:18:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"CC"},{"updated":"2023-02-14 19:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 21:07:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-14 21:07:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-02-14 21:07:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-02-14 21:09:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REMOVED"},{"updated":"2023-02-14 21:09:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-15 11:58:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"0ad1ea90ce09186c33194ec8727edb11b6845096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-14 19:18:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6fd9d6744549c0a1e839fa8111c233593d9ba401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:59:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1a6aca9031407a81ee0518fd8cdba2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c26fd8329cb49c684d38e1a67fc3b5d : SUCCESS in 38m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91add51a6f614f2bbd831489444ff926 : SUCCESS in 17m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42b5c0c626d04ae6b061547c9799e4f9 : SUCCESS in 32m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22cae73034ea492a8773494a4ead7df8 : SUCCESS in 24m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69319061e2f7460692eb8ce0b852dd15 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9ad3f171c84bb5881712a54ab784a3 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"11c67f3c02004359071e214a21a89e98bb7899a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 20:02:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec3bbc3b36641bab0de62e9769f24c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/072d833ec13242a8a0fdfa50a203b32e : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c649f3a0e4b4e47b18d5de939bb824e : SUCCESS in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"686e0ca64c68a624961dec1fbe9fc6ef70caaa65","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-14 21:09:25.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_13861\u003e.","accounts_in_message":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"_revision_number":1},{"id":"3606e624607de40dda68c032b104d303e143327e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 08:58:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3590f435717ef5c6f434726d82c54005da6515b8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-15 09:23:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb4f5d41e8fa4fd64380e514836bc7102539190c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-15 11:31:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d019052b355f382b852e3d874df843660567bef9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 11:31:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17e68799b4a48007cd54456111bd6ac3b106e548","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-02-15 11:58:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93152b1036b120ed11a9a866977bf6b9d9c7565c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 12:26:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d68d343a5714af3b451e0af9cc5c005\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2791e94e154fd5b21ffc828605f677 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc92e225954f4f51bc820a2183d36821 : SUCCESS in 26m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e468ea00f45344e59d5157f1ea0e473b : TIMED_OUT in 41m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8311de221624b05a41c616cf1e26c61 : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8cb8e30087b7d53cd42762dedf181b105b76f59a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-15 13:48:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"472d8adbe54dc2e6eaeef470a42d42d8f28896cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:07:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4171c2ff931943c6b337bb952c097bd8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c2e405efe2b4f28a00b98e7bce0172b : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f91e7fd9703f4ac687f3aa0738c870dc : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69d79f0e8538ffa5085d35f339e07c381dddf963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:48:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df8e4568c8294a00bd58c7b20f5306f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9adbeaf49da45a9b8f303e0896ff8f3 : SUCCESS in 48m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d8fd0a598a432dbc4de5d9e7be8bf3 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea73c5fe96904e348bf5e370560a599b : SUCCESS in 24m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0aed59be56140fd965118500540d78e : SUCCESS in 24m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/441313ec72fc46a58aa015aeb4d6eb3f : FAILURE in 27m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5dda93071e4446a9d4555d382bcee2e : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":1},{"id":"200230bd92e02a46facbbcd68073d5c4453b1c68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:49:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb5ba50f10e6e0b82b5a74a62db5e907877c7019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:46:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aa696156ea4442d89786cfc49b336f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131514c0f4f34bf99d8b9e73d194b5cd : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636562ffce244e20bc36afeaa36cf6b7 : SUCCESS in 37m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fda13bbb8494aad8d238d3429b93049 : SUCCESS in 25m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e732d9ec1a924105bda87d0d4cc37e05 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f9004d192069d8dffc43aa748f981c855326d984","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:46:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e77fee6c9133b8c5a407f63804950cdb411cd15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:49:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7553fef3740346c2a6f92733a99abe70\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9875f0a5bd4fe69b96657a49212275 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"288daf9974d6776a27f3539da2f8a168831cf144","revisions":{"288daf9974d6776a27f3539da2f8a168831cf144":{"kind":"REWORK","_number":1,"created":"2023-02-14 19:18:59.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/85/873585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/873585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/873585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/873585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/873585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/873585/1"}}},"commit":{"parents":[{"commit":"6116ce9a03821d25ea9ce1f4ba3f8f041f47b0ce","subject":"Filter subnets by \"enable_dhcp\" flag using the correct type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6116ce9a03821d25ea9ce1f4ba3f8f041f47b0ce"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 13:30:51.000000000","tz":540},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-02-14 19:18:59.000000000","tz":0},"subject":"Fix deprecation warnings in pkg_resources","message":"Fix deprecation warnings in pkg_resources\n\nFix the following deprecation warnings.\n\n  PkgResourcesDeprecationWarning:\n  \u003cMagicMock name\u003d\u0027execute().split().__getitem__().__getitem__()\u0027\n  id\u003d\u0027140417024565696\u0027\u003e is an invalid version and will not be\n  supported in a future release\n\n  DeprecationWarning: Creating a LegacyVersion has been\n  deprecated and will be removed in the next major release\n\nChange-Id: I23540114120f6ea52754116cfaaeac35e09543b4\nCloses-Bug: 1986428\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n(cherry picked from commit 76cf6b4a9e9a8c8f7b01f44d787b66b9b894331b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/288daf9974d6776a27f3539da2f8a168831cf144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/288daf9974d6776a27f3539da2f8a168831cf144"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
