)]}'
{"id":"openstack%2Fneutron~873626","triplet_id":"openstack%2Fneutron~stable%2Fxena~I25145c088ffdca47acfcb7add02b1a4a615e4612","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2006496","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-17 20:53:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25145c088ffdca47acfcb7add02b1a4a615e4612","subject":"Improve scheduling L3/DHCP agents, missing lower binding indexes","status":"MERGED","created":"2023-02-13 17:08:31.000000000","updated":"2023-02-17 20:55:02.000000000","submitted":"2023-02-17 20:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873626-bug/2006496","meta_rev_id":"afa312e4e7724c81457990a02619b1749b79ef2a","_number":873626,"virtual_id_number":873626,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 20:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-02-16 09:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-02-17 16:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-02-17 16:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 18:44:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 19:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 09:54:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-02-17 16:36:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"34de76c604bc88c4b79ac1a487199d3bd350e526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-13 17:08:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be580a5968e80148587b618e48dd280e4ba22fd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:44:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fefea55af7fb40778af3ae963f23e5c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94c75eedbd734f95855ef5eac395ee4d : SUCCESS in 59m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69d201b5b8d544d2b7cff78a7fd0fec5 : SUCCESS in 55m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cea5cef84e1d4ca8316b298ab1445121ea119b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:12:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/132a58e3257041c89a48c97a060888db\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/5a690a7680964c61991e1cd8ab79d292 : SUCCESS in 44m 02s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/15b5d68268e4411db3f5393302b9b393 : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/97b3afed703345f68dc126044fbd0313 : SUCCESS in 1h 00m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/232e0ce87d1148d29238467e9ee98370 : SUCCESS in 1h 06m 58s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/f629991853204bd197bbe019e4409387 : SUCCESS in 53m 00s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/766df7bed563478ba898e120d2a39b1e : SUCCESS in 39m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a497cf60ad1e49d78e459c7fe69db666 : SUCCESS in 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6b59e0d9e4497eb524ea09254cabfb : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d679efaf662c46adaeeb5af673e53db2 : SUCCESS in 29m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2654a20994574648b85679f7d529f5fc : SUCCESS in 27m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e99209c515e4ad3b81e404a5e301f5a : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411b3d591fe74f06a628b577c6ae9bab : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5240c171c91457da9e256ced258345c : SUCCESS in 55m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b59b3bd890494413952ca27d30fb6fae : SUCCESS in 1h 27m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc291664487f4555b30720029148147c : SUCCESS in 1h 11m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ced231b40479488ea883144e7e1b8e69 : SUCCESS in 1h 08m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/530949e70cb747e3a206dc8e780fa3c5 : SUCCESS in 1h 00m 50s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/82cfedae6e634dd2892f94c54b6c5d7f : SUCCESS in 1h 25m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6fe8e98ac348413a9b940211fa04b9ba : SUCCESS in 1h 11m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27a5c1891636468284107cd96ba54ad4 : SUCCESS in 1h 06m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9083e7c9c47a426b85f1628fcdfaa366 : SUCCESS in 1h 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43c3c6301d481dd532b1ea8ed610c897af2dfe4e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-02-16 09:54:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"548f25240795f0aa82b676c74eaa3ca79ebf4c46","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-17 16:36:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3046046164f089747da7de84d4089ec70f0570a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:38:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7a877832fafdaefb4c67c0b9feb34477869a428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:53:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c8c7b07656b403c85b16fcb24108af3\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/7606196aae06467e8f066d6fec35bc4d : SUCCESS in 58m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047721285bb74138a0b3664ea19cd014 : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c340423b79034ef4bfcd194f0cf96e55 : SUCCESS in 27m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff5bea6b2bb43bd8951408580220eca : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558180660a824e748e635843227e278c : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d50eb4b5b6c43769ee896e2a582e90f : SUCCESS in 44m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92cfcb6286f24b8cad04d976e499a7a7 : SUCCESS in 1h 25m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8ea926f3ab394a789d8832b061fec190 : SUCCESS in 1h 13m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9739068ca9fb48d1b68620a19bfacecb : SUCCESS in 55m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e06737d93a824543a57f54f1b9b89f8a : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b01dadde05f1729cff389b4679a2c5f1bfea51e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:53:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"afa312e4e7724c81457990a02619b1749b79ef2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:55:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/384e686a936d466b84a87c05e85082b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00eb4a49ca1045358352cfc8efd013c6 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bbe4390410af36859a0d1c871d20a5777f1a134","revisions":{"5bbe4390410af36859a0d1c871d20a5777f1a134":{"kind":"REWORK","_number":1,"created":"2023-02-13 17:08:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/26/873626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/873626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/873626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/873626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/873626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/873626/1"}}},"commit":{"parents":[{"commit":"7d12e285c009d2ca5a5a97d0096468081be68cb3","subject":"[OVN] Allow logging all traffic related to an ACL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d12e285c009d2ca5a5a97d0096468081be68cb3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-08 12:14:19.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-13 17:08:10.000000000","tz":60},"subject":"Improve scheduling L3/DHCP agents, missing lower binding indexes","message":"Improve scheduling L3/DHCP agents, missing lower binding indexes\n\nThis patch is covering an edge case that could happen when the number\nof DHCP agents (\"dhcp_agents_per_network\") or L3 agents\n(\"max_l3_agents_per_router\") has been reduced and there are more agents\nassigned than the current number. If the user removes any agent\nassignation from a L3 router or a DHCP agent, it is possible to remove\nfirst the lower binding assigned registers.\n\nNow the method ``get_vacant_binding_index`` calculates the number of\nagents bound and the number required. If a new one is needed, the\nmethod returns first the lower binding indexes not used.\n\nCloses-Bug: #2006496\n\nConflicts:\n    neutron/common/_constants.py\n    neutron/objects/l3agent.py\n\nChange-Id: I25145c088ffdca47acfcb7add02b1a4a615e4612\n(cherry picked from commit 5250598c804a38c55ff78cfb457b73d1b3cd7e07)\n(cherry picked from commit 0920f17f476ce8b398deea3e54e9f90b5251cfc9)\n(cherry picked from commit 7dcf8be112ed205a6c694c1f3549e08b4234d82d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bbe4390410af36859a0d1c871d20a5777f1a134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bbe4390410af36859a0d1c871d20a5777f1a134"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
