)]}'
{"id":"openstack%2Fneutron~874112","triplet_id":"openstack%2Fneutron~master~Ie2f5ff169bb3007694b9858465125775cfee40f4","project":"openstack/neutron","branch":"master","topic":"bug/2007166","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-02-20 08:48:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-02-20 12:37:45.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-20 12:37:45.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-20 08:59:44.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Ie2f5ff169bb3007694b9858465125775cfee40f4","subject":"[ovn][ipv6] Skip test_update_delete_extra_route","status":"MERGED","created":"2023-02-16 15:50:09.000000000","updated":"2023-02-20 12:41:05.000000000","submitted":"2023-02-20 12:37:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874112-bug/2007166","meta_rev_id":"d0d2a10e9201cc74a69bd3709bbbfa2ff0d620cb","_number":874112,"virtual_id_number":874112,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 12:37:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-02-20 08:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-20 08:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-02-20 08:59:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-02-20 08:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 20:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 08:30:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-20 08:30:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-02-20 08:30:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"008892538a15bb2d25d7e5203677427626ce5432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-16 15:50:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6337aba9f3f582901e3b498857002ae0c3ae285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:58:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52441e73d8814b19937bb34deec86a33\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/ab1aa7f81b964ebdbc636bfc5e1c60fd : FAILURE in 1h 09m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/47e3bdb6253441c5894cefb09f1dbc0b : FAILURE in 1h 22m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/9e948f3afc4340269057876735ab2955 : FAILURE in 1h 01m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/4772e792464e40e3b728e88f9564dc21 : FAILURE in 1h 02m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/cecd28e8aa2a4224bb5fa255c97acbcb : FAILURE in 55m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/082b4e8566544b6cb03ab32fe31532bc : SUCCESS in 1h 35m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/18432d796b7e4ab786ccea0e2b435fe4 : FAILURE in 1h 16m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/29bfd929a7794d88b5cede188e2640b2 : FAILURE in 1h 21m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/4719af29c3fb413d96773b5e05d20f61 : FAILURE in 1h 17m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/c14eb378dbc14406baa5c146b6e0d204 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9729c96d35ae45047576729e45a4e6a151c74f92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-17 04:11:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"036fa5ab4f0673672ddd07c3834f0d0e50568466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:12:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2028fbe8ffb544fb9f9e2b5b24cb6789\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/0e21fa67d84043a4948170992f8758e1 : FAILURE in 1h 52m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/53c51bb842c4485c9fd7d2654b12ce61 : SUCCESS in 1h 13m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/2b9a284e78a34060a859db8081cecaf2 : SUCCESS in 1h 08m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/85f9d230bbe64986aff13080f173632b : SUCCESS in 53m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/7821064ab1e640458b72adf00faec6af : SUCCESS in 1h 12m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/0801e5cfc4654ee4afa6975dc739168b : SUCCESS in 1h 10m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/252d3c1f91714a90a3271284529a5ab3 : SUCCESS in 59m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/3420b8faffec446eacdef0c25575fd80 : SUCCESS in 1h 29m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/ad25ac0945c6471988f1325856b9e0e4 : SUCCESS in 55m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/f2502b3ea35541cdb88c195854adeeda : SUCCESS in 1h 12m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/f703d869e56a420d827acb6faf9f2978 : SUCCESS in 1h 06m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/adc813407315434f9c6c611f8f3fbe16 : FAILURE in 50m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/cf64ccfb446f44d2932a36d6bb2e40be : SUCCESS in 1h 02m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/6842558ba0284f549fd6013c648e8303 : SUCCESS in 1h 04m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/3e581185ea0243e7af53976aa6c80bea : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"178fdd702a65c7d2ce2953fcfb9667d339129b2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-17 07:15:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"254d8d420e486036cb46174f4276a3cadbb1250a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:30:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64277b55e251457ebbf8dc9b3f6d2cf4\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/3feea7d026094fc383a983cef5a703ad : SUCCESS in 1h 16m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/cc46cf9560e34e5b8d3ab04d65c1a132 : SUCCESS in 54m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/13e58c3513e2453685c42764884999d6 : FAILURE in 1h 22m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/323c9ac9c9eb4c8aa561bea5f5f3da3b : FAILURE in 1h 23m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/27a303811442474eb7ac66525eb46d4e : FAILURE in 1h 16m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/664efe031974481485fde3ffc1daee8c : SUCCESS in 1h 26m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/f4153a0b9f8c40628ef401a2afe3f40b : SUCCESS in 1h 21m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/723a7f868ad44a95b7f31ed4f427892b : SUCCESS in 1h 19m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/e7e3092e28624be58ce5220fefa6e92c : SUCCESS in 1h 28m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/30bf0c9008234a1aa457691e608cb150 : SUCCESS in 2h 07m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/a901afd20c854085b1595355d11e3354 : SUCCESS in 1h 38m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/5eb6927d556c416bb65806fb955967a1 : SUCCESS in 1h 23m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/958723fa31ad413a8268d6afa562153c : SUCCESS in 1h 13m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/8223cccc85fd4245ae12423d1156f140 : SUCCESS in 1h 20m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/0fafae9c1ad646e4a7c75682166c0e57 : SUCCESS in 1h 26m 47s","accounts_in_message":[],"_revision_number":3},{"id":"56e821d8bcfb5d495d300a8943aaadf267ab99f9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-17 09:36:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82fa6cb36a82c9bf3ffeae0b847363bdba406a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:30:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7baf233c64e843e1bf3eb4d100f29e19\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/91970ef657044bacb80dfb921d9d18f8 : FAILURE in 55m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/05d5536ae0fd4e66837b9a8a4bf5282f : SUCCESS in 1h 06m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/5b5517744fce40c4a68b08a8696d86a6 : SUCCESS in 1h 12m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/7a82205f1bfe4f29ab7a4f3c02e59806 : SUCCESS in 1h 17m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/ece596938bc54c20b766da05ee8cc27e : SUCCESS in 1h 02m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/3e4fd666621f4816b57010081a4d18d5 : SUCCESS in 1h 09m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/a20101cd4c524e16b013afc74a654355 : SUCCESS in 1h 12m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/88b2be7bb23a4c51a55019aa3c512eaa : SUCCESS in 1h 05m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/b694cadb226040ecaa7cf6fd9c9fc2b2 : SUCCESS in 1h 15m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/6d64417c6ee54b0d88cba50546de22f5 : FAILURE in 53m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/f28528e037b340eca73d3fb001522356 : SUCCESS in 1h 08m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/f9c391b3e04f444cbc9d8f7efb07042e : SUCCESS in 1h 21m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/ded89c988b17491489aed916396362aa : SUCCESS in 1h 46m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/57e8310605434e0c97e53031b7cf9706 : SUCCESS in 1h 00m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/913ed1b3e0744b2fae22c5a22bc839ec : SUCCESS in 1h 10m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f0b7714e77442d079165d929da71552899b93d11","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-17 11:53:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c66d8b0e4c10aede35be62f3b1226aca2cc00d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:39:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d17965b1fc4f4677bbbd85e6c2c231d2\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/baf55acdc72147a98e1a9d2dd4bba892 : FAILURE in 1h 18m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/8c09ac87c53344d0b064d57b72f48f2a : FAILURE in 1h 00m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/da0f6ccdb4cf45e8badf933c350cdd31 : SUCCESS in 56m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/101b05a72cf54ee5ac740461004cb5cd : SUCCESS in 1h 11m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/c6202e547c4e40838430b921c292860e : SUCCESS in 1h 11m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/401f794d87f34b808ea04bdbba879fae : SUCCESS in 1h 14m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/cb1eb7652c884c1dab40199ce531d7ad : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/3538fab4c7d34b04bbb89f8ce98b994b : SUCCESS in 1h 17m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/d6f5c17055dd450fb467a4413a1c05e2 : SUCCESS in 1h 32m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/73b60e2ec21546f1bf2d9243215935b9 : SUCCESS in 1h 22m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/68719c2e893642cb85798d1977d07962 : SUCCESS in 1h 23m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/21004f049800467ba079f63c6ef04b9f : SUCCESS in 1h 13m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/42bc8c0280224636badea9d57a206c0a : SUCCESS in 1h 38m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/b13cc390e234416389c36420179b1222 : SUCCESS in 1h 00m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/4dcb09dd43754484a1345ae71d953850 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7132bcbd18877c582c02261a06c0ab3c82e9c75a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-17 15:50:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"99815a89175087384b7977f8b489a417a25d9a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 17:39:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cb9bbf64d514c2e8ad5c84e9f660e2a\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/b41442436c4148f08d554648eaed894d : SUCCESS in 1h 13m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/074d93094e53458b8df89c8c9a9a96b9 : FAILURE in 1h 06m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/f06d36d3952c4548868ff0ee8c3cc94e : SUCCESS in 1h 29m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/5c5ce7480cd8447dac95de369e6fa71d : SUCCESS in 1h 18m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/bd7fe92afefe496baf9c7075ceb17b1b : SUCCESS in 1h 23m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/263b80e02f904cedad458b2d0e023bb9 : SUCCESS in 1h 20m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/69346010e0714ac8b7af711a2cb238f8 : FAILURE in 1h 24m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/a750c2d78afb42198edbd42b8afadac3 : SUCCESS in 1h 19m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/39232e0034bb4095a1efdac26fdd3c65 : SUCCESS in 1h 00m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/4706395bb53641eb855c3951114d5d87 : SUCCESS in 1h 15m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/a990b429532244619d21fa5786645603 : FAILURE in 1h 11m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/c53edaf9003c455f99ddeffa6c6dfae3 : SUCCESS in 1h 08m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/7ae9552fc1174ead8ba51a8091ccbdd3 : FAILURE in 1h 17m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/16f503bec4094037a1edec7323ca5797 : SUCCESS in 1h 06m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/94998f99256c4585af94ff670e819fd0 : FAILURE in 1h 17m 10s","accounts_in_message":[],"_revision_number":4},{"id":"06389be95ed559141396ac5a4ad87364c0795741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-19 06:24:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"14bfee09bf947d8112496bb14f7914cb40bb3939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-19 07:54:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a1f0354018c477eb7ea5b236e3faa6c\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/41a50b34664940cbad17702a578fe1bb : SUCCESS in 1h 18m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/ea815a24a42a465892923660e84ada7b : FAILURE in 1h 17m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/b7faf08357554072a747cae83c2c9c7c : SUCCESS in 1h 10m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/bd8395f19b73473da416001d1adb9075 : SUCCESS in 1h 08m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/eefe5d681baa4cfc94deedaba8b45af3 : SUCCESS in 1h 08m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/c8daeee2b76f4005b1be382f06629b28 : FAILURE in 1h 15m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/19bed450d33a43108d0abfcbb872c682 : FAILURE in 1h 12m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/52683c12078c4cea9212c335d37a835d : FAILURE in 1h 11m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/e6abdad4db524730a9001e8fc29da82f : SUCCESS in 1h 14m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/6c92a1a256774d35826b333a1b27a9d3 : FAILURE in 1h 20m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/eb06859756aa40c29918bb1eb4231b75 : FAILURE in 1h 16m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/fdfa103e136941159df7a48632a41c52 : SUCCESS in 1h 14m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/b13e96f57d614b158d888eb0df712f57 : FAILURE in 1h 11m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/0a8ca4e53e244064901c4f0b78bdb5e1 : SUCCESS in 1h 10m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/7280f8fd50e24aeabe63b678d344bb0b : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":5},{"id":"c54d84fe29468d1243a946d7a777bd3b82c87c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-19 12:28:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ad5074bc49750fadfcd5813cbdc7d07cb817d9d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-19 15:00:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027ec8d0fbcf46ffbc8d5c64ffb4c2b8\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/b605537b519c4d10bd84dfd95cd8f8c7 : SUCCESS in 1h 17m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/b28ea719ead14103a8cea73179da7428 : SUCCESS in 1h 18m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/127c272c4e53447cb2f2b81f218d43e1 : SUCCESS in 1h 26m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/1a58a7e37eed4bb7ad3821bf87b34713 : SUCCESS in 1h 07m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/b2b680f96c9b4ac885eba126fbf75786 : SUCCESS in 1h 11m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-1 https://zuul.opendev.org/t/openstack/build/ac6478e04637425994d864dd5ebc64cf : SUCCESS in 54m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-2 https://zuul.opendev.org/t/openstack/build/00f026c08927444395a069a7bf872bcf : SUCCESS in 1h 13m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-3 https://zuul.opendev.org/t/openstack/build/9555e1fabdec44dd82e5eb1afa7501a4 : SUCCESS in 55m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-4 https://zuul.opendev.org/t/openstack/build/2885f77f192d4421860e824f3b8e5d15 : SUCCESS in 1h 09m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix-5 https://zuul.opendev.org/t/openstack/build/761a41ff935b475a83c3814486741bac : SUCCESS in 51m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-1 https://zuul.opendev.org/t/openstack/build/edc14d7a6b6b4aea978e33b82875281a : SUCCESS in 2h 24m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-2 https://zuul.opendev.org/t/openstack/build/8e0bc65b0a99476b8ab3adc4919af7c0 : SUCCESS in 56m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-3 https://zuul.opendev.org/t/openstack/build/79c2fb0bffbb4dee83ed277887012140 : SUCCESS in 52m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-4 https://zuul.opendev.org/t/openstack/build/e5a18d6ccfb64828b15e0d6f234799fb : SUCCESS in 1h 10m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release-fix2-5 https://zuul.opendev.org/t/openstack/build/0966578c1a294f5cb91afcab21a30a37 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":6},{"id":"8efd3cacc5844d7cc4fab177fcb63b444a6e8a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-20 05:49:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"aad4229ff14bd47e349e73c4a70d69a4b35c6cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 06:55:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba2384e007f401dbc2c68b917859fa1\n\n- neutron-ovn-tempest-ipv6-only-ovs-release-1 https://zuul.opendev.org/t/openstack/build/63ca2c56d90b432c8d85af90070b6673 : SUCCESS in 56m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release-2 https://zuul.opendev.org/t/openstack/build/b1e61da89153431190b25b4379e24613 : SUCCESS in 57m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release-3 https://zuul.opendev.org/t/openstack/build/555169a86e90441b969ea019c1a1911a : SUCCESS in 1h 00m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release-4 https://zuul.opendev.org/t/openstack/build/0834e2703d7744f782ba00275c5b5e86 : SUCCESS in 54m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release-5 https://zuul.opendev.org/t/openstack/build/c0ae18c13c1f477caf49c91cd7e14e33 : SUCCESS in 55m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release-6 https://zuul.opendev.org/t/openstack/build/e85495662b4f4556a86d50b16a085900 : SUCCESS in 53m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release-7 https://zuul.opendev.org/t/openstack/build/fe0341509167412c8def2b0e3fa40554 : SUCCESS in 1h 01m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release-8 https://zuul.opendev.org/t/openstack/build/f1877699e0964e888ce88f1ee61b1cfc : SUCCESS in 1h 03m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release-9 https://zuul.opendev.org/t/openstack/build/8e432a7cc65d4b2d98257ec095c6d0c7 : SUCCESS in 56m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release-10 https://zuul.opendev.org/t/openstack/build/b76cd0b417414ec5bf8254e53f040619 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":7},{"id":"7fcd529b09ff2aab4362474b8d4fcb5330ae066f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-20 07:14:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b1cc03111127e985313263d2beac2cc7eb60ecc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 08:17:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b7b9458d66f48249725671c21e71010\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee342b5e12c44a00be64d667c74aa8eb : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77d4be721fd4d82b30a1f49f0a0830e : SUCCESS in 11m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1418b3c076f141178d72e28f0dab02a4 : SUCCESS in 21m 51s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ce7b3a9c5b94e01b40cf3bd6463ad8f : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":8},{"id":"9305456a6f294794e716c934181401a014ddbbdd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-20 08:48:24.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":8},{"id":"a875d3dbf605abd97efe473d3989df2c6000707f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-20 08:59:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1df5e1589fe7b2d06979a95625577193b0bbba8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 09:00:26.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f72add87a9ffd814c0507acc839f7cffe8f0255b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:37:45.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd41e4a83664ec8822499516a3e0bff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1fbfb6f302a4953b54fc7c5e4498b09 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422ef4b7daa841e1a7fa0d1f77ce5aa6 : SUCCESS in 10m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e15c09015cc3411fb539c9e21bbed456 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":8},{"id":"d27087a102672d3c467482cde32d234470cd2ddd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:37:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"d0d2a10e9201cc74a69bd3709bbbfa2ff0d620cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:41:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3feda17f4445bc9a9eff68d9b1d85d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3cae8f6bd54909a659aa4c22cda769 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f991fc2fbef6b69b97d373e7acff8038b2da3028","revisions":{"c9540fc088a0122c7b4da03089b189b07eec99ea":{"kind":"REWORK","_number":1,"created":"2023-02-16 15:50:09.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/1"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:28.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9540fc088a0122c7b4da03089b189b07eec99ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9540fc088a0122c7b4da03089b189b07eec99ea"}]},"branch":"refs/heads/master"},"1bfd593ec0b0486677b56cc0fabf2e8e3183b126":{"kind":"REWORK","_number":2,"created":"2023-02-17 04:11:40.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/2"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-17 04:11:48.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bfd593ec0b0486677b56cc0fabf2e8e3183b126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bfd593ec0b0486677b56cc0fabf2e8e3183b126"}]},"branch":"refs/heads/master"},"4e495e6faea5e9c0c2745f75eaa5fa0a9f1dfa85":{"kind":"REWORK","_number":3,"created":"2023-02-17 07:15:59.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/3"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-17 07:14:16.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nStatus:-\nGood:- 22.06.1+\nBad:- 22.03.0, 22.03.2\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e495e6faea5e9c0c2745f75eaa5fa0a9f1dfa85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e495e6faea5e9c0c2745f75eaa5fa0a9f1dfa85"}]},"branch":"refs/heads/master"},"8e44925d8f9992a822dbef31e44024cb040dcb25":{"kind":"REWORK","_number":4,"created":"2023-02-17 15:50:56.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/4"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-17 15:48:10.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nStatus:-\nGood:- 22.06.1+\nBad:- 22.03.0, 22.03.2, branch-22.03\n\nSuspecious test:- RoutersTest.test_update_delete_extra_route\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e44925d8f9992a822dbef31e44024cb040dcb25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e44925d8f9992a822dbef31e44024cb040dcb25"}]},"branch":"refs/heads/master"},"6c4fad7f278fc1333dabf0be5db25ffc9c138e3e":{"kind":"REWORK","_number":5,"created":"2023-02-19 06:24:43.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/5"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-19 06:24:57.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nStatus:-\nGood:- 22.06.1+\nBad:- 22.03.0, 22.03.2, branch-22.03\n\nSuspecious test:- RoutersTest.test_update_delete_extra_route\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c4fad7f278fc1333dabf0be5db25ffc9c138e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c4fad7f278fc1333dabf0be5db25ffc9c138e3e"}]},"branch":"refs/heads/master"},"ce2a1723fb6012cb9a5a58f2314a7c2e11ad134f":{"kind":"REWORK","_number":6,"created":"2023-02-19 12:28:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/6"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-19 12:28:47.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nStatus:-\nGood:- 22.06.1+\nBad:- 22.03.0, 22.03.2, branch-22.03\n\nSuspecious test:- RoutersTest.test_update_delete_extra_route\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce2a1723fb6012cb9a5a58f2314a7c2e11ad134f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce2a1723fb6012cb9a5a58f2314a7c2e11ad134f"}]},"branch":"refs/heads/master"},"7dfd453742a60c4554b96928a3662a703e6c453f":{"kind":"REWORK","_number":7,"created":"2023-02-20 05:49:08.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/7"}}},"commit":{"parents":[{"commit":"502c0253de6b47cc096998b10dc88b679c7fe878","subject":"[ovn-ipv6] Temporary exclude ``TestNetworkBasicOps`` class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502c0253de6b47cc096998b10dc88b679c7fe878"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-20 05:43:51.000000000","tz":330},"subject":"[DNM] Check issue and fix with ovn source installs","message":"[DNM] Check issue and fix with ovn source installs\n\nStatus:-\nGood:- 22.06.0+(with commit[1])\nBad:- 22.03.0, 22.03.2, branch-22.03\n\nThe issue is random and when it reproduces, any test involving localport\nran after test_update_delete_extra_route fails, and all\nfollowing test passes.\n\nSo until issue\u0027s fix is available in branch-22.03 we\ncan skip this test.\n\n[1] https://github.com/ovn-org/ovn/commit/eaf9832be2\n\nRelated-Bug: #2007565\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfd453742a60c4554b96928a3662a703e6c453f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfd453742a60c4554b96928a3662a703e6c453f"}]},"branch":"refs/heads/master"},"f991fc2fbef6b69b97d373e7acff8038b2da3028":{"kind":"REWORK","_number":8,"created":"2023-02-20 07:14:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/12/874112/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/874112/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/874112/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/874112/8"}}},"commit":{"parents":[{"commit":"7695f48f1376f2ee95486f599eadce0c18a109da","subject":"Merge \"Do not log full network on reload_allocations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7695f48f1376f2ee95486f599eadce0c18a109da"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-16 15:49:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-20 07:14:55.000000000","tz":330},"subject":"[ovn][ipv6] Skip test_update_delete_extra_route","message":"[ovn][ipv6] Skip test_update_delete_extra_route\n\nThe issue is random and when it reproduces, any test\ninvolving localport(metadata access) ran just after\ntest_update_delete_extra_route fails as metadata\nflows are missing.\n\nThe issue do not reproduce in ovn-22.06+(since [1]).\n\nSo until issue\u0027s fix is available in Ubuntu jammy,\nwe can skip this test.\n\n[1] https://github.com/ovn-org/ovn/commit/eaf9832be2\n\nRelated-Bug: #2007565\nRelated-Bug: #2007166\nChange-Id: Ie2f5ff169bb3007694b9858465125775cfee40f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f991fc2fbef6b69b97d373e7acff8038b2da3028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f991fc2fbef6b69b97d373e7acff8038b2da3028"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[]}
