)]}'
{"id":"openstack%2Fneutron~874130","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ia4a68bdccecfbcb9d1aa49e2b14e06d139891c0f","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2003553","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-02-20 12:37:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4a68bdccecfbcb9d1aa49e2b14e06d139891c0f","subject":"Do not ignore attributes in bulk port create","status":"MERGED","created":"2023-02-16 15:59:14.000000000","updated":"2023-02-20 12:39:47.000000000","submitted":"2023-02-20 12:37:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874130-bug/2003553","cherry_pick_of_change":872880,"cherry_pick_of_patch_set":1,"meta_rev_id":"8930845883f4baee6db3679e94da1e232d3380a1","_number":874130,"virtual_id_number":874130,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 12:37:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-02-17 11:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-20 08:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-02-20 08:54:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 15:59:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"CC"},{"updated":"2023-02-16 16:22:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-16 21:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 11:29:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-02-20 08:54:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"6cb626d01b184864097325269a04e518d44c2b5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-02-16 15:59:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"3a9a856949afa6eb88bccccddbe514c68ed84cd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:22:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c61fe652db474cbc8281c1e1884adf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/130f81cf51d3475a80c71384177910ed : FAILURE in 19m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3592db3aebb04d92ae1e6a746bf813cd : FAILURE in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"037d1debc3000769c4c39821c2021d92f9cbb938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:19:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/921eb3d5521140e09a86ab007d633f49\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/df57b2d44a5a4307a374045e58258989 : SUCCESS in 56m 04s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/0fec1584182c4ae8bc4de9be0b0d4fdb : SUCCESS in 1h 14m 28s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/753a1078100f49888fcd927dddb3d341 : SUCCESS in 56m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/1a3b8688c1084c3187aa65e44d5ce89a : SUCCESS in 47m 19s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/62472bcdd1c84838a106fdc3fb97b399 : SUCCESS in 58m 25s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/56fcc577ce32401094d0d21910e32e46 : SUCCESS in 25m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cff980cecbb4c47a3c918a0296f4a35 : FAILURE in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/605c39138aca4669af675c85eb50e8d1 : SUCCESS in 13m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcaef457b1b44cfd823e467680592493 : FAILURE in 26m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90538cea399c441c8377f24d45cd5ba8 : FAILURE in 26m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480adbe4dbef421fb51c968347dc741a : FAILURE in 24m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bff19432eae41d59b3d858aa449b9b9 : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59004e82b13d4418ae44bea8f6cd731a : FAILURE in 43m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea753724a6794942a7d77ffff6716614 : FAILURE in 1h 56m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/971579e45d31404ebd0b19915cde0108 : SUCCESS in 1h 05m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4d2fb2f8c094505b43d5f5bca45ac9c : SUCCESS in 1h 00m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/98407db184cf40b7a9afe0a35b3c4773 : SUCCESS in 1h 10m 41s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6e851f93189c4e7b9c0459dca43cff0d : SUCCESS in 1h 21m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/893d588d01b34547ba967c7e750b94f3 : FAILURE in 41m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cfd05ec853f41a89abe91fc7447c75a : SUCCESS in 55m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65792a168f3c48de8fbe986ae79da091 : SUCCESS in 1h 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b103353e06f4236ab1fdf66c82b860df15c8f805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-02-17 09:18:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"778f6d900ee5b74b09e2ea0a579094d8f9c641bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:16:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb5c68643694493a90e0ee771facd75b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c96b0b9e124424190ce563272cd91bf : SUCCESS in 48m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58a5800a791d4a8a87d31b5221f88163 : SUCCESS in 47m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b52a381fe93387d15251818c6eb56f75b3e77738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:55:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f1d0200381f43da855a87d1aaec3a73\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/b96c0b102234411ca249182f5e6ea7c7 : SUCCESS in 56m 57s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/9226cc8c3c4b4efe8ece01ed128b4519 : SUCCESS in 1h 20m 51s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/36e663e4ddc144b7923470ea5246386a : SUCCESS in 1h 32m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/41159cbdb16f4a9fa9fb2a0e6e6ec6bc : SUCCESS in 1h 15m 55s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/41a0be5c5f4e4e278df38398303b567e : SUCCESS in 1h 26m 58s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/afa5e745171e47f39472727c28199af5 : SUCCESS in 28m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a3dcbe49823437fa4c8984f9079b3aa : SUCCESS in 35m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b694789f1e4b49c2a6cf9dc14f3ca3ae : SUCCESS in 17m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60a80e57feb7405e929a88056f9a77a2 : SUCCESS in 33m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad0c1917ed1a4fca8646fe55cab52f27 : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d590795156e14b2fb22bd99092d23418 : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766f06c1ded440efb7a3a7d8dc11d1e3 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbd58eb41a2746d0b1a8341d76c13089 : SUCCESS in 58m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ef140b34ac344f8aac0af26b330b56f : SUCCESS in 1h 21m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a736ac16dbf94fb299bd6d1a46ee1117 : SUCCESS in 1h 19m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/15d2dcd9da8e4bf498eeb531e0561c64 : SUCCESS in 1h 00m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/efcc685744c3406fb4dffb669b2cf8d1 : SUCCESS in 1h 05m 21s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1771b8b185be4de4b6ad45852d602d76 : FAILURE in 57m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f589e3981eaf4e2fa0b717cba2c751d0 : SUCCESS in 1h 13m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b799b40d3a747f499a0086ebeaa6744 : SUCCESS in 54m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/031bd1f4681349f8b05dd0aedebed7e2 : SUCCESS in 1h 03m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad1f0e04762df56fe951fc9c4e2b52df5d718f82","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-17 11:29:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8aaf7335b334d4406710f33cf3d511d197c07a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:12:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bd4e7b15c554a249ebf534705822832\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a7dbaac5bd742d9a334a0a89260a2e2 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ec1afd661b94cafa5415688f22870c4 : SUCCESS in 40m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"61b1d0d8fd840de04df86b125ecef587e21bd462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:16:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af3f78dc94a34ce8809fc68ab800fc71\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/9192fcd6de9f44148d2212c5b093f2d2 : SUCCESS in 43m 28s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/990fa23b40d747fdadf206b7dc774be5 : SUCCESS in 1h 11m 48s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/a4730b45c5034e01bb41df9dbc264a70 : SUCCESS in 1h 11m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/a40ab1f8555749da82d2a08d2a54f419 : SUCCESS in 1h 20m 00s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/3aeec9728405442ba9cbd023b874ef2b : SUCCESS in 1h 07m 41s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/8ed887375b1a4cc19a5df9618b33c47e : SUCCESS in 36m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f2a3b03abe3442998850dec488e666a : SUCCESS in 38m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9709d4e8eaea4dc782ea7d606e73b2e2 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ecfbb9cbf0e42e5a5e61be86a8ab8b8 : SUCCESS in 24m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a81ceef5f3f84185b131ece7540ad33d : SUCCESS in 24m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94ad6bc15e174175b71e0faedb7ad2c5 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3077edff7d954fb2909b913eab24ae34 : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09ed29c379554bb9980f66a0174b755d : SUCCESS in 40m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ff96ae1363043ef8556bd4b09055621 : SUCCESS in 1h 20m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bdf57c0ee9844338b798404d82216017 : SUCCESS in 1h 10m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a13222b9ea7b49d29ecbb9f989bbb91a : SUCCESS in 1h 04m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/544b525f7d994ac592992005eda9273d : SUCCESS in 58m 26s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/675bf1d1387c40e68eb17e62b8a68ef4 : SUCCESS in 1h 07m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d9bbf4e6f26443c8688e5827f66c314 : SUCCESS in 1h 18m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/844a9f60802f47039e6b89154dcfcb58 : SUCCESS in 1h 00m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0682539917a844bda292ec6ea3bf7c8e : SUCCESS in 1h 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef3ca3467bf8748fc51db3e39f995cf32558e898","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-20 08:54:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"52888c8cb9ce319e24aad082ac66cd6fe6ba805a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 08:55:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"613758ed26d29e2980deb10dfc3fb0991a83275d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:37:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5242c118cdc4c338d15949dccb68fe6\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/916a55768b264cc7a805f1ea021b022d : SUCCESS in 1h 08m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727fba2e4f644a1caca1cc08cf18e72d : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ae2550f69cf4bb4a7ffffa5b6e99e26 : SUCCESS in 25m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faff4b35f89a48e5a7da3b9ec9bc9574 : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6d534f45154db58a3e833f6f47505c : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/adf22cb28bcc424cb7230a1200312a83 : SUCCESS in 1h 00m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/da69e7afc1264734881be30279888375 : SUCCESS in 2h 04m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc4838588e794e718883edec2c664328 : SUCCESS in 1h 22m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb1b88eb5d1941c2a4043c30a06fa36b : SUCCESS in 52m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca92a99357924160ad49ffce05498eb6 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"02169edd9e3d65d23240f58d082a45f3a3daf48b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:37:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8930845883f4baee6db3679e94da1e232d3380a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:39:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb85c36fdce40baa075b16a6df9e978\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1facb9f73f2f4693a78d9d6f2c182554 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"719ef724f5c732f23f17b9bcf558be9a2fe0e53c","revisions":{"708d06cdf857416bb22467df276da265f66d175e":{"kind":"REWORK","_number":1,"created":"2023-02-16 15:59:14.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/30/874130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/874130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/874130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/874130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/874130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/874130/1"}}},"commit":{"parents":[{"commit":"288daf9974d6776a27f3539da2f8a168831cf144","subject":"Fix deprecation warnings in pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/288daf9974d6776a27f3539da2f8a168831cf144"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-01-20 15:26:21.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2023-02-16 15:59:14.000000000","tz":0},"subject":"Do not ignore attributes in bulk port create","message":"Do not ignore attributes in bulk port create\n\nWith unit tests that would have caught the bug.\n\nChange-Id: Ia4a68bdccecfbcb9d1aa49e2b14e06d139891c0f\nCloses-Bug: #2003553\n(cherry picked from commit ed68ba4a4cb622a67a4b28065da7f1ef7777e0dd)\n(cherry picked from commit 1b086f1c1fd0879224e61b05e3d4bb54486ea62c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/708d06cdf857416bb22467df276da265f66d175e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/708d06cdf857416bb22467df276da265f66d175e"}]},"branch":"refs/heads/stable/xena"},"719ef724f5c732f23f17b9bcf558be9a2fe0e53c":{"kind":"REWORK","_number":2,"created":"2023-02-17 09:18:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/30/874130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/874130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/874130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/874130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/874130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/874130/2"}}},"commit":{"parents":[{"commit":"288daf9974d6776a27f3539da2f8a168831cf144","subject":"Fix deprecation warnings in pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/288daf9974d6776a27f3539da2f8a168831cf144"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-01-20 15:26:21.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2023-02-17 09:16:57.000000000","tz":60},"subject":"Do not ignore attributes in bulk port create","message":"Do not ignore attributes in bulk port create\n\nWith unit tests that would have caught the bug.\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/test_plugin.py\n\nChange-Id: Ia4a68bdccecfbcb9d1aa49e2b14e06d139891c0f\nCloses-Bug: #2003553\n(cherry picked from commit ed68ba4a4cb622a67a4b28065da7f1ef7777e0dd)\n(cherry picked from commit 1b086f1c1fd0879224e61b05e3d4bb54486ea62c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/719ef724f5c732f23f17b9bcf558be9a2fe0e53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/719ef724f5c732f23f17b9bcf558be9a2fe0e53c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
