)]}'
{"id":"openstack%2Fneutron~874776","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I5470f61a5e0b55bf682526208c3f57dc0ca6ffd5","project":"openstack/neutron","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2023-03-01 21:22:41.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-03-01 21:22:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5470f61a5e0b55bf682526208c3f57dc0ca6ffd5","subject":"ovn migration: Turn validations off by default","status":"MERGED","created":"2023-02-22 14:13:03.000000000","updated":"2023-03-01 21:23:55.000000000","submitted":"2023-03-01 21:22:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874776","meta_rev_id":"d211e012c2bbbcb860a010d197f9511e85ec8d7f","_number":874776,"virtual_id_number":874776,"owner":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 21:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-02-28 14:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-03-01 14:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-03-01 14:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 14:13:03.000000000","updated_by":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-02-22 15:26:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 14:54:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-01 14:58:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"e636bbaa60ed2cf5cce80808dccaa17f39d7a3bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2023-02-22 14:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"836ebd58637941ecc1664aea7e9a4f09cb32e3c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 15:26:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1c1145bd674d91aae73e63be656a3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2061f07974ee4912ab14feea9adb7cfc : SUCCESS in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0db525edc644c358c4347ac7f7ab12a : SUCCESS in 30m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b37d61592046d98bf1e92e8fce783c : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41b87384e9234eb4ae15c4fa4440a67c : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"605aaf0dcdece1556be71eaff481e78d5ea7b2bb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-02-28 14:54:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb82df8700481dab11f755f4c7a9a47490084a7b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-01 14:58:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22900d5c7182f0abc8c54bb641f1040e5ceda8d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:59:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61205bf32e8372dfb72cbf48fc395dd9972fe509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 21:22:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d40d83cfc104306989596015609744d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ccf7bb51cae46fab55e69af5e5a8f4b : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c25e1d347244654a6cd152e64c99c3b : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3026c4d7c2ef40dabf411a7dbcccc7cf : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e6a29076d520a6fc1034f343722cbfdadea3492c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 21:22:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d211e012c2bbbcb860a010d197f9511e85ec8d7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 21:23:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77677d5ab60e444ab967249cc864efab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74479e44396d454899010b4b08ff967d : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c86a264ae6e43819bd63b2d0c7a3ca1 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"291c85f54dd43c7fe03b33e56171626e89900edc","revisions":{"291c85f54dd43c7fe03b33e56171626e89900edc":{"kind":"REWORK","_number":1,"created":"2023-02-22 14:13:03.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/76/874776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/874776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/874776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/874776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/874776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/874776/1"}}},"commit":{"parents":[{"commit":"94c2c92f8c2310bb3c8148e5c9ef503f1a3accf0","subject":"Merge \"Improve scheduling L3/DHCP agents, missing lower binding indexes\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94c2c92f8c2310bb3c8148e5c9ef503f1a3accf0"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-03-23 17:34:14.000000000","tz":-240},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2023-02-22 14:12:43.000000000","tz":60},"subject":"ovn migration: Turn validations off by default","message":"ovn migration: Turn validations off by default\n\nThe validation is intended mostly for tests and don\u0027t make much sense\nwhen running the migration in production because likely there are\nalready running workloads. This patch changes the default to False so\nmigration validation must be explicitly asked for.\n\nChange-Id: I5470f61a5e0b55bf682526208c3f57dc0ca6ffd5\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit 0baf8841eea402b7b8505ee685b1ac1a3f346df5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/291c85f54dd43c7fe03b33e56171626e89900edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/291c85f54dd43c7fe03b33e56171626e89900edc"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
