)]}'
{"id":"openstack%2Fneutron~874797","triplet_id":"openstack%2Fneutron~master~I3fcd0458d20f20ce40378f90f073f37c41400865","project":"openstack/neutron","branch":"master","topic":"2023-aa-l3-gw-multihoming","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-08 23:40:38.000000000","reason":"Change was submitted"},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-08-08 13:18:02.000000000","reason":"\u003cGERRIT_ACCOUNT_7730\u003e replied on the change","reason_account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"}},"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2023-08-08 23:40:38.000000000","reason":"Change was submitted"},"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2023-08-08 23:40:38.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-08-04 13:18:34.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-07-31 20:33:09.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-08-08 23:40:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fcd0458d20f20ce40378f90f073f37c41400865","subject":"Add extra router attributes for ECMP and BFD","status":"MERGED","created":"2023-02-22 18:13:50.000000000","updated":"2023-08-08 23:41:59.000000000","submitted":"2023-08-08 23:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":67,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874797-2023-aa-l3-gw-multihoming","meta_rev_id":"2f12be971fb73c44d6e0fdf87eb07a5fb034fb41","_number":874797,"virtual_id_number":874797,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2023-07-25 15:58:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-08 23:40:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-08-08 13:18:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-03 16:10:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2023-08-01 03:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-08-04 13:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-07-31 20:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-08-04 13:18:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 18:52:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 19:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-22 21:19:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 09:09:53.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-05 16:01:33.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2023-06-22 17:41:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-23 06:44:08.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"CC"},{"updated":"2023-07-04 12:28:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-05 08:32:29.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-07-12 06:23:09.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2023-07-12 07:02:53.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-07-25 14:07:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-01 03:56:21.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"dea261a000d54ae256b4e3e73747988780f69ba4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-02-22 18:13:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"460bb1b1492450018002567f9ed4569259260d60","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-22 18:52:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/97/874797/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7d8c66d50ce1b4674a7cfe65a30dc5806c833d99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 19:04:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2778d601fce945c89cb836a97f27e054\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11f5ccedaadc44ad824567531e98ddf1 : FAILURE in 45m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4b5e9fbe9c04846b80cf35bc2c26348 : FAILURE in 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed5887ce8d1937ac021361b99c6372328a85b235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 21:19:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5401be570664edd996916e1901eb807\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/65a315b9a75b47529f6b8d157f0c5afc : SUCCESS in 1h 22m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2154dc667d734596bf49703890efd13f : SUCCESS in 1h 14m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f952f59fa1584eee830b6fd7bbf5bf25 : SUCCESS in 1h 09m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55fed0b50730457eac0ca8c80f8298bf : SUCCESS in 1h 15m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c1a8d81f21cf4d73b253e527e7dceeae : FAILURE in 15m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/52900ca2a9cd4f589bd2e7b5d5248641 : FAILURE in 17m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbde94bff71d4f0383f765ecb61e0465 : FAILURE in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20824564b6ac422591cf565fb42b7cb9 : SUCCESS in 12m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/679283a20937458290f115e5f15cf276 : FAILURE in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9d4582bd5714a439097579135919933 : FAILURE in 22m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c8df47e89a4069b7ce78d2efee1e9a : FAILURE in 4m 35s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/211acccce95c43e38892ce72af297bb0 : FAILURE in 27m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19b858419d5d4bc39fae46b006975334 : FAILURE in 43m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ac39dbc7d3f4fc0aa602c4d51c4b386 : SUCCESS in 1h 33m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dd188f0bade461f8e092ea76b15585f : SUCCESS in 58m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/07e9c8460f6d4c0eab615f15149f9e1d : FAILURE in 47m 43s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/14e8e09c6292471b88353a0a19f4c350 : FAILURE in 54m 37s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b301e421f5ae49c2bd18e9afaa9d0135 : SUCCESS in 44m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/47e06be0414448dfa555c7957fbe3a81 : FAILURE in 18m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d08da585572445cb8b88981318f30c4 : FAILURE in 28m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/93544cc849b847ec863f3bcb9dbfb23f : FAILURE in 17m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d58dcf6435b641bfa409268a6b9638c0 : FAILURE in 16m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/824d9c0c6ff947629dd98b0fa1d99628 : FAILURE in 17m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/128b7101388c471f822fd355309eda33 : FAILURE in 17m 43s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2bdf40f519e44a02ae87fc96b18b456b : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d85f09db9eb29db681bb7f307d5e02aeccfe2da8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-02-28 15:01:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"87e5a76dabab66f158c957c1087aca211c34e3b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:25:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cfc52d3ed61482b8c5ee3abea399f22\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5de24cb3f5994797926355543082b424 : FAILURE in 21m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86e0d8a3dfb74c2c97ffa394120b192c : FAILURE in 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3533a355aa006fe78389495744c14031d329cca8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-28 16:48:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/97/874797/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"cfc52dc7cc622b7ad213f5839288ba244365a85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 19:15:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/171d8e395acc43378781855d6892752e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d5b3bc8711e7414eaf99e8f79ca36cdd : SUCCESS in 1h 22m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3165d7e6d7345e297c6860b280d091c : SUCCESS in 1h 25m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38e41d789d70499195cd2895b2bc910d : SUCCESS in 1h 13m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c72711b79a61404081138e2dc2553aad : SUCCESS in 1h 20m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8271862839ad4030ad9e1b803f0faff8 : FAILURE in 17m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2918a157a30d419a9d5b9890c64616bc : FAILURE in 18m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67c4363858f34b3f89da246c99c6d6ab : FAILURE in 32m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f92ce891b8254381b05b9f6243cd24f0 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2bbc826aa3741fbbb846f5c7446bac5 : FAILURE in 24m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7f0c9870a334d1783e161d69ea2458f : FAILURE in 28m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab42ef3d3aa423eb2caae0dba1632b6 : FAILURE in 3m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e53015196e1f4522b4db1583faeedc4e : FAILURE in 18m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c52886989cf942489631f7f9632f993d : SUCCESS in 46m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb949b61ea8641a8a3add0d9512ccab1 : SUCCESS in 1h 31m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8793489470004619ae4d316e6c0f045d : SUCCESS in 57m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/752296f992df4cb2992b44aec22063a9 : FAILURE in 52m 07s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0a6db6b01d3d478698135a0a365650b4 : FAILURE in 1h 23m 35s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/36a4795dfc6640abad4493347e9afce4 : SUCCESS in 1h 17m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc144d26ff3d4dd785a2edd3fc78c29d : FAILURE in 26m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/43ff721377294019b3364e8ec9206d79 : FAILURE in 27m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2df43362aa464d79814e316272493be2 : FAILURE in 22m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/33334931259848aab7ad14d48ca7426f : FAILURE in 25m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/21498954fe0b435a8284a24a6b3f9865 : FAILURE in 26m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/904cc602fe6f43a6a93ea74d2df4f2ee : FAILURE in 26m 14s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/0e6b32709c144dd68778653ce526f114 : FAILURE in 25m 12s","accounts_in_message":[],"_revision_number":2},{"id":"91ef21aeccc004e84e2ddc2a1e296046a6efd14f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-02-28 21:14:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"80fca9f50dfe5b44d83f3a567fdac693974bcec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:01:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3050af128c94c308c22020c8a1c63a7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cfa30cb870d45aeb267e1e3011e73c2 : FAILURE in 44m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a56ba64bffd4fd58023ff0939ce4ddb : FAILURE in 38m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"449ac34bce1254ed8ce1b09063e50b16f164122d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-28 22:33:14.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/97/874797/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"42f34705749c981dcb55f0d9d50bd124d9d0d2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:50:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a58d64342f844dd2bfa4adf6455b588c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/036f56b5a4c7412a98f5ea2a431be253 : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/61d9066e74f84dec904f6cfca50b876a : SUCCESS in 1h 09m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d12fcdf690b5481cb6ed39cace56dcc8 : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e80c7a5590374071b9e6eaeb017765c7 : SUCCESS in 1h 34m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/409b61cbffda44e588f558412651b16d : FAILURE in 19m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c988c56a910452ebbd53704a58663b4 : FAILURE in 19m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac360d36376642f89ab966cb82c8ad8a : FAILURE in 31m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ab30e7fdf54866b455ea7e68ba7ad9 : SUCCESS in 12m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32ba984c4adf4f0d91d4fb97330a78cc : FAILURE in 22m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5375178e5e24f868ec11c2529103c27 : FAILURE in 21m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973f204f5ef24abfbee4256fbb55e96c : FAILURE in 3m 53s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e0ce7de8805a42b1869d3884a91803f0 : FAILURE in 21m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9110962ecfb14eeeb8027cb00f7da5b4 : SUCCESS in 38m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f53fc029c79f4d32beea91875e5830e8 : SUCCESS in 1h 25m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c23f7c8d5314c578a49b17bf2a407f8 : SUCCESS in 42m 35s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4ddd6f0746164d86b90bd4e6d32ec7fc : FAILURE in 53m 36s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9b9e9cdb33b64bd092c50b9b75bb942b : FAILURE in 42m 20s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ed7e90c9b7ae421b9b2a84adf2ebad1f : SUCCESS in 1h 24m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b980e9ee16b340b1b0a8f5205dfa2e7f : FAILURE in 14m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec86ad160865455db1368ce45d07c671 : FAILURE in 18m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/15d7217dab8347e6a79f5113c4fc7903 : FAILURE in 21m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f83f1993187649318a1b6acc379b17d6 : FAILURE in 22m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc4332b96cbc489098e3001a72d01645 : FAILURE in 25m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b5f99b789be44bf84c44fb832595fa8 : FAILURE in 22m 14s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/01e81ea9b4f441a4951977ec08f9c2cb : FAILURE in 22m 20s","accounts_in_message":[],"_revision_number":3},{"id":"732bdee57c296249e8fc86f75c9a424ddfe1e82a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-03-17 13:28:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e2517f010c9544ba69a5dfe47099d84f8bf91491","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-03-17 13:28:46.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"d9d90e7c5587e95efef4e4a12e6fc801a7e3468c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 13:36:38.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aad9829534f494f80236dadf0267a65\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f664f17f43e245cd9f1a7708785a3d00 : FAILURE in 3m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f8a47d5a7e444b28ef8c59d2fae6894 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70e5a5c1127e4d328e81d3101e2704a4 : FAILURE in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08e350a04b37d48cda791e16922877651376209e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-17 14:34:46.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"98129e06a39af0678f46b2108d196109d405b4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 14:40:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b1cbf738f774e96879aecea5fdb0f61\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/240c3799db334d059aebe652349b207d : FAILURE in 14m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0a8ec6ae979d4c7888ef408aea121909 : FAILURE in 14m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c16c1a566c44b578d6ae164ea0a7367 : FAILURE in 14m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e6d16b78b03642c88c7a7f2a34a8a56d : FAILURE in 16m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/14bda784a37a4988b64bcd233c3a350e : FAILURE in 17m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ede154cd050a4d3a95bc2db853b52841 : FAILURE in 18m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/033a6244e9b44b52a10567425b3f299f : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107fa27a68a046faa9789c2e37b627b3 : FAILURE in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e006f7de89488e930582e9d880b13a : FAILURE in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4101604b8a7449e5b981e0d13629934e : FAILURE in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac5d7a7af6cb40a88b5dde82b98e32d8 : FAILURE in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7951e0574db54ad091aa5c318b007164 : FAILURE in 3m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/48add41cdbb84bd7a5019a14afa60e82 : FAILURE in 3m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0028266bc5c54e7a8ad60ae831dc2daa : FAILURE in 10m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4d81d5e975d471892e2f6b1cfa19cc4 : FAILURE in 9m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a377a3bbc8e6495aba911cf80e9e0f50 : FAILURE in 53m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7c3aed96ac5a4c1a9fa7937780e70624 : FAILURE in 38m 36s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/15f3b72efc0a4fadad6804c0d84b286f : FAILURE in 1h 05m 27s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/182f81ca3903482087935abdd7818d4d : FAILURE in 37m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5fbe5d5db9084d5cb63c729bcafc4d3e : FAILURE in 33m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2754d93567ac48cf96d5c1dd5998133a : FAILURE in 15m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d570fc78b7ae412688403a9d646d3dbe : FAILURE in 21m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8c3c4a5f789e4f0b892e2da11c5e04dc : FAILURE in 26m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d8a219e874ca4881a7e35dc1f9491f6c : FAILURE in 31m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ca33aaa8f0c64f44bed38577c3472075 : FAILURE in 31m 49s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/73c701c35988483991b58b94e3b19ff7 : FAILURE in 22m 47s","accounts_in_message":[],"_revision_number":4},{"id":"58c08c1db1fbf593466ddb55ebd43fdd4dc56af4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-03-18 01:53:53.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6e253e11d400f5336bb73944d07720a640c65716","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 02:42:00.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd590db40ddf406bbc3b1a0e3e2990f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb48f3df07e6455ea01238c3b2251f35 : FAILURE in 25m 13s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8198ac44c5274ce68db9c3f24d8b69eb : FAILURE in 46m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fb4070b4ff3448bb6a7dea634869b39 : FAILURE in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"42fe07e9884b04769d9601dbf0ed530ad14b6098","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-18 03:24:20.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b75ce5bb0503fa35a6b6643e90b0533e4fd5086a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 04:20:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b7416caae744b6ca7aad41586b40da3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d34a7f354fd04344ba418bdedeb53719 : FAILURE in 14m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b13e4bdf005f481fbae3899a586124a8 : FAILURE in 14m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66be0bcd27bb4374a34c6fa23b92c000 : FAILURE in 14m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4921afa17134682b37b825f8b084229 : FAILURE in 14m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2333b3341e1484b8983cd7b4173218e : FAILURE in 17m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/39a25021f41e4dab9687835dde5c8bae : FAILURE in 18m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e782e369596046cbb7a68e6f5d373154 : FAILURE in 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676bbc19aac44c45837c728e26aa0c47 : FAILURE in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b97ec1973164863bf0f495769970eb6 : FAILURE in 35m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c46e3c855e2c4769addc8950785bc103 : FAILURE in 29m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/825546a13b694f349d982f374d22cceb : FAILURE in 29m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b00792d9584daa8e3ea8e6642823c6 : SUCCESS in 12m 17s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/50df6333760a4659a945a4c92c892e21 : FAILURE in 26m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cadd821a86904ac39094671fe9a18600 : FAILURE in 1h 03m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ce4b6289a6c46dd96921cf6897c06e9 : FAILURE in 2h 24m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae6f099abdb6426f88feb08fb618d8ca : FAILURE in 1h 01m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b8004b149e644bf2b6bb4eba3726c66a : FAILURE in 1h 00m 49s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c03948cea6a241d9a55d0bd352412944 : FAILURE in 1h 01m 23s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06bbe7fc75a7447bb5295f9325ab0f1e : FAILURE in 56m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/66eb687baaa34907b1c444fac538c16e : FAILURE in 28m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fdc98ae256034b9b9f58b6194bda992c : FAILURE in 38m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9f081c07a08f4f4594e4df3838e6b985 : FAILURE in 26m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b84c18049cb8429590ddfaced2414711 : FAILURE in 25m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d16e5177b9744d31b9f102d8a9f211a2 : FAILURE in 20m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0fa0ec7d48e344fcbc1d474bfb8418dd : FAILURE in 26m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c3ddc53fe3e745e7b9245294dee67d46 : FAILURE in 23m 57s","accounts_in_message":[],"_revision_number":5},{"id":"27888e25d30eed1cdcf433b20f16e00ade9e165a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-20 08:29:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f9d1a03f2926ec2aa3665fcc35c19b06533a0f20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-03-20 08:52:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8026caaaf2b18d35a977d1030b8d38f81f384f58","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-03-20 09:09:53.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"2192541960dd8048dc99e4a428a54f5ffcb0917b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:42:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/555c6f1914294b058cd430162dea72ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bf3b8b5569c47ffaf1c806fe1840c93 : SUCCESS in 20m 20s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b226117f30c4461cb9d71d808a47f476 : SUCCESS in 48m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bfa36fc72ac54aa4b0b817586929790a : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d9a0d16c92e0d4857a677117605c5d46bec8471e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-20 09:47:31.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"e1cadd4887227955e17368a4205177738e0e2745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:03:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f968407cf4a14b00bff3a06d6de78e74\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6c5a42be12ed4e45abdbe79cedab3d54 : SUCCESS in 1h 34m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/97d90b8d7f7d43cb9f118d2cc64e6003 : SUCCESS in 2h 09m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95684e4625954496af7b000e80b1ece1 : SUCCESS in 1h 25m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5e54bf14186e412382eb15eaf801c97b : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e0afe0334774455db9a6b44e9c163929 : FAILURE in 22m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe6ac96cfa3c445ebcf876280cb33a56 : FAILURE in 18m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b621244f6f0944e49e26ef97c938a272 : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ad50caf2d14869b14bfb10069f3786 : SUCCESS in 15m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d15b7041c9c8442a94bdd7453015afca : SUCCESS in 31m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe43bf91344e496084f5cb10a355f75d : SUCCESS in 24m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fe7bf4ce8b147b5a83d0f669b4cac82 : SUCCESS in 30m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8875c53804a64d4a93c787b02cf41129 : SUCCESS in 12m 00s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ff0f94a089e544328e6584cd1d59770e : SUCCESS in 25m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/598ef39ccf2b443d88b73c003733c229 : SUCCESS in 1h 11m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e2c4228843e48fdaf599cd7be9697d3 : SUCCESS in 1h 34m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/599036dfd06a4d69afed539297c1861c : SUCCESS in 1h 22m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3f2348f96a0142049e0d03b28217f933 : FAILURE in 43m 26s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/894f27dc349c45c8ba217c0aadc6167d : FAILURE in 1h 00m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/04aedbc68a4c4c909c55015db0bc8420 : SUCCESS in 57m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bbf16ae4bb04467bb613a73119de6c4e : FAILURE in 28m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/015a1880f73a48deb7b83757a158ad21 : FAILURE in 30m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/426beb979858484380c1ab91a6c7c3fa : FAILURE in 15m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/44d6e2db707b47d0968a4f1016493b67 : FAILURE in 24m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91b319bda4fc4b84b99f9e5fa72117c3 : FAILURE in 19m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b609a727947a45769bbcfb6174405ae5 : FAILURE in 28m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b7966ae4370f484782701c1230b7e968 : FAILURE in 15m 37s","accounts_in_message":[],"_revision_number":7},{"id":"7ea899cf1ea01f64ef83ee7e4377aaf965ff5324","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-20 11:44:31.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"c554a9e4078313b4192db52ee4a8f0dd813b91dd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-03-20 12:02:49.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"f2ec02994ce9129ffa5617b6c7857130405d2278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-21 09:15:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"232bdd709a87866776a1e9c4b2bba887acac06b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 10:15:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd54d83dacb4970a5953b7e666faeeb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa97a24c74c14c498db9cad4acaff5ff : SUCCESS in 53m 13s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab6ebf1ec9ea449ea508871d7a5c7216 : SUCCESS in 54m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6284e5efffd04ff6ac5e8a065cb2f4fd : SUCCESS in 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"05445372e6063556978f1de2568f61bde20ff435","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-21 11:14:19.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"4f1d71cd0cff0c8fcf4662c0955057a43417ddcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 11:25:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/433ac91890b3411d93f5abf9465768e8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a6d5865ea6ae4fc2901b38b6aedfdf85 : SUCCESS in 1h 17m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15d3a5dc0a6144f18a83913ad781cc95 : SUCCESS in 1h 20m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a6083cc0a3e452c8e0505fcee5a6f8d : SUCCESS in 1h 15m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/653050c9d9424680ae541d1f34bd2f13 : SUCCESS in 1h 18m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f84c8407b3d148ecbcfb25006530c0cc : FAILURE in 17m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/493734b98ec340038c18916e66392a22 : FAILURE in 19m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9272ad3e56a48e69662cbfd444563b8 : SUCCESS in 45m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910bf04e907b42c3b243c0defe7e258a : SUCCESS in 13m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc494ac4be174c44b12a2c8e2e2bbc31 : SUCCESS in 27m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87ef1aacd24d45a1890c580871cfb368 : SUCCESS in 25m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e95049bc649240278ac423e3e88c2473 : SUCCESS in 23m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb742b92571543dfa8c31b1e314a7824 : SUCCESS in 12m 09s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f5e197b4c6c54e40ba32439391dc63bb : SUCCESS in 27m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c18408787b2e459291d2f9591b6bba86 : SUCCESS in 1h 01m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fca36f09ec074820a266212817d95398 : SUCCESS in 2h 06m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4c04bd878ad49328489d9081b2029e4 : SUCCESS in 50m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/661d5bbcd9704122ad010cc3891beae2 : FAILURE in 1h 01m 07s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2353c03f00924e1088edc115eccd16be : FAILURE in 1h 00m 27s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dd429f8fa34d4d0095908353fe09a295 : SUCCESS in 1h 10m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ecf5307c9384ec5b5bfd8987a524552 : FAILURE in 31m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c638b529348941278721c88c1126faf2 : FAILURE in 21m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ef61b7dd33145a5b707dec57f0b4f2e : FAILURE in 22m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d4d229c57cd545588c4f9cae1d9fe424 : FAILURE in 28m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ff7e1a8edf34aa78f30418dea99547f : FAILURE in 27m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bfba3223c56b4cd4bbc2623a2a4cf42c : FAILURE in 25m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2abe6ae3bfa64db188f316a288afed04 : FAILURE in 25m 52s","accounts_in_message":[],"_revision_number":8},{"id":"6f2eb1200b12f55697bd2841126d7071b42b70ca","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-03-21 14:48:44.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"8b6193ba9527503398c9eb7d627e10a79ffd29a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-03-22 13:53:50.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"53f3ce9a9501600acf90be3d37f57ecfed65a6cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:21:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba5cf10a01546dcac77b4446c85d4bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9445663688954ec99ebdd620b68cadb5 : SUCCESS in 22m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c256a02ac5a4ebd80492cd02593ed83 : SUCCESS in 24m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dac5b28f181e4a8f8062a84aca4e1d72 : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f526e5ab03b130ea8d3d92ba65c1a3104a985643","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-22 15:13:47.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"48d7d54a04e7253953285c1fa33a56c09f485e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 15:57:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f7a0332dc6649a3871213aea29dcf82\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0a4c88c3d3ba42938cdedf76212570c2 : SUCCESS in 2h 00m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e0b49ed9392b4f20ac8b7e84566c523f : SUCCESS in 1h 32m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0972c4ceb5bd4816835bf7ee9f84c8a3 : SUCCESS in 1h 38m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e4d9cfb0df5402a922c26f1e325eee2 : SUCCESS in 1h 32m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/787cc5e3c9ea4655b87fda804cda1b2c : FAILURE in 23m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0857cab08bd344fb8e319e2222f3489c : FAILURE in 32m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf6dd43ca37c442ebbd1405850302e50 : SUCCESS in 46m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5bd26b214654ab8b3ec864a95528d0b : SUCCESS in 14m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b9ed79a26845ecb101ca94c73bdeb3 : SUCCESS in 26m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/665ab5f7f6f5493ea310c5b8344bcfc2 : SUCCESS in 31m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15752a0f3ce840aabd2fc94938018de7 : SUCCESS in 29m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba064a4b7694cff81e173afa4c2294e : SUCCESS in 11m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b190a6d9e66c4a4b83ce10d00417b85e : SUCCESS in 23m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/69152ebfacf34a0aaafab54e5b5a7dd4 : SUCCESS in 1h 01m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c579734f3af47cf822ca9bd64fe1225 : SUCCESS in 1h 44m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d5a794eae5049c98d2c98f9f62695c9 : SUCCESS in 1h 06m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/57ad943fa3f2437db8854e0adcfe0fe3 : FAILURE in 10m 04s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/18bd9922ce4d44f68319b3162d268399 : FAILURE in 27m 34s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d68bc4769c724d89a129cfc08024dcbe : SUCCESS in 51m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7cfeaa75486643b08409a0811d941c72 : FAILURE in 16m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/59bd55ce8a374495956f5e3175b4ca7f : FAILURE in 19m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d188dc7b055648e7aae30cd6424994bd : FAILURE in 27m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d3a634f6959a4a36aa11a3579159c0a4 : FAILURE in 25m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1686ec448d914fb98abf89a4c70c8f1a : FAILURE in 34m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4895fbda77aa453da4a4b075a6524299 : FAILURE in 21m 05s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3b977bc4feb04e45992ba67a31d7a22a : FAILURE in 22m 53s","accounts_in_message":[],"_revision_number":9},{"id":"714bd245e9702bdbce8491ceff36eba9486acc48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-03-23 14:00:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1ff7f5f67250dd41aa03489c52f53fa9e3b98193","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-23 14:28:44.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"adb96b8acf1c0a9d0f22f25c6bd5ea0f657f6239","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 14:43:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d7308f4ff524befb7a6f55501970ee4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/363de591d469466e88538da18d5bf469 : SUCCESS in 19m 34s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/defeff3d75724afe86b7fbbec481d235 : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2e8796ac51443a2a24cbc56c5ede14e : SUCCESS in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"786020988f03a70785dedfa5e68e0fbb9947445e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:25:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/799d07a4a71d48d888c2d929a2b2c0ad\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c5ab9970daa246d4845d670c9040ba76 : FAILURE in 13m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79214b080c8b441499e40c301817cc4a : FAILURE in 14m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ab57a68d9e6484bac12ec98558b4b54 : FAILURE in 16m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/366aeadd79ca4eb79bdf7c4b0e51419d : FAILURE in 15m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2275352b6fe40cd9d2abf58dc0f275c : FAILURE in 18m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0cd13cdd0f1485eb2a4b6a7c43aef1d : FAILURE in 20m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/882a5ba910814be4b82dc7e12ae6e825 : SUCCESS in 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d82c418361543d8bcbd403e23b4d06d : SUCCESS in 13m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50c5c00b18b84b2aadcd54835406f7e1 : SUCCESS in 28m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eac98aba97c421fa4f1dc51d119b92a : SUCCESS in 24m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4d3f6f79cda41248fcee6006a997307 : SUCCESS in 24m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4830c2113ebc497a9b5bdbe11bcdf6df : SUCCESS in 14m 01s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5acd23781a1c4303ad453d4ee6698bbe : SUCCESS in 24m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/92803d965bc54cceb2286f9a9c57f70b : SUCCESS in 1h 01m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e7630a00a08481ba2201bb3dc15e39f : FAILURE in 1h 18m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e35f8a02fde946248f6e994fa8138d54 : FAILURE in 53m 12s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a2349f7e1bbb462ebf7261f86df1ed18 : FAILURE in 21m 26s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/67322422af4c40f8b7256a20fddbb2c3 : FAILURE in 25m 11s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3426c93d5b8f4a9a8165522a08fd50c5 : FAILURE in 56m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a2fb49d22cdc420e8f9ddf24282ba3b8 : FAILURE in 27m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d892a7c2bc84acfa035e9b843069c59 : FAILURE in 27m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9427151a528e41a09b4691c57878d7eb : FAILURE in 24m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/12450e1956ef4aa7a09c98e8a7bd1f29 : FAILURE in 26m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb972670f5064d2ebc58f9f25d6ecf76 : FAILURE in 27m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e605392fe45f4896be1e593ff8eb859e : FAILURE in 34m 20s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d7da357ee9374e28b781612bd08bc94e : FAILURE in 25m 18s","accounts_in_message":[],"_revision_number":10},{"id":"9b8c3c129ed21c247a0934a0c6060a14f5d3b81f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-03-24 11:44:31.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"dc8de0f5ed31cf9621da084e58ae11c503eff14a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-03-24 11:45:06.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"ef06fded9ae430c22987cee9bc78fdb180d5a71d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-24 11:54:00.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"8e40a51a4d2be0011f2249c73e45869daebd68a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 12:20:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1362eaa83394330834141c4fa2dc9e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/981953d736614ec7891bd6d0d0fa6863 : SUCCESS in 22m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42038bcdfe7348c5bf7fe02c723eceec : SUCCESS in 23m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee33dab386254b3dba8d4afa74284cd9 : SUCCESS in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1305f8ccf415649ca0f416b307e04925e45ebc24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 13:12:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b031672fc7424eb0ae527e62affe4996\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a9a4b0c26caf4e26aafe50ca5480d8c2 : FAILURE in 14m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ee7ba180d58c49768cda65eb8370eaff : FAILURE in 15m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f85eb357682849498b16096498de7a30 : FAILURE in 13m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/80c279edc89b4311bdcaad2443c8e502 : FAILURE in 14m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5f58c78ca6c449719ba4b1386c66fbd0 : FAILURE in 16m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/94e7c695c5b54dba98c2d1ad85c84232 : FAILURE in 17m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bcf7ce5d537401898cef5fdb0ef41aa : SUCCESS in 46m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c90f2cd233470e8d98a92b6212694b : SUCCESS in 14m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f7e785c76324d05b6523b769f58822a : SUCCESS in 36m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d52f20e0ee44461593d6498826b5b88f : SUCCESS in 29m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efc0c9e5b25d4ddd8d571e7f5e952d91 : SUCCESS in 36m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6dce78e10c40b5a0037a33ac744aa7 : SUCCESS in 13m 16s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/370f3555d86345babb05d615625a3d87 : SUCCESS in 22m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ded28e825933404683e93ede763c86e2 : FAILURE in 57m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/205b9eb0dfee4204b41194044c904b1c : FAILURE in 1h 14m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9403eac4354f47c4b48e640e0978fb23 : FAILURE in 55m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0add8cef99714cd8884a483110d8b5f1 : FAILURE in 13m 37s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0468fa50ed0144c5b8d8e34c93357373 : FAILURE in 15m 37s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/43b7b113169c4c50851021d2967737a1 : FAILURE in 1h 05m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d04f3d0b8c894a5ab565c32a3fb653b5 : FAILURE in 18m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/17b2a697bd11433bb8ff2f75a2462b93 : FAILURE in 31m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb2b28862bd844ef90f2391d89f9bf23 : FAILURE in 23m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b60a954349c943fea5b11385eb420e09 : FAILURE in 24m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b71964b535724a458d681f1f1523d9f0 : FAILURE in 31m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba2f80ff17194263ab95ec09957923d6 : FAILURE in 17m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ec347118a61b459cb6393b648e4e407e : FAILURE in 26m 29s","accounts_in_message":[],"_revision_number":11},{"id":"fbb418537f6cf948187c9d74fb89a68703e3b8da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-24 13:32:59.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"488f1ca9ff7c46b26eba2c22bfa336be1b2bea82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-27 09:19:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"28572f7eb5f5bc50a22de422a56cf6280e74fdc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 09:43:16.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb4d5da98a64be5b7d9c36369309023\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e758171404814625903366d29ea25f2b : SUCCESS in 21m 12s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06bcdb1f75dc4e72a68d76d6d4e24e2a : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/666a9056d44c4582ade280117fc9290d : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"13ea9593c1cb71df1e6f809cc9858ce2eddf0fc5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-27 09:58:07.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"af0180e1d63de503e9d48572e499ad89e7e3ba3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:45:19.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/873593, https://review.opendev.org/870887 are needed.","accounts_in_message":[],"_revision_number":12},{"id":"1dbbeda9142c70c9b8ade1fbac2a580ec058cf38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-27 11:05:32.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"ebf3a0f4aa565695d1df84a8e92a56edad9cfeed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-27 11:42:54.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"fb56f9009a478fa3b2a23ea414e86c088237f8f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:51:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c40dcb131d5439b93621eaae0f8cc14\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14ee82e7a331484bbfd27cf1906f165b : SUCCESS in 20m 13s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1aece5d2573941c6a8450ab9b4232413 : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/afb38270c32e4a8c8a7c4899b07fed76 : SUCCESS in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0bdff7705759fa861e9a425102e3d39e6700a732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:58:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54771354d40a49cd954891a12478c75d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/40ecbbae67794d9a845e378fa85d9896 : SUCCESS in 1h 15m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/40825a0d760c4e80bde948b5fca89ea1 : SUCCESS in 1h 31m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0ccfadd38c84d638c6aa3ec9fc6093c : SUCCESS in 1h 18m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2cdc292b8ae847d1bf69084af1996b5d : SUCCESS in 1h 05m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87b864d8c8b2401a81326be179fcf05b : FAILURE in 16m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/948cd85b80df4578a94c6cb38b948a13 : FAILURE in 17m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ca208680d6f49c7ad3a4a299a05b471 : SUCCESS in 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aeabab7b7bb4354b04eb0e6073aa0ce : SUCCESS in 14m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae1ebda34cd647ca9f0295a727efb0d4 : SUCCESS in 26m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bfb2c22a0d3421f878dc9afab3a6624 : SUCCESS in 28m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8ec53694b6c443c9870e86558054c75 : SUCCESS in 20m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c2a7ec827e40efb129aa1b4ea19a4f : SUCCESS in 10m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9ebcb7a05b9544a58df1600b602baabc : SUCCESS in 21m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00d25e8425e3458781ab7f883c2ea17b : SUCCESS in 1h 02m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a3f704d90f143509bd178aa6cad2f4f : SUCCESS in 1h 46m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a10f4e6e4fa142e09882814e67f03d55 : SUCCESS in 37m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/660f12d3bf894acdb27d28ff4802298d : FAILURE in 19m 57s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7e7aa562c8fa4dc884079f676d7d4b18 : FAILURE in 22m 45s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f2586ca7bca24e23bf744b91db5130d7 : SUCCESS in 58m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/831fda24814e4a15890163af178cbf7f : FAILURE in 24m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e98ea0914fc24a4885f4f6b9ebd83dec : FAILURE in 28m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8fc009ccacff4e149e45e9d996e5003b : FAILURE in 24m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2a24ec4a1f0f4fb5a6ad4560eee9bde9 : FAILURE in 22m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c5181bc0f6446d8ab7506beb8a44a36 : FAILURE in 14m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1a54343e37954ca2a253712c77e3bd8d : FAILURE in 27m 36s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e5503c205ad14e72a15fd7d6acc2ee4e : FAILURE in 22m 59s","accounts_in_message":[],"_revision_number":13},{"id":"8a10ba73e1bd04c7a3efc70e7c844acd0c74127d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-29 10:13:55.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"f6a673a9a9ee1d6393e2eaee0458c9cd850b4882","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:44:10.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/637ce480165e46b7ae550b0f9bdc623f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9008c0710a14c329fd0d367a8fa1e44 : SUCCESS in 25m 02s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9aea7a4a3dd4a658315b90ae858b2b0 : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48ae2b5561ef4911a32f09ebbcb31514 : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d70818e7e99b92edfebc61ea6fad0c5a26b29c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:56:48.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fc69e9c83f04d3699beeb4dcf967318\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2cdb0861a89e4143a03c978b5d2b49bf : SUCCESS in 1h 30m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2f5dcc051ab3439eb7a669e4950d7054 : SUCCESS in 1h 16m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d24caa8b36f4533807ffd45fde3d0dc : SUCCESS in 1h 31m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c2fe8fb006ed48f78071862385ec3ee8 : SUCCESS in 1h 23m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85ee42f793e94e0e8a28dacb2640368e : FAILURE in 16m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3ecf4328b2649948fe2ee5826ca9f67 : FAILURE in 17m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aea11c2c73ce413c95ca5b0783646a6b : SUCCESS in 39m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fa858a2a1b420f8780a6a654ca6f03 : SUCCESS in 16m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d412049513d46da9df3a1204735e977 : SUCCESS in 27m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13eaa0a0f9df4395bf64c3f1cb227afa : SUCCESS in 33m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/750eea989f184d9f9abd3eddd898ce38 : SUCCESS in 28m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a3bd27df1a46c1950d7f904a9e1cf3 : SUCCESS in 11m 20s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fb781f6ceea14284a5c351ac46638957 : SUCCESS in 29m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/10d70edd02cf4de9bc26ac88094620ca : FAILURE in 1h 14m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf05f5804eed4f3baf612e81549df023 : FAILURE in 1h 34m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc86b9285c114d0bba011c5efe724200 : SUCCESS in 48m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4494623dd01946a5b9493cb04cb550f9 : FAILURE in 19m 50s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/049d1cbf24ec4a8d9de631d49b8d4710 : FAILURE in 21m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/99bdd36f3bf3451ea942069a5af54baa : SUCCESS in 1h 17m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/12714c94c767489499506bc5228794e3 : FAILURE in 26m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d06d804c0c744720aa362332cfe92569 : FAILURE in 25m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa18f358df764c349a09d82ff6c06ee7 : FAILURE in 25m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df3cdd71c3be4045914a5c71995d0f54 : FAILURE in 25m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05522ec5138d477086df3a07e4ba128b : FAILURE in 28m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ad4141b36004bd1a1f689ef1fa121ff : FAILURE in 27m 35s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/05c3022588944cf4ae68171ca4806a00 : FAILURE in 16m 50s","accounts_in_message":[],"_revision_number":14},{"id":"0bfd482c7f7dbede431b4ac677a3964ec2d7cee5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-29 14:08:21.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"13ee191c7c7b345217623104ffadb25cc10ca6ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-29 17:58:28.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"60df38d0def84a08c5cce25bba29bb95b8da1014","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 19:22:41.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bd7927a9cd842c5895febf4e86108d9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe8145f04511425c90ea0f66bdcf30a0 : TIMED_OUT in 1h 21m 07s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6523859ee3924db495a935b0b56eb07c : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4d0a9a8b5de45a2b4ddef11704f690a : TIMED_OUT in 1h 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"bd1a3a19d39656e28fd8423ab398d8d8bf51911e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 19:48:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6613a3bd1f2440c83d3ba495a06638b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/954151e358c54a568435f9b1965ea7bd : SUCCESS in 1h 42m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8fa96e17365040ecba3cc78aa1d3a3f5 : SUCCESS in 1h 17m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0506d8bfb76649359dc8b1812a168ff9 : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/80a2d9ab918f40d09e1629913f03940d : SUCCESS in 1h 07m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d595c83fe38b4b66bbf0e37e538b879b : FAILURE in 16m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4f2a9470839480f9554a0eec048e97a : FAILURE in 17m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33c1b02e07e74a0ebaa72d76d58ec162 : SUCCESS in 34m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1a36794e9d45a9a23da0605329fc68 : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21f5d59fcb54950947759be7aa575aa : SUCCESS in 31m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f95aeff6df0b40b699abc6bc48cbf68d : SUCCESS in 25m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11f166837fb5452fab94f974e4fbe8f5 : SUCCESS in 27m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4471cb0b8445beae70634e5fcb3d4f : SUCCESS in 11m 35s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/862bdc50aa1f4f259f45188080421d8b : SUCCESS in 21m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/646ffaf84e5946caa02e9b35b8495eac : SUCCESS in 46m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd0c3c8deedf490b83b2e204bc29e2c0 : SUCCESS in 1h 44m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3cb1eafcdc442f18a1fc99e32db2969 : FAILURE in 32m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c1aed347fc2345648480182e722631b3 : FAILURE in 18m 07s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/40de3c8e4d6e48a3bb161c9002e4ee51 : FAILURE in 37m 23s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/790a9dbbe3e34d3d95b1f90d23aaefb2 : SUCCESS in 1h 06m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad95a4209cbd463b94f9fa2658796b71 : FAILURE in 15m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dd5a039c697c4f41820673a4dbbf4241 : FAILURE in 19m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/87daddb63b244d629bbcde4873ddd141 : FAILURE in 22m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f6e3fbc25ee462ca9d2d277ffbdb9c2 : FAILURE in 33m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8683eebab0ac47dfba6eff6d0cde3d36 : FAILURE in 15m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/81054a9d06ad4d249a88ff36d85b714c : FAILURE in 23m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/230b68a12013429babe366a275ea819f : FAILURE in 25m 17s","accounts_in_message":[],"_revision_number":15},{"id":"dcaf50c1dda3bc39df66bf0b683b448bb4cbf8e4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-29 22:49:30.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"725d0a79f5945215dd8e24afb2c53f8c58e8b2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-30 20:10:10.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"eb5e9642fd54fff5aed462b59e9d5a3aabd5d847","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 21:14:09.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f29c14152cd4a0aab538ed722bc00f0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a5f9e2584a9463c80e63eccf2d62027 : SUCCESS in 26m 08s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4b665ad34654b0b859f66e645fe7d47 : TIMED_OUT in 1h 00m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8f1a3ebfc80448ba8edba217b8b122f : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"59a23050bfa4fd51be4d48e26c6c535e1fcd042d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-30 21:31:49.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"229d3826f5a64d8b2c20eb0e42fc9382f2533a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 21:59:23.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c41a811f420460d99216ae089fbb741\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f7555b17fb494d599e583fbbf8194477 : SUCCESS in 1h 34m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5df96dac20af4c249fa2bbd856a031ac : SUCCESS in 1h 35m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4e0dec3e6364af39e3a0112101cf521 : SUCCESS in 1h 39m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4c1f7e35c9854906aa9ebf6bdf9360fa : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/00f0bf9be4294e6e8fb897ef09522cc0 : FAILURE in 26m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bdeb96fc25504ff083ce4ab5421555aa : FAILURE in 21m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b35964ebe11e4fbb92efcf03c92fa2a6 : SUCCESS in 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ffbcade22b4f2ab59aac7aa06852aa : SUCCESS in 15m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a8565f513d344438cb6e72471dd7be9 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db861fcee30c488a909c4d1cf7f42950 : SUCCESS in 36m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d83ecaae7f01452681090de8439ac106 : SUCCESS in 26m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf519ba49204137ac859db8bdd9d744 : SUCCESS in 10m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9e55c96e34c3405faea54e59e036650a : SUCCESS in 23m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60f79ebdaf784bed8e9b8200e202f50a : SUCCESS in 52m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e596d65cc1f4c29a18863ba21f987bd : FAILURE in 1h 30m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ece575783c4e4f73b2e959fac66770d8 : SUCCESS in 55m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6076198a668544f8ae1604d64b105616 : FAILURE in 11m 16s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9860658a3bf54def92b7bf385bfe1872 : FAILURE in 12m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/914eeda7dc1b4801a93deb51c6eb17d4 : SUCCESS in 1h 02m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1301d690d83f4d0a8769ea941f8fb4ba : FAILURE in 18m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/112e4ba220c648f7b0e5bc23c9f0a2ed : FAILURE in 15m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63910e85c4234ccdbcda6ebf1c2d4def : FAILURE in 15m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8509ca19e33445138ecbffdc73a7ee84 : FAILURE in 29m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f924751ddede47c2a932b7726472124b : FAILURE in 26m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6425bb862c034ed88cc0ae8a3755cda3 : FAILURE in 15m 36s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ad5843c72c7c4b4a86a33d77b6cb84fa : FAILURE in 18m 14s","accounts_in_message":[],"_revision_number":16},{"id":"914fbcc2b4e28a46c77e394444b891d2f84ac300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-03-31 20:05:06.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"f6e3815db18a60893fc44eb4f7896bd2b4f95546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 20:41:29.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851b3d8ad8834ec2978100253084bf69\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/275781af0b174afa9606a5b22c2fd1a4 : SUCCESS in 25m 10s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50abb2ff29f04deb9cb3bc8fa4433165 : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/616f62e601a34df2817e2ba17d2e7a1f : SUCCESS in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9a8505fb3b1181816ac1947f3ef247725972b5a8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-31 20:43:38.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/97/874797/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"cc96a721c98c14799a5bc3e6ce8f36ecb21c8230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 21:58:37.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ef44ba4ba9a4a7d83752f69250232cb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a61211b053dc4cd8ab0822296bd48ff3 : SUCCESS in 1h 51m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aae3774e0dd14fa68c8380dabcd97cc1 : SUCCESS in 1h 13m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9f73ce3168d40eaaff6285269a95426 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2da2654c93514707838eb25b0d66d28e : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0cf05fbe8d9f4761b77bf3c478e185fb : FAILURE in 20m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10fe00f8db3e4f5f9488d5a5e4840958 : FAILURE in 21m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cee887befb24ed89abc5b345cd09050 : SUCCESS in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e882de9089284739a331b4043a1546c6 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe9b18f74454b67b761d545d5fb1867 : SUCCESS in 28m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cb4e47d009f4e25847af239fe1188e7 : SUCCESS in 24m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/108ee1a856e2477fa2b9798935294d03 : SUCCESS in 29m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ab1c378830422f90823f2a0ef79b57 : SUCCESS in 9m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/925b773270bb46828ac47bf087e74034 : SUCCESS in 20m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f63d479cfec4238880d33693e4df665 : SUCCESS in 55m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30d045b48926447a9ceb7e21b57d5d03 : SUCCESS in 1h 33m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf87926f568745c5afa8e6a6cf943a0b : SUCCESS in 58m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eb172927c8b64fc2bd87b8153724e909 : FAILURE in 10m 58s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a4b6e729455c47279cf5007f87294600 : FAILURE in 23m 50s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/75b1dc93c2c44182b28e6880d9a0907a : SUCCESS in 1h 05m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e0192c802fea404494c3b506515c09bd : FAILURE in 16m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/663b706edd3146359014da8f1662e3b3 : FAILURE in 26m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/624ecc3c81d9464a803fc63a03bfb323 : FAILURE in 21m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c57dd273cf0b4c25ba62ddccd446f29b : FAILURE in 17m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fc82ab5551c54c3aa7eb10ac135e63f1 : FAILURE in 26m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c57b0259ebe47079158a3f7f106a467 : FAILURE in 25m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d628b695c52e49aab1c1cbc0c8e8f92b : FAILURE in 22m 26s","accounts_in_message":[],"_revision_number":17},{"id":"d76174973f95d31ff0bf2ec6e78c1eb375c8111f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-04-04 10:05:05.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"ef15728c140bb00d1fdc90609e4cd6cd35292395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:51:32.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9c7b990f0af4db1b535d0e70a630ac1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9995ca2e7d5248ed8d0ea1ef29c28656 : SUCCESS in 42m 47s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e24daa1da32643ed8e519a682eac5b84 : SUCCESS in 33m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56d996343d5f4c98a8ad182aa34428e1 : SUCCESS in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c5488bc1796d0414b1f00f50818d3cc0e65bc335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 11:58:51.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad02b12ab6a445d49cefce1756756f44\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e9e5ec99e94a445fb6caa1ffc24fe198 : SUCCESS in 1h 51m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8d03619f156747c28f6831c0023ed2cb : SUCCESS in 1h 13m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8896a34117d4097bd5780ba6fd18d17 : SUCCESS in 1h 49m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7a367c20b7af4fe18ab1428f622bfc85 : SUCCESS in 1h 17m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/938879b0c50941ed9381a60c5db3dad1 : FAILURE in 17m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07a3af7e4f3b4481ba455915d27faeac : FAILURE in 19m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f733d9f72c7a404f9bc5f83c5fcd33bb : SUCCESS in 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d5a63b677440489a12d8e388def33b : SUCCESS in 14m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e73f004a14294c99a3be398435833364 : SUCCESS in 27m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68ca5eb9ada942e4b0024d814cb1af4b : SUCCESS in 26m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f77e8638cb3b418bb3c66f510858dc77 : SUCCESS in 27m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc16bd80adc408abbba7dee18a6ca1a : SUCCESS in 10m 58s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3b9971e1486f493e96d558bf0fbfb561 : SUCCESS in 23m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54603c460da64c5cb42a66afb246b225 : SUCCESS in 1h 04m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2da4c1bb41ab4c7ba6327662047ef532 : FAILURE in 1h 20m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/669324f9c9bd49f08329fceeedf83ff2 : SUCCESS in 54m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/797529b4f43b40a68f6de32d797518f7 : FAILURE in 10m 01s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/673cfc43c4ea45bca85b223ca9376d8a : FAILURE in 21m 48s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6f7c89eeedcf4bb290790bf21f0fe715 : SUCCESS in 52m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/149a8973c64d403aac2d5e8590d9892f : FAILURE in 31m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99a2d7ddc40d44caa7fd6902658d8727 : FAILURE in 30m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a94faae8769c415fbebf6b7e627c1ce1 : FAILURE in 14m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c6ce7c6e2148400f94ab168715498b09 : FAILURE in 24m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37c2a394f79e40da9401fd9d5fc10199 : FAILURE in 16m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ab7ee303e0f3477c8902d480addb3f30 : FAILURE in 23m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b83964e52ff34ac398ff66d1c71c1a4d : FAILURE in 25m 39s","accounts_in_message":[],"_revision_number":18},{"id":"ab5542d55fd1bbbe8c389f99814f7eaa68a28001","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-04 12:00:39.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/97/874797/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"55ff20861c6924044d787756fb37efce29d2eff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-04 14:28:42.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"5c3544d6ef181516eed21a87608a1c56e6751001","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 15:17:21.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ba52aea5854d03aa1f1c2ea65b3764\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b676f345dcad41cf8c2c5d7eb4d1e0a0 : NODE_FAILURE Node request 300-0020902951 failed in 0s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76dd4ab71dab44ce8169ee03d5684180 : SUCCESS in 28m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2687d699f98b495394e3cbfcd03991c6 : SUCCESS in 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"48a84564e94a289727bf5c6f0c1d633a514dd0ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 16:21:02.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/873593, https://review.opendev.org/879462, https://review.opendev.org/878527, https://review.opendev.org/877712, https://review.opendev.org/877831 are needed.","accounts_in_message":[],"_revision_number":19},{"id":"70f55de7a0030c9303c0d6ee942a909c1416b1c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-04 22:23:54.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/97/874797/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"cc25af5843e28edd85fe931e38451aac47a9155c","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-05 16:01:33.000000000","message":"Patch Set 19: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"36eaefe5f245f78446174ab5f9d762fb8322019f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-13 18:52:43.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"98f69fe4540fa12297c9c693e2da1e20f2327ef3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:18:02.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/340982b2906c44478a8601b73498f0ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3322257375ac49b69425583fdb0cbe89 : TIMED_OUT in 1h 21m 27s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bab1d1657af4044be0a8742c3469ddc : SUCCESS in 48m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0e5690033524358b758c83c79edd84a : TIMED_OUT in 1h 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"9b5ef24a3f62f2face472995014c7286c9cff7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:09:41.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9865ed9e641d443f8c4bff43eb9eed4a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4bed3ff4bfd946028e8d60b0479ca49c : SUCCESS in 1h 29m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e7889ecacbb441281969037f420cb76 : SUCCESS in 1h 22m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a2b764fe9ce2423c940b153a7bfd2b9b : SUCCESS in 2h 04m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ab446c057e3a41d4ace5544fdd5e5288 : SUCCESS in 1h 28m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a647974da933414c88f192c726ecd226 : FAILURE in 24m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8208cca97d9945c08fff1b1e46d2ad98 : FAILURE in 22m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dee4732ed7d41ab8e2abceb79ff2c6d : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0b5715160a4af4b0205ab925bfc743 : SUCCESS in 14m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd637841e75547edb29691906efc1afa : SUCCESS in 25m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c3931b6378d43ce8247a6f3f0614336 : SUCCESS in 24m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54fbbd1d0bdb40c4afc41acf2eb27948 : SUCCESS in 27m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25405130d8454225b6c1308abf694fb2 : SUCCESS in 9m 56s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f8581c30c2044d0ab32f6bf29900189d : SUCCESS in 30m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98ab2fb3a8714855a18fb8b1be5b8015 : SUCCESS in 1h 03m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8d2d5c874cb420789eb2997aaedfaa8 : SUCCESS in 1h 44m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10aa2bfccac646a6923f66cdeb86c850 : SUCCESS in 1h 01m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/240705e5ac1a4ab597d1150d1def1964 : SUCCESS in 58m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3aed335e4d624c39a4261a2dca87220f : FAILURE in 24m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a46526cfc364439b2554bc0c8714cae : FAILURE in 36m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/983a94d93a6c43f48ec38e5ede631a77 : FAILURE in 16m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/abb8d423ae0e4ad781cc328b7a1a96c0 : FAILURE in 23m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0ee562244c8d4dd283732e5e77c4f246 : FAILURE in 28m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9068a00b228f4637a320acfe46b8698e : FAILURE in 27m 32s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3fca59e30981495d8fc7640867112105 : FAILURE in 24m 20s","accounts_in_message":[],"_revision_number":20},{"id":"a382f4a8a18b2b76733d7efbe68eedcc2a58061f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-13 22:29:17.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/97/874797/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"bb76e5a10223763f0c307600cce85d947fe6a916","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-18 18:41:05.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"7f86fa76142ba4be4684a4fe4a45ebe0b9b4f201","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 19:08:42.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7b198a3b0848e9a02ef3e8a3b6d836\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/736b609e0aac4dcdadc57aacf9e73a60 : SUCCESS in 24m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98ec2645247441788beef081ae990963 : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73a46d285a76452b9ffa5c96ab6ec84a : SUCCESS in 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f454edafe433a7464a9d252c0e9b6b6a86aba803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 20:44:03.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/013c329944294269b7409667ceb2298e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/abbe14d7f0de43c6a69251341bfb519d : SUCCESS in 1h 17m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cfcd5b79bf51451693c43cab9f185656 : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05bb5d653ec9421bac645ceb6c32c4a3 : SUCCESS in 1h 47m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4985e08386674d92b293bdc113b68fa4 : SUCCESS in 1h 11m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ef6d8e9bda5d4002ba37cc0b6e3cd12b : FAILURE in 5m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4a9b5ec756df449dbe0e5ddbc74cc8de : FAILURE in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73e0c3fe31b445ae8913a31b270e7e30 : SUCCESS in 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61caba8bd8848cd8664288a424458e5 : SUCCESS in 13m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22b23e9fca0d44caba12949ff2bc19e5 : SUCCESS in 27m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73784eda31ea472fa93b01d2856fe45c : SUCCESS in 25m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae7aa40db7c8438896dc8b74232264ed : SUCCESS in 28m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7544e2bad9e140b69feae870cd219155 : SUCCESS in 10m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1ad6ac798deb46a087efce24a9c0ab6f : SUCCESS in 25m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e69bed40bea84bd3be7906eb88a23893 : FAILURE in 46m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfef988576db4a47889e0f9611e8a4c3 : SUCCESS in 1h 45m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe206c4b0690474e97c92276e1298593 : SUCCESS in 56m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c0d4f426a1c04750b84f36b45053a35a : SUCCESS in 58m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43c1dd05259346b28a942de602f1f765 : FAILURE in 19m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/82256d9ead3646e7b363d1b8687a418d : FAILURE in 29m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/28dfb9dbc3e747fdba967fcbf919dabe : FAILURE in 17m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/296c66fc09ea44e7af3876dacb1d3069 : FAILURE in 29m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7123b4ae33394e90a94ffc83a2cb991e : FAILURE in 27m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/72bbfb80abab4cf79f6daddc31f86a00 : FAILURE in 17m 29s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5c812c8165d24e71b485c298eda1f6fc : FAILURE in 25m 33s","accounts_in_message":[],"_revision_number":21},{"id":"f92392edb1eae79480d8a05877a038df2c4e8a3e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-18 21:13:35.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/97/874797/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"35b7194f7bd66a1c15efe69b2c37a762b659abfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-27 14:35:11.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"d346c213ddbb5c987281da2a507c73d5abe92dfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 15:01:25.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64be0221f330449f9e90404d76d836bc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f0fc7bad29444b1850b83ed0b782bc3 : NODE_FAILURE Node request 300-0021077992 failed in 0s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc51af0563c7430489b5d4993403e0e9 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e54f9f91ef394385a3b7338f17ea2a72 : NODE_FAILURE Node request 300-0021077994 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"695074a207b089474e27badf8912607f2b5c5e72","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-27 16:32:36.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/97/874797/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"022afbdca2fab2a4a7396f72c0651d6fdbfdcbfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-27 17:29:02.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"014f2d00344ef37bdadca26b3b042633a1ebfffd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 18:36:24.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92436c33e0ba4c449ca3847e6fa32feb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc88357a422a41d7beee2278e660eb3f : SUCCESS in 1h 03m 58s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91327c8c69b844ad82f4b68e21ba2fb0 : SUCCESS in 1h 04m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c38a62404d7d44e2a429b902ce82a4c0 : SUCCESS in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"ae5e3caf250a8efc9553c385e16df81bdc5e5fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:50:11.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d8fafad42634fa5a9d32537b19bf8ee\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9fb31d509c29413491e7b0f1f33c921a : RETRY_LIMIT in 5m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/478cbfaad3ef4b35a59dad584d3c5469 : SUCCESS in 1h 02m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/76fe33b348544244a8efbf4d7d920e51 : SUCCESS in 2h 15m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e45813450c844bc0b7ee9dc29f98a562 : SUCCESS in 1h 45m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/503bd1afd72c4e18804af766d2da16ce : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2702644952f4027a875a90d89598c64 : SUCCESS in 44m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05652fab26f5452bbc90a8125213b189 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202247ff4a0b439e853363efe25a6e34 : SUCCESS in 15m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1fcab79160d4f2a80d9bcc2bcfa8089 : SUCCESS in 26m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11277f333d4e4055a403e02f6bbde15b : SUCCESS in 25m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ceceb11d8544fbba005c2e2eab521f7 : SUCCESS in 26m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1efffaad4b4f5dbb8e68914b4294fa : SUCCESS in 11m 36s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3c143fffe29642dbba597a3ed82d1dee : SUCCESS in 25m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d41f57f4407c4e939ff8c201e72b5138 : SUCCESS in 59m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd0356f5794d47d08e02b9b025986fc5 : SUCCESS in 1h 40m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b76132da637a42a992456bc0da7f8e27 : SUCCESS in 54m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9638e8fbccfd44d1ab1fc0cfb94831aa : FAILURE in 57m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a5a7bd4d81843628d378076c678f581 : SUCCESS in 1h 33m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/138f4f34e11f458487ed4f92a991fb8b : SUCCESS in 1h 14m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3c0f52c1b8af4d0bb1a2f9c847df9ee4 : SUCCESS in 1h 12m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c3026064619f491eb87bec46930e2426 : SUCCESS in 1h 50m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b665206d301452bad950fc53e39a600 : SUCCESS in 2h 04m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3140fa16ff0549d3b2e8f592ce860190 : SUCCESS in 1h 15m 13s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7874730a78e04e5bbc4fa12fd7990f78 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":23},{"id":"64b551f30bbd6a97e9842e25997ddb65ddcf733c","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-27 22:12:52.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"e824dece3f9be2bb67b406cf4cc77931466bf9a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 22:35:00.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97b6790ad6224d4faafbc056cc46e888\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ed624ec433945cea3d51d9135ca70f0 : SUCCESS in 20m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b379feba80c4cd790e7d2c97475697b : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab561352139e4f39bd89417caa7f3d11 : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2c9b6a667aafa75bb835fd23ad5ae48f50ac6d89","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-27 23:07:41.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/97/874797/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"9e4b360a3382410e61df66bb05f7bf79f0d319a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 00:24:37.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2789200e9180410ebd65653c473eb442\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9c64fad0d0c44acb9665505f85693bed : SUCCESS in 1h 42m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4d0e81a61fc0407abd6e04a49ba23d19 : FAILURE in 5m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21a38839814d4b769b3ba5023bbc13b6 : FAILURE in 4m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bdf64707a8a345538d5a7666eea4d4dc : SUCCESS in 1h 37m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee21a30443954cd988a079c1c23b95d4 : FAILURE in 14m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6b5d5750bb74e2a8a40a705213b788a : FAILURE in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cac025a19e474eababcd8f5db4e9e589 : SUCCESS in 42m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea30cc4d772c432e8e271c5a07cc8dbf : SUCCESS in 13m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/890bdabce4464609a421726e74c69e33 : SUCCESS in 24m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/586d12761058499c9f3da567402be797 : SUCCESS in 25m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c815d742be5456492504c7c74285c48 : SUCCESS in 26m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c4e4a3c4e847608bc65944b9e574b4 : SUCCESS in 10m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d28e468635eb473d81f95c73c16451b8 : SUCCESS in 32m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7bcd7b3a0904257a11a3f381fba3008 : SUCCESS in 1h 27m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2be13e0839414fb39fbde4d01a3cacbd : SUCCESS in 1h 47m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb980da83e784a08941c78b4f7e9d757 : SUCCESS in 1h 03m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dfdff74bc75341b3bd14c5dff723041b : SUCCESS in 57m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/39579267d9d94de9b2d89a5d6b2b6505 : SUCCESS in 1h 31m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/925e4143013a465c853c03caf1305288 : SUCCESS in 1h 29m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f4afb80d105c46c89a70c7c96addf03c : SUCCESS in 1h 21m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/81de124b6e3543258103c7ee9973f42a : SUCCESS in 1h 55m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5f07a8c2a7e471cac45a015445f100f : SUCCESS in 2h 03m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9665c9b574974720a33e224341d205c0 : SUCCESS in 1h 19m 56s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fd748a5d6e9b4b1f8d88a8d737f5fc9c : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":23},{"id":"405910d4cb0a362139e571b81678ffe3ff921481","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-04-28 13:45:58.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"d56ca06ab6befd0f76107a8357f976d9a03a0409","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 14:33:50.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7889eee48aa45e8bfec6de85f3169db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98ca95497d754f938a849d57cbdc5558 : SUCCESS in 45m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19ff277f143846ed9877cb10ef2f718e : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b492412534b49ef8247d7d9186f3938 : SUCCESS in 31m 03s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"677b783e33f68993f99e01d065d265db15279284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:15:03.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef868d48c05340ae872c3c081bcb6486\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dfb0f88af5154c53b2786bb36af8a8f4 : SUCCESS in 1h 38m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e0d6fb4d64d4afa91d1150e4c654d16 : SUCCESS in 1h 52m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c9d2ac7eccc947e981ba0ccf853e6cd6 : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4ca47e1ad4ac41debed488301421ef52 : SUCCESS in 1h 42m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2b1fbae5ce7f412f9f2dc879f8102639 : SUCCESS in 2h 04m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d44107b383bc46c7a63743ebdb65b532 : SUCCESS in 35m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69e54206f4f4435b9829483b40131a1e : SUCCESS in 39m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d824bea79d9943eb9f4859886b69ac06 : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07225cbe6e614bbf96defcf78cb8dee2 : SUCCESS in 35m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/805577c12996443d8b9b330745cc7f17 : SUCCESS in 55m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c39b0759f8f24fa59bc19d82345bc4de : SUCCESS in 55m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864f4c119a58449bbee1bd3f1ea74f63 : SUCCESS in 12m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a221c3a47a0f4ef9ad080136cffed24a : SUCCESS in 51m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f346356172994ff4aa1744f49cc6522f : SUCCESS in 58m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c10646182e644ade89b3c8121002e6d7 : SUCCESS in 1h 52m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/624121b83c2e48ca970c9503afdd9e21 : SUCCESS in 1h 03m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3bd3087c5175479b846456dbd7418cd9 : SUCCESS in 1h 03m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9487ccba8a8473bb310cb5656566456 : SUCCESS in 1h 19m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c4478a9511248dd8cbeaf5a86c15086 : SUCCESS in 1h 24m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/748d9ef562c144df8dbe7402ce310154 : SUCCESS in 1h 14m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dda5ea4f268d4d16be1cdaffcd4aef6d : SUCCESS in 2h 12m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3c43bd2b65e4c6084da95be140e21b4 : SUCCESS in 1h 46m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/41c49fbe947c446dbceeee76349823d1 : SUCCESS in 1h 14m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/41eb7b07f17447778e1f0c8e78f03dca : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":23},{"id":"21597a97214115569bdfd95f2467b07682ec5c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-02 11:14:48.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"33177d65591997eba3038dea68909e4093f56471","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 12:36:13.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac7b24f15404d42a78607b5f67f4b6c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4f421c141c44d0986ec89c3c5faa04a : SUCCESS in 37m 20s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/deb93eb831da4971bfa1c883f3783aaf : FAILURE in 1h 15m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cda75bdc58247c0ba02dfb1bf68e17b : NODE_FAILURE Node request 300-0021111291 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"95b181b9062d51d06d0036d13473656521ae83b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-02 13:13:38.000000000","message":"Patch Set 24: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"c7f573a13b441ebe63acd9a6e6c9d643d4faa64e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:50:20.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1615c940b2604de08539e01e3e43fd81\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3027a5148d954aed8063b9ae0c236c19 : SUCCESS in 1h 48m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a651c5dd58e64aafb3a6eff4b026efd9 : SUCCESS in 1h 44m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0819672058a1429a9b7bb85ea42b3d89 : SUCCESS in 2h 02m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a86af95fea804dcda7946616a63e5e4e : SUCCESS in 1h 32m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0dc6f369c614c638aa1b59bed7c34fd : SUCCESS in 1h 49m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91abe6aa2c14468d871b44f384b62cb1 : SUCCESS in 40m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27f9fe1e252041ecb9ace750ced8938a : SUCCESS in 37m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336b4c8ec852454b9dabe2173360a127 : SUCCESS in 15m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da22dddc9df54894906bb64e0aaaf425 : SUCCESS in 29m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b80e8f3242624d6da743949b3aaf98be : SUCCESS in 26m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33a386398354452d8f0d2bdb481eb26b : SUCCESS in 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34135d2fa8bc46078c637e374e9ef6d3 : SUCCESS in 11m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f9ebc67f071b4416a1c91238c70f4943 : SUCCESS in 24m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e0156825d584d93b918d2807d5f30c3 : SUCCESS in 1h 00m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc43c301bcf34fe2b0b9b862e50b8b31 : SUCCESS in 1h 46m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71336aef5f974e6d867fb5613e359480 : SUCCESS in 1h 00m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ab059cb9b694c3f8c7419d948d25644 : FAILURE in 44m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5d85d281f884f55a51990bf5215f555 : SUCCESS in 1h 48m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/07ee69778ee84ffa9d22df2c53c54bcd : SUCCESS in 1h 19m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8a62b88c77294398b1d2f3ab62d45c4d : SUCCESS in 1h 28m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/71868bf4db2c4eaea5b8a2de09ff2baf : SUCCESS in 1h 42m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ce906410ccf4185a4ffe7368b1587d9 : SUCCESS in 1h 43m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/35eecd4e4e70433d94253245998a938f : SUCCESS in 1h 10m 31s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a778fe5aeb2a452b9ee82d1134ef814f : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":24},{"id":"180e37ef5bbcbcb8dc896ff6dbfa3c3444238447","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-02 15:23:22.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"8b300e69e00ad64a89e91b8c70ba8b8eca02ed09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 16:21:48.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6291cf076b8346b4be00977501df1763\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac1c32d189004b08af2b085be4b825a8 : SUCCESS in 55m 08s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9b8cd3938984735b5dd5baa31e57842 : SUCCESS in 56m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/55cf62f0027c42b09c5e2f437e763a5f : SUCCESS in 55m 17s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"e163803d9706e0a9253de4c2de4d0a5770526a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 17:57:40.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc1739b3cac543fda6ab5867dddcb1f2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b8c9aef04c664da0b89383c23b3e1f22 : SUCCESS in 2h 04m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/84ad7e52d33643298c8976b2c2d7cb7a : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd07f789b4394c059594903c3886f6ac : SUCCESS in 2h 00m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/737ea75ce788491c978b824a8b93e11e : SUCCESS in 1h 32m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a5dab24b5f04044a27578047a64fefe : SUCCESS in 1h 35m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af1d99759442441a8e4a01b8f73ff9ac : SUCCESS in 34m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79471cd12485410fb7b5f000a2b18244 : SUCCESS in 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be79b1ed1bc94fc8b15d0b1a24e5da54 : SUCCESS in 14m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313024f1eae1469fb383bc605c8efea4 : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f0cf81ee81e487ebc5c8be47953c825 : SUCCESS in 24m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ce1be0d6ac446d8b17ab69661ceb800 : SUCCESS in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f873e325b184f19af313f7b75d1461c : SUCCESS in 9m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c4e26a9b23264f2190be0844f52b2e38 : SUCCESS in 22m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2fd4c6c7b1f4d35857358fdd998a50c : SUCCESS in 52m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26f82adf09fd4a29b4f83cdc63e6a74a : TIMED_OUT in 2h 32m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5a6434e47dc41949ac7dd14b29a713e : SUCCESS in 47m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b223f4ebbbf459d99ec1d98c7cf6cb1 : SUCCESS in 55m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a8ffa92fd92048eeb88c60391237f97a : SUCCESS in 1h 29m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/66f8c0a7403c43dc8009e26343971c30 : SUCCESS in 1h 14m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b88f60c1c04648f28a1ebf7e769ae2ef : SUCCESS in 1h 26m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bfc7698bfc254152875e6b5210866439 : SUCCESS in 1h 49m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3dba089b5ba5463a805fbbd14a0193cb : SUCCESS in 1h 55m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2e1c3887ebc34775b978eaaf82df773c : SUCCESS in 1h 18m 59s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/28f77ad3aed0411186708e8fd778e3a5 : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":24},{"id":"db0454994ba8d456fa46ad0a60b53d7bc98868ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-02 18:30:22.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"fd51ae2a55fa5d5ce89be9aee539ebe6ba1712ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 19:38:11.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/805150d20e9c4febaa3b56add150014e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fb6c1afbd384dc0893b2e438c7c0694 : SUCCESS in 1h 04m 37s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c86d6f9cfed140dbbc5b92982d8a9c8b : NODE_FAILURE Node request 300-0021115208 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/217466ef66a74b969313d8735b68b529 : SUCCESS in 45m 59s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"da72b681f0ab83c1c1a5068d7e8762c400b90fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 21:32:23.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/477d8cf2dcfb470583f23f795bf6996b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/975382f5284f4b5eac19fc78ada483a6 : SUCCESS in 1h 57m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b582cfcaae98465d8f0d00a9f2e7ce08 : SUCCESS in 1h 36m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c864441397674e7fa053d47934b0d8f2 : SUCCESS in 1h 54m 11s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2070df19ad3547d1af24d9e3ce38d8c2 : SUCCESS in 1h 33m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ccb1eebbed974beb80252b608677ef2b : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a071252de378470094248d8ed58c6072 : SUCCESS in 41m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/565eac4226e14d898f28d98029214a60 : SUCCESS in 48m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d46c512b8942ee992f8170b9717e3c : SUCCESS in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52fc45629f60419ab77521a0517623d8 : SUCCESS in 29m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb82a90116964f929b3cb6601048f81d : SUCCESS in 30m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5becadbc0f44989ae2103702b4c38b9 : SUCCESS in 30m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedf28a4143f462c81a94931f593bac5 : SUCCESS in 10m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fffc9faa66a2491696e42c50b84db3a7 : SUCCESS in 23m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f07903bccf84dbf81235c4cf0086b69 : SUCCESS in 56m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/39d96575a0744e38ad55ef013d6946eb : SUCCESS in 1h 45m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72e31f4df0f54fc7ada75e7321ab4b75 : FAILURE in 1h 00m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e4223e9e64d45c6b275e7b0c295e31e : SUCCESS in 1h 04m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1fc81964ea9f4affb455fe69663480eb : SUCCESS in 1h 21m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b49eff44af784476b38dc2a7f5f5a6f3 : SUCCESS in 1h 29m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c403efe0655f48bd9dfc952e5529ef14 : SUCCESS in 1h 07m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a9a70607c5104733b5fb29fd034e645a : SUCCESS in 1h 59m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fe996cc8fcd4489a59a7b5e31202ec5 : SUCCESS in 1h 58m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/58e7e16d50244b7ca95ed67f2938dd0b : SUCCESS in 1h 41m 12s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/17d5653ad7884f3fa3434acde3b47b18 : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":25},{"id":"80cc629b31b595ab4926bfc48cd4e09f6416e5e0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-02 21:43:08.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"bd67dcaca64520c764c651fe9f972cdf0be0a44a","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-03 09:22:23.000000000","message":"Patch Set 25: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"ad2b32280697572d2e6619e0b40de08b82a8c76d","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-05 23:53:50.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"e54a329895d3dda1c8cca753cb80d3ebf13cb5c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-05 23:54:12.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"a56a52ba03d900aa8d2d979d89bdde68fc19ae82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-06 00:29:55.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486ca5afe831413788a4d9e5a1084d49\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b1a3cef22fc46189c0d6fde88b978ce : SUCCESS in 28m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/deea91356f39450c975341ebbc47406b : SUCCESS in 33m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e8bb2518d7a43d3b4ed798a5920591a : SUCCESS in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"e1b1298a790b5b3933edfc452ddd153fc2855b0f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-06 01:11:24.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"d051d5668228419db47d8ca7b6ffef324bdf6210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-06 02:35:34.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d31dd79c8ab1462baab20654517b5293\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/19c102b138bd463383ed70780ec315fb : SUCCESS in 2h 09m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/31cde9d2558c4183ab5b19d54686f460 : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d9e6943cc004db0af441f084bf3da19 : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/16d9e16879ca416d85dc9b4be30fcf6f : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8397d9dce0f24ee184db530434479be9 : SUCCESS in 1h 48m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c162b7387274482be3c996616f10e2c : SUCCESS in 45m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c3cc4543fe248cdac3bd1a96ce031ae : SUCCESS in 43m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a46673978c04382a5a32f96115a5869 : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31302b3d58474218853a2a7e2a29fe20 : SUCCESS in 25m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/557bc89e1f114dd58026c68502958b3a : SUCCESS in 40m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/942f9a53e6fb4a52b6cb0b109ceaf8f1 : SUCCESS in 39m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b0142f08cd4419a5d97e411921bbd1 : SUCCESS in 15m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d9bb860b378f4029b4c3e1325c58029b : SUCCESS in 28m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/484a490a3a1f4c59bcf9972cadb4ba81 : SUCCESS in 1h 01m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/77c864cf13d34908aa04664cbaa89396 : FAILURE in 1h 04m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fdc9c6529674b87b189f8a48b5f9509 : SUCCESS in 1h 00m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/79f98854cb71437cb252b0f2d418420c : SUCCESS in 1h 13m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2996b3c4abfd4710b18522acf1c95d10 : SUCCESS in 1h 18m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dbefc93cd351474cbdfb0fadcec4f311 : SUCCESS in 1h 26m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cf9cfa108d774e0390d1b91b29356bcc : SUCCESS in 2h 39m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/32f4d282752f490cb0aa5a3f0b9f0e5a : SUCCESS in 1h 59m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2d7db26765c401ea0b18ade4c2399cc : SUCCESS in 2h 03m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dbadb387dd4245daaf519a97466559b0 : SUCCESS in 1h 11m 44s","accounts_in_message":[],"_revision_number":26},{"id":"00231a15a908df2aec4160978ecd45e49e40b4d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-08 13:20:13.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"0bf79cbd518f6a1ceacce1237f6e43d7691b29fa","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-08 13:22:49.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"334f8cbfc0c2d7fa876f2b4edd5a06292f005859","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 14:15:17.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/081fba454ca04633bb328982d53e78e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f841c685207f4d2e92af97ef9f4a8f9d : NODE_FAILURE Node request 300-0021163872 failed in 0s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea18cf9c1c024a939ca5518ac385c121 : NODE_FAILURE Node request 300-0021163873 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02b9f4caf558421aa67409073de2280f : SUCCESS in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"ff93e859fc2850158db225a08fd7976b1a1f744c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:21:23.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a2460a85794e81996e6e77c727e2fa\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d078b8f7e5fb421baa90c7225f6912ce : SUCCESS in 1h 38m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d875930550824b4587db5c5fed12a97a : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f31adb73ce6c4300bbbf9300b94be962 : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96179850e42b4b839733048fa6016ead : SUCCESS in 1h 44m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1c10c82df54a4bf6b92d1237d694977d : SUCCESS in 1h 41m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4768b8edaf084809bbd2d90b91147576 : SUCCESS in 41m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c154c58d9fed43ef9169f9b643aa1cff : SUCCESS in 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28db0219c8114f06977269452c5dc79d : SUCCESS in 14m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5231c3cda7974743ba5a2fbfa3b447dd : SUCCESS in 24m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f16726766e0543c5ba308d7424f46870 : SUCCESS in 29m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76574d63a898455f84f9188850dee09b : SUCCESS in 24m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8231252af14e10820a92908bd14a3a : SUCCESS in 11m 36s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3a2488708a784e378410d1cad0e24c39 : SUCCESS in 25m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b39b08d784154a60812d90b76ca6e64e : SUCCESS in 1h 10m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/654297e14ca247cb9d82d4141726e65e : SUCCESS in 32m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d316309a169424da5261a84521ae0ad : SUCCESS in 1h 07m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/02cde3c4573341b9a8ceb0a6202af5a3 : SUCCESS in 54m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c692ea70f424ae69fd079efbcfabdc4 : SUCCESS in 1h 45m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eda5b44f0c294960947654402d8ea48f : SUCCESS in 1h 23m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a23f88f7556e4fc586a5edc74c398eba : SUCCESS in 1h 30m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f28edaeadbdc4446a0d7c6b2a1a7ba48 : SUCCESS in 1h 51m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/828e85de13ac4cbcb0ffb378ab689606 : SUCCESS in 1h 49m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fbfa439ca6e84d02aaf2e96798d9cbe7 : SUCCESS in 1h 38m 47s","accounts_in_message":[],"_revision_number":27},{"id":"4304c4f31e9a52d5fafb4a6c4dddb76d5c64393f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-08 20:30:29.000000000","message":"Patch Set 27: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/27 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":27},{"id":"a687b47aee019e4ec0033729affd786acc2c29dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-05-12 06:17:10.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"15fea84aba6ded752d16412f1b365e0d52d479fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 06:40:33.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a454db6ed93451b94391fb2c91a70fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1e3be6d21404435ab30c5e6017e9e09 : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b4cdf1008eb4b5bb0d9c27ffebe518f : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73ced1f7b2364d53a1686001f57b0b7d : SUCCESS in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"5718538810e87d0def33efbb3c1e6bd0b10a9da9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-12 07:34:38.000000000","message":"Patch Set 28: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/28 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":28},{"id":"92f25be4278f6eeeaa3d446963b026a1620cfc48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 08:33:13.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/215b28e26f5c40debbaa7c41b5be5445\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/72a559115d5545ffaaaae20650bbc580 : SUCCESS in 1h 54m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/47fc7160f95342bbb3cf15352f178276 : SUCCESS in 1h 28m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d4e65f1d9921464cb597d8e85e61c135 : SUCCESS in 2h 12m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0753cbd2be7045a096a7fa0a766546e5 : SUCCESS in 1h 50m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/87eb4c06313e4fdaaaa41ea3725d3f07 : SUCCESS in 1h 57m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b95fa416ef0423598d83932b3c54fd3 : SUCCESS in 40m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2315a227f4d41c694b39b0dfad33795 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71fda42e882f4724bbec7937a3e2cb1d : FAILURE in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/332ef5cf3cc84d02b3bd250ad93043df : SUCCESS in 29m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76897ca4da0e4848a326ff9aa39ced9c : SUCCESS in 25m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07d1f58978bf43659b78e41610a0a5b0 : SUCCESS in 29m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e4c3ea83714dccae110d0b221198aa : SUCCESS in 10m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9763c73a104344ffb772c1424ba74994 : SUCCESS in 20m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e2f5f6c0df04256a062f91b727bc665 : SUCCESS in 44m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b616336257d9492d8e3f3fa3107f56ea : SUCCESS in 38m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bac660f900fe43eb884dba370e0611a1 : SUCCESS in 1h 08m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/492a38f53c3d46e18cc7fdfd42c5b0a1 : SUCCESS in 1h 08m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/92865b70f2614906b626e693ec5b80aa : SUCCESS in 1h 40m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d5303d06fd1f4551a4a192875e36888c : SUCCESS in 1h 10m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/71b5f418e8b240718bb6c717cf84d0e4 : SUCCESS in 1h 23m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/295a7000e4f8428a98c4b196b2046a90 : SUCCESS in 2h 09m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d516d94ab9b4681ab280b7e7c3c5eb9 : SUCCESS in 1h 36m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/79b68f6fadd34dd484c44b157d3a14b2 : SUCCESS in 1h 18m 48s\n\nWarning:\n  Comments left for invalid file neutron/debug/shell.py","accounts_in_message":[],"_revision_number":28},{"id":"5b5119c777290292c0b5b6c3fee9007872913606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-16 21:45:54.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"de251c64306b1b843a1f9075bbf931ad01d58412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-16 21:55:01.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"42adc67aa189b36d65de4f357a20edf9553c4af2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 23:00:05.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f524fc5947a4a44b4026debd55198b2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/647c00ca4cfa4c6daac04819d8a24784 : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c30e07e8ed924d4898e2e57cb1243595 : SUCCESS in 48m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5da052b684de48928d49e33f0bf8e63c : SUCCESS in 1h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"9beb6311079b66465abd7b77cbda2cf36c85e7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:08:15.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2398a53f4bf44a1a8ad039ae2649b495\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ab50132da7c84358a23f826087f16a46 : SUCCESS in 1h 50m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/adf5e74ff10845de89ae388ce9192162 : SUCCESS in 1h 36m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89f23c2571f548ad83940e195fe52888 : SUCCESS in 1h 59m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14644c223c5943d1b534527c21e821aa : SUCCESS in 1h 50m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/df3d7f7d4c684ed8b3b63fb0f6090dfd : SUCCESS in 2h 00m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce1103704830465dae04567f4665ae46 : SUCCESS in 38m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcd81609d4f846f38d937208b7772809 : SUCCESS in 43m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7789821ec2a4d4f9c2c5182afc4962d : SUCCESS in 14m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a55b85f95d5472bb9379eaf389279d3 : SUCCESS in 33m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72f9bbc5f258408d8c66bce7d3dd408d : SUCCESS in 32m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de363ecbd277499dac5fc988c57bb59a : SUCCESS in 23m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7a90e1982949eebf92f93e1fd70b6a : SUCCESS in 10m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/89835221d0984508996cb1bc6c3ea02e : SUCCESS in 26m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a07a7a704dd446cd915982b6a7cc5f8e : SUCCESS in 1h 10m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d0ff5f000884e68923cd3ba010b5aa6 : SUCCESS in 39m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab54481869aa404a9c640f11ea8c8eac : SUCCESS in 41m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3eedbc4598bb4d5eb331affd4992b8b2 : SUCCESS in 1h 03m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7923c30455e548a997a667deecca69d9 : SUCCESS in 1h 13m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b8c40ed1606407a83e91c16c2faf2b1 : SUCCESS in 1h 27m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/385d663f3b7841f8bcb8fb5dd3ab38fe : SUCCESS in 1h 26m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b1d79cc436d748d0bdb16cb8ef5cc7af : SUCCESS in 1h 25m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4340ff305d994a0f839a114bf28e9940 : SUCCESS in 2h 07m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c232f67dce424cdc8e45a7307b9ac88b : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":30},{"id":"34f68ee73b75209a37a161eb282b2710b21d3b2a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-17 00:17:26.000000000","message":"Patch Set 29: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/29 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":29},{"id":"a4748459c8537b72ee4a6e612b8f1be9d0dc11eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-17 05:05:47.000000000","message":"Patch Set 30: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/30 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":30},{"id":"088be053c1095a37e39389b3a223a6546edfa20c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-19 09:48:56.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"e560b9d53fc58be13fcdc1a17cc9f60fbea6f225","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 11:13:16.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730795eae3a14963bc4cbfc6bb1c66c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e8e161571bf4a979abacc67a1afa711 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8eaeec181161415787d39f45766beb24 : TIMED_OUT in 1h 21m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4694441905d04633b6079188bb6a7db7 : TIMED_OUT in 1h 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"82356461ca2bff323f85aba2fcb73fdce67ab717","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-19 11:29:32.000000000","message":"Patch Set 31: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/31 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":31},{"id":"baaf924911bff7fa1be029f6bc67c987a3711ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 11:55:32.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15db3e8d22a946228551cbd3becd6b6e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/08f1e5b200d74729aeb0061fe5142045 : SUCCESS in 2h 02m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e48af5d492484322ae1d827b4bdcbc82 : SUCCESS in 1h 23m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ebbc3ee963d04db4bf7a6e6db81f195e : SUCCESS in 1h 29m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c2151a275f034ab5a7c5b485421442b2 : SUCCESS in 1h 42m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a61e68d238f24cc78559b7fe6ab2434f : SUCCESS in 2h 05m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00915d09d4cc4faf99148a804024a04d : SUCCESS in 38m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc2d95a0052f49d0a774f06c14489749 : SUCCESS in 37m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b10a418eab4ef4ac1304032ff5c0d8 : SUCCESS in 13m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25f3dd13d0304d97b9b75d9a9d315018 : SUCCESS in 29m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4a1c759305f40489c70c632d059e961 : SUCCESS in 27m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0bfb6623e1d4f62b24d5743f89d7e16 : SUCCESS in 39m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7ecfe81c2d4555ad7203f176e80faa : SUCCESS in 11m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c60552cfdd9a4816a8e20af1cf240ce7 : SUCCESS in 25m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96d460c0f949489e93a9e370c8b2562c : FAILURE in 53m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/660db708f63e4adabd54b2e77098511a : SUCCESS in 25m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c87f3c6f4ad34a7daba052f09ace23a3 : SUCCESS in 55m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7d59fca0f3d4b609b4761c6a428f57a : SUCCESS in 1h 06m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9677166ba374422796ea43c9ec1ab337 : SUCCESS in 1h 35m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fde2dad5bd2b445fb409886b60a66c53 : SUCCESS in 1h 33m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1e6e618391f645179e97b7899cb5284b : SUCCESS in 1h 18m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/77deb31e74d5487d80642025e599e158 : SUCCESS in 1h 35m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd2d9c9fbbac4b9dbe53fd75b723c6aa : SUCCESS in 1h 49m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5391c8a2ac2241c7b736056a30dda279 : SUCCESS in 1h 25m 01s","accounts_in_message":[],"_revision_number":31},{"id":"2edf8e5c4659cfe89d942ca53a2f424631c76cf7","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-19 14:14:00.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"09c08a16de0106108a2930c2f6fef46e98b68b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 15:07:02.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d881cb7f15de4fefb3214603e274f115\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27081971733045b0a3e0da0b7ec6214f : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d829dc3be2164bb4876a6f6ba83189fe : SUCCESS in 51m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8dc6b89b42e5407593ad511c0a9da9b7 : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"7c8f2ad152104fcef29ad1a0e42e06356351a306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 16:34:21.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9649bed5002420e95f1ffce77e2dbd1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/573eb42d613748ac9dc2729b4e15f523 : SUCCESS in 2h 16m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4f158c7f43147a0bdd3cfd7236ee1ea : SUCCESS in 1h 59m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7d4c71ac1a64ea9b30597b186a367f9 : SUCCESS in 2h 18m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5346bb6897304f1cac8af8d75895f014 : SUCCESS in 1h 35m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c86a38b0fe7643689964e84867102e61 : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1557f6d85c54e27aa91b8dbb17e3f04 : FAILURE in 41m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/700bce0f5ef54a7fa34e7b54958638ba : SUCCESS in 42m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ab1425144c400588793a5120f3533f : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82a76cb71ce744609620975da8b04b9c : SUCCESS in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70b6c3da8b7f48bca142426006d3c0ed : SUCCESS in 25m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb60dd997cdd40cf8f917fdaec0eef21 : SUCCESS in 37m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0741868946f34ca8926d7a0976fb7a35 : SUCCESS in 11m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a1703f4604bd49329236c0e388774012 : SUCCESS in 32m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ebead84f9154be1bd0877584615cb19 : SUCCESS in 1h 07m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71636d227d044378b66e7c0691dbfed5 : SUCCESS in 43m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7732fb68de6e4ce4b4b4017c154b0653 : SUCCESS in 44m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6e1bf19e90e7420bac34f0de51f9b681 : SUCCESS in 42m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e84bc9e684549388a57ccea7ae269dc : SUCCESS in 1h 47m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e6ba7a89dc3a4143a313eb07134347ea : SUCCESS in 1h 25m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/55bb2dd72da04b4ca0af2ef7ba190f90 : SUCCESS in 58m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e27c2c60bc7f4d8ca0f7a8d30e4656df : SUCCESS in 1h 32m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/944d3db59d1c4aefb9fa793e8f062a69 : SUCCESS in 1h 37m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4bcbceb1e6974019b44e752faac3fe17 : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":31},{"id":"356c80bceddc54839dfed42de7bae0bb77ae896f","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-19 17:52:36.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"3470cb8c6bbbf5ea1bef79598f4b6d1b886b373a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:54:36.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d28ac57514234bbb8794408b357cc58a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8c09082057f4b95be9c14b9093b941e : SUCCESS in 59m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/495ac9c9596f427e88d3009e5f5790b1 : SUCCESS in 58m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97b9f4828355491583eabcc183171003 : SUCCESS in 58m 28s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"dcea2b9ce6e0353effd4f2ae37c3e5714089825f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:02:36.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa1ad6c4a614fe88094719d2b4f9b2a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/37b00c10ecca4659a0e5a7cd7aef4815 : SUCCESS in 1h 50m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/168ef09f750c4635b8328557e4d8e1fc : SUCCESS in 1h 56m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1338923e844048e5918a5fc04f950418 : SUCCESS in 2h 08m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d92c90ca86c344b2abdca883d99d0568 : SUCCESS in 1h 45m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/009b71096c4449cc918b72b2c060ef8d : SUCCESS in 2h 03m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04142bca90654ed7ac46b7b9dff8a764 : SUCCESS in 32m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7a5f1eed08b442db91174f652525467 : SUCCESS in 42m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f865f1e49954c00888c1f7e87c30194 : SUCCESS in 13m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0489457d6d1a4ff8a2bed0fbdad2adbb : SUCCESS in 26m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f07871c369742b88fe8f687ff44ae7d : SUCCESS in 22m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb74b66df84a4219ae21284488534b6a : SUCCESS in 24m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16ad135af6b94cbbb64739120c01683d : SUCCESS in 11m 16s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/646bbaa595ba404cab39d032b33e0f62 : SUCCESS in 19m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3e39372036c4c92bced6a844eca49b9 : SUCCESS in 1h 06m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23bbb79c8d04412ea61aea6c58a2f5a2 : SUCCESS in 28m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a38a60b3514d476d886139abc0f9b4bc : SUCCESS in 49m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/112d862c58a34111b57f5b4d06900c66 : SUCCESS in 48m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b903fcb7883943d5bab3d87255ccb1d8 : SUCCESS in 1h 33m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/74e4d335bc824fb29053ccb7a4254b5e : SUCCESS in 1h 27m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9d62194774514491be5d19aa232a4c1a : SUCCESS in 1h 17m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/de4c39b33514415f91f120f3a12fc3fd : SUCCESS in 1h 25m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/89fa8a68f0d84ce79a47aba6ce5e7ce8 : SUCCESS in 2h 01m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2e875500818c471295fc9f2bc53c2c99 : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":31},{"id":"97b9ed6d91d3ae204728b70264795db933ec77ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-05-24 23:27:47.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"ca795fea33ef8c39da25f5d90873968ea68c4594","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 00:01:29.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23269f5a8000483e8412569ff3b9c33c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83a2d60b4ebe4be4985765a7cb0c17a8 : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7eda1988a489424cb2a4e77e3e160b6e : SUCCESS in 31m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93aeda83be7b47829fb0935731e04d62 : SUCCESS in 27m 36s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"8c961a29bd3a9a9d93e402f9e07f9bdafabd618d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-25 00:08:12.000000000","message":"Patch Set 32: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/97/874797/32 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":32},{"id":"fbc39468624aec300739fb97a4a48464de3ec379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 01:46:33.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338402d164ca4ea9bb2608e886754caf\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a459dc585ef04524a6cfeef1722a02d0 : SUCCESS in 1h 24m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eccc672c5d9f49bf81e851d617d477c9 : SUCCESS in 1h 23m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5239d1472c9742f3ad3723422b120014 : SUCCESS in 1h 27m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f17cee6639074bf4a69df66d0e1cb289 : SUCCESS in 1h 19m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8a5e0ca593b7470ea6ac071bb15f774f : SUCCESS in 1h 26m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8d0430fc8b34465bde52b6de6254466 : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c26c7795d3d048959efc4d841381628c : SUCCESS in 59m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a111aa03978642bf821482d75e45445c : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761d3ad821cb411dbaab2dd07921a072 : SUCCESS in 28m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/358d26c9651b477e89e52fbd67832731 : SUCCESS in 30m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/723db325b67e4754b1415e854208b6d3 : SUCCESS in 25m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a89b7b3c24746d9ba08dae64bbbca57 : SUCCESS in 10m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cefed954101b438983727c34851ac21d : SUCCESS in 20m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67c0d5a05a2745c3b2e0a38d50f202ae : SUCCESS in 56m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/402f9638d11f40d493b89741efa4d4bb : SUCCESS in 1h 40m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36044d16dfbc417496b5a924257de52e : SUCCESS in 57m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/395eeca63a894de588903d7b4203eebb : SUCCESS in 46m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/76721800599242fca3eb61d7db9f95de : SUCCESS in 1h 26m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3ad415a2caf4a8591ad138250b47fed : SUCCESS in 1h 31m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4f84d248bf8d42fd9b2a0a44465e46e1 : SUCCESS in 1h 12m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8d3853c6dad44dabbc92ae34220a32cf : SUCCESS in 1h 31m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a356bd3a50394e4ca08b8feef382351e : SUCCESS in 1h 48m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b4b31bab1434fa092e5fa3fd2d0288b : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":32},{"id":"a6de46a1b69d471a04bae79d2324131752a3d1f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-06-12 09:30:14.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"4df21989532de47beec40cf3702d9b7260813efe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-12 10:11:10.000000000","message":"Patch Set 33: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/97/874797/33 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":33},{"id":"b02054347f12bfd014d0db1a334607654d23f286","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 10:32:53.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f6608b7336247698de6d1c96e38c0a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d16ed33e888543808c63ecdc185728a3 : SUCCESS in 59m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/214bded2ddf340a199d08523431da328 : SUCCESS in 59m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b0532b7413f4b3ea9c8a891bc14fc6f : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"16f7ff59284b7b893b602ce937953ae013cc7272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 11:33:18.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb17599feafc4eb583c0c508dc9b0bd5\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/abc5febac832441798005c11c6964315 : SUCCESS in 1h 50m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/43d997ed37944e038b3772a86f2a350e : SUCCESS in 1h 16m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ffe6d4ea32b74c88987e4314b7d34bd0 : SUCCESS in 1h 40m 29s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4b5618c1714f480185f4c901b6d762c7 : FAILURE in 20m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ee9bd0cf49344e0914537a3fdeafee8 : SUCCESS in 1h 26m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf7fb658b838446d9946fc2678339f66 : SUCCESS in 33m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ebba643de454ab39bb76d7ca21ddbba : SUCCESS in 42m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0145b7a6384f442487d875f925101c06 : SUCCESS in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1db5fcfc86343f0b77f632425e15769 : SUCCESS in 26m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e82fc83c670e4a0b9184530329c7b518 : SUCCESS in 25m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebefc4c4eca249dd988f197d372d635b : SUCCESS in 27m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfe041bacb14c3c81c1a52735099fc6 : SUCCESS in 11m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/52f152a8bebc47c5821446030a269f5f : SUCCESS in 23m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62959de001b844a69329b6af580aa0f7 : SUCCESS in 47m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ad0f97cb6104ba19aa255838d2a23e5 : SUCCESS in 1h 54m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f54ca640e091443caf066c46ebbca962 : SUCCESS in 1h 02m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2d8a48fb6a824a72abe5729d70d54adf : SUCCESS in 41m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/70d9802991b44d869a599bed2984d9ec : SUCCESS in 1h 13m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/34aa5d7f7a30417dbcfaffb3ab219f2e : SUCCESS in 1h 21m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cff45b8329df4f5a90b7c74c7c3a7bb0 : SUCCESS in 1h 00m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a85d5b061ed2416fb7680fcc7dc83ff3 : SUCCESS in 1h 54m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd5a3ac3cb884d31811208c0a6970393 : SUCCESS in 1h 34m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/91a6e4c3eceb4bc88040b0b8d8780f42 : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":33},{"id":"2624bd565f333cf9a600e36aaf18aa298727041d","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-06-12 12:45:32.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"23f60ce13513ab9ad65547c708760edd0c9aabff","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2023-06-12 12:45:39.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"dc2021da3752ff8b68590620a808b93ab50a5db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 13:08:20.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd1dfd99a41459f85ddaa243823ad63\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fead6cdf81ae404497ce1a7818f921c7 : SUCCESS in 21m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ecd97b92bb8491abf09205876c52c73 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43b76b71d5a84713951d35a0bccd4e50 : SUCCESS in 20m 58s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"3007e11f996e04e8b85e1850042ca0ff0c2d857c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 15:07:19.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0d8edb0f834feca00084d2b9565375\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d72c8f7bda314e37b1d8a3321fe5ffbd : SUCCESS in 1h 57m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef9282af6c0d4f22a5aed6c4d19a1e41 : SUCCESS in 2h 19m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ae12e43fb274245aa5ce0b695d09726 : SUCCESS in 2h 19m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7531ab6b2b9a4bfdbd43576c264e7ba6 : SUCCESS in 2h 04m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8469d56f61fb4eec99af5847e366d237 : SUCCESS in 1h 55m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5292104079614faca28ace78cb45f036 : SUCCESS in 38m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be2515f61ded49bc875e26cb53f48760 : SUCCESS in 40m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e8e0facbdf474da8f31ccb519b8d80 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af32957a1e04a9bbf99b150d20ac9cb : SUCCESS in 31m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75d93dbdacf743b3a16860aa801e1442 : SUCCESS in 28m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c02f00a7b204f1099c806dede1f88ae : SUCCESS in 31m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaae382bc5b6449a8ce8c4190bd5da92 : SUCCESS in 10m 06s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a7e859ec66734b7d940fcbf98a8a93ba : SUCCESS in 21m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c78feeda8ef74f0990a9007e30df2705 : SUCCESS in 1h 04m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/53b0454e354b452897220597140c2dfb : SUCCESS in 2h 05m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/412d6a71dec04889921ffbc32dd00dd5 : SUCCESS in 1h 00m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d68f8af7353c461891fafb3af07e4e0f : SUCCESS in 1h 04m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ef24a47488cf4c21b2d3270c572f7971 : SUCCESS in 1h 21m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ffefff9c62d8428791883eb75ea49fca : SUCCESS in 1h 24m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c0c158f246174d758fcd97477398983f : SUCCESS in 1h 16m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf7d6835227e4a8ba2faea73d76f507b : SUCCESS in 1h 55m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f347e27a04ef4457b86b19614b372c3a : SUCCESS in 1h 35m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/32118cfe103d4d54bcd5d08282b6425f : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":33},{"id":"06496d3396c31f575a734be66b1864dda01cbeda","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-06-14 15:58:06.000000000","message":"Patch Set 33: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"7e123dc2a2e1b03ea5622b8aa7a9bcc46b8c9b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-21 09:18:23.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"6e4935fd60d5bd08b4856f0f7f033154edec4337","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:45:29.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5135af2994a41f8a7743bd4e35c15e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/721da1856f9b49bda0c56bf963623dd4 : NODE_FAILURE Node request 300-0021504824 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8495b5df934c443abd80820adb2aff0f : SUCCESS in 25m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73ec5099876f49bea4d5ca07ebe1317a : SUCCESS in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"1443a333e8f8480c4ce5ab6f8ecdb00c29ec60da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-21 09:57:26.000000000","message":"Patch Set 34: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/97/874797/34 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":34},{"id":"a853daacd237e817b7f047b0da8a1cfcdeb752ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:06:30.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d05f2c0890641968798f19fc5388b2e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/83da5c0cd73f441bb1b58f5abe3247c5 : SUCCESS in 1h 55m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c6d4b29d15574068bf8c12b1a5ccb2d4 : SUCCESS in 1h 24m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eddb4b161da14658963586619c0c079c : SUCCESS in 1h 25m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf6bb4944638404b9acc9b7dc634890d : SUCCESS in 1h 41m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac9bf23e6a66417da7d9c9977177b9d1 : SUCCESS in 1h 44m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/331fb41e055f4dc69020ec0ec0944a53 : SUCCESS in 32m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4addf61ecf54496f9ca38ece65378a27 : SUCCESS in 51m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/230ddbcfbea34b43a24a145c4606cdfd : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bdfe5097ef046528a782b5dcf508bc5 : SUCCESS in 34m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74595e4412184f1f91fc3fe73caa71c8 : SUCCESS in 33m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bbd44c475c94b5592675e1c7c764667 : SUCCESS in 34m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59925491c044090af5b9db5b8e189aa : SUCCESS in 10m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fc40423c1d264753afbef4c60e21061e : SUCCESS in 26m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/86c58ac44af64ee6b35ee75ee4537435 : SUCCESS in 54m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5fe3982a32743a9aed212fad82d588e : SUCCESS in 2h 05m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c6ac641bff74f97b2a992926eb007d8 : SUCCESS in 49m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa262cff18204295a5493fd50ab1416a : SUCCESS in 1h 14m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8a7d376acc1444c48d6b61eaf106212e : SUCCESS in 1h 34m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/553bb99353aa4ade8e06b8d35ded6a75 : SUCCESS in 1h 30m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7a01e243e10e47f5bf70ac4e94c10e74 : SUCCESS in 1h 18m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6553219477c845ceb4c2e9422c52e3ea : SUCCESS in 2h 11m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac44d4a25aae4a418e5e8d5ed7f30dba : SUCCESS in 2h 25m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c9450c7a983f41ec89fb21b207e649e3 : SUCCESS in 1h 27m 32s","accounts_in_message":[],"_revision_number":34},{"id":"d81e80599a00cc50ed2f902fceaa36525b35aa0e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-22 17:41:54.000000000","message":"Patch Set 34: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":34},{"id":"56d1f54f560cf51a8f364cacb41af8de5914ec06","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-23 06:44:08.000000000","message":"Patch Set 34:\n\n(3 comments)","accounts_in_message":[],"_revision_number":34},{"id":"960a8959d6cf4b655d56f45d06f9383862df5fe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-23 06:44:41.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"b2905cfde2455b9e98b3be7fcc6ef4da198097a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 07:36:08.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d4ee025c8d4732b89a61b7e2cf8ba4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5c93ab2028f43a7bf1c540a22ae6ffc : SUCCESS in 49m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a97f1f9fb4f4bb5860401c9c8f8e5e1 : NODE_FAILURE Node request 300-0021525674 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bcf975d26c414fb0bbad63cce0f9b55b : SUCCESS in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"c24952efa00a56db0ffad8de3734761c40c238ee","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-06-23 07:46:38.000000000","message":"Patch Set 35: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"a873c47cfd3fe020a2eb1a6e34c10b34d0343229","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-23 08:02:46.000000000","message":"Patch Set 35: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/97/874797/35 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":35},{"id":"4d2b8c2d6f054a53e5ee361eb6b96dc3b77c6429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 09:56:16.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83444d2dcd32448db43bddd3c9742115\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/13ac4bb7094247d39d08bd771cd0c050 : SUCCESS in 2h 05m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b9cdd894e49d43dd8eeaeadad9d8b4e2 : SUCCESS in 1h 38m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e58cb1d24b94715af9373f89a60477c : SUCCESS in 1h 54m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c98c4d397de44e1990ae4f017b139975 : SUCCESS in 1h 55m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b178ed7e2d946228e97fea4ce04c4e7 : SUCCESS in 1h 44m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/456248d38cfa4fa19bdacf19c0381fd3 : SUCCESS in 40m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fddb0160370e48e9883d9b800b7b2bfb : SUCCESS in 47m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db2ce170c1f1413cb8d6f61f029df78e : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdca154b92d24fee9b283051f7257fe1 : SUCCESS in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/039d3c405147476c98b23bc18439fcdc : SUCCESS in 26m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c709d10e5fd44740b831b592a630c2a2 : SUCCESS in 26m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1336f4fcac4503b4f8547be73fb4d7 : SUCCESS in 10m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9617a8ae31b54b09a594f1bbdc76f1e3 : SUCCESS in 28m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/587c5ff8c63d46f386fc4bb4abef16b5 : TIMED_OUT in 2h 23m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8c617847bda4486a45567efcfedf67e : SUCCESS in 2h 02m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/53f57d910b8b4a7fb0bd92fe9cb416f7 : SUCCESS in 1h 06m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7e4dcd52d0c42eb939f1eb5d45be58a : SUCCESS in 1h 04m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9a8b46aadb7f4f098a2d75478034f2e7 : SUCCESS in 1h 40m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e20ebed406ab4d628f7ab61bd0b5d841 : TIMED_OUT in 3h 07m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7674168a6834a12a862e57401e5f9fa : SUCCESS in 1h 18m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d1730c7fa50b4368aff8d2ece0b90b2a : SUCCESS in 1h 59m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a0c1ffe986945eb8b0c7c515dccfab6 : SUCCESS in 1h 32m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a5c2c1bb25140ca8d657804aff1f473 : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":35},{"id":"8f426e64b5b70c96c1a333e546e266f8cf976d19","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-23 10:08:52.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"5ae168330206557f538bd0eee1bd01b9e942a229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 10:36:31.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1caa65cc4a140d6abfa0f82eb4d235a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e82c09180e0b43f9a3ca831d1a7ac611 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3f9c37bf4a44d2b81d53edb145d8a2c : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3b58fd25b61406c8589d57b20ecf236 : SUCCESS in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"ddba72a8823455fd5bbac027966f95862141b988","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-23 11:55:40.000000000","message":"Patch Set 35:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/97/874797/35 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":35},{"id":"78e62e07f425168001ca41ba6f3d1c5c4850c904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 12:38:29.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11d6d0a056ff4ae586d35a3492ba8b04\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4a42a104a5f2459b81541336031e162b : SUCCESS in 1h 37m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b222ed2415bd404f8b90a888f923a088 : SUCCESS in 1h 21m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9ed7892f52d54b9abc5fbe263d3d1832 : SUCCESS in 1h 28m 11s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/351bf4735457433582a3c8aa59d76c2a : SUCCESS in 1h 23m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9a6be3465eaa48f7af4de1bb590be44a : SUCCESS in 1h 34m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62e221ccc42b4e488c76d548ee5ef348 : SUCCESS in 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35361c6071474e98b4393adfbfa89463 : SUCCESS in 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6fe0850b8754ae5bbbf07a7486f8ad1 : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9376d95c44a143d6a848331b0fd5fe46 : SUCCESS in 35m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d42cfff3a064720aaa22e9f5caa5ada : SUCCESS in 27m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/648787eb246e434aaf3bd8324df1e3c8 : SUCCESS in 29m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca55fd670af468a9e1554a1b6a63823 : SUCCESS in 10m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/10026563f3064f96a93fb36909ee2b1c : SUCCESS in 28m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/78a666826c1948f982597d1e2c1865c9 : SUCCESS in 1h 01m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a3c81ed57104c18bf22671fde296e25 : SUCCESS in 2h 06m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c64c2164f293466f8c6cd6653e8395de : SUCCESS in 1h 12m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0981f637526c486987f5d8332268de91 : SUCCESS in 1h 00m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d60910e51fcc4ece8a20928665a01af5 : SUCCESS in 1h 36m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/71cd47a59cfa474fa151a9928d547ccd : SUCCESS in 1h 25m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5fe6ae68d8a842c1a6e0ff10a2bf3568 : SUCCESS in 1h 12m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bfde5e254c2a417380b30e887c13b618 : SUCCESS in 1h 53m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6c4e9d6e42143ffa10f3c1abfb88ca3 : SUCCESS in 2h 21m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ac88ae51e564a94bd9510abb047bcbc : SUCCESS in 1h 42m 43s","accounts_in_message":[],"_revision_number":35},{"id":"d49ae484384a1e947e079df3747915e267fce0ab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-23 16:30:44.000000000","message":"Patch Set 35: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"763a03ebe44400b95a8b8d29e300d55a85e83f12","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-26 10:43:47.000000000","message":"Patch Set 35:\n\n(3 comments)","accounts_in_message":[],"_revision_number":35},{"id":"05bcc8bc628bde25bb1865fc15f3e8006b17d3ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-26 10:45:12.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"28edc5cff1f4ec0249956626e70ef887ae689f85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:18:06.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/191b68cc71904a258b343a222c2aea4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63098b0700ea4f77be412954a750ad01 : NODE_FAILURE Node request 300-0021543390 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe02e9d4bc1a493e9d98e0f34705f84e : SUCCESS in 31m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce3168d15ae245b3a7172f4b6d6bafc7 : NODE_FAILURE Node request 300-0021543392 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"64bc60d33972a1efcde6e97c65dbb6af2b374229","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-26 12:22:02.000000000","message":"Patch Set 36: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/97/874797/36 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":36},{"id":"3c646f271834182e5977c20a76dff878cc389d99","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-26 14:47:04.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"53ecbb00f75c38cf91b58897bede91f4b36eb588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 15:03:54.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6aa0cee0572c4d0d8d2617f19a3f3ad5\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b0d289b7ad4e4c779d3ffa2f5b9ff4dd : SUCCESS in 1h 38m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d5e2fe363822484cad38e2df0bef3604 : SUCCESS in 1h 22m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9334a60e6a1443549be1187f40634e64 : SUCCESS in 1h 50m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9fb98417da2740d7990f3a9fd08fe77b : SUCCESS in 1h 19m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/989870ffb05c44fc90d9fdbc029b7cb6 : SUCCESS in 1h 22m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9766f2d61f634a3db201447073755531 : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/125c157d4e3b42ca827acb9142fb1a3e : SUCCESS in 42m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8b79e39983460c828b4e677bf36b57 : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f059345e2124596b95be1f210dd6c7e : TIMED_OUT in 1h 01m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4834e4c551dc4827bdcbb2379ab6c36f : SUCCESS in 25m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c132c0f86e8a4cad880360428ab539e6 : SUCCESS in 25m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d171b39f0ef402abd2afb6abb1f83b7 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fb4f2be1445453ba247a3893617477e : SUCCESS in 7m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f4144f41a8784aaf9f49884be0445e96 : SUCCESS in 28m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/69730bf252fa4cb99565c37925a67c43 : TIMED_OUT in 2h 16m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d48c416e16047f3b52a3fe2a2834f14 : SUCCESS in 2h 22m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/73bb6834939546c1806c85dc292fc77e : SUCCESS in 55m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/280240b2a74849eb89b2114f3d014d9b : SUCCESS in 1h 22m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/63d0daf962d8400691033d1443b7b8aa : SUCCESS in 1h 39m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/64f2508bee654528b414fde9b057489b : SUCCESS in 1h 44m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/adb2a2f397c54786bf5b7e69b2d54269 : SUCCESS in 1h 15m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b8b3237704da4f52a8758f458631de40 : SUCCESS in 2h 01m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/857ab14c9da5464f9d5b6f7d2bbc1ef0 : SUCCESS in 1h 57m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/db47448cbad74267a349f9a8ce3dae7b : SUCCESS in 1h 52m 25s","accounts_in_message":[],"_revision_number":36},{"id":"8e7b6f0cc61371bca5ed82846936362dd98882ae","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-26 18:23:41.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"83aaa0878145c8c788f1225c7a830f0f2be93c77","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-26 19:33:32.000000000","message":"Patch Set 36:\n\n(2 comments)","accounts_in_message":[],"_revision_number":36},{"id":"e5ed9a8a9809b7f8e8ae43aff39abd0d6adffffc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-26 19:35:24.000000000","message":"Patch Set 36: Code-Review+2","accounts_in_message":[],"_revision_number":36},{"id":"55a9197df7d86797b1dbd6f8f0d228a0fb8dfd5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 20:09:02.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afdec5b359554b9e98771a946282c07e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af5fdcafad9e405e8a96ebc41b8d86f8 : SUCCESS in 31m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12d164045cac4dee90e0cbfe2a073491 : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da37a58fe5e340718a99fcb90b71f774 : SUCCESS in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"8fcf5dcba5603137071c710fdb375ed99c8d7ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 22:18:59.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c374c230ed64ea6ab54f07760959ed2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/13e19d0db4974433bf91d8e7361120be : SUCCESS in 2h 38m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d66e0ed44e2c4c3c90214f82dd887621 : SUCCESS in 1h 51m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ce79130837e4f4297ed49787fee4b6c : SUCCESS in 1h 40m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c4a3e24f86cd440ea375489d20b5f669 : SUCCESS in 1h 51m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bf5143a33e4b4c758ac738880b9a1584 : SUCCESS in 1h 53m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eadcb44bf470402ca2df57d86d69fba6 : SUCCESS in 51m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81cb623e94db44949d6cca8e7299e7fa : SUCCESS in 32m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f0c129c6364579bf5a435c067e41c5 : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/829b005874474dbd845eea3e4500f721 : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c463b17b9c5146269cd68d1495e2ac7f : TIMED_OUT in 1h 01m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4a05338d4be4c14a5d031cb2d42ebb8 : SUCCESS in 26m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827ef74a001b4869a3753a9764ffc49c : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d25d653975846a792d95e3e987cafda : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3b260ab3bc9b45e2a7ace0cc01b0fab1 : TIMED_OUT in 1h 01m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6e51b59f53146319877f9a2998af23a : SUCCESS in 50m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f96de62641724829afa8b41a747944ba : SUCCESS in 2h 00m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/09ae158785c84c15a17221cd5c6eb93f : SUCCESS in 1h 01m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5029017986d84532ac2e49ef3221b4e5 : SUCCESS in 58m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ba632bc1651549ccb5ecb30e15c72d91 : SUCCESS in 1h 34m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/997738bf8a5a44fd9a197686dff0727b : SUCCESS in 1h 32m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/70422898df0d46018cf63f20d008afc3 : SUCCESS in 1h 18m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/69477873f98944f99dfd35cf6a52da4f : SUCCESS in 1h 27m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f615c67a51f46d7b971f02265645561 : SUCCESS in 1h 41m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1531074806994616850e47967e4340b4 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":36},{"id":"8c2f17b3096eeb194b83e7674c09fa8ebe423b61","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-27 06:02:08.000000000","message":"Patch Set 36:\n\n(3 comments)","accounts_in_message":[],"_revision_number":36},{"id":"5ce3bd41930f06c731327135ce3749c9e126f111","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 06:28:33.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a8d861f6c7a4f64a1feefbb79d99ffb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08cf266b111546cd90d734610b058e2a : SUCCESS in 21m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a03a026cc28a4e4ca2b5fbc433510e82 : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/00ab2f0252a94e88b19c77327871450b : SUCCESS in 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"6d0baf7f55dc3b22d1bec9c6cf24771df23ad324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 08:35:54.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0adb81871a0f448d883a91f4785e5d1c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/317b38de7f764cefa207ca9581646a54 : SUCCESS in 2h 05m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c46a7737b534a66a3fa6643596f56da : SUCCESS in 1h 59m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c4253807dce64095b58d73e418a0be5a : SUCCESS in 2h 25m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b2f9f5ccb7b42a5a39f3a54000ae1d5 : SUCCESS in 1h 42m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/025a48639ce8472d8cfae1023f42ef83 : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/702f707872e940b5aacb07c3f1919b51 : SUCCESS in 32m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff460dc68af6411e8d05d8f6dd05eab7 : SUCCESS in 49m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d338b46a734d76b18a8c2f14ce2d00 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebaf4e21d983409aac5066700c230909 : SUCCESS in 27m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2d681da796549018806358262118abd : SUCCESS in 37m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04d0709dd4f147d3803cbf9ae6d4bcee : SUCCESS in 21m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4f29147cb944d5bafea39ebec09640 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0b14d7406c249cfbc3413811bbfc089 : SUCCESS in 10m 06s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/785cd7acb9644333a083cd1675a0a42d : SUCCESS in 20m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3032aeaefa3d47dfbea33fecda441deb : SUCCESS in 1h 01m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7942efbd86e49de886b4ceb0616feaa : TIMED_OUT in 2h 32m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/085d30508b9844c8a2bd7f936d67b554 : SUCCESS in 57m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e1e9742cc3c24110abef56ec86d65367 : SUCCESS in 40m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e6e4a261b74442e38f0dfd3394f8fcf3 : SUCCESS in 1h 32m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c601b062cc64bc0b2a43dcdb814264d : SUCCESS in 1h 42m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a95c2935eab8487c9fec4c4fe0213d79 : SUCCESS in 51m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/770815b3e187479a87014e31b1525116 : SUCCESS in 1h 29m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0c60fe29cd64879b93ed11a86c96f00 : SUCCESS in 1h 35m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cfb3688a2f6d4d9faefd460235d9c950 : SUCCESS in 1h 34m 29s","accounts_in_message":[],"_revision_number":36},{"id":"d2f83e6cd111d1b93f3c5fa64f5e54852ad1a99d","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-06-27 08:37:59.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"a3f752e29c7bafde56532ecb9019f41d71369bdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 09:03:11.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac01f94115bf4fa2aa445fd5e159494c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c023d22634404158b28131cd10328d1b : NODE_FAILURE Node request 300-0021552089 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48b9fe71b14b4e9e96ee1170c4342293 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e361d978d144e5aa60bc4f73237868a : NODE_FAILURE Node request 300-0021552091 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"5118d107a36d86f716e03155cebe8ed5d6569588","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-06-27 09:11:37.000000000","message":"Patch Set 36: Code-Review+1","accounts_in_message":[],"_revision_number":36},{"id":"17f095ba5d57782e19263801a14be0176d325f7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 11:47:44.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb7bd8fa44d644e291e5a3df6b85bdf7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/839961d527a54dfd9d37edc2c8b60525 : SUCCESS in 1h 32m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1c950a2e57fc42f2b62f62a1bceef6da : SUCCESS in 1h 09m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2bbc77cc47dc4192822b99d3fff71af1 : SUCCESS in 1h 41m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d15425985ba4f85a6896fc68b869f23 : SUCCESS in 1h 16m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/96aa4fcb432b4f8c91275701e45bc6bb : SUCCESS in 1h 25m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c469f47d80a45b2af4492f0c37c0db8 : SUCCESS in 33m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdafe51a08e641cdbffd1ce7113e00d1 : SUCCESS in 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897262bb0a7c45b389098e7d7bded21f : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/686fb72004714b8ba8e1ac1a293eb814 : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c136ee9f9e0426da274b7be8b54f736 : SUCCESS in 29m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d2609c3808b4f238d5c50985c490ea9 : SUCCESS in 29m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b3d1e30aab4040abe81a81eb60c3e8 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f484bb3cfa4ca58926aeea295da1f3 : SUCCESS in 8m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b2840f8021224928ace8aa074cb57c77 : SUCCESS in 25m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95e031ceae044a6fac12ab5f12c1f34b : SUCCESS in 1h 16m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f50954f719f4ce1a4b76febb7bad7e8 : SUCCESS in 2h 00m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4aeb3d57cd40406790efd321d25c193e : SUCCESS in 54m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/285b980c6c4841c0aae1bb3f232ddc44 : SUCCESS in 1h 12m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/19feff93c711414e831b5bbd132d3a94 : SUCCESS in 1h 43m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aaba1eb0f55c40e7ba7f269257fd382c : SUCCESS in 1h 34m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a5fe15b8e8964191ba526d776c728975 : SUCCESS in 1h 15m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4a656414bf854fb08190b41fad89757e : SUCCESS in 2h 14m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/40e0d33758cf47699c0c0d6738f19fd4 : SUCCESS in 1h 47m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c019061fffd544cda1aba8a53096f8a0 : SUCCESS in 1h 44m 46s","accounts_in_message":[],"_revision_number":36},{"id":"5cb997cf047123826f6f653b751d0a9c6d3f2711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-03 14:47:50.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"89b119ef2e51d5c07b087604fdd1af64e5385a5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:18:23.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc49e110df8441ba214a23e7bb926f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/440f08b3801b416187f91c1ecd1a1ba7 : SUCCESS in 28m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/feb1901a6876484b89b814fb71af9137 : SUCCESS in 29m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3dc5463833e454e900e0fc27faf8a7b : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"29d84114075c2f0ca44919b97bc80f496cbb8abc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-03 15:27:08.000000000","message":"Patch Set 37: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/97/874797/37 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":37},{"id":"289ec3bedfbcef473fe169b7a3fbe84e09d84787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 16:57:59.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb24077d423041838c0d0cc2f634ff40\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e7ab75b2284e41479b15f6e1cacb1ca0 : SUCCESS in 1h 27m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3e5e23904d54eedb965e6dd6e69aaf6 : SUCCESS in 1h 32m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e430af07e3f435ba11e40a1b7d58536 : SUCCESS in 1h 18m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ec3b920f046243a2ac1f60f9c9aafcd4 : SUCCESS in 1h 38m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/416f8a0b17e842f4896b82314c930644 : SUCCESS in 35m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/004be2d04b7c4016a4da27d734c4fb57 : SUCCESS in 47m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260ca3578edd417a86628d7d302ab896 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f64bec1525fb4040a465b9cddbab4b8d : SUCCESS in 30m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0a6f7f92e24453696ed1899b6fdbeaf : SUCCESS in 29m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7426dc6c1c14c3a9959251dd1a7f2f7 : SUCCESS in 32m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e24f1bdc2b94d7596547efd1faf66fb : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e8c695adffb4eb29d28404da7acefd6 : SUCCESS in 10m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/700c203be1244a8c8dd78679d7c744fb : SUCCESS in 29m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d69674ea85d84060afe6d8ee759634d4 : SUCCESS in 1h 09m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3c89b8eb3ae49d2a546288fca95df25 : SUCCESS in 1h 58m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a996702061ee49859ccf6f4b065bd908 : SUCCESS in 1h 12m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b7e82300bf4a47ada7757c7ebdbf001e : SUCCESS in 1h 14m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4116695fbe22490aa440fc798e9f54cb : SUCCESS in 1h 27m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ce2e9b9b1784ea68430777dd3e95e01 : SUCCESS in 1h 22m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0f463003b4a14db09bbc67f27bb6536f : SUCCESS in 1h 01m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8825f02ae32f423e85215d5efb18d607 : SUCCESS in 2h 00m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2d61dabd8dc44cfb78f8f6f79d90094 : SUCCESS in 1h 42m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fd65601b85844c0087db8c568b0a3fd9 : SUCCESS in 1h 39m 14s","accounts_in_message":[],"_revision_number":37},{"id":"ba805e9938c230d40d1b2710460b97b280cd6b03","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-03 17:04:58.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"f5dd31954441b25b06bb00591e20e38b3f118c73","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-04 12:28:02.000000000","message":"Patch Set 37: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":37},{"id":"821927a8b6308d9e90e3d4c34c0fcb3d015a15bd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-05 21:41:55.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"526f5678dcff1a900545b40f26e1819ebe0cc156","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-06 06:49:30.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"19fcd79b7ac5fb60644206e478991db353397be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-06 13:39:40.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"d8998d3a02ecfebb5bd28c24efb2dcd6e70eb3da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-06 14:19:30.000000000","message":"Patch Set 38: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/97/874797/38 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":38},{"id":"d78c528bb91df09103dfa937daaf985b43aa1277","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 14:21:15.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11e30f20c39434098b5742cef27d887\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38426da6bb0d4e498a08d60e9489569f : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f82716051ab4bd6a44215400c2ba639 : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89436f2dab5f46168aed2f04063e1697 : SUCCESS in 37m 12s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"84ce442e67bf5e71bed9bf81640f2f8aa497dffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 15:56:16.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41841b5c4ca04f44b2dcbb77a001d07b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e42a36f3cc8d4b8e8b4827528a16221a : SUCCESS in 1h 37m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8b8ad5e530f4a0081375f2f2fe81c45 : SUCCESS in 1h 38m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/02e5a4bc60644399a46761f2842bd5ff : SUCCESS in 1h 59m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f351f9655f242ef9e2a0547ec11de03 : SUCCESS in 1h 53m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a0fdf6c91d4f4ecba8c7cc67c27c1774 : SUCCESS in 32m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1910cf87d7248ef8e4bec7bd034defe : SUCCESS in 44m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15b17edd92a4297b9a9dfffec3a10eb : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aea7f99943174be0ab61b743d3cadc03 : SUCCESS in 29m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ffd2656020742a7822a0c49dbd7d6f9 : SUCCESS in 31m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fff8a20c2e8d4a179d676cfafcdf8621 : SUCCESS in 23m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c87cf8f02bb740e49b95772324648034 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc31eec25d54436ebf4ed4cb692a3e8e : SUCCESS in 13m 07s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8fccb62118464dd8aa98ac1a7f99368a : SUCCESS in 21m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53238e89941a42e19a29cb895e359e3e : FAILURE in 45m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3d87624a57c44e2b96e2738df5780e2 : SUCCESS in 2h 06m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/43a839c1ff1d484ab3f1cea80b42efd1 : SUCCESS in 59m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c055d23f46914b2bab7b1c0fab3b877b : SUCCESS in 1h 00m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d7883392b67c40f4851cdc58c326b64e : SUCCESS in 1h 13m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5818731ae57a48f6a104e02ab7638dfd : SUCCESS in 1h 44m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7f69ee2c53943498f9b7f1349e3043b : SUCCESS in 50m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/879124eb30ed4f39b2bea4a578d94f24 : SUCCESS in 1h 41m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d836a17e050c45719d6195af90d1d9a6 : SUCCESS in 1h 43m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf9a25035b3f4e098261355ef64951c7 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":38},{"id":"386acd141b6c10bdfb5d881c7a3804bfd87416e5","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-06 15:58:54.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"c1ed09498ae7cbd9f3527abe3afe6c9056a19b5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 16:23:29.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38af1b19cefc49c194963591dffc9704\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43de37775c8d457d95e2d92c81ef3530 : SUCCESS in 20m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6754455a9b249ec976533acf5b8885f : SUCCESS in 21m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a35863c4c65e4ed1a9ead9af7eff6e7d : SUCCESS in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"c2bdb55fb4e032e9fffd0909787c3ca6aebee1cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-06 16:33:29.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"4ea1df6801548f02d0c7dd77cefcf8e0d9aaba8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 17:18:01.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5ff4fcbc2a841f3b7810eefac94feb7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b85333d1539247cea771b4fcdfcc8317 : SUCCESS in 36m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e683a07ca7a1494f90a38855cf9b9c9e : SUCCESS in 29m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7837a11c8004dec94c4267017baf948 : SUCCESS in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"bb2be48c09fb87b42265a6e9180614fc05624575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 18:55:52.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e258e63c02a443e39c54807efa198239\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/587799d41f64454e9377adce85b038a5 : SUCCESS in 1h 26m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd89ca57b0744424bf33411428f3eed9 : SUCCESS in 1h 45m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3d176a94a854a7bb3ba34a53a8e8645 : SUCCESS in 1h 22m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d36a5ea7a9de45c49d46ab58c368e0d3 : SUCCESS in 1h 29m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e2dae85e83c4a06a0b8911990c95747 : SUCCESS in 35m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fba99108cf543578b3c0ff96aaa5ee5 : SUCCESS in 38m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9dca7a746ad4b48b0ba10793a23a48f : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48d10078686648c0ac5d56eafbe71b19 : SUCCESS in 38m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79cf9dfcccc24016be1d148f40029b71 : SUCCESS in 38m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e32c18e6556741ca8f3e8b404e654f49 : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85297f2031d148a19beb2dbf478ee585 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19e5bd316255422082e5862005e4692a : SUCCESS in 11m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cb04f3c0e1c9441a85b7450d72ee5405 : SUCCESS in 23m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/530a59ed1fbb475ca66b4ed093d863e3 : SUCCESS in 59m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ef415a40150473b8f3a62812409a113 : SUCCESS in 2h 09m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2defd070a7bf46929de7cce7061a8c28 : SUCCESS in 45m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/61e4f14dbab844df86ddd4dea1e9fdfd : SUCCESS in 45m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a9652cdf5f304f21b1d1399a79279db1 : SUCCESS in 1h 30m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f74c231e80c34871b43528fbb9d80bc2 : SUCCESS in 1h 45m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/adf567f8518c453bae848daf45f6f5bb : SUCCESS in 54m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8c76c6a098584814a59bb7a6830f6a79 : SUCCESS in 2h 17m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fbf5872f2dab417f9f19a028ba529a9e : SUCCESS in 2h 09m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4fa4f91ab4944f8f9eaae95843e1cf79 : SUCCESS in 1h 52m 28s","accounts_in_message":[],"_revision_number":39},{"id":"53162e0d007dc6dbd6aa7b81f94f825a50876344","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-06 21:07:19.000000000","message":"Patch Set 39: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/97/874797/39 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":39},{"id":"ea5dd0f1b181c33ba5818e378df674002b0d3055","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-07 09:53:09.000000000","message":"Patch Set 39:\n\n(6 comments)","accounts_in_message":[],"_revision_number":39},{"id":"d6c3e79f94c674e537c5ec87b9ee27d934270be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-07 09:53:38.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"3812e4c9e60c17d06b82e114cd8b39fca0793b1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 10:27:49.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8fcb830b16d42d5af0aa3a22a5d505b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ed85bd2a507489c965eed37255cfb1e : SUCCESS in 31m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1cceef8a88742bdbd9dc32cbfe08058 : SUCCESS in 31m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf23beff3616438db62523d171675786 : SUCCESS in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"16e2527e45b52e4097b7786f0fa3ceedbc203174","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-07 12:10:04.000000000","message":"Patch Set 40: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/97/874797/40 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":40},{"id":"01a574268eb5fb23b544db43ede248935c7ba90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 12:24:35.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/341df43b152c4cd4bb21065775abaaae\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c499a63cc57a4464a4fb5afeb91f78a8 : SUCCESS in 1h 24m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/123dc941851b47bc8bb7eec73a48f631 : SUCCESS in 1h 31m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c0f728a317a349b4908238eea99e40bd : SUCCESS in 1h 18m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aa333de129244004a482786f1da8c73f : SUCCESS in 1h 40m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a311a85c59bc48e2bd428dc6ce995c84 : SUCCESS in 31m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3736d24d961c4599a065fc54636938c4 : SUCCESS in 44m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da3ea2314fc04ebca06e539737987e39 : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d00e29aa36242bc8faa87469b55936d : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e341bef0688043318329772ed9b896c2 : SUCCESS in 38m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bec9cf67599499a97de279e6c6a3d61 : SUCCESS in 29m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e1bf0c2f684fef9f7264571bf0f969 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b47d7988a8cd49d790cfec53615e3db5 : SUCCESS in 9m 41s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9544b35941ac44cfaab72fe0bb4e135b : SUCCESS in 24m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae28294023684dc8a533d582d7a5f293 : SUCCESS in 1h 11m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e508e653e1174f27866ef62c1c38bdd4 : SUCCESS in 1h 56m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bb52dab6db34ad89e8b0361aa9d7023 : SUCCESS in 1h 08m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2ca056db4f1b4c30b6637f8df9ea66ee : SUCCESS in 1h 04m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/69a4a1e8840a4ae289bf4f6087c4008e : SUCCESS in 1h 11m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3251fd1601ae4544a1526444e265aab0 : SUCCESS in 1h 31m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7b20c72f13964f2185e3da0e2faa3508 : SUCCESS in 1h 19m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/644a938361694d179dd1f9eb5bee49a6 : SUCCESS in 2h 16m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/80d6e2626c19492faff88589685e4d81 : SUCCESS in 2h 23m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2f10b2906a847a78e792e6cbe1c2152 : SUCCESS in 1h 38m 01s","accounts_in_message":[],"_revision_number":40},{"id":"343e07740f9705d227d52f0f363f51bce7065e10","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-07 20:21:02.000000000","message":"Patch Set 40: Code-Review+2","accounts_in_message":[],"_revision_number":40},{"id":"ddf8d7326ce28cfebf50e54ac51840139caf0efd","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-07-10 16:04:32.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"3b2b6cf83bc47114d4b834a25599c078cb4afb29","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-12 06:23:09.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"8790f76c3e51ed72655307054c1a187fb15463d5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-07-12 07:02:53.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"b2dc5fbbc420a29c638eb370e51529b8c89d50ab","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-25 14:07:58.000000000","message":"Patch Set 40: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"80761140bf74916025779455a7fc8e37e5c86553","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-25 15:15:53.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"18ef96047bacfb178e284b850efbede10b3a9475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-07-25 15:16:00.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":41},{"id":"c048ba3f58091d0bf885505c23905cc4353d1e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:16:09.000000000","message":"Patch Set 40:\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":40},{"id":"68b5858c67e7f71e2de28803444c158c63da2b10","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-25 15:58:41.000000000","message":"Patch Set 41: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/97/874797/41 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":41},{"id":"4dbb5ac5adaab3abe802df55dd282c1969961db6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:39:54.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34859bff0fdf410696738dd885b08f64\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/912cb084058d473bbb67218ff9ff0f6d : SUCCESS in 43m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3eb8a2dc3704344b3a58f9b498f3625 : TIMED_OUT in 1h 21m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b1311b2e35244048f72daeb91421252 : SUCCESS in 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"0882f7ac32cfc40c1e87b537a548b97158a74d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 18:37:46.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ef5ff30c4e483c88357d0dcafd7ffe\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fab600374e2048c09f3fe0906af99d5c : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4231f5fcbbd74c4ba1eb533704610073 : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e18dd050cba049129d768080a1b81b3b : SUCCESS in 1h 18m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca5ae244b4604cc4bb66ef6860e6998f : SUCCESS in 1h 37m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1108720d381141e7af3d8a948d10afa3 : SUCCESS in 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96f57b955c2f4c9bb5c3d3695bb6902d : SUCCESS in 51m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4fca056d64404195a9168d1140fecf : SUCCESS in 10m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b7de61c8b174f398b6c02dd42136887 : SUCCESS in 32m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab9735cdeed54911b5c71464bf36af1c : SUCCESS in 30m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ab73955b3c44db898dcd6df45e7878 : SUCCESS in 32m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1a5a374934439e97447565979d2fde : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23862d7c17fb4fa88d7cb894548b43f8 : SUCCESS in 10m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/737881f2f2014feeab11935d0226461f : SUCCESS in 27m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcf353c50b6f4f8b9e7cf24e8622e602 : SUCCESS in 49m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a74355b2bc244e7884899ac459c36faf : SUCCESS in 1h 58m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4517d2091d26487bbc10ecfbe6be2953 : SUCCESS in 57m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eb0a2624f50448e59ff74660a64a26ad : SUCCESS in 1h 02m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/07ddc093f4fe44658747ca864a770ceb : SUCCESS in 1h 25m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/053eb9e635524472938c4284a294462f : SUCCESS in 1h 26m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e31a92db3d0d42788a2024b8ff1fd5ff : SUCCESS in 1h 20m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/684ff2067484489d98cc2cd3ce6d2af0 : SUCCESS in 2h 13m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c9d4d24286f4a66a133372ff7674aee : SUCCESS in 1h 40m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ecf20bc3e6a44d5d98d996f56ebb9edc : SUCCESS in 1h 44m 29s","accounts_in_message":[],"_revision_number":41},{"id":"85b8cb70b551b7b3d693f25596cb5bafc7c6b170","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-31 20:33:09.000000000","message":"Patch Set 41: Code-Review+2","accounts_in_message":[],"_revision_number":41},{"id":"85eb7abf3e267942bc2bca61d8e069d40b673775","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-01 03:56:21.000000000","message":"Patch Set 41: Code-Review+1","accounts_in_message":[],"_revision_number":41},{"id":"176f617e8b23b58aed67c2fe758bf6a621bc55ce","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-03 16:10:29.000000000","message":"Patch Set 41: Code-Review+1","accounts_in_message":[],"_revision_number":41},{"id":"1d253332b692bc6e893775cb7cb9cee2cf760ed4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-04 13:18:34.000000000","message":"Patch Set 41: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":41},{"id":"76d2aea698f29398016ec7e8f88d924132771d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 13:19:10.000000000","message":"Patch Set 41: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":41},{"id":"dd3586cdd0a06bcd3db55a1f1a978db51839ad63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 15:51:14.000000000","message":"Patch Set 41: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/068a06c5f1174c46a2fdc6a80f8172af\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2e7cfa64828e48ffb8ae5514979b0f57 : SUCCESS in 1h 56m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c2435d865b9c4a428808cb844e38aaa0 : SUCCESS in 2h 21m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd2aa243beca405e8c6a21fbb9b46f89 : SUCCESS in 1h 57m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd7435c4117a487c80ed79f34fd46713 : SUCCESS in 1h 52m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b8eb079a34414d9087b153cdf3437f : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1186a3bca994c6fa6e526aceea7d49b : SUCCESS in 23m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01b621e50122405dafea854919449fa4 : SUCCESS in 36m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6d50a5872024e2181ff540c409bff16 : SUCCESS in 25m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1118c45fbcca42aab2c57f5ba18417ec : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49fe533b655b4a528b911e0fbfaafa5a : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dc56d58769b46e8ac900a652a65a52a : SUCCESS in 49m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bc046d7004f42588920ba9c1a343c37 : SUCCESS in 1h 46m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/35658afee8654bb9ba70e402ce74410f : FAILURE in 1h 17m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c74dd2a0c3804e1799ac551c6fa0bd8e : SUCCESS in 50m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f276ec231ca04ed3ae6f586499d51051 : SUCCESS in 1h 08m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c3048d39a7ff4f2e9217985135ee775d : FAILURE in 58m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2b4a1e2eb904aeb85aad63e2ccb6736 : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":41},{"id":"9de335538bbf32a6eeb8b5f4f9ffaba4e496b980","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-04 16:09:20.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"e8b99cecb976cb5efaf3cac3749550f986cb0d5f","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-04 16:09:28.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"ee6f661a3cb00d0bfa8212ce0412c4808658d1af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-04 16:53:55.000000000","message":"Patch Set 41:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/97/874797/41 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":41},{"id":"e0af998aa932df6261d83ae27ea8cf22c10d9ef4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 17:06:01.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef6fdd12e364fe98ceb27ae8456d21c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f561cf1bcc648c29876f6d233897830 : SUCCESS in 52m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c16969f41d14d7caa6782871b9fbae5 : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a37253d718047458d7728fb34f4b7ed : SUCCESS in 54m 36s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"39838c604c5bb1c0f18a6799ac9eaa6adb5cdd70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:22:09.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/239015d4d97f43d79c08e446c7f1c50f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/13ccbd648fef428d91d9272353a9daf5 : SUCCESS in 1h 38m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3c1e56a16f0a46208686329a6ad374f2 : SUCCESS in 1h 50m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b9d151ecc3c44b1996f30464bbf94acd : SUCCESS in 1h 40m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/06d047e6aa3948adbfdf65da7b97754d : SUCCESS in 1h 58m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4739d97c34a44e88b15313ef07b80e45 : SUCCESS in 37m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f77dd93bf6b486ea0886181c7999491 : SUCCESS in 38m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2254296d60545f482b6c428c84e3538 : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa8269783acd4f0d9288b2eac207f218 : SUCCESS in 27m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/136a025b239b4788aff46c7ebffb8a7f : SUCCESS in 27m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1d1a92f78d84d98a565bfd9d06f3f11 : SUCCESS in 21m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1853b3f8b54796a2864f23d5f615d9 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/340631d358214f29811a46e50eb79272 : SUCCESS in 6m 09s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cb0c3668e23b4342933b2a2fa0823608 : SUCCESS in 23m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bae5e1a8c3df4e849db3d9122f0f7973 : SUCCESS in 1h 01m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/86242cb121ff477496408d360c56bfcb : SUCCESS in 2h 06m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/47ff9835fc7a4e77a6f0182491e1875e : SUCCESS in 51m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f5eaf030c9e44d0af6b2e9afde2b43c : SUCCESS in 1h 02m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23fb51bb89c04583b191379da815e8ed : FAILURE in 1h 08m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9be3e61c194b4530bb695a483187d09b : FAILURE in 1h 05m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/93939b3867fd403d99a48c76b91f5eef : FAILURE in 42m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9d5f8a8c7cac47458559e4531274edbc : SUCCESS in 1h 22m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2ee43ce230845019b783daf780f47c9 : SUCCESS in 1h 42m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b79d4477b884650b037b37d869802e7 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":41},{"id":"3ba30449973cac3ab17a1a65560181ebe7e45c43","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-04 21:12:23.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"ed541d1fcf1bf2e6e0c3a10a18cff2e5d196491f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-07 01:40:21.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"b8bb25493ed41a2043e28b5e800ad564925ab529","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 02:10:14.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45610824c0e54674ac986e5cf0d97056\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94ffc00d14fc445086189e079c12eefd : SUCCESS in 27m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98daae86a9ac4826be7dc147dc9a1028 : SUCCESS in 28m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8e0f57c363d4b74a6c73d60e0170bcb : SUCCESS in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"cc73924578be26c8dfda9b7f3730cbcf0e2c0dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 04:14:05.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84d210faf43e4e20875075dd65d38e75\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a3cebb8e00164abbab4aa7d5ab6348e0 : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/13b602c638eb4099888d4ff889ada9f6 : SUCCESS in 1h 57m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1549508fca13450d8655a958a834b468 : SUCCESS in 1h 42m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2bdfd1da5f4149be924f85ab90e4fbb4 : SUCCESS in 1h 41m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/756008ead7264669b5208d6546b31e72 : SUCCESS in 35m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec8ff17b097c41019dbda8913135acfa : SUCCESS in 34m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3e5e558537498db9862c481acbbe90 : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb305ccafa94f51a1b935eb0cbe7c0c : SUCCESS in 25m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3d23a9950045f49b9bd3313c3196ba : SUCCESS in 26m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650a12efeb6143eb809f3931299a4937 : SUCCESS in 27m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69b6605c90144d1a5f592cc4b89b6f3 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d7979385e6d458686520540f143e658 : SUCCESS in 7m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b0131aa73b1442f583b947ed0cf14a00 : SUCCESS in 25m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b20dc8dc468434280691248ab9e3dd3 : SUCCESS in 47m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69bafe5f80ad48769a7958f9af7946a2 : TIMED_OUT in 2h 32m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b730739558c4df8b9e15e962cb5adc8 : SUCCESS in 1h 04m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c955a32a308742f3810065e77a6af073 : SUCCESS in 45m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6ac3e6e5cfad4a34a952e8bd7e08f398 : SUCCESS in 1h 20m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ab0d49fdc4154b1caba60c2437538c46 : SUCCESS in 1h 07m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21987fcfbbf148ea96fef9723a9afbbd : SUCCESS in 1h 09m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2efa341a53594f61a0473096bf4e40b1 : SUCCESS in 2h 02m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/39098dcd865c4035a122df73b37b75fb : SUCCESS in 1h 59m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/280cf394bffd474084f3aee65818bba7 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":41},{"id":"aa9eccdbce9a5baffc3c34a5679c54fef9f24375","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-07 16:02:30.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"046fa603a94d8b5829cf0a724b74a078687e509a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 16:55:04.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a64cfc63ca4bb1be5b60c8c24ee046\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59bb7393a09540819d71a0ccbdf39875 : SUCCESS in 20m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91df98738d4d4f899f109c7bef3f6590 : SUCCESS in 30m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1fb30632ed64911856732ee401c831f : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"640d6fad5085a2261b05a9b9b5f1bae29a122e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 18:22:50.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9264e46d8b14813ba325b5ceb02a5c0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/87c841e97202407bb4b6e588b0a0e5cb : SUCCESS in 1h 24m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a8bc6f7fc7a469799da56b346060d67 : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3804af51493d49f49e6bf57eb031ca80 : SUCCESS in 1h 14m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d84246170c8e4187b7206fb0916f4dac : SUCCESS in 1h 28m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6f0e885dd3c431198595878c44605aa : SUCCESS in 35m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a85f17860da4dfabc8b0cb7899b5d3d : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4915df4e49f04fb4a7317b380b7c45ce : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3526961b691f4240898be651dbfeda5f : SUCCESS in 29m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e3e1ccd3ccc44f98e91bf333db34279 : SUCCESS in 29m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72f128631262460ea0aa096e13b8434f : SUCCESS in 28m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2ef79f2d9246edb938c6b3dc1973d5 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e750e347ed47cba460b6f385aa7f15 : SUCCESS in 12m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3fbaa202332a4596adb55618afda50d3 : SUCCESS in 27m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95269d39b91b4cd8993126a4a49e0ed9 : SUCCESS in 1h 10m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5970bed46285444bb5051f7216d488aa : SUCCESS in 2h 11m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/932209958a3945c395c37dc9eccb8d06 : SUCCESS in 1h 04m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2301c16f0a3146b595d09236110c2f8e : SUCCESS in 1h 14m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/378ed0154570410f9ba997736ddf55a6 : SUCCESS in 1h 38m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/feea33de626c4ebcb3f7b8cb705d8516 : SUCCESS in 1h 15m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cb578706d17944728e9b8804c67fa771 : SUCCESS in 1h 12m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b334ef2946c849488ac72ba0dd30b7ee : SUCCESS in 1h 24m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9e1d9503f9c47ee8ce8d4f09dc4ba03 : SUCCESS in 1h 27m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b64c9445296044d295c8c5c5beca4717 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":41},{"id":"61a8fe4cd8daa9b3fc3fba8acfc8e6f5729bbc09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 18:23:22.000000000","message":"Patch Set 41: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":41},{"id":"21bb770e7840510a603610b4032e7bc524d159a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 23:56:38.000000000","message":"Patch Set 41: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e579ca1396a4078bb7473c42a37c089\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/006045c892e046f7a3ea9c7d1ae55007 : SUCCESS in 1h 37m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b55bf55b29b42008e692a0c1dfcffba : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b9c19b5a11164f1498c504f5b18ac402 : SUCCESS in 1h 40m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/713b6e62e9e649c49a3bd062f633e6aa : SUCCESS in 1h 25m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f6f6ae2d0741059c39310ebe4b6673 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fad80f24ab7436e9da97fa96e148ef5 : SUCCESS in 27m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/891c7f89c13645d0b3df104dfbf17d88 : SUCCESS in 28m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1e614f6fdb84db0b28ffa0fa3b7c207 : SUCCESS in 38m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8df135218e54c05b13f05001324417a : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86f7b7f0b1f54bd2880fa12a83ddd1d6 : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7477b97c53b4b9cb7c0a18ad87ba0c9 : SUCCESS in 1h 00m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76259848b31f4764bb10bf856aa82844 : SUCCESS in 2h 27m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2aff34170e4348f2b94a3bc9fdc19121 : SUCCESS in 1h 10m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ecd0168608a469fba25ac8791bfc8b4 : SUCCESS in 59m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ed0c792633448d5a81b331f3adf67dd : SUCCESS in 1h 01m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8162e79eb5724b2382f4d4f80b391caf : FAILURE in 49m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/83727f8cdc6c4e3fa406408475c99980 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":41},{"id":"2fd47f404d8dac2c5dbc596283f67f6d772a2d5b","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-08 10:49:35.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"f1187bc9212e827f17b985d0a9470c6b61132f92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:39:39.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c891751df31e41e0b4b09f8b1b3640cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73c22760ee9540ac80291a05b6062b02 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7eed41bb67ef40f69484d81dc8019480 : SUCCESS in 48m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fbb7a9a3b9047f1ad4e3095edf5c8fc : SUCCESS in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"a6edc8afa09483470b9950156bafe9de94eba0d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:05:31.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d02d977c49014b2c9f57872b8cafe115\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/92f442ff40e744509978a8fe0b010fb6 : SUCCESS in 1h 33m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab5100d7484440c3aef7b62cc9a446c9 : SUCCESS in 1h 40m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fe3d12ac754a488a803046ea5f4d7498 : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8a3a20d47569477594d72f0269c02445 : SUCCESS in 1h 31m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27413d1a2b4c4b66b9338d798d6a8d58 : SUCCESS in 33m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1994324908c43fbb60601f538413412 : SUCCESS in 1h 02m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dceaf2aa14c1454eb893987c57342b4b : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ba4dc34e7cd4feb83e49a9f573c7c46 : SUCCESS in 37m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def0e780d4c54f8eaac8e5b5933aad6e : SUCCESS in 28m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cc2311003b6499a83013c08ae87c525 : SUCCESS in 44m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6edeb188748d4cc89ebfe03cc1aac5d2 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f35a328463ec47299825926b71b4254a : SUCCESS in 8m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/54427f151d7c4848b528a1b9ec2049a8 : SUCCESS in 25m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e6e0962297245f8be3613881b2421b8 : SUCCESS in 1h 03m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3d08de5eb6743a081cd61ad5275b502 : SUCCESS in 1h 39m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a533ba2a9cd94d93a58ff550a1a70c79 : SUCCESS in 57m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ebd822cf2f724b68a722a6e334277f88 : SUCCESS in 48m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d4f7fb1fcda4af5b58ace76cce683bb : SUCCESS in 1h 35m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/07f83047b3404098a6702ab5300c7a4c : SUCCESS in 1h 23m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eda57b20aee8483f9cb4864d2ba0c87b : SUCCESS in 57m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ce775a9c0b4244f88aaaa2d8123ab624 : SUCCESS in 1h 32m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd92c70498204318aaed20f63e747f3d : SUCCESS in 2h 00m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4036d49a16104aa0bde0bb82b48ddce4 : TIMED_OUT in 2h 05m 06s","accounts_in_message":[],"_revision_number":41},{"id":"dfd055df74292f5da4571c2f5e823121ffda9b57","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-08-08 13:18:02.000000000","message":"Patch Set 41: Code-Review+1","accounts_in_message":[],"_revision_number":41},{"id":"40bd8be68cc736ab9183bdc523aefa9598483fe5","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-08 13:34:02.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"7e5c430f137e83e03378d1c7e06fde425f949e6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 14:55:58.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20511a27400149609de6048b6e1dc327\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f3c4164ca684cbaa999c285344e6ea8 : SUCCESS in 1h 19m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9380dc6165344ad298875a9b10d11745 : SUCCESS in 1h 18m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f25dbd3927f42dc84580300207109cd : SUCCESS in 1h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"820ad4157a3f687c8275e1ab90874db9bf3d530a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 16:28:12.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad0333e54a7249879b4b3fb366c596fc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/42132cecad554b34820c807bcd2a11f0 : SUCCESS in 1h 27m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cba8c853c2fb48f2b012ccb29e3db02e : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5879550acd24469c85e358814798d8e3 : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72923aed44f643239ec6365fab76e61e : SUCCESS in 1h 30m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f1e939f7c4e9409aad5d49e609426675 : SUCCESS in 35m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9933a4015f45443686275c06a5388814 : SUCCESS in 44m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2dbaf41ab4541fe88163b399b22c48f : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d963cc5e34e4f6dba71efd4e28e128e : SUCCESS in 30m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e922392770e458d86e56190e7fe831e : SUCCESS in 30m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/112f248f337545e8bfccea15af364496 : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2c3cb3c76f490eb91a30845d4a454d : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f4990c94a8141e88c48d403f42ebd6e : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3943e374a354468d999feff8898c18ea : SUCCESS in 28m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/861282f66e7a4c6bb881a60355eb34a7 : SUCCESS in 1h 02m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b0122b3506144888926ee8b3dea6f15 : SUCCESS in 2h 12m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8df6afd6a6343eb8acd278aa2cb5a0e : SUCCESS in 48m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c2cec53fbb5b428da58b3558d80de857 : SUCCESS in 1h 05m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1dd6d42bab1b4c08adf1910bc64446b9 : SUCCESS in 1h 33m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2539fca33881446b9142cf168a128d10 : SUCCESS in 1h 21m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ab543fc1195e4a4cb73361d0de168375 : SUCCESS in 1h 20m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8dda2c31d0d7418f82d3accd08b9439c : SUCCESS in 2h 05m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f65060d2caf4cb9a2f70dc2bea9549b : TIMED_OUT in 2h 34m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f79db0e886b44d48a5db62ac5fb8926 : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":41},{"id":"c28c9e9e3207ad8dc02c8c5277c730171d7953e3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-08 16:31:47.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"6d305ce9363287ae6e1c902b0ea619f381ea938b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 17:10:54.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ea36939904495bb6c657d376b93b77\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17b64f26f6884a8aa462cc291ad82279 : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38642763247644318458371b4a344bbb : SUCCESS in 38m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a5166832f554648a59a02dffbefb4e2 : SUCCESS in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"2bfe9ec7b85f47f237b97105ab3d25993c1e4b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 18:57:29.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db283ea243c74d38883b7bfac1981ed2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/328ab43e9ca442dd9f76c6946c3118d2 : SUCCESS in 1h 51m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da489f46fa9c4241b9e7ca99148157c3 : SUCCESS in 2h 23m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b75cd4736b54f58895f59844ab2e3fe : SUCCESS in 1h 38m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ae91776d8ab4f8e828b4388c64c823b : SUCCESS in 2h 04m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/097a15e66c424391a02f26517cae0fd9 : SUCCESS in 33m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce0fd4aaba4404788503a0dd23d99e3 : SUCCESS in 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8875526e8edf43bcae026b3055800a1f : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca5440427a37408e94ed8a2d26190310 : SUCCESS in 28m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edecc8ede889450692fdbfa21eabbe46 : SUCCESS in 28m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6e6725c5c6f4299b15137e602f13297 : SUCCESS in 27m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db953b2985340cd8fdb9bf4fa324393 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a580e8bf7d445e6beed566377790e21 : SUCCESS in 11m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/76e7be558efe4b59ae293d61f855cbf6 : SUCCESS in 27m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b8e67b826244b88bdd82824b758837e : SUCCESS in 1h 04m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f49977678c249eb8dc3f75148f24f0e : SUCCESS in 2h 01m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c15463a7b7244061bfc2e106c1385b67 : SUCCESS in 43m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cb2087d2f38643328650a4b43b7726ef : SUCCESS in 1h 05m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c716edb9411f45f28120ffea341eacb4 : SUCCESS in 1h 38m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/68d7ba613c3c4bc3aa55c0ca321fc08c : SUCCESS in 1h 12m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/040deb8c14bd4c1bafe7e2ca94ae67d3 : SUCCESS in 43m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/db067f58b0a14ae3ab02ecbeb3341d80 : SUCCESS in 1h 19m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c4aa1474e2c426c98979287adcb17da : SUCCESS in 1h 34m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aee58ecf05dc4432ad82b405c895e260 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":41},{"id":"3baa3e6c0ef29ba2d69622de2587de7d0ca2bf0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 18:58:10.000000000","message":"Patch Set 41: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":41},{"id":"e29484c837e175d812db8b113d41370682304011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 23:40:37.000000000","message":"Patch Set 41: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/249ce676e61a4edd9fe1a7eb597722a8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e984dd1aa1dc416296c2da5ebbcd7555 : SUCCESS in 1h 42m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/faaab0b9c43a499fb15bbfed3fb2f7fe : SUCCESS in 1h 51m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ff782c08d08c4637ae1f28625f6c1711 : SUCCESS in 1h 58m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/507b8489340d484fa9bf1864362a03e5 : SUCCESS in 1h 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09013d3992b6430cb84876af24c1d98e : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd749716a2ae431b9042b6735cf547a3 : SUCCESS in 27m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/377fa4703a5c47f4b414244e9c65c7cc : SUCCESS in 27m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ed64030011f4d6a8e5fe9ca0959d5f6 : SUCCESS in 26m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b4df5a08664a84a3871e325f578a41 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/637521fd9d91497b9607becc8f530c7a : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a81758af89d407ab095c56c14fb7e23 : SUCCESS in 1h 01m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cac6b691750746cdbd968e99a9b10485 : SUCCESS in 2h 00m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/47e522a655bb46e4b9c2d35f74318acd : SUCCESS in 1h 02m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/837733a9c1bb4fd3bdc1838362667a98 : SUCCESS in 1h 10m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b52d66c5b705406aa19a59114eecbad6 : SUCCESS in 39m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3a3adeaf9f7b49b0ac07a07e9a8318aa : SUCCESS in 43m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7d6ed860068e4d229f313c18cab78f23 : SUCCESS in 1h 31m 05s","accounts_in_message":[],"_revision_number":41},{"id":"3074b5ebbd1d0a44eacc37b8c1581d36a82b17d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 23:40:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":41},{"id":"2f12be971fb73c44d6e0fdf87eb07a5fb034fb41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 23:41:59.000000000","message":"Patch Set 41:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee958fa733e4ad78ff5a6070daf8167\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60288b72bda4bb091bba01f527a753b : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3703093dab143e997bfbc3ce42141de : SUCCESS in 44s","accounts_in_message":[],"_revision_number":41}],"current_revision_number":41,"current_revision":"89702218db2476a17f6ff36cf51b909db563d887","revisions":{"0f1c4b07e2b82fce6f474c1b3a5314e1e3f62113":{"kind":"REWORK","_number":1,"created":"2023-02-22 18:13:50.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/1"}}},"commit":{"parents":[{"commit":"9477628f29316cd4f9bead6acc04ba01d1ad5e2f","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9477628f29316cd4f9bead6acc04ba01d1ad5e2f"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:13:28.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nhttps: //bugs.launchpad.net/neutron/+bug/2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f1c4b07e2b82fce6f474c1b3a5314e1e3f62113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f1c4b07e2b82fce6f474c1b3a5314e1e3f62113"}]},"branch":"refs/heads/master"},"aa25b380feb7af5f610868fccc1d8caec76245cb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-28 15:01:33.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/2"}}},"commit":{"parents":[{"commit":"137437f6a409c3244edbc4fd815cf4f0557f8f16","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137437f6a409c3244edbc4fd815cf4f0557f8f16"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-28 15:01:03.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nhttps: //bugs.launchpad.net/neutron/+bug/2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa25b380feb7af5f610868fccc1d8caec76245cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa25b380feb7af5f610868fccc1d8caec76245cb"}]},"branch":"refs/heads/master"},"e22044e2aeb3600fecda77535910dcad4a9a7d15":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-28 21:14:30.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/3"}}},"commit":{"parents":[{"commit":"00dcedbeac328efedb6d79d9d54e9535c77ca997","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00dcedbeac328efedb6d79d9d54e9535c77ca997"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-28 21:13:22.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nhttps: //bugs.launchpad.net/neutron/+bug/2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22044e2aeb3600fecda77535910dcad4a9a7d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22044e2aeb3600fecda77535910dcad4a9a7d15"}]},"branch":"refs/heads/master"},"83414093442fdfcf01e9fe62c18b7f5971be86ac":{"kind":"REWORK","_number":4,"created":"2023-03-17 13:28:15.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/4"}}},"commit":{"parents":[{"commit":"4bff3720898d1dd8a6802a427697127fd6f5f38b","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bff3720898d1dd8a6802a427697127fd6f5f38b"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-03-17 13:27:50.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nhttps: //bugs.launchpad.net/neutron/+bug/2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83414093442fdfcf01e9fe62c18b7f5971be86ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83414093442fdfcf01e9fe62c18b7f5971be86ac"}]},"branch":"refs/heads/master"},"aed5546cf6b2cb141d5703559cb3f0f3ec5ea31a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-18 01:53:53.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/5"}}},"commit":{"parents":[{"commit":"398750c3fc699865bc06af07782af9d51dd96574","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/398750c3fc699865bc06af07782af9d51dd96574"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-03-18 01:53:33.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nhttps: //bugs.launchpad.net/neutron/+bug/2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aed5546cf6b2cb141d5703559cb3f0f3ec5ea31a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aed5546cf6b2cb141d5703559cb3f0f3ec5ea31a"}]},"branch":"refs/heads/master"},"24ef4216192336746f11a73d34ed9de56d7cc444":{"kind":"REWORK","_number":6,"created":"2023-03-20 08:29:28.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/6"}}},"commit":{"parents":[{"commit":"f59e8f55c6f30630c65d88098b41e31bf8aea038","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f59e8f55c6f30630c65d88098b41e31bf8aea038"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-20 08:29:07.000000000","tz":60},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24ef4216192336746f11a73d34ed9de56d7cc444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24ef4216192336746f11a73d34ed9de56d7cc444"}]},"branch":"refs/heads/master"},"5c7456c6f3a90b59b06278c5556458412947164a":{"kind":"REWORK","_number":7,"created":"2023-03-20 08:52:17.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/7"}}},"commit":{"parents":[{"commit":"9800ea64cb9867a7f7306cbaa6b9c9a03606eb12","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9800ea64cb9867a7f7306cbaa6b9c9a03606eb12"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-03-20 00:09:06.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nTargets 2023.2 so the relevant updates are done as well for that to\nwork.\n\nhttps: //bugs.launchpad.net/neutron/+bug/2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c7456c6f3a90b59b06278c5556458412947164a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c7456c6f3a90b59b06278c5556458412947164a"}]},"branch":"refs/heads/master"},"ed2ab558230742e0d42f969745f1c493dab83bf7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2023-03-21 09:15:53.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/8"}}},"commit":{"parents":[{"commit":"e25bfde4d982fd536daef56413c0e1bfdb6decb9","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e25bfde4d982fd536daef56413c0e1bfdb6decb9"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-21 09:14:04.000000000","tz":60},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed2ab558230742e0d42f969745f1c493dab83bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed2ab558230742e0d42f969745f1c493dab83bf7"}]},"branch":"refs/heads/master"},"2cdbcee1b9536c6c81cfd514afcf0c6e92179928":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-03-22 13:53:50.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/9"}}},"commit":{"parents":[{"commit":"dcca516aba328b209f78bbdc430d281387b60364","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcca516aba328b209f78bbdc430d281387b60364"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-03-22 13:53:14.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cdbcee1b9536c6c81cfd514afcf0c6e92179928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cdbcee1b9536c6c81cfd514afcf0c6e92179928"}]},"branch":"refs/heads/master"},"38c96cf62b08a8eda8f2a410d25aa7721617fa00":{"kind":"REWORK","_number":10,"created":"2023-03-23 14:00:26.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/10"}}},"commit":{"parents":[{"commit":"dcca516aba328b209f78bbdc430d281387b60364","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcca516aba328b209f78bbdc430d281387b60364"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-03-23 13:59:38.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38c96cf62b08a8eda8f2a410d25aa7721617fa00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38c96cf62b08a8eda8f2a410d25aa7721617fa00"}]},"branch":"refs/heads/master"},"4a8de95bcb8c052ce770ec2e1cc63054e2fe0e08":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-03-24 11:54:00.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/11"}}},"commit":{"parents":[{"commit":"9e8d4ab155f5cad2e5757868b192aa55266ca51b","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e8d4ab155f5cad2e5757868b192aa55266ca51b"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-24 11:47:07.000000000","tz":60},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a8de95bcb8c052ce770ec2e1cc63054e2fe0e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a8de95bcb8c052ce770ec2e1cc63054e2fe0e08"}]},"branch":"refs/heads/master"},"eb311590594060b40422ffac54892ae3c2076dee":{"kind":"REWORK","_number":12,"created":"2023-03-27 09:19:51.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/12"}}},"commit":{"parents":[{"commit":"9e8d4ab155f5cad2e5757868b192aa55266ca51b","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e8d4ab155f5cad2e5757868b192aa55266ca51b"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-27 09:18:34.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb311590594060b40422ffac54892ae3c2076dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb311590594060b40422ffac54892ae3c2076dee"}]},"branch":"refs/heads/master"},"96deba1e4ab4c39a7c51ddaa8cf040c9788cd2d8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-03-27 11:05:32.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/13"}}},"commit":{"parents":[{"commit":"e33750da3d1fbba5e14d0d92f91fe965f8c14f4b","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e33750da3d1fbba5e14d0d92f91fe965f8c14f4b"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-27 10:58:32.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96deba1e4ab4c39a7c51ddaa8cf040c9788cd2d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96deba1e4ab4c39a7c51ddaa8cf040c9788cd2d8"}]},"branch":"refs/heads/master"},"ff28f30fb5394e3faf8b21862791fa0893375900":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-03-29 10:13:55.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/14"}}},"commit":{"parents":[{"commit":"35d1889bc18b39ff9ab974467227e852062d3384","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35d1889bc18b39ff9ab974467227e852062d3384"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-29 10:13:07.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff28f30fb5394e3faf8b21862791fa0893375900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff28f30fb5394e3faf8b21862791fa0893375900"}]},"branch":"refs/heads/master"},"947ef8e396fa003d36b76a5476d6415c6d9290ef":{"kind":"REWORK","_number":15,"created":"2023-03-29 17:58:28.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/15"}}},"commit":{"parents":[{"commit":"2ba3cae8cc741970711cf5b9b1a81786fcbbb226","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ba3cae8cc741970711cf5b9b1a81786fcbbb226"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-29 17:27:29.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/947ef8e396fa003d36b76a5476d6415c6d9290ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/947ef8e396fa003d36b76a5476d6415c6d9290ef"}]},"branch":"refs/heads/master"},"88bffd6ea0dd025e1c83bd67bdd43f04c9e0d8b3":{"kind":"REWORK","_number":16,"created":"2023-03-30 20:10:10.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/16"}}},"commit":{"parents":[{"commit":"0e8b2aae53323b13b7248782bd6436a6b6a795fa","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e8b2aae53323b13b7248782bd6436a6b6a795fa"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-30 06:28:27.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88bffd6ea0dd025e1c83bd67bdd43f04c9e0d8b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88bffd6ea0dd025e1c83bd67bdd43f04c9e0d8b3"}]},"branch":"refs/heads/master"},"17825bf95b451c366bcfca83fb9386e9c3706bbb":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2023-03-31 20:05:06.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/17"}}},"commit":{"parents":[{"commit":"61c091169cdcd39bfaa2d7ec765101c99406e6dd","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61c091169cdcd39bfaa2d7ec765101c99406e6dd"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-03-31 20:04:55.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17825bf95b451c366bcfca83fb9386e9c3706bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17825bf95b451c366bcfca83fb9386e9c3706bbb"}]},"branch":"refs/heads/master"},"a47b160c4e9590c6401f76854a65a84ac80c581f":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-04-04 10:05:05.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/18"}}},"commit":{"parents":[{"commit":"cbbb4442df5fcb287c7a88dd33caa33f2435db45","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbbb4442df5fcb287c7a88dd33caa33f2435db45"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-04-04 06:01:42.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a47b160c4e9590c6401f76854a65a84ac80c581f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a47b160c4e9590c6401f76854a65a84ac80c581f"}]},"branch":"refs/heads/master"},"bf0439aeb3ace96b5832c4ba0942f3b7c9aa9a10":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2023-04-04 14:28:42.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/19"}}},"commit":{"parents":[{"commit":"5cebc13908ccdf6e0db36b35316f2029cbb61586","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cebc13908ccdf6e0db36b35316f2029cbb61586"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-04-04 14:20:37.000000000","tz":240},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf0439aeb3ace96b5832c4ba0942f3b7c9aa9a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf0439aeb3ace96b5832c4ba0942f3b7c9aa9a10"}]},"branch":"refs/heads/master"},"61be5ef16bc88ca80ba03d3541612b870a0add3c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":20,"created":"2023-04-13 18:52:43.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/20"}}},"commit":{"parents":[{"commit":"b6be6100769f3a2982bf18e2437fdb4395d92b66","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6be6100769f3a2982bf18e2437fdb4395d92b66"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-04-13 18:52:24.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61be5ef16bc88ca80ba03d3541612b870a0add3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61be5ef16bc88ca80ba03d3541612b870a0add3c"}]},"branch":"refs/heads/master"},"1ce7490014628b8174653e1aada48ed6d7669ee9":{"kind":"REWORK","_number":21,"created":"2023-04-18 18:41:05.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/21"}}},"commit":{"parents":[{"commit":"8e6aae043fa17b833aa19c120b7f897c5c5be7e8","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e6aae043fa17b833aa19c120b7f897c5c5be7e8"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-04-18 18:30:57.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce7490014628b8174653e1aada48ed6d7669ee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce7490014628b8174653e1aada48ed6d7669ee9"}]},"branch":"refs/heads/master"},"0a9e7315911fd166d76006575fbb988304cc96b9":{"kind":"REWORK","_number":22,"created":"2023-04-27 14:35:11.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/22"}}},"commit":{"parents":[{"commit":"006804558e698d042ab7be6b7cd50d2dd72614bf","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/006804558e698d042ab7be6b7cd50d2dd72614bf"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-04-27 14:08:58.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a9e7315911fd166d76006575fbb988304cc96b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a9e7315911fd166d76006575fbb988304cc96b9"}]},"branch":"refs/heads/master"},"d84f023a4577061cbc24a72209080ffc16d5a7ee":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-04-27 17:29:02.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/23"}}},"commit":{"parents":[{"commit":"75a8df50a861e2a123618fbb92533a4ac250f304","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75a8df50a861e2a123618fbb92533a4ac250f304"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-04-27 17:28:43.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d84f023a4577061cbc24a72209080ffc16d5a7ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d84f023a4577061cbc24a72209080ffc16d5a7ee"}]},"branch":"refs/heads/master"},"912f8176f869d66dea5a8c0358bc8343726e280a":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-05-02 11:14:48.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/24"}}},"commit":{"parents":[{"commit":"ff67bb780821ac19dd365a7fc56f546bea9d017f","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff67bb780821ac19dd365a7fc56f546bea9d017f"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-02 11:13:52.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/912f8176f869d66dea5a8c0358bc8343726e280a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/912f8176f869d66dea5a8c0358bc8343726e280a"}]},"branch":"refs/heads/master"},"e55a416bb215d5c282676d90023be554fcaea2a9":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-05-02 18:30:22.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/25"}}},"commit":{"parents":[{"commit":"3f7b6455bec61017de9f64453a77724256954164","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f7b6455bec61017de9f64453a77724256954164"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-02 18:08:51.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e55a416bb215d5c282676d90023be554fcaea2a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e55a416bb215d5c282676d90023be554fcaea2a9"}]},"branch":"refs/heads/master"},"d4a9bb05e0f208d1c75d326e3ba680efa66bd96c":{"kind":"REWORK","_number":26,"created":"2023-05-05 23:54:12.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/26"}}},"commit":{"parents":[{"commit":"b9a18ec65b627d71a5f435ea9098f8355e30b51c","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9a18ec65b627d71a5f435ea9098f8355e30b51c"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-05 23:53:29.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4a9bb05e0f208d1c75d326e3ba680efa66bd96c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4a9bb05e0f208d1c75d326e3ba680efa66bd96c"}]},"branch":"refs/heads/master"},"750e58ec015a96811afe0310f7b7c53f45edf17e":{"kind":"REWORK","_number":27,"created":"2023-05-08 13:20:13.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/27"}}},"commit":{"parents":[{"commit":"96d3e96b95c9053da6129a4c7c1c2cd54f2802e1","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96d3e96b95c9053da6129a4c7c1c2cd54f2802e1"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-08 13:19:57.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/750e58ec015a96811afe0310f7b7c53f45edf17e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/750e58ec015a96811afe0310f7b7c53f45edf17e"}]},"branch":"refs/heads/master"},"88caaf73d9ea9926506635c4763d4b0d6d37822c":{"kind":"REWORK","_number":28,"created":"2023-05-12 06:17:10.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/28"}}},"commit":{"parents":[{"commit":"d1448362a6f0dae8de1f188abd7d5872d71eac67","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1448362a6f0dae8de1f188abd7d5872d71eac67"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-05-12 06:10:44.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88caaf73d9ea9926506635c4763d4b0d6d37822c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88caaf73d9ea9926506635c4763d4b0d6d37822c"}]},"branch":"refs/heads/master"},"6d8f70f1a2e3d5cdba358b7ec4bb5a53e2457545":{"kind":"REWORK","_number":29,"created":"2023-05-16 21:45:54.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/29"}}},"commit":{"parents":[{"commit":"8a47c9df41ab9263d936eb5d9410cba69c607d63","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a47c9df41ab9263d936eb5d9410cba69c607d63"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-16 21:28:38.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d8f70f1a2e3d5cdba358b7ec4bb5a53e2457545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d8f70f1a2e3d5cdba358b7ec4bb5a53e2457545"}]},"branch":"refs/heads/master"},"1da5bce477f1e970fd40243fdbc16638eb66637e":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2023-05-16 21:55:01.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/30"}}},"commit":{"parents":[{"commit":"083cb9a1429e85779dd42b9e0e9d9be9dbbe1a40","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/083cb9a1429e85779dd42b9e0e9d9be9dbbe1a40"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-16 21:52:44.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1da5bce477f1e970fd40243fdbc16638eb66637e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1da5bce477f1e970fd40243fdbc16638eb66637e"}]},"branch":"refs/heads/master"},"0eeee3fc628c67c0e4595f71f034c6aa9e93eb97":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2023-05-19 09:48:56.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/31"}}},"commit":{"parents":[{"commit":"ecabbff1ba2c6d4bba540332e8e81384fb145e8e","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecabbff1ba2c6d4bba540332e8e81384fb145e8e"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-19 09:42:41.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eeee3fc628c67c0e4595f71f034c6aa9e93eb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eeee3fc628c67c0e4595f71f034c6aa9e93eb97"}]},"branch":"refs/heads/master"},"bd5e8984615bcdafee656cac114f5001544bb7d0":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2023-05-24 23:27:47.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/32"}}},"commit":{"parents":[{"commit":"a221764751de05e42069f1c097b1025bd9c4fc52","subject":"Allow Multiple External Gateways","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a221764751de05e42069f1c097b1025bd9c4fc52"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-05-24 17:41:04.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd5e8984615bcdafee656cac114f5001544bb7d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd5e8984615bcdafee656cac114f5001544bb7d0"}]},"branch":"refs/heads/master"},"55e7e7ab6978b0736add549e4cb4b482d9d58451":{"kind":"REWORK","_number":33,"created":"2023-06-12 09:30:14.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/97/874797/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/33"}}},"commit":{"parents":[{"commit":"6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e","subject":"Merge \"Allow Multiple External Gateways\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-06-10 20:27:53.000000000","tz":180},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e7e7ab6978b0736add549e4cb4b482d9d58451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55e7e7ab6978b0736add549e4cb4b482d9d58451"}]},"branch":"refs/heads/master"},"98ec9dae8d77b5d8b70f5d614acc46ded51391e4":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2023-06-21 09:18:23.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/34"}}},"commit":{"parents":[{"commit":"31b425953d7611958a3073854004e6e1a648c0f1","subject":"Merge \"Delete the \"Chassis_Private\" register when deleting an agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31b425953d7611958a3073854004e6e1a648c0f1"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-06-21 08:11:33.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98ec9dae8d77b5d8b70f5d614acc46ded51391e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98ec9dae8d77b5d8b70f5d614acc46ded51391e4"}]},"branch":"refs/heads/master"},"1fba624f4c76e6c29a83af9d73bcd1c4ff12dba2":{"kind":"REWORK","_number":35,"created":"2023-06-23 06:44:41.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/35"}}},"commit":{"parents":[{"commit":"7f460ab1afad563e2285693796ea30deefd05714","subject":"Merge \"dhcp/agent: fix \u0027get_metadata_bind_interface\u0027 driver call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f460ab1afad563e2285693796ea30deefd05714"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-06-23 06:43:48.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fba624f4c76e6c29a83af9d73bcd1c4ff12dba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1fba624f4c76e6c29a83af9d73bcd1c4ff12dba2"}]},"branch":"refs/heads/master"},"64513526801227935ded4d148d93846d402e065a":{"kind":"REWORK","_number":36,"created":"2023-06-26 10:45:12.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/36"}}},"commit":{"parents":[{"commit":"b86ca713f7d422489f2526c1a94ffc8393b3ed75","subject":"Merge \"Explicitly define the subnet creating a new port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b86ca713f7d422489f2526c1a94ffc8393b3ed75"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-06-26 10:40:47.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64513526801227935ded4d148d93846d402e065a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64513526801227935ded4d148d93846d402e065a"}]},"branch":"refs/heads/master"},"00ce497412fa2eba61da10bc9c3b5215a1140bdf":{"kind":"REWORK","_number":37,"created":"2023-07-03 14:47:50.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/37"}}},"commit":{"parents":[{"commit":"91a292a6d111736b28d11b5b63ce3677cee7e825","subject":"Merge \"[OVN] Expose chassis hosting information in LSP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91a292a6d111736b28d11b5b63ce3677cee7e825"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-07-03 14:47:22.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\nDepends-On: I2618475636b2bb9bfd743a62f5d4859d4f68a547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00ce497412fa2eba61da10bc9c3b5215a1140bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00ce497412fa2eba61da10bc9c3b5215a1140bdf"}]},"branch":"refs/heads/master"},"d4d7f7d087ccc878743c94673143e6984322e33b":{"kind":"REWORK","_number":38,"created":"2023-07-06 13:39:40.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/38"}}},"commit":{"parents":[{"commit":"6b9333d8dfb78881a4368cb8716c48c45d49857f","subject":"Merge \"Switch fullstack/functional fips jobs to 9-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b9333d8dfb78881a4368cb8716c48c45d49857f"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-07-06 06:52:11.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4d7f7d087ccc878743c94673143e6984322e33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4d7f7d087ccc878743c94673143e6984322e33b"}]},"branch":"refs/heads/master"},"20e200089ea53441d79657847f8c1584d43025d0":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2023-07-06 16:33:29.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/39"}}},"commit":{"parents":[{"commit":"bad15e38e2220cbc88fa6697b42a347dab733196","subject":"Merge \"[OVN][L3] Optimize FIP update operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bad15e38e2220cbc88fa6697b42a347dab733196"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-07-06 16:22:07.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20e200089ea53441d79657847f8c1584d43025d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20e200089ea53441d79657847f8c1584d43025d0"}]},"branch":"refs/heads/master"},"572d4308bde25eff5dbaf74e830b72368e9aae2a":{"kind":"REWORK","_number":40,"created":"2023-07-07 09:53:38.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/40"}}},"commit":{"parents":[{"commit":"eee897323ca5d202e9d8f00861db42191aaf3cc7","subject":"Merge \"[OVN] Prevent binding a virtual type port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eee897323ca5d202e9d8f00861db42191aaf3cc7"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-07-07 09:49:41.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/572d4308bde25eff5dbaf74e830b72368e9aae2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/572d4308bde25eff5dbaf74e830b72368e9aae2a"}]},"branch":"refs/heads/master"},"89702218db2476a17f6ff36cf51b909db563d887":{"kind":"REWORK","_number":41,"created":"2023-07-25 15:16:00.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/97/874797/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/874797/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/874797/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/874797/41"}}},"commit":{"parents":[{"commit":"656897f32e0deebd030d357c2e3a6b14c045e9fb","subject":"Merge \"[sqlalchemy-20] Replace Query.get() with Session.get()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/656897f32e0deebd030d357c2e3a6b14c045e9fb"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2023-02-22 18:11:31.000000000","tz":180},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-07-25 15:03:34.000000000","tz":120},"subject":"Add extra router attributes for ECMP and BFD","message":"Add extra router attributes for ECMP and BFD\n\n* enable_default_route_ecmp\n* enable_default_route_bfd\n\nPartial-Bug: #2002687\nChange-Id: I3fcd0458d20f20ce40378f90f073f37c41400865\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89702218db2476a17f6ff36cf51b909db563d887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89702218db2476a17f6ff36cf51b909db563d887"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
