)]}'
{"id":"openstack%2Fneutron~874909","triplet_id":"openstack%2Fneutron~master~I3dfa2688f7288dda43883b3cfba45e5f15f182fb","project":"openstack/neutron","branch":"master","topic":"bug/2008227","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-01 08:09:27.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I3dfa2688f7288dda43883b3cfba45e5f15f182fb","subject":"[sqlalchemy-20] The Connection.connect() method is considered legacy","status":"MERGED","created":"2023-02-23 10:55:14.000000000","updated":"2023-03-02 02:58:24.000000000","submitted":"2023-03-02 02:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874909-bug/2008227","meta_rev_id":"cdc20d0f92f5f87e9a0f10e9e664ef79e876e6fd","_number":874909,"virtual_id_number":874909,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-01 15:16:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 02:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-01 14:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-01 14:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-03-01 14:36:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-01 14:36:51.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2023-03-01 14:36:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-01 14:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 11:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-23 11:53:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-23 12:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 15:55:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-26 03:41:39.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-02-27 02:24:45.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-02-27 08:36:44.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-02-28 12:06:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2817e98473a109bc60c851cb9c79e8d03abbf718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 10:55:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d290ac2d00201ab3d023ea836b92a725995ce21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 11:40:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01893e014069481e815efe5a9b2322a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d3656949a714336b15224c7dc4f54ab : SUCCESS in 42m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1181fdfc44cf4d149fab6f3a02d02391 : SUCCESS in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a0d30828f4ffb0aaae8290cc7ca968400aec648","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-23 11:53:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/09/874909/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"eb45284a8aaa40b7065dbe836b63ca623881164f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 12:59:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/007da1089f7244af92397d2485f66128\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8634be8e46cb4ecd9bc9c33c8e4f538c : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/620f119bf214440fb9454e82a4bdf5bc : SUCCESS in 1h 13m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de5e520892f64e9c81b53654d517c80b : SUCCESS in 1h 11m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3ef25c89833c4b40b36fd72e24b0aa95 : SUCCESS in 1h 16m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f1a687ffe5c749939798281919cacf98 : SUCCESS in 1h 28m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57464b3ebc90400b8409f4c6f42f2568 : SUCCESS in 27m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e7ddb04402344b1aebcc2f0687de597 : SUCCESS in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8886b83b6d448378964f5736451fd53 : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ebfa030639746b18e0b33ffcf3e09da : SUCCESS in 20m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/412f7f0e0e5a4ff791dd2a010e43339a : SUCCESS in 26m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d127b39563743a8900d4dc6037e1e96 : SUCCESS in 10m 56s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/20a7c913f4ec4b02b58dc6a5060ceeb1 : SUCCESS in 23m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4432284ea55f492380c8147080e1013d : FAILURE in 48m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc3961703e5d412d948b4a8b20115e5f : SUCCESS in 1h 45m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/984db849ea81411fa6df5587c411fca0 : SUCCESS in 1h 03m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4372d74be3704ae5b855ebbbb2808faf : SUCCESS in 56m 51s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/947043b4f30249e18ac9d9b9b3af1086 : SUCCESS in 1h 00m 46s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dc02b43ecc1f45bc897d12a7880cddba : SUCCESS in 1h 19m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/57a97e474d984c6187cb2d059916cb66 : SUCCESS in 1h 34m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0786e1a2c5ff4d45a2e573daa8b3e5d1 : SUCCESS in 1h 30m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/936e0d0fc3e843afbb8c8c9723aa0189 : SUCCESS in 57m 24s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8517b1623e5c45f7a63a29b6db031a3c : SUCCESS in 1h 58m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/13db62798dca4e5e8efcd2a77d7c3c14 : SUCCESS in 1h 46m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c5a2282f6b942929abde5d6cb0d3e9a : SUCCESS in 1h 20m 40s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1aec8eb8f4c54a2f9de172fcae7c9928 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6dd74b087262b72245faa29a573092c64101f174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 16:17:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a32300aa19697555915b354b85e5729cce1719b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 17:03:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c87a138c9e6456a9eed8f69d49717ac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1dcda74ef1743969fbd039af5d50757 : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dbf2d2df1714fc8b13729be10dd09d1 : SUCCESS in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54f65a69612f0bd964ac0bd16b5f5ad3ebd8d9f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-23 17:48:20.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/09/874909/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1bed20fba079c3db98bf61907038a0bd369f7f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 18:18:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d559172b994ce3865010ff95a8f6b5\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4312b96354c74b2aba5bc036473c3326 : SUCCESS in 1h 58m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1a3a78fbcf9c47a6bc26ad36ec565329 : SUCCESS in 1h 28m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/622aae35c8a34fbeb6bd1f64cb289529 : SUCCESS in 1h 34m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f2732917e5204329bbd79892bbde6dff : SUCCESS in 1h 51m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e05b1acf28434e72b5a3650d130b7648 : SUCCESS in 1h 45m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12d8aeb14f564d2eb4e5db5e3045ef24 : SUCCESS in 36m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50bf5d5fb34d46f8805a420ebe00dbdb : SUCCESS in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d27cdc1a5542ceaadd5d50541182ee : SUCCESS in 13m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe1c71fd1d54c03a233b7c860ee9888 : SUCCESS in 29m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c31f72e6cf6d447c98e94f411a3f388c : SUCCESS in 22m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65302486f81a425a9e157fe2d08a2ae7 : SUCCESS in 9m 44s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/14c124a318e84600be037303f3310146 : SUCCESS in 22m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/465dca70b9e64a24a1862d68defe1dab : SUCCESS in 43m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/355dbe83265149ee9bf4823900aca65f : SUCCESS in 1h 31m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d51854989aab432ea86b3edb5a35d387 : SUCCESS in 53m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cdd84248207f4c63894cdd74c6cc7066 : SUCCESS in 51m 43s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/78d32d70cb0a45fa8db5072376d70244 : SUCCESS in 1h 07m 13s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/90896867446d458a991d68b6532d1700 : SUCCESS in 1h 07m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/03eb3b1840a34ee092f680f0b22b5e6a : SUCCESS in 1h 19m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee15a9090cc847f3a112c3779c86dd59 : SUCCESS in 1h 22m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c8b38f85e0624e0daf9121dd7c08cfe4 : SUCCESS in 1h 00m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c1c14aad1ed249128b2c362d48a62413 : SUCCESS in 1h 48m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/412502caf52b4833b4ac2f650fbe0a52 : SUCCESS in 1h 39m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/960cf5cfaafc41b791476080d21ad7f5 : SUCCESS in 1h 12m 03s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7bce56ddbb9e4e05bd986449194fe54e : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2966034794c2bb6a597a7bfe2870fd7dd7fe05bf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-24 15:55:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4ddcd799ad0a52be2ee6d208b861242757aecacf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-24 16:03:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f64b0abe39ca4ebf9b07df7568aa4442b4577e5f","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-02-26 03:41:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"18f511d7b3c239cfe42019f0cbdb488689745df8","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:24:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8c489dfb11fb5ff03b86a0d81f46dffe526fbbb4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-02-27 08:36:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97f325e8d8df928e9c49f3fbcf75cffcae580dd5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 09:03:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00743e4e5d498c774d80d9e1a0d42ad823cac632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:28:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a6c9ae922ef41ab80de959bafafaa68\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/688479a4f15c4a619046ca8b46b3f01b : SUCCESS in 22m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5beec78803f4225a0b16ccd6fe150be : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1706574ed5a78a46305c987165581366fdff259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:28:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e90bcf9cce0ba2ef4690dae0e389dceb7ed35ee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:06:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18ed0b89c58346f395820064b0a99f12\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6a524146216046cab45d7184c958fcf6 : SUCCESS in 1h 27m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e1e328bf07364d0ebee07a554011d55d : SUCCESS in 1h 06m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4264cd4c948a4162bc6edb88e5bafc76 : SUCCESS in 1h 30m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2966b19b20184cd1b633623389c06e57 : SUCCESS in 1h 08m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd2d6d14c2f24522a352edd60b369dfd : SUCCESS in 1h 10m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6847ceec151c41668ae1655c08d0d1f7 : SUCCESS in 27m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df90c03507bc4c358e020edbeac28de2 : SUCCESS in 33m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597a3347a2c241e4bd16f5eb17e55e02 : SUCCESS in 15m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86001c9216d549bcaf285cfcfa5f078f : SUCCESS in 24m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a5d66dca6a54ce0937212a19ad2cbe4 : SUCCESS in 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4558f335fdc4411e854c041e5bfdb5d5 : SUCCESS in 9m 39s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e29a0cca41ea4ce8865398fb101779a2 : SUCCESS in 24m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d9818f47cd2342eb93decae486e3d602 : SUCCESS in 48m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c51c27724c64cd495571786271545f8 : SUCCESS in 1h 38m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b1813a2fe294571a20d9b44d871e750 : SUCCESS in 56m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb9556dddffd43ca98fd6dbf3624817c : SUCCESS in 1h 03m 15s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ecb6dad840eb488690f4674395656975 : SUCCESS in 1h 06m 18s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3aef81fee89940adbfd66d6ff5b90630 : FAILURE in 46m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b11684a386824a5cb7dad5d137a66a61 : SUCCESS in 1h 22m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/10b2281471384ac5aa13802ce22d1b40 : SUCCESS in 1h 19m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4c1e09d518914260b155e52fc4d80099 : SUCCESS in 1h 03m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/29d8b40756e649eea1f661c7f1e0d038 : SUCCESS in 1h 38m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3e9e699d08da4a708744601f3e10fd96 : SUCCESS in 1h 45m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0d2a5b7e506d4b00ad6ab9784c669f7d : SUCCESS in 1h 23m 39s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a743de216fea45dbbd377b4ec8761b37 : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3c8457640715342a1c7a48e6afc3c668ac63b098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:25:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/849ead796e5149deb1bfff9d3f0c5768\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/56f45703afc1413e859797c9fd3f171f : SUCCESS in 1h 17m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4caef5956dd04ba7bd5557c62226b078 : SUCCESS in 1h 22m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c8e3b10af9e54aeea3f4664710cfc7d1 : SUCCESS in 1h 15m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d01f78e27d6c40d2a8eef74029f1822e : SUCCESS in 1h 14m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9d4129a09e324493be608c9ad3e67d81 : SUCCESS in 1h 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86489618e63a496ea3bc0279dd964af6 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c85cc45f38d44fbac5bf4048fca6d95 : SUCCESS in 24m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd7b3d463ae24702a0af281b7ed2da29 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ffbb5473284c3a8eafb87e30eb750e : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3889369dea1e4e45a2080144cbc25611 : SUCCESS in 50m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/619cebf4ed804db9ae1fff8f320f92c9 : SUCCESS in 1h 49m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d45d347ba5864ac5b0a115ebca8bb105 : SUCCESS in 1h 28m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/be615a997b014804a470259a21c02c39 : FAILURE in 55m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a9b5787bb45d4f6486dd8231d6454865 : SUCCESS in 1h 01m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a431f6d45717480195ccb8096c48c293 : SUCCESS in 1h 07m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0deddeadf854b0394339f9452269995 : SUCCESS in 1h 24m 13s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5f7c3ebbe6e643d4ab4e630bb3ef9ebd : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ea4b098b298d4d8eb0e042a7c48c048cb5dc8c42","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-28 12:06:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b9b39e288f150eebcec5a1489bf4c835865e832","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:56:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcd59394795f41948c85336f0aa6c0f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d75aa698d61e4b94a2f8a2e9f25103df : SUCCESS in 46m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/542d266450474c4ba33bfcdc7122a042 : SUCCESS in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07a3488898ecb9d547678452d671d2a477c26cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:03:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d0a33d1a6044b2bbd6b54dad62ab35\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dbf530cdf8ae43ffbf7a76cdf3dcf8e9 : SUCCESS in 1h 49m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ad6630cb9d049b6916d00f8e614365d : SUCCESS in 1h 44m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3744267c30d84c1db67a160e18167d56 : SUCCESS in 1h 48m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e692dd8f06d54861958f15becf6a12f2 : SUCCESS in 1h 45m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd7179be7f334746a7d7cf78e696d398 : SUCCESS in 1h 51m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b863f1532647426894e44389f5d128cc : SUCCESS in 25m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6008438c21eb42a993aa2a63262fba71 : SUCCESS in 44m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd91ce0fb6c481387ab15d92611798c : SUCCESS in 12m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71a33e4f9691472191591c035c915e4d : SUCCESS in 22m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66b333d81a674affbdb12484a6c23549 : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa55359de7704002bc44c0d4a0786147 : SUCCESS in 10m 53s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d2f2fa26559d419fbae072cd2bb466c0 : SUCCESS in 21m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca11153957a24d92800d15f4408a48c6 : SUCCESS in 48m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a3173c7811b41a9af114e9ac294a862 : SUCCESS in 1h 31m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74cdc765c46241f091e13df907b23271 : SUCCESS in 57m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/89fdf7cad77049a9b2284090f4fa1060 : SUCCESS in 56m 23s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d3d0c3b6db6f4532bb221d029fe7c844 : SUCCESS in 1h 06m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f7ff95af49746db81003624d4f9ede0 : SUCCESS in 1h 05m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fad6aebe0d9d47368b3d55606d4697ed : SUCCESS in 1h 22m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d882ad83647a4b8cb8da24f27149ca6e : SUCCESS in 1h 16m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3e2d4ae39dec40dabe5ca1f0751de8a6 : SUCCESS in 55m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/158f722d6dda4f3796e4afb8c342e2ee : SUCCESS in 1h 50m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4bfb67c6ea234c7a9abebe579268d293 : SUCCESS in 1h 54m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/09f069bfa7414fa5ac4d3d92fe08c163 : SUCCESS in 1h 17m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4b1738e8f65e478bb06a69feabd2c51e : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":2},{"id":"463b9a85fc7230d898281701b2c3fea741190a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:04:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"493e9f498714b9dcb0da0a6ab4b762310aaffe97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:39:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8438d1b40db645fab81a7dd186f8b57a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fc2161253de04e4dbb607379cfb5b9b2 : SUCCESS in 1h 16m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f008335c1112482fbda4cec7abc2b844 : SUCCESS in 1h 26m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c75b1c75ca3f4e4d860519564963ce64 : SUCCESS in 1h 17m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e2c24727d0c47058c861f89424345d9 : SUCCESS in 1h 16m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4d120618555340919d975a633dde1f0e : SUCCESS in 1h 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e1359a2a494428afcf806a0e9e01c5 : SUCCESS in 13m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e971837000549898818c880298d952d : SUCCESS in 24m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2b26f23a1d34725b814fa113ce0bbda : SUCCESS in 26m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ef894d454b4485b2871a49c0adf1a0 : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e244bb919c94481586489ccf3699dafa : FAILURE in 40m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c7bf04050534a5c8fe4cc36da2c1ee1 : SUCCESS in 1h 32m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/09e6f18513464a528822778e09900793 : SUCCESS in 1h 26m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/001b588ceaca43f8a637966980a5913a : SUCCESS in 1h 01m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6b7dbff36d14c5e8f880e3ec79f2233 : SUCCESS in 59m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5b8329ea591e47779fee9f492675ef0b : SUCCESS in 1h 19m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/254cad98404f42c7944e9515994ac1c4 : SUCCESS in 1h 18m 11s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/44288df9053d431f9398f6f8da7429f4 : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c2e460021fc2bfb94a60c8b26fbce6e8c7935c9d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 17:47:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bce15ed9cc941a135b79dae4d501b8ca2dd17f4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 18:09:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/236375b1d491454fbe9ba7648c7e4fb9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d89ceea9610b4a1d8b63798aac10eb64 : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac7c18cd0c9246af9922d148b5ca6a0d : SUCCESS in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0a5ab6aebee917cb78de02bffcdf2eb5f76e306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 21:57:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c36eca942c134aa6ab8fe9731fcd8bc1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/10407c29aea94670a1c8b71474ef5d36 : SUCCESS in 1h 16m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a7ffbd154ab48339166054ef5554087 : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30ccbb6444204a2c9ade39d45a9efbe5 : SUCCESS in 1h 28m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cabd261e92ea4903892b676642470701 : SUCCESS in 1h 11m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1375df91931e472c8e8118adf4505adb : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67ec97a40fb34461bf0816c9a8643119 : SUCCESS in 30m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b28f0c694d14d148326254ab4db2591 : SUCCESS in 37m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f472b84a1f42490abeb1e73b08893bd0 : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c84e6b83e7a4d098e0259228f49abae : SUCCESS in 23m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbca69871fbe40f6880e1b344fa4ae70 : SUCCESS in 35m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a855d933951845b59d47ad612f66af75 : SUCCESS in 9m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d38778738841431eb6b404eb638e0e64 : SUCCESS in 23m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa0f3204d2774a7ab3fd7617b82fc321 : SUCCESS in 42m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c364b0ac9b494cecb1f8527571929f25 : SUCCESS in 1h 36m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bad578a35b047ff94a73d35980e865f : SUCCESS in 1h 00m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c1fce3704452467f90a75be1f92123d4 : SUCCESS in 50m 26s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c750f688a0cd48908d5a41eb53806598 : SUCCESS in 1h 04m 48s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3eaf328070b341a18a86251f72fb1bef : SUCCESS in 1h 08m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dfacb3097d4d48a5acb2a99afc1ef762 : SUCCESS in 1h 36m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72da4b1c737b4529b0d446d953275ef2 : SUCCESS in 1h 20m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bce8224464c444edac468c5cafb230ae : SUCCESS in 1h 04m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad9aae0e57254ad185e4ac4f9c5b2642 : SUCCESS in 1h 32m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef8f2af3f37b456cbb8ef7af09dce124 : TIMED_OUT in 2h 14m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a338234ba6864c158dba263699407643 : SUCCESS in 1h 32m 38s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b59c7d22aaa34edd80e5ac647e5d1194 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":2},{"id":"995a2e9bd0672fc9947c0e47aef554b0ca7bf62d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 08:09:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0eb153fe51b1df96cfefdc7d618d0040f296108a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 08:53:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53beb29f7e224a678c83968a98901d29\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d762a61f16047c8ac3ff922c17cfa96 : SUCCESS in 42m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c590968f9e594bfb83570ccbb0b2b4ac : SUCCESS in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a46c9520b80c42b1382c4ce80ce94e1098e2f54d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:18:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a861047f14fa4495a19ce4389d0218bc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1637ea52a0fd4bec976345f2e31b7984 : SUCCESS in 1h 21m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b0c0f511bc524a1d95017f891b0e6856 : SUCCESS in 1h 34m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b48187a8ab3a47a4b6e40f592316039e : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7b2d4e571e84964b8b626f8aa1787b7 : SUCCESS in 1h 19m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/17334fe87bae4085836b7b7bff85a578 : SUCCESS in 1h 18m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ee7e626da344ea288a130e56210c3a0 : SUCCESS in 25m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/318e98020a144234b54333a64173f25f : SUCCESS in 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1c2ee131ca409d8ea60148504441af : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d77d865649da400e92da1538cb5f2e00 : SUCCESS in 29m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dab86be3f7e94e8a84aec49da8157497 : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a4fa0ada5a4cc893e3dc96f6116b0f : SUCCESS in 11m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1819107e161d4258808c37bd7826bd57 : SUCCESS in 22m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f4da77450794ece8037a459eef38619 : SUCCESS in 1h 03m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/54fd2b06ed7141519fd9a24cad3b822a : SUCCESS in 1h 23m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d941d71cf3d4e3496598439c782a17a : SUCCESS in 54m 02s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1d374ca5ba1948f0836b60ecbb489f37 : SUCCESS in 55m 09s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ca75dab86c6b4a5f8bcf1539f1881e77 : SUCCESS in 1h 12m 39s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ae50e7b5ba743879459e5599ea1b75e : SUCCESS in 58m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32b4e6cd98844207a10492f1f73ed381 : SUCCESS in 1h 09m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c33ada541233489aa600f6137f46f6c1 : SUCCESS in 1h 28m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2138334f8d974d96ae149cd10c37066a : SUCCESS in 1h 15m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/990a0bb83b0b42d5a1d4218d60fe5be0 : SUCCESS in 1h 51m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d844f65228214c43a72d69005d6a997e : SUCCESS in 2h 03m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/be1d060cee89401fb6651627c3a7bd30 : SUCCESS in 1h 22m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/8e44f7ddee8d4a80bff951a02d26a406 : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":2},{"id":"195a3b1600951fcfb11fdb5fa9be3403c80687b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:19:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e3c6e1d9b6978f9f33205c3661f7b076bdad4ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 12:52:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14bdba37c012473ebbfa552cae21bc29\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1a52eb0af61f42d2866076cfaa098b95 : SUCCESS in 1h 21m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3db407ed9a594f4b87fb92d27928e2db : SUCCESS in 1h 27m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/60e9efdd63a24b2bb7ddab1b9a935b84 : SUCCESS in 1h 22m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01e783cfa7e74a6da5f364bc015f2c74 : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ffc970c2f59f43c9a826baf71719e212 : SUCCESS in 1h 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140905ca942245a982d402ce8995c31a : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd64f3acf6b422083af76b59b804e88 : SUCCESS in 28m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38dd1102f7a442609b02cd81c498dbfc : SUCCESS in 21m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269676a5d3fe44f283f03b2a3339178e : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/23b1457d53af4fe8b4fc6f0a5334e331 : SUCCESS in 53m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23f488b45b3b4247a8dd797b05e93bc7 : SUCCESS in 1h 24m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1094437f4e0a4b838a6528e5382a1147 : SUCCESS in 1h 35m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3efe63ed43b24887b447fcbf93f07de6 : SUCCESS in 1h 08m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/790d396c644f4e10b31b4488181ab822 : SUCCESS in 56m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a866368b50264fdaa8785fb93cb18673 : SUCCESS in 1h 13m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eb76e9ddb8d4413798762302ab1b5139 : SUCCESS in 1h 12m 24s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/06e6aa68b9a9484cb7ddc84322c68e01 : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"16114d6e29a8b28cbd6d85160e2a2d146cc37fee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:36:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21cce6fc51ba747218e23cdc2629dfe9f602c30d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:36:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d68d5ec888757292bec25c7017a612e209e6cd97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:36:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ae08fd61b216d3c8aa111d7a1b06288d4c2b7b0c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:36:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4e0e8e6151e95ff95389ef4b4032f072f0a7bfe2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-01 15:16:26.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/09/874909/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a88f13fe2f0fc30556dbf49df305f5d735cfa2a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:21:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e860226abda8417489e8c9be9f0f0585\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bb90b04b660415a894dc1a030f64c41 : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff498c3ef9f64598a24aa9b8aa3312a4 : SUCCESS in 41m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2ff762ae18a960c2f14a1355af1e7b2bcc9500fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:06:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1d9221ebf64f5788667a4fcbda27f0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c57631a5e0604abc9114ed3c8c1ea54e : SUCCESS in 1h 27m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6bce2326d1cb40df946d149bf19670fd : SUCCESS in 1h 19m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dfc993e3de1147d89461ab6eb683e740 : SUCCESS in 1h 18m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4a18bcf8322a49fa9f6d7aea841a9db0 : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a4fd10ca616416392ac2ca28f3f18b8 : SUCCESS in 1h 11m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea4e6afd92ba409c81d93715bf8f8525 : SUCCESS in 27m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e7103d93225493783cab7cc441f0671 : SUCCESS in 56m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e05b0683580042aeba8e7cf2c1c5c3d0 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaac3a4471bf41ce8a05a2c2fb70421f : SUCCESS in 27m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f0b54be4a2c4cc0bac1b8388039f9b4 : SUCCESS in 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/457f48a1f79b444fb8fe5a8fab9f8415 : SUCCESS in 11m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/251fd6bbd2d044a7a6aa94b0fae06bb9 : SUCCESS in 25m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b4bad345fe54537ba7630905f53c8ff : SUCCESS in 39m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/527756bedbe947e491565365c767d3ee : SUCCESS in 1h 31m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7066ae53303a4607a93a6eee44c47866 : SUCCESS in 1h 10m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3600581e0d124493a6970d2c521dd807 : SUCCESS in 1h 03m 47s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d7c91ac3c4ac42169700f35fd9a1f61e : SUCCESS in 1h 09m 04s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/73bdc1ba83b4400abc13b4c95b6daeae : SUCCESS in 58m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0750a50fc26d4f59a7ffd96702845728 : SUCCESS in 1h 30m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/663c8bb24277499a963f89f610dd56b9 : SUCCESS in 1h 26m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/82f3f33cdaa847c7b98728125e9de85f : SUCCESS in 1h 16m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0ff3c916f0f845bfb78e6573029db095 : SUCCESS in 2h 03m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c2484aa304e34422a8309a938a4af8a5 : SUCCESS in 1h 43m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bb6e717f246c438abb4a13ce3c061bf2 : SUCCESS in 1h 20m 34s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1486d27814644b06bebb62c00fc14059 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":3},{"id":"05ff2a96131b94df6b854e916ca44040196acfc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:08:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b927793330de03b6280cc0bbe84dc69971577497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 02:56:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f021e69f1d0c49e3871d5e223fa16f86\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/39689c2d4f2b4b14bb5b99056eaf2b3c : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b4def24eea5465eb59a3818107212f0 : SUCCESS in 1h 32m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/83b46f7138fc4aa5ab85e03ba2050be3 : SUCCESS in 2h 03m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f53c8f0e68d42c1a554ec9326aa3377 : SUCCESS in 1h 43m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c1a5476a42e64441b57ec1d42cd98236 : SUCCESS in 1h 46m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b561389f94f46cfab1ae8823754073e : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9c2f5215f134b3dba51b489ba6e292b : SUCCESS in 24m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72218bfe37674c3f8ad49bd71e2e5fd7 : SUCCESS in 30m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c1bdbd041e48c8b56ccf260e5c2192 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd1ce60aec2d4b94841cbdb676d05adc : SUCCESS in 53m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0af2731d47d487a99ba0643b02e3b65 : SUCCESS in 1h 43m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/84b46959e41f4a36a87d24d43a60a6ab : SUCCESS in 1h 27m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cfef8cc86084c149f7c4404416816d1 : SUCCESS in 1h 01m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/619e29d8d54940948d8e2211701c7400 : SUCCESS in 1h 04m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a21bcda6e49244e98d4f7a7dee4551b5 : SUCCESS in 57m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a8b218fb7a945fab4ec134563414eca : SUCCESS in 1h 24m 18s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e0654a658682431b90abee0555750a0e : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3e13acdd5cc8fc166e46447e48f68767636e788c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 02:56:35.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"cdc20d0f92f5f87e9a0f10e9e664ef79e876e6fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 02:58:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cbf638c16374289912dda73b070dd2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab569205c8544289b6772d36b7bab19 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41846a6e8bd5fd55cd2661d50c51d31956fa1156","revisions":{"6be3349dc68535bec0b9a80ba296f1c83ba49f92":{"kind":"REWORK","_number":1,"created":"2023-02-23 10:55:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/874909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/874909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/874909/1"}}},"commit":{"parents":[{"commit":"662972c4f59fafd912b0ad69f5ae2d854ce854b5","subject":"[sqlalchemy-20] Use the .begin() method of Engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/662972c4f59fafd912b0ad69f5ae2d854ce854b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 14:26:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 14:26:57.000000000","tz":60},"subject":"[sqlalchemy-20] The Connection.connect() method is considered legacy","message":"[sqlalchemy-20] The Connection.connect() method is considered legacy\n\nCloses-Bug: #2008227\nChange-Id: I3dfa2688f7288dda43883b3cfba45e5f15f182fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6be3349dc68535bec0b9a80ba296f1c83ba49f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6be3349dc68535bec0b9a80ba296f1c83ba49f92"}]},"branch":"refs/heads/master"},"e6595a4107e196b27e6bea9f49238a608ba54fee":{"kind":"REWORK","_number":2,"created":"2023-02-23 16:17:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/874909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/874909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/874909/2"}}},"commit":{"parents":[{"commit":"662972c4f59fafd912b0ad69f5ae2d854ce854b5","subject":"[sqlalchemy-20] Use the .begin() method of Engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/662972c4f59fafd912b0ad69f5ae2d854ce854b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 14:26:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 19:47:09.000000000","tz":60},"subject":"[sqlalchemy-20] The Connection.connect() method is considered legacy","message":"[sqlalchemy-20] The Connection.connect() method is considered legacy\n\nCloses-Bug: #2008227\nChange-Id: I3dfa2688f7288dda43883b3cfba45e5f15f182fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6595a4107e196b27e6bea9f49238a608ba54fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6595a4107e196b27e6bea9f49238a608ba54fee"}]},"branch":"refs/heads/master"},"41846a6e8bd5fd55cd2661d50c51d31956fa1156":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-01 14:36:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/874909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/874909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/874909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/874909/3"}}},"commit":{"parents":[{"commit":"fb08c34561bebf537a728c8043f643b0fffad771","subject":"Merge \"Check port.tag is not DEAD_VLAN_TAG in ``DHCPAgentOVSTestFramework``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb08c34561bebf537a728c8043f643b0fffad771"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 14:26:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-01 14:36:51.000000000","tz":0},"subject":"[sqlalchemy-20] The Connection.connect() method is considered legacy","message":"[sqlalchemy-20] The Connection.connect() method is considered legacy\n\nCloses-Bug: #2008227\nChange-Id: I3dfa2688f7288dda43883b3cfba45e5f15f182fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41846a6e8bd5fd55cd2661d50c51d31956fa1156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41846a6e8bd5fd55cd2661d50c51d31956fa1156"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
