)]}'
{"id":"openstack%2Fneutron~875537","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ie903fd3f645f7428854cf6b2fe48a1edf342cc32","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2007992","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-02 10:47:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie903fd3f645f7428854cf6b2fe48a1edf342cc32","subject":"Format correctly (dialect\u003dmac_unix_expanded) the MAC addresses","status":"MERGED","created":"2023-02-27 17:04:58.000000000","updated":"2023-03-02 10:48:32.000000000","submitted":"2023-03-02 10:47:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875537-bug/2007992","meta_rev_id":"7ba11c8adf583daccf56b5151de1986c59ed5b9d","_number":875537,"virtual_id_number":875537,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 10:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-03-01 13:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-03-01 14:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-03-01 14:52:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 10:35:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 13:14:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-01 14:52:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8356a030a06e68fe183252566b982e75ae06e6a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-27 17:04:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51d52dc83abb473975417db8291baf6e2fe6bbb0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 09:43:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc329da7246681a22ddd42878d523a1962ccc537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:35:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/520e60d5eaf64486b5573a89cf4ec6b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58644720eb345f28edf27e66f8dc658 : SUCCESS in 13m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a540369ddd9141818d4dd7ee5a812407 : TIMED_OUT in 41m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5dcc00183b94a07ad6c05a6e0a964b5 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0527160eed534ba88b652b6bcf47fc8e : FAILURE in 41m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2a6e6058cb9af2b34f3d00130b5a4a3337534827","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 17:49:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0afa08f19b4fb56e256d7892f3cc104bb7c9f859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:57:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642bfca945f84f13be4044937260b64f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ea649ca45242c097c43c568564e9ad : SUCCESS in 13m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/325477b86bec480c9aa506a25436ff5f : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea50c7e9a25c4520898b2d29b1e623b2 : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/53d586599ff343d5ab83c1feac73bb9d : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8725633980042e87119befd87f4dc76a64fe531f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-01 13:14:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93c8a50726c061d05098e86baf290ecad33f9660","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-01 14:52:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"494d82a34017c46607b21f4261116189a052cdf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:53:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3102e272ce3842c4d607b78911bab7978cef540b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:32:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25273ce4f67b4c1baa8900584fba4793\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636f83c9646e41f1a69ba05114ce02dc : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/170989e90618417db5fdf684508466aa : SUCCESS in 23m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1730dc677724d129d08adfba0dbe32f : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a862d03920a54d968fe07bf8d99c58aa : FAILURE in 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3278649ae616682e921c7e5226632123e8851999","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-02 08:01:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09e0155ecb27bcf71453401240a6cc0c53ce85dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:14:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13caa154a1944b38f05fb7e85eb9680\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18298639850439c9e8e6666452ecb5e : SUCCESS in 14m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ee1636a02c45b3a35ae6e8d0ef399b : SUCCESS in 25m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa7c17e50fd4cdf93c54f522265cbe3 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/685ea31e53f746c6ac3cf0b8b593feac : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":1},{"id":"472eb590609f70d9c8341c98fda912f2d62da081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:14:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d52a91addf9cab12ccad08df20588f9d031373e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:47:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716670cf6bfb4e688bdcdc474dc7bdcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc2798a8e91647298edf024217542761 : SUCCESS in 13m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28450bee50af44bc8c9acdbf8d5579f9 : SUCCESS in 26m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2569fcd1df6406393777276e599b489 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/be0b6942b49f47ecb922950d4d2c6382 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aedb57972aa1854c25108c278eaee666d69321ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:47:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ba11c8adf583daccf56b5151de1986c59ed5b9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:48:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4558228ac94d8186af3ba45851958a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b4359b63344a8d81e6b55e2399f9a7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ea2b540aeeeeef01c5130bde0a563626c394416","revisions":{"5ea2b540aeeeeef01c5130bde0a563626c394416":{"kind":"REWORK","_number":1,"created":"2023-02-27 17:04:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/37/875537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/875537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/875537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/875537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/875537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/875537/1"}}},"commit":{"parents":[{"commit":"dd45d5baa0bf8b269233b3d166aa3c48766f4cb5","subject":"Merge \"[OVN] Bump the port revision number in trunk driver\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd45d5baa0bf8b269233b3d166aa3c48766f4cb5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-17 08:55:50.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-24 18:18:38.000000000","tz":60},"subject":"Format correctly (dialect\u003dmac_unix_expanded) the MAC addresses","message":"Format correctly (dialect\u003dmac_unix_expanded) the MAC addresses\n\nFormat correctly (dialect\u003dmac_unix_expanded) the MAC addresses in\n``DHCPAgentOVSTestFramework``. Before this patch, this is the leases\nfile the DHCP server was using:\n  Done building initial lease file\n    /tmp/tmp3_m7_fqk/tmpiadm5q0e/9afb0444-e28e-4582-abb0-031d6ed500e4/leases\n    with contents:\n  1676775199 24:77:3:7d:0:4c 192.168.10.11 * *\n  1676775199 24:77:3:7d:0:3a 192.168.10.1 * *\n\nRelated-Bug: #2007992\nChange-Id: Ie903fd3f645f7428854cf6b2fe48a1edf342cc32\n(cherry picked from commit c61ce3447e2aa3689fa7545170b4e0d94d1a4380)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ea2b540aeeeeef01c5130bde0a563626c394416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ea2b540aeeeeef01c5130bde0a563626c394416"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
