)]}'
{"id":"openstack%2Fneutron~875552","triplet_id":"openstack%2Fneutron~master~Ifd44dcfb57cdb8c0ebf976fd7c8da4009bdc1a3c","project":"openstack/neutron","branch":"master","topic":"move-register_common_config_options","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-02-28 14:46:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd44dcfb57cdb8c0ebf976fd7c8da4009bdc1a3c","subject":"Move register_common_config_options call to DietTestCase","status":"MERGED","created":"2023-02-27 20:51:43.000000000","updated":"2023-02-28 14:48:16.000000000","submitted":"2023-02-28 14:46:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"875552-move-register_common_config_options","meta_rev_id":"6dc16905062de945ca3937edb61d242c7c9faa4f","_number":875552,"virtual_id_number":875552,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 14:46:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-02-28 11:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-02-28 09:57:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-02-28 11:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 21:36:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 21:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-27 22:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 09:57:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-02-28 11:56:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"70066d728b6bd74a3bd1270f10b4a24d4da62bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-27 20:51:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7cedec1a62b4f866d33047c3454cdfe8cd8d1e9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-02-27 21:36:52.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/February-2023/Arista_Tempest_Testrefs/changes/52/875552/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"da9828831db0afe27942ee5a78691ba39763e6c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 21:38:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8a80b9b9a54e6385f0491c1aa13cd2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53ec5edecfd74dada1475cc78bf4840f : SUCCESS in 43m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69cfd82d2b9940549688d9e8d2636d9d : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83f8f99fb4b030723076eba922d9a69631a34508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 22:25:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2827f0f61b4996b9ed8729a0497f6f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/718629d5ee1d478aa3362e97b3b49b6f : SUCCESS in 1h 11m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7244640fedf141b2bc7f04db3e0fece0 : SUCCESS in 1h 17m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20205fe9ccfb4679a1ea678d4dde86e1 : SUCCESS in 1h 12m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/664d3c40c2ae443cb03d4f274f2798b3 : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8d9530897e364cd09aa4b35a48b283a9 : SUCCESS in 1h 19m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/59be3e2aa0e14e1780cfe232d96c2742 : SUCCESS in 27m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9097fbd74a8471abea30711907f80a4 : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc8843d1bfe415cb0a3e72075ce097f : SUCCESS in 14m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bda9bebba1c4ce4aa788f31433774e4 : SUCCESS in 26m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e5606ce3aee496da44de54e15a942e7 : SUCCESS in 26m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b059c69098904c6e9bb325b17d3fa1c6 : SUCCESS in 10m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/56af45265d6946af940e968ae10edd2e : SUCCESS in 27m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a65677c105c43f19fc8660e4365648e : SUCCESS in 53m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e4b22f3ad6c41b2ae591ec97ba841f5 : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2f35a3fd24d2b2e6371b95e6525e7394fefdad51","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-28 00:49:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a8744a624bc987b06d8a3844e54eb9c0df2afca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-28 09:57:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81dc146c05b9e3df2ee6b0feef931e0f88c58996","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-28 11:56:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1530e04904570a207bdeb1c0ec7c59c4628d6d64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:57:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18f0e5bdd82511e809c539e46066e9692dfc5b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:46:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47aeacabcdc748078cacde44e103ddb4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/20b2e74ee3ec40138ed4de8116a668a9 : SUCCESS in 1h 32m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/69905d24c45f4d1997a938a260b44040 : SUCCESS in 1h 11m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9923ac39cd5d4470aeaea2f3874020f2 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fce87cb9bb0b469ab7e25b9475c52252 : SUCCESS in 1h 13m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/44f76ddc4c1c4d2a8c64910f1f17017c : SUCCESS in 1h 06m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f99288c5e844d2195cd5d939664ffc3 : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fbb9cc5653e4da78302f6abc045b465 : SUCCESS in 31m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/184c2543e03046ca96e2c8229194cc9a : SUCCESS in 23m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787263ef89b741b592e425638e729ae0 : SUCCESS in 13m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4498ebe6daae465d93ca36a515f27b10 : SUCCESS in 49m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5784b7d13dc454da070da8dd4b3eeff : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":1},{"id":"468f4b510bbb2b61fb5bf4c1992d7d87b12ca781","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:46:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6dc16905062de945ca3937edb61d242c7c9faa4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:48:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3910717e3b99480dae8f4908efa021e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bb9148e59c4dabb8ecdc23702c1255 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdf0c8bdadf7b4a348e5e8494598bb584ca3b02b","revisions":{"bdf0c8bdadf7b4a348e5e8494598bb584ca3b02b":{"kind":"REWORK","_number":1,"created":"2023-02-27 20:51:43.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/875552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/875552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/875552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/875552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/875552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/875552/1"}}},"commit":{"parents":[{"commit":"f2093d98d73abcbba673fedc6d638cb72b6b0ad1","subject":"Merge \"Reduce number of routers created in ``test_gateway_chassis_rebalance``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2093d98d73abcbba673fedc6d638cb72b6b0ad1"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-02-27 20:45:47.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-02-27 20:45:47.000000000","tz":-300},"subject":"Move register_common_config_options call to DietTestCase","message":"Move register_common_config_options call to DietTestCase\n\nRunning the quota unit tests generates this warning on\nthree tests:\n\n oslo_config.cfg.NoSuchOptError: no such option\n  notify_nova_on_port_status_changes in group [DEFAULT]\n\nSince this seems like it could be common for a number of\ntests, move the register_common_config_options() call from\nthe BaseTestCase class to the DietTestCase class.\n\nThis could have alternatively been done in the setUp() code\nin the TestQuotaDbApi() class.\n\nTrivialfix\n\nChange-Id: Ifd44dcfb57cdb8c0ebf976fd7c8da4009bdc1a3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdf0c8bdadf7b4a348e5e8494598bb584ca3b02b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdf0c8bdadf7b4a348e5e8494598bb584ca3b02b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
