)]}'
{"id":"openstack%2Fneutron~875682","triplet_id":"openstack%2Fneutron~stable%2Fxena~I75a37fb9b49a421e4524da6b56ef8362ceb6107b","project":"openstack/neutron","branch":"stable/xena","topic":"ovn-agent-heartbeat-timestamp-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-01 19:19:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I75a37fb9b49a421e4524da6b56ef8362ceb6107b","subject":"Normalise format of OVN agent heartbeat timestamp","status":"MERGED","created":"2023-03-01 10:21:00.000000000","updated":"2023-03-01 19:20:57.000000000","submitted":"2023-03-01 19:19:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875682-ovn-agent-heartbeat-timestamp-stable/xena","cherry_pick_of_change":874926,"cherry_pick_of_patch_set":4,"meta_rev_id":"ad4aa2ca4f588943626cb42d2754e07bf768da23","_number":875682,"virtual_id_number":875682,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 19:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-01 14:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-01 13:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-01 14:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 10:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-01 12:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 13:16:18.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-01 14:32:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9e01df5f3eedce7786039fa78bba3be96f59dc07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-01 10:21:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9a22a83ebabc8864b880287353cb0541532253b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 10:52:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c05dfd309884759b752246ecd9183e7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5320eb0957374c678b17042e35142821 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba4a13520da841b2be0d3cd2650ea57d : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3861c7abec08f893541f019b31066da0d6444c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 12:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c3db431e1fa44448d69ce9d80ced56f\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/05191b95b5cf4c0db1b6c502edb31eed : SUCCESS in 55m 38s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/0a50db4e69c34ccc9b8f66639c3840c7 : SUCCESS in 53m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fa626dce4e146868b4ded633d9873bf : SUCCESS in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ce369a29b35427c8487774f765bd591 : SUCCESS in 14m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbcdbbe3f9cd441ca18d654abac6aee5 : SUCCESS in 27m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d6cbbf0df614dc59d92d5f5c99ed25e : SUCCESS in 25m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9989e90b75fc4496b6e3c7501f58ac79 : SUCCESS in 28m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6ccfe2b70940ca8780814fea6853c8 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/012bc761b764461d8fb2e6469ebd7a80 : SUCCESS in 6m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/608a3b587d0b46eeb7194230c89dbddb : SUCCESS in 45m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca65677786d64cb8b8d14ef9f33e6a94 : SUCCESS in 1h 00m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d317854a5dba490c9b4c66ccd36ae3c5 : SUCCESS in 1h 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57bce0a7539a7b3adbc4f65e2eeaf19f709ddd86","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-01 13:16:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7062bdebba3a70d2f5560781b65ece0fab01a92e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:32:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a4cf5bd83cfdabf3df9d58853acc0c536e3b8d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:33:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa986aa5e6b3e3ff763a8bbb0698d1136d917763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:19:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4db5dad116dd43588e6c88a1e4c9115c\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/c00399daa8b54c5b8065d947ad7f4a7b : SUCCESS in 48m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06b08e1c36045228a7b8b7260a358fc : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/576247f03658401c9682b65d46f7a2e5 : SUCCESS in 24m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad1ad28f4c6240da81509d3336ba1e06 : SUCCESS in 23m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c640ead9fd374ee2882eda57e659bcd5 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cef1b64bc1ad4d27b461df2c6290dbf3 : SUCCESS in 7m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4221581c30f4cab9439d9e30a9666d3 : SUCCESS in 43m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e7f796905234766beacb6f5a306559b : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9b6203496ef736ebc7adcca62da5687bd3f2e38d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:19:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad4aa2ca4f588943626cb42d2754e07bf768da23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:20:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a68bab664748a080cb3c9f8a620a31\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876db1d408374b0ca470ed9ae6f3122d : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/724c7b8b849348b08df3629eb30ec4fd : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19ad0f2853cf6a72b8d1c5d2e914301e52532d3e","revisions":{"19ad0f2853cf6a72b8d1c5d2e914301e52532d3e":{"kind":"REWORK","_number":1,"created":"2023-03-01 10:21:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/875682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/875682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/875682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/875682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/875682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/875682/1"}}},"commit":{"parents":[{"commit":"7e7eff2d35b43e2a885dbf1eb18e8674945d8c96","subject":"Prevent router_ha_interface port from being removed via API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e7eff2d35b43e2a885dbf1eb18e8674945d8c96"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-02-24 04:34:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-01 10:21:00.000000000","tz":0},"subject":"Normalise format of OVN agent heartbeat timestamp","message":"Normalise format of OVN agent heartbeat timestamp\n\nA recent change [1] to show the real heartbeat timestamp from OVN agents\nhad a side effect of changing the timestamp format, which now includes a\ntimezone:\n\n    +-------------------+----------------------------------+\n    | Field             | Value                            |\n    +-------------------+----------------------------------+\n    | last_heartbeat_at | 2023-02-23 14:12:07.471000+00:00 |\n    +-------------------+----------------------------------+\n\nThis unexpected format change causes some clients to fail to parse the\nresponse to GET /v2.0/agents.\n\nNormalise the format of the timestamp to remove timezone information.\nAlso remove the microsecond part, which was not done for OVN, but is\nabsent from other network agents.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/844179\n\nCloses-Bug: #2008257\nChange-Id: I75a37fb9b49a421e4524da6b56ef8362ceb6107b\n(cherry picked from commit 827fbd01c306ce292e46b0d881bc3cc804202285)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ad0f2853cf6a72b8d1c5d2e914301e52532d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19ad0f2853cf6a72b8d1c5d2e914301e52532d3e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
