)]}'
{"id":"openstack%2Fneutron~875683","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I75a37fb9b49a421e4524da6b56ef8362ceb6107b","project":"openstack/neutron","branch":"stable/wallaby","topic":"ovn-agent-heartbeat-timestamp-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-01 19:19:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I75a37fb9b49a421e4524da6b56ef8362ceb6107b","subject":"Normalise format of OVN agent heartbeat timestamp","status":"MERGED","created":"2023-03-01 10:21:25.000000000","updated":"2023-03-01 19:21:06.000000000","submitted":"2023-03-01 19:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875683-ovn-agent-heartbeat-timestamp-stable/wallaby","cherry_pick_of_change":874926,"cherry_pick_of_patch_set":4,"meta_rev_id":"ea13c6df727e66a0836421825e6c071f1339f6ba","_number":875683,"virtual_id_number":875683,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 19:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-01 14:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-01 13:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-01 14:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 11:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-01 13:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 13:16:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-01 14:32:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7fa920738e643c08975d586dc51e2d300ca1c611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-01 10:21:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3810ead86c8c35360dd420cc1846e2e0b33dbeb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:10:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170f211d15674e628dc8ebfcc7c646d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/876ce820b2634dbaa9c4c9047db8d20a : SUCCESS in 47m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85eb0518c2cf4b3585e59a63fcb2581a : SUCCESS in 47m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaceb427e492da6aae5e0b89957721e2bd42e772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:08:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6431b7067b4defb4982a9935be2b58\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ae013f95917b4d6b91dab8c72b0f8cf1 : SUCCESS in 51m 34s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/ef4195247b1449b49c09d79509d6338c : SUCCESS in 2h 09m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/679e5abda77f4405b4668ac8a7461eb5 : SUCCESS in 38m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc2b0348c2e4ae4b77233cc35f12d90 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1532e33723184606bd7960dadab08ed9 : SUCCESS in 33m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd8c66575f4a419f9356c0835cd650be : SUCCESS in 28m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/857ba17d8d6d46b18e91a2c8b470c6cd : SUCCESS in 29m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6caa298208184ccd9ca21234bc174cf2 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eda07031ad743abbfcea7b4d8ad1047 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf5431a7f57a47e08e046cc2cf1cd803 : SUCCESS in 46m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cb1ccc6456df47b986c27ea09bf0e1b3 : SUCCESS in 1h 01m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3216c1fccc084dab8019f4eeac26cb4c : SUCCESS in 33m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c52aff2185f54002a5b42a47e3164f8f : TIMED_OUT in 2h 13m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/250249de546b42fa907928ec90f11316 : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4cf95ea9151107dd781777883fcb08f0a24ef302","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-01 13:16:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1f73339d1d0ffb10eadf03b8d77abaac887bb71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:32:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a295f46cfc0b7f6082f200639e1f67c33d87b4e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:33:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5224de8df3f7b144af7ea0d3f5a5f21d3958b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:19:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd8c8f35cec4da689640f52d8dd3cce\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/e64b90d2860a40c5abd216fa0dd3c68a : SUCCESS in 58m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323822ed2d9f4ae09bf928c7ba0d34c1 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81c882a0e61c424a9ea3d9c5af31fc32 : SUCCESS in 29m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72e584b956e3425485e41662635cdb3f : SUCCESS in 28m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5037d2a07a6b4e788c67abbafd761cae : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4138fe8709747828d5ec931b6d42df0 : SUCCESS in 8m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58776f68dbe2457ba38fac47c6f4af2f : SUCCESS in 45m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5211a7a7ad5f4824a6a0766a82d8a0c4 : SUCCESS in 1h 03m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/496d7e4ec0bd431593671767dfcd3b48 : SUCCESS in 27m 57s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b39f15bcf0794121a2a4d7bf06193842 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b77574636650b9c0252b05cb004366991d23162f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea13c6df727e66a0836421825e6c071f1339f6ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:21:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecfea5dbe71043d88565a9400b0380c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d40ff7c7454c028893476684bea8da : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfe544b90002431d8b5b4b2b6b3c2400 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f285dc23515811433fe3f3c65214f706541edbc7","revisions":{"f285dc23515811433fe3f3c65214f706541edbc7":{"kind":"REWORK","_number":1,"created":"2023-03-01 10:21:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/875683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/875683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/875683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/875683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/875683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/875683/1"}}},"commit":{"parents":[{"commit":"b7eec38ffd78a60cd985ec7d2ccd683a0e0db2e7","subject":"Merge \"Enable qos extension_driver in ovn_db_sync\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7eec38ffd78a60cd985ec7d2ccd683a0e0db2e7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-02-24 04:34:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-01 10:21:25.000000000","tz":0},"subject":"Normalise format of OVN agent heartbeat timestamp","message":"Normalise format of OVN agent heartbeat timestamp\n\nA recent change [1] to show the real heartbeat timestamp from OVN agents\nhad a side effect of changing the timestamp format, which now includes a\ntimezone:\n\n    +-------------------+----------------------------------+\n    | Field             | Value                            |\n    +-------------------+----------------------------------+\n    | last_heartbeat_at | 2023-02-23 14:12:07.471000+00:00 |\n    +-------------------+----------------------------------+\n\nThis unexpected format change causes some clients to fail to parse the\nresponse to GET /v2.0/agents.\n\nNormalise the format of the timestamp to remove timezone information.\nAlso remove the microsecond part, which was not done for OVN, but is\nabsent from other network agents.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/844179\n\nCloses-Bug: #2008257\nChange-Id: I75a37fb9b49a421e4524da6b56ef8362ceb6107b\n(cherry picked from commit 827fbd01c306ce292e46b0d881bc3cc804202285)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f285dc23515811433fe3f3c65214f706541edbc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f285dc23515811433fe3f3c65214f706541edbc7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
