)]}'
{"id":"openstack%2Fneutron~875689","triplet_id":"openstack%2Fneutron~stable%2Fzed~I7856aa20947e75b394e2cf4c66f48764dbbd425a","project":"openstack/neutron","branch":"stable/zed","topic":"bug/1958225-stable/zed","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-09 09:00:58.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-09 18:39:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7856aa20947e75b394e2cf4c66f48764dbbd425a","subject":"[OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API","status":"MERGED","created":"2023-03-01 15:43:12.000000000","updated":"2023-03-09 18:40:33.000000000","submitted":"2023-03-09 18:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"875689-bug/1958225-stable/zed","cherry_pick_of_change":875937,"cherry_pick_of_patch_set":1,"meta_rev_id":"d62db22afbf9b54794edb5aeff563e5621e6772a","_number":875689,"virtual_id_number":875689,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 18:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-09 13:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-03-09 13:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-09 13:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 16:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-01 19:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 08:07:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-02 15:02:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a357436790aad214855f5091f8fb594c1c617899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 15:43:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"de63cfae3b5fab4729237d315a6360ef4602c1b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:08:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7b80441817c480dbb21e56b370ddbc9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91533573d1b444b3a9f445fc8055a652 : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c6ba5eaf3f749af917d20f966845f37 : SUCCESS in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"564bc0bb073dfbbf92b6257da37d245718a14c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:01:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8876b815fede49569590f5bdde540239\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/dec228e69c3a47a1835f636c281991d0 : SUCCESS in 1h 20m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84a2518dfe624ceaaf35ef28a47bfbdc : SUCCESS in 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5b094d5e194c479a064c87f4be98ad : SUCCESS in 16m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9e8805170424e9e80d594fceb2f5569 : SUCCESS in 24m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/422ea3a43fee4c65854453009bb157d1 : SUCCESS in 28m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a664c0254e0d45818d59b49ee4466191 : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8fb4ecc7fe4ffc94c088770c199ae3 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49da0eeed219462d99632968000566e5 : SUCCESS in 55m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76845111177f46b9a816ca129b314194 : SUCCESS in 1h 08m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0ff8d1febbd440ba630a2a05037aeca : SUCCESS in 1h 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ff8fb5ea505a106234a094cc846a0bf07dcfd31","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-02 08:07:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43b1d448a58cc091d06702f3f25e08f6571ad7eb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-02 15:02:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b9e5a65bafffaa68c0e694b93bf7e093985a941","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-03 11:47:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0107974a6040e99f5768ef789b36194c8e7694a7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 09:00:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef836e01464bd7e62e34ce2a8291980204a01f3b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 09:01:14.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"4220dabad1091106b452fc315a2056883da96887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 09:01:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dcaa664d129eb927f28050cc7927e928406bd8e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 11:48:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24632dfc930ec5958a8f98aff765a06b3724dc1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 12:02:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36c3c65007bd4c608d176734842e81c5\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/1d199f7c0c5e46e9aa95dff02833838c : SUCCESS in 1h 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804bf9819975474eb40b848f25073e54 : SUCCESS in 12m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9489b2ee37f3411da29f6cf995fbaeaf : SUCCESS in 22m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b561b73ac84041dabbb4f61ee194d43c : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5788f3cb434440d68ddfd9b70d2edb43 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e05abe8ee3d74f7793a5a6e869be5cdc : SUCCESS in 55m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b2725c023ae4d639e3e4c4ae73f1314 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"89b58fe96d77cd41018bda9b56a7c126907a6da8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 13:46:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"101fb75ca6dcf17de1049017cf7b72bae4ee8c31","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 13:46:37.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"20fb5a19972d3e154fefc6f653853c7b3f762944","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:41:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a4489ad0154abe8f884bc7cca650ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c350ed2413b448b8b27088ce840d534 : SUCCESS in 51m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85265dff911348c08c7dd70f05797bd3 : SUCCESS in 51m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eef4ba91e0d6279b8daad54c77187b49cb1bfe8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 15:26:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b100eb07714447fdbc3204b9152fe384\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/39c3b3398f2c4bcc82b66a1e5bed7f08 : SUCCESS in 1h 23m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9248f2aee5ec4dc983d7a3850c649d71 : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47cf07fb0b542108b021a0ac31d362f : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adcd7b36516f4a2da06bcf1233223d5f : SUCCESS in 24m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4435f30e5267441a8dc5be9c899571c5 : SUCCESS in 26m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0edfa170c6d4eafa1209f37b3eb82a3 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa32b1594b341e7a9de7ce66dab8d19 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7355d7ba33824fbeb3218c5717a6a512 : SUCCESS in 53m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6292bf3816bb4e2ba1f4bbda47d58be7 : SUCCESS in 1h 01m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e1147c703b74a64b86df5ba5a865cec : SUCCESS in 1h 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7601c4fd28978125d539b57ceebc9a9bd84a7f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 15:26:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a4e6d93877587d9089019b60da340290c7e167b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:39:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72cc8d69363f4edb931268d895018fec\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/7dc433136a164d5683a25637677c8d2d : SUCCESS in 1h 40m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ecf101b4b147678c60ceb09583752e : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22118ceac127474bbca70d85184885c1 : SUCCESS in 27m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25e7f8eaae5340628ddf056d1fdc81f1 : SUCCESS in 26m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83c24ac82c94cebad995b15fc90d39a : SUCCESS in 13m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/56148b4cb55145539bb0d779f4025e3e : SUCCESS in 47m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91a2eb8447b946c3bdd872e651646285 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8e2a6442dd28f705b7bfab41d595d8d023301eb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:39:17.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"d62db22afbf9b54794edb5aeff563e5621e6772a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:40:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47767964afc2419ba9f89aff82fda715\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5899c1b2980246da8e91b32d1d29c151 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d78bcd5ba2ce85b763e82d190a34687b7af44da","revisions":{"a63194722e1270854c5398b82ec90561357a8304":{"kind":"REWORK","_number":1,"created":"2023-03-01 15:43:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/875689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/875689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/875689/1"}}},"commit":{"parents":[{"commit":"a13b4a925443a1c8888f75c7015e07bf5b4dde65","subject":"Merge \"Remove policy rule for get_network:router:external\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a13b4a925443a1c8888f75c7015e07bf5b4dde65"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 15:20:27.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-01 15:43:12.000000000","tz":0},"subject":"[OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API","message":"[OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API\n\nThis method description was missing in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/825073\n\nRelated-Bug: #1958225\nChange-Id: I7856aa20947e75b394e2cf4c66f48764dbbd425a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a63194722e1270854c5398b82ec90561357a8304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a63194722e1270854c5398b82ec90561357a8304"}]},"branch":"refs/heads/stable/zed"},"7d78bcd5ba2ce85b763e82d190a34687b7af44da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-09 13:46:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/875689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/875689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/875689/2"}}},"commit":{"parents":[{"commit":"ee28d85db943ebc6c525fad97b225236df054016","subject":"[Secure RBAC] Add shared_security_groups rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee28d85db943ebc6c525fad97b225236df054016"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 15:20:27.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-09 13:46:32.000000000","tz":0},"subject":"[OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API","message":"[OVN] Add ``get_gateway_chassis_az_hints`` method to OVN API\n\nThis method description was missing in [1].\n\n[1]https://review.opendev.org/c/openstack/neutron/+/825073\n\nRelated-Bug: #1958225\nChange-Id: I7856aa20947e75b394e2cf4c66f48764dbbd425a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d78bcd5ba2ce85b763e82d190a34687b7af44da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d78bcd5ba2ce85b763e82d190a34687b7af44da"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
