)]}'
{"id":"openstack%2Fneutron~875739","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ief031801c1a3e3dd64e6cbf65e27f04f2bef9cba","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2008270","attention_set":{},"removed_from_attention_set":{"28722":{"account":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"last_update":"2023-03-01 09:41:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief031801c1a3e3dd64e6cbf65e27f04f2bef9cba","subject":"Prevent router_ha_interface port from being removed via API","status":"MERGED","created":"2023-02-28 14:36:50.000000000","updated":"2023-03-01 09:43:09.000000000","submitted":"2023-03-01 09:41:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875739-bug/2008270","meta_rev_id":"c7ce0295140f9af698de477da23bc8aa6a3657e7","_number":875739,"virtual_id_number":875739,"owner":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 09:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-01 07:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-02-28 20:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-01 07:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 15:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 18:43:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 20:22:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-01 07:27:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"0b06562b1adc778aa7bfecdb6eb59470cb138cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"date":"2023-02-28 14:36:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32141b7d7360655db07b9e9eeffbae9ece36d2a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf33c539cf724f7e9c36fed688be47ac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcaaa2737cbe4cc382161db023a19ca3 : SUCCESS in 44m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d1aae4a72e042e0a9fb05141fb7eb12 : SUCCESS in 47m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24b901788f62ea0e8da18f3f98073bc0ff9467b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 18:43:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91bf7ba799ec4b65ba891c966cc05cc2\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/4c0f4e85dcb049a4afa32f2588a6e659 : SUCCESS in 1h 00m 06s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/814781dd5e9f42728295eded7f5056ac : SUCCESS in 1h 11m 25s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/90388745d55f42b7a80a8738ff6ca160 : SUCCESS in 1h 01m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/756aac5358a6455c9bb257ac528c30d9 : SUCCESS in 49m 09s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/0499901898d44bc0b2c63e1ac903bee3 : SUCCESS in 56m 28s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/fc40ea346610485d8b7a6980354967c7 : SUCCESS in 27m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0e1225c24b145b78cad7a5a528e9f66 : SUCCESS in 44m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29e8da8d16747c792b91733bcf9d266 : SUCCESS in 14m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf6459eb176c44b3ac8bcb3a96d30f7e : SUCCESS in 27m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7703e5589de4c08b275cba05b96f6e7 : SUCCESS in 26m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252b34600c99410590388cf404827301 : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5877d12c016147adb027b4b474ddb9eb : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80c1e6281c28405abaa78f4589f7c8a5 : SUCCESS in 53m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e981ff99f5e44097a11fcaa88f3b6166 : SUCCESS in 1h 30m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7e760bd1942f419cb8adb293cedf9491 : SUCCESS in 1h 12m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36a2dd1826e542f6906f491e584880d7 : SUCCESS in 1h 02m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/385ade89d69f4eeb86480d4714d446ea : SUCCESS in 57m 37s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7aeaf7dc7c8647dcb610d6586f281ef7 : SUCCESS in 1h 27m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d7cb636dd6f4fb2804b94d5ebe1d3da : SUCCESS in 1h 38m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc598fbe47dc4bd49c3850a2a7974a1e : SUCCESS in 1h 00m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1abe3d65fb34c4cb578c79066a76a17 : SUCCESS in 1h 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f22d1919ff5b88ba6d94b86429844fb138ca359","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-02-28 20:22:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77584f79f26c7271eeca897d44b6477e4007b888","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 07:27:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f042966b2763a894c9f697b0bef9977155ef8cb7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 07:27:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0085fcb04dbb22213e1458be0bb18dd9643dd01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 07:28:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1236747189173a66124248649f74da0ac8947b7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 09:41:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1478f7bdb1544e5c82789549ef37495a\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/dd8ea14b57b74234b4d691809b8b4217 : SUCCESS in 50m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bfac4d40aa04b2da53d1d44b53e5104 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4098183e67a241c2b4eb4065be357292 : SUCCESS in 27m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c306dc24fd344b408fbfe0bc15170b2e : SUCCESS in 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4edcf75bc7424ba624196054776f31 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/29a60dfdcf61458e9f644e29bdc113cd : SUCCESS in 57m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a09e32345edf4e6c83546d5a89e20ff2 : SUCCESS in 1h 53m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0084a1e9dba94d64800910112fe454be : SUCCESS in 2h 06m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5ab82d9413044c8b31bcda075094dc7 : SUCCESS in 53m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d68668a3e7f4287924a0b0a79bc5392 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"aaec77254e551f1125c7d5d4ebd6e2e86a17f977","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 09:41:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c7ce0295140f9af698de477da23bc8aa6a3657e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 09:43:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b0b0a6bcfb94ab39c2e34b4b521d376\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6d3983887c4b9aa4c7a58cd6759dee : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e7eff2d35b43e2a885dbf1eb18e8674945d8c96","revisions":{"7e7eff2d35b43e2a885dbf1eb18e8674945d8c96":{"kind":"REWORK","_number":1,"created":"2023-02-28 14:36:50.000000000","uploader":{"_account_id":28722,"name":"Anton Kurbatov","email":"anton.kurbatov@acronis.com","username":"akurbatov"},"ref":"refs/changes/39/875739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/875739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/875739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/875739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/875739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/875739/1"}}},"commit":{"parents":[{"commit":"de1a6db5c36b041c608a31bb2863714406772d4f","subject":"Merge \"Enable qos extension_driver in ovn_db_sync\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de1a6db5c36b041c608a31bb2863714406772d4f"}]}],"author":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2023-02-23 15:07:11.000000000","tz":0},"committer":{"name":"Anton Kurbatov","email":"Anton.Kurbatov@acronis.com","date":"2023-02-28 14:36:05.000000000","tz":0},"subject":"Prevent router_ha_interface port from being removed via API","message":"Prevent router_ha_interface port from being removed via API\n\nIf someone removes the port with device owner router_ha_interface,\nthen we can get unexpected router behavior like doubling\narp response packets. This patch prohibits removing such a port.\n\nCloses-Bug: #2008270\nChange-Id: Ief031801c1a3e3dd64e6cbf65e27f04f2bef9cba\n(cherry picked from commit e68e4162cebfa8e5d8f70d4220119d5ca54666bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e7eff2d35b43e2a885dbf1eb18e8674945d8c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e7eff2d35b43e2a885dbf1eb18e8674945d8c96"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
