)]}'
{"id":"openstack%2Fneutron~875989","triplet_id":"openstack%2Fneutron~master~Iac1472f7f896ea434deacb6d236ab469f4f6ed56","project":"openstack/neutron","branch":"master","topic":"db_sync_sgs","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-17 14:58:15.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-07-17 12:58:49.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-07-17 14:58:15.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-07-17 14:58:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac1472f7f896ea434deacb6d236ab469f4f6ed56","subject":"Fix ACL sync when default sg group is created","status":"MERGED","created":"2023-03-01 22:33:19.000000000","updated":"2023-07-17 14:59:25.000000000","submitted":"2023-07-17 14:58:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"875989-db_sync_sgs","meta_rev_id":"1858dd634a43c81559684b801d1154ccfd99e045","_number":875989,"virtual_id_number":875989,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-05 05:42:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 14:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-07-04 10:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-17 12:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-09 04:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-17 12:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 22:35:00.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-01 22:35:00.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-03-01 23:11:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 23:22:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-02 00:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 09:46:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-04-09 04:23:25.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-05-25 19:57:58.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"d3504b36cee4514130d0a837cfb2257bde90342d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-01 22:33:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6906faffd38c2e07f3155d8b24b38f18cf140cd9","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-01 22:35:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74c8735627a46e5d788d2a265b30ba1c7ce74da6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-01 23:11:16.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/89/875989/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3f9a9fb1ed9ae86b3a811f2096386c0fad432ab2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:22:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a22317e351a457e9d65fb7a917f5ef3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c37c40d56b8461fb645a3b0876b9f99 : SUCCESS in 39m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47584ee083734a05a9ebc2408b92628b : SUCCESS in 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83277e75f9e83ca49f2bf5b87b68961218b96e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 00:14:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d72dd81c66cc4b2dbc0ffb2890f9a3e3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c4026a6c548a4a5f8d53cddb40067443 : FAILURE in 1h 38m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1288ba6c563f4f91925e7a5e477a91a3 : SUCCESS in 41m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dd08a1eb1b4a338a0554ffe18f88bf : SUCCESS in 14m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92ef28c2ff3b40b3aa079b371e7c0f90 : SUCCESS in 29m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcc153ede03d4ee8907c17990642fadd : SUCCESS in 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fc181ff38f4a5099067ccec50990a2 : SUCCESS in 12m 03s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7893b88e32a2406b8792b5511699bf92 : SUCCESS in 28m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e6e33ccfb0445cea2c1efdee77aa05a : FAILURE in 40m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6852d5ec80504885a7c376c9f637ed25 : SUCCESS in 1h 08m 46s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f4019e4f6eff4b0fa517a0353c4f48d3 : SUCCESS in 1h 05m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38480df6645f4caea47ed6d8744a72bd : SUCCESS in 1h 35m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/27ad5d53308d489baf9eae8aa861940f : SUCCESS in 1h 17m 58s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a92f137f5e3a46368f0096c129ff6f2f : FAILURE in 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8ef07732a2ae2acf658dbd73e47dac7c273c9cb2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-02 09:46:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6c8409a6cfba3148d67018d0bfc76bae5dbfe0b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-02 19:19:16.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3c3f72f53d614e0f6307404ae1e06a31098adc51","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-09 21:40:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8462170971b5023033e4d5e57b648925e95e369c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 22:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecfdc7f4127a4f18b3aa7114c5b4ffe2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4e6ae66a20840f586661e26dbdf4f20 : SUCCESS in 49m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6588f73286244b2cac17b0493392bb59 : SUCCESS in 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96ab4b3b8602299524fba3c28b50ed6ed6fd8efe","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-09 23:00:28.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/89/875989/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"24e7c969581444c9e3c0acc2e90a931135ad551d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 23:23:59.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/367c98c5adb94be5a5c0482dabb1bcfe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/158cc75559ab4273ba9e6945b4e74d22 : SUCCESS in 1h 21m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25b08de278384fb18c83779da5dff091 : TIMED_OUT in 1h 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b51f0305ba4b91bb1bec2dc703499a : FAILURE in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d9e4137d6ae419cbd5fbdbf20eca764 : SUCCESS in 34m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ebffc960e004ebab02eadce5f7e9faf : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad9cb19a46a946eb8a4e6c176a9227f4 : SUCCESS in 11m 20s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/409c2266592244c59e9f20a9ee057445 : SUCCESS in 23m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c375aa12dc414e5d9d6a9738f5893f2a : SUCCESS in 59m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c20c7279aa68433883654eec049144c4 : SUCCESS in 1h 21m 13s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e913bd2f3a91404fb732a20268b2d795 : SUCCESS in 1h 13m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae1d165b3bce4e96840807c3b275dbd8 : SUCCESS in 1h 41m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b89d0919a96404e8eed6944c4f43b48 : SUCCESS in 1h 19m 10s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/db8c83b228fa4ef0a9b674d2c4efb8cc : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7123c94e045564af8dd4a09024ae0ed94af7354a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-16 20:37:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bec943ce64f90798a5ccc02f4e5a42a4ff80c9c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 21:04:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d1dfc28db7f4d419d10b08220fd027a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56b9d48d657844f89196d952464e8fc3 : SUCCESS in 19m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e5940c6f3404939a33f15380266d9a1 : FAILURE in 26m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a044f3c42edf4b88bc1d95aaf631daaa : SUCCESS in 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b1a79906a5debe7e3106e4da6245cda181d2df01","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-16 21:34:13.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/89/875989/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"bbffc8697b5cad9f72765368626c893022eb3eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 22:14:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e09529b0a994b8186cdd6e99c9dc099\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/140a25f6544342058d07d208718c0671 : SUCCESS in 1h 33m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8b7e538b2384f7c8f9c76586293163d : SUCCESS in 43m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb5f1d59eeb4e8a9ec68f9628cd20c7 : SUCCESS in 13m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/913b273d687e42419252aa1a741670c2 : SUCCESS in 20m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6c6df75becc410c93895f7fb77f0240 : SUCCESS in 25m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86fb7e3d2ccc4ad8aa38c309d6cbd1f4 : SUCCESS in 19m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67eb97d5a39d4b2d9e137831967ae799 : SUCCESS in 10m 40s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b8e4078a11ec47e9975b6266d5b22be7 : SUCCESS in 18m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35828fc6d39344739e5f4de8da0d5ce0 : SUCCESS in 49m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/31135e51fc454d9fbab4caf6f21e35ed : SUCCESS in 52m 07s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0ffb772188a401f8ce8d7bba458d55e : SUCCESS in 1h 23m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7325291aefac40e782e05e8897395ab5 : SUCCESS in 1h 31m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e93c28c6a21d4ee28f34ff0fce74849f : SUCCESS in 1h 15m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/8b9a7c5349c848699d0f72468bdf2031 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":3},{"id":"0a266c46f3cad7c44dbc298a93d42eb976a1773a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-05 02:12:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"073945d005ef79cb921a74b55133161e745a6a67","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-05 02:13:33.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"a338b3edaf03aa9fd5aa73bf9b93a977010b2e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 03:22:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2887b59f7b4c4b058ae40b389dc7d38f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c6ad64777c740bb9cb3200705aa1a18 : SUCCESS in 27m 38s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1258c1cf0588416bb140bf5cca7c7063 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/040f0f79a6514eb690d7670993bac8c5 : SUCCESS in 57m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4749914f693f6ff7d30346ec67dbee9fa438dacf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 04:34:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffe6abdd6c047a2ae698425f862eccb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a1d86262c46491eb486f579eab4d16a : SUCCESS in 1h 27m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a71e0e2acab43a1be297bb287993a8b : SUCCESS in 36m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9371f697574f46926aa4dfa654482b : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3cb289b8946438994911fd654778833 : SUCCESS in 25m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a737096fa7bb4f93b2a301d3a0c4d800 : SUCCESS in 26m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be23d34f07ae446dbe86db201e490d5d : SUCCESS in 25m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1a07df73234c7eb2dbba5b35422ee7 : SUCCESS in 11m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/dc5b09ac14774e5093fd9e34b3c78c5e : SUCCESS in 23m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85d66bb3ca37414db2be6bbfe6556d7c : SUCCESS in 1h 01m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7493ed937d074710964926b773a263c3 : FAILURE in 31m 22s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/137a1b6a5aab4c1a91aa43c2a426110b : SUCCESS in 1h 23m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af2df722fdb24f9da1f05235e1cf013f : SUCCESS in 2h 01m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5465bf86ba6b4a6ba4ed3b0626381490 : SUCCESS in 1h 26m 29s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/6e26884c1af840aa9c2e4ccd62d9e448 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":4},{"id":"7ba165281dc8b3b68d5b465c23a5655c0cdb490b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-05 05:42:21.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/89/875989/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"604688714bfc5bb8e9aa8b38c6972a222abde093","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-06 18:34:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52471b6361f9f8baa01a1fd9c60f98502e196f9d","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-04-09 04:23:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e6b36edbf72acaa7525f535c8becfaf40b890f92","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-05-18 16:15:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0fd78a82739ae31e17a40a5d23070577356b1080","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-06-16 18:02:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ca34d90bd0195a8a7b6e9b5cceaadc343a51e86","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-04 10:55:20.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"418240a511c6e535cf8ab6081bc5a062b1bf5e45","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-17 12:58:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d907e614192593313abb1d64c1e38bf3806c23a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:59:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2e453b750001ca038b6e52a410d899fcf41a584e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:58:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79cc54d8a73f4cf18584fa59db44d223\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4b2a7886071c46a8a115630605e78e59 : SUCCESS in 1h 57m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf436640c3f4c518a4a920b7e4b0271 : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a73811bb11a44120a69919ff958bd807 : SUCCESS in 29m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb765b55ac249258cab243a069f3a90 : SUCCESS in 28m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b74ba05f03344128153cc49cbe36299 : SUCCESS in 30m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d37f0dbeb654bcd900af95e874c4c72 : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf5bdf3b8742430b926e0079985b3a0b : SUCCESS in 1h 08m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dcf513a91eab47f983d729bccbe02dd4 : SUCCESS in 49m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4fb9b0d79920499a80731bb6ef917405 : SUCCESS in 1h 26m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ea81bfce6351358541e65753f3644c2d2c49a988","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:58:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1858dd634a43c81559684b801d1154ccfd99e045","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:59:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc89824104047559eabfe7f26913e4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05dd9388ce6549fd8c9a37bb2df9c520 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"33cf2cdc83a8cee9ee075eb371f779c3d356cf48","revisions":{"2b827cab6fd98e2c49ac58a2ba75138ce9b32e95":{"kind":"REWORK","_number":1,"created":"2023-03-01 22:33:19.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/89/875989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/875989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/875989/1"}}},"commit":{"parents":[{"commit":"fb08c34561bebf537a728c8043f643b0fffad771","subject":"Merge \"Check port.tag is not DEAD_VLAN_TAG in ``DHCPAgentOVSTestFramework``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb08c34561bebf537a728c8043f643b0fffad771"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"subject":"Sync port groups and ACL closely together","message":"Sync port groups and ACL closely together\n\nSome sync processes like sync network ports can implicitly create\nnew default security group if the default security group for that\nproject/tenant does not exists. However, if security groups were\nalready synced that will cause NB DB write exeptions when ACLs\nare synced, see linked LP bug for details.\nThis change ensures that port groups and ACLs are synced closely\ntogether.\n\nCloses-Bug: #2008943\nChange-Id: Iac1472f7f896ea434deacb6d236ab469f4f6ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b827cab6fd98e2c49ac58a2ba75138ce9b32e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b827cab6fd98e2c49ac58a2ba75138ce9b32e95"}]},"branch":"refs/heads/master"},"91a4fd66cf3c6f21c6dd958046d0080409fb3ea9":{"kind":"REWORK","_number":2,"created":"2023-03-09 21:40:15.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/89/875989/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/875989/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/875989/2"}}},"commit":{"parents":[{"commit":"ef4217c61662e7d613905d1ceadcea56ba779c16","subject":"Merge \"functional: set dns_domain config option after its registration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef4217c61662e7d613905d1ceadcea56ba779c16"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-09 21:39:27.000000000","tz":-360},"subject":"[DNM] WIP FIX repair of an empty network","message":"[DNM] WIP FIX repair of an empty network\n\nCloses-Bug: #2008943\nChange-Id: Iac1472f7f896ea434deacb6d236ab469f4f6ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91a4fd66cf3c6f21c6dd958046d0080409fb3ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91a4fd66cf3c6f21c6dd958046d0080409fb3ea9"}]},"branch":"refs/heads/master"},"b416f983a32bbc690a6a705176c261ba056ed191":{"kind":"REWORK","_number":3,"created":"2023-03-16 20:37:02.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/89/875989/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/875989/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/875989/3"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-16 20:18:26.000000000","tz":-300},"subject":"WIP Because I am trying to figure out best way to write test for this. Fix ACL sync when default sg group is created","message":"WIP Because I am trying to figure out best way to write\ntest for this.\nFix ACL sync when default sg group is created\n\nPort group not being available in NB DB during ACL sync\nis bit of a corner case but possible during the ML2/OVS\nto ML2/OVN migration sync. It can also happen in ML2/OVN\nonly enviroment. See my detailed description of both\nscenarios in the linked Bug.\nThe easiest fix is to just retry ALL port groups sync\none more time if ACL sync cant find a port group row. This\nadditional resync is really quick.\n\nCloses-Bug: #2008943\nChange-Id: Iac1472f7f896ea434deacb6d236ab469f4f6ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b416f983a32bbc690a6a705176c261ba056ed191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b416f983a32bbc690a6a705176c261ba056ed191"}]},"branch":"refs/heads/master"},"33cf2cdc83a8cee9ee075eb371f779c3d356cf48":{"kind":"REWORK","_number":4,"created":"2023-04-05 02:12:44.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/89/875989/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/875989/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/875989/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/875989/4"}}},"commit":{"parents":[{"commit":"b1ca854791fe391ccbb7ac3a019dc9b9d46a3995","subject":"Merge \"Mark \"test_multiple_agents_for_network\" as unstable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1ca854791fe391ccbb7ac3a019dc9b9d46a3995"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-04-05 02:06:37.000000000","tz":-300},"subject":"Fix ACL sync when default sg group is created","message":"Fix ACL sync when default sg group is created\n\nPort group not being available in NB DB during ACL sync\nis bit of a corner case but possible during the ML2/OVS\nto ML2/OVN migration sync. It can also happen in ML2/OVN\nonly enviroment. See my detailed description of both\nscenarios in the linked Bug.\nThe easiest fix is to just retry ALL port groups sync\none more time if ACL sync cant find a port group row. This\nadditional resync is really quick.\n\nCloses-Bug: #2008943\nChange-Id: Iac1472f7f896ea434deacb6d236ab469f4f6ed56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33cf2cdc83a8cee9ee075eb371f779c3d356cf48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33cf2cdc83a8cee9ee075eb371f779c3d356cf48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
