)]}'
{"id":"openstack%2Fneutron~876049","triplet_id":"openstack%2Fneutron~master~Ia3da9bf6d6113edacb0e21de9a51bd60d5e2465b","project":"openstack/neutron","branch":"master","topic":"2023.2_bobcat_db_branch","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-11 20:24:38.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-03-08 07:43:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Ia3da9bf6d6113edacb0e21de9a51bd60d5e2465b","subject":"Change the release tag to use the release identification","status":"MERGED","created":"2023-03-02 09:54:33.000000000","updated":"2023-03-11 20:25:47.000000000","submitted":"2023-03-11 20:24:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876049-2023.2_bobcat_db_branch","meta_rev_id":"ef928d3806a69f7f88b4756eaa117fda41b4215c","_number":876049,"virtual_id_number":876049,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-03-02 15:23:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-11 20:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-07 08:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-08 07:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-03-02 18:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-07 08:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 10:33:20.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 10:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-02 12:39:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-03-02 16:57:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-03-02 18:33:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-02 20:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 08:32:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"83323713afaa6600a10a0bcd12bc5798e053b4c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 09:54:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8b395f441c50f9b8ca44f7fdbe050fa55ac44d1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-02 10:33:20.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/876049/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"63fd42efdd7b2a25213772fa7d1fec51c774b853","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:50:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a53ec35c7df4a8eb715c34c598f0985\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9da93658eab940e4a1bd8ffc0e69fadf : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fde784969e38462097701fbbd775299d : SUCCESS in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"252f2ca8dbafb6fac713d0808079d0f267734047","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-02 12:39:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56800ed5a1ccf85b67c94e58bd3ca1757a55951b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-02 12:45:42.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"15dc21db5a1355572ebccd62ad81a4377dbe696d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 14:02:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d002a82466202ef8424021062dde75def2226b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 14:04:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45371fd1333141fc407cd1d313c20335bcd54c4a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 14:26:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c82a1fe4df0af4d4ff1be5f05823639521989d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 14:44:10.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f1b00429bc4bc248d65d73b1ed948b085960d887","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-02 14:44:13.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/876049/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b0b3e237f4720920fed800583e52996d9443ef57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:07:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d3dd4f9ab042dabf3ce7723ae4c6c4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6bddd23dc994f22a5f6a7e5996287c3 : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b57ea6b3fa9a4a98b4ed566ff2eac91b : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb70598b579ade04ceb50c1562d59005b546ce28","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-02 15:23:15.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/876049/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"80eb83e4a52cec2b7d07a40215807c2efc13e19d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-02 18:33:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f5f31a70134bbbe2d1419ae3155d57c515c5e35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 20:04:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f1d5cedd2a409a86fac2d91b2ab5bd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8a93ee1ad52a4f16b625a691c3756844 : SUCCESS in 2h 12m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6def6d0f83864a15b49163c78799a719 : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/caa8985065e449169b7aa917f144dac3 : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cfbd309c8e264662bfe163d2044fdab3 : SUCCESS in 1h 42m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20424372924d4498b3b11d6fb153b0ee : SUCCESS in 1h 40m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b0201fbcd2e41b7a4e9817c17c041bb : SUCCESS in 31m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99deb185697c4f6db1fca957ee1ade11 : SUCCESS in 37m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d0f2941de6438e83c91d89df0a0e60 : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/456732eecc15450c9c873f0754ecccca : SUCCESS in 23m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49ba8493bfef4d268eb953b9fd618962 : SUCCESS in 22m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093f72c182004ba8965819805515efb1 : SUCCESS in 11m 00s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/455e8f4ef33a483eb552de7a0204862c : SUCCESS in 20m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/daea536dd2724f1c906bd32a47b2676d : SUCCESS in 47m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee403ee8bf594fc7b233a84f3df8ada2 : SUCCESS in 1h 44m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ca8adc2b69142aab5ba0ace24d15663 : SUCCESS in 1h 02m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6dfceef4f9914f7f851e741d6dbeb110 : SUCCESS in 1h 06m 44s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a8666302d2f246f682787627de74c59c : SUCCESS in 1h 12m 52s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52c5d51ebe304447a9ec13a219c6347f : SUCCESS in 1h 00m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2ae8d8fa01de4b86abf98643cc806836 : SUCCESS in 1h 37m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3761404b231a4b7f8c37fabf4a59fbf6 : SUCCESS in 1h 13m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/94cc84a3f7df4abd9122ea3cb879987d : SUCCESS in 1h 00m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d3f6538ff294799a4c420834a1fb0e5 : SUCCESS in 1h 37m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/68544780ef7b48a09a2d9873b93d302f : SUCCESS in 1h 46m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/58911baedbaf414d890170a666946410 : SUCCESS in 1h 10m 05s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/393100685bdb4d9b98f1fb00af11b2d3 : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":3},{"id":"dea13eb45ac27627f83c163eedde42d264a91749","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-07 08:32:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51bcbff1e1ce3616544445c79d6e737ee0b8ba07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 08:33:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ffc835e550b1fbc43c8295e4b483f6f17533a6cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 12:49:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d25cb2479b481cac895cb14c7b02a3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fa7b36a42a7c4f85b7807b46802d9f11 : SUCCESS in 1h 10m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/52fd88c577444d07bda7cdb9f3507c9b : SUCCESS in 1h 22m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/957afd472a2b4c9ebfb75f1b677a5a15 : SUCCESS in 1h 16m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ecfc6c788924755962d8064eb52b9f4 : SUCCESS in 1h 08m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3f5fe4bd8574e7daf168d6a1b67594f : SUCCESS in 1h 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60e2739099941bc8c38a28d932f4853 : SUCCESS in 11m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41f5f01c005945ea842709025e696fa5 : SUCCESS in 24m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/add6c39c136a429da40bdc3914091022 : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ff7fd0de914ceba86c83a747ea48d4 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/203467c68f304ebfad47b1aa462416c8 : SUCCESS in 48m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a20464fe402477e9f1085c37e40d656 : SUCCESS in 1h 46m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5dbfb1c662f2481eadd7b95934b26126 : SUCCESS in 1h 14m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20ecc233feb04b379771eaff29f3c45d : SUCCESS in 1h 02m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/711e0447509c46b09a768859380b5a1e : SUCCESS in 59m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/16a7d0504f384a1590daedba9df4a3bb : SUCCESS in 1h 05m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d636677f13284942bbd7dccf1101cca9 : TIMED_OUT in 2h 03m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/167eb5d8d3e14584bd57a3b78efd9f5f : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":3},{"id":"0e954d8932ae1c899a085566fbbeb8adbc3d5107","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-07 19:43:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7d54c95f4e85b27ca7ecb9b5fd66fd9040b7581","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 20:02:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca6ec225c27459eb5cfb83b029447db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6954209721944054b9f87ab7c6c33720 : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e28a5a7ac67c45e6ad14bd659fe10ff1 : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"246cd4ee26af537f2b08273bf935ed23a37ebedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 21:46:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5882b670e4ff4901a2fdf4b9d479b2c3\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cbbe59fbd2d845ffbc4481d7ca70cb69 : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/849fcfa428cd4efc86aa8a6dcc61d0fa : SUCCESS in 1h 48m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4d6f151b6d8147a1b4848f617c595b30 : SUCCESS in 1h 36m 40s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b55940a199ae4984b8dee97aa2a82dac : SUCCESS in 1h 28m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0945413673c4d3980181bb0a80cf5f1 : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/baab5e0c57dc411eab6de03cb26c461f : SUCCESS in 38m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d74d6d5d88674c109a3e5fb1fb375532 : SUCCESS in 36m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0abfe47e1d3841649f7e15831144e074 : SUCCESS in 13m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f68c01b9bde4f77a1389d60811f6279 : SUCCESS in 25m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b74aed97cbbf4ea2a05d722229a523fc : SUCCESS in 33m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92744aeb0dc0461aa03e7f2407ac49e7 : SUCCESS in 10m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cc2d6433bed040a887e7bdf2047a7027 : SUCCESS in 26m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/854f7d3b1b854fd58b328c59381ea7fb : SUCCESS in 41m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6a757c14ebd4a4c896754b9cd27ed5e : SUCCESS in 1h 30m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/67746f989b98497cafd736ad0e376c12 : FAILURE in 1h 03m 44s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af60c9e3d45f4eedb15447a2609cd01b : SUCCESS in 1h 03m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/00ab22710be142d0bc3d27dad78950e0 : SUCCESS in 1h 13m 32s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4c1bd03399ea4c5d920beeccace119ec : SUCCESS in 1h 01m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b31a0bfa2ecc4f7aae9657c16dcdd071 : SUCCESS in 1h 21m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de09a102e3f945148a98998cdd0c58d4 : SUCCESS in 1h 09m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/78ae8cf680da47fc94aa0b91a244f479 : SUCCESS in 58m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/19e2259b1eb148bcbc3e6e6348050d2c : SUCCESS in 1h 29m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe85625cf3e74b1886560d2bc1c3f9ea : SUCCESS in 1h 53m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee836d82dbf34f6bb88962d10d636776 : SUCCESS in 1h 08m 03s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f4e3ef0139e04619ae26a9ca8b270a45 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":3},{"id":"a46d405ed81cc523fa82fbcb62a0a8b9c7de0d78","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-08 07:43:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"491e7f7871737d1801b56041f19cdcf8a12169ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 08:02:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37a18a9bf5824eec99ec6f8c12001bb7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84c56e683fd34edb93af282fec0e0f3c : SUCCESS in 17m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b149b74a324044bb990060e30c8be3b7 : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a7b3dfbed462f8b56651c207009270c6131ee29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 10:03:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f81182395a2b4914ad83356108d3f3f9\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/084396737dfd4f43992aef174a23eb40 : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b91162ea1ca940a38ab50490e68fcbb7 : SUCCESS in 1h 28m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6144c771489f42aa81d05ad0ef692434 : SUCCESS in 1h 12m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21df186ef49c43848e2d7679a9ef91e3 : SUCCESS in 1h 13m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b3cc7048d0e540808183688ff7adebc6 : SUCCESS in 1h 15m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7884a3ec4a847b0a6df978886a700bc : SUCCESS in 34m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04de9f901603430b8e4e2cf28b42088e : SUCCESS in 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043cd23833b24e63b02e0a328981ef84 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9a87bb3640e497cbbe654f6464d7e86 : SUCCESS in 22m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6bd42c85a544a4a911f53fa0c4f2190 : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9427ef4c2b0421eb99a83ea1be4600e : SUCCESS in 10m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bfb5390c2264489bb3f319ac5911d868 : SUCCESS in 23m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da14c5ccfcf84dc58291e7d46ed7b02d : SUCCESS in 1h 02m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/653788281bc0412390f2a1936128d67d : SUCCESS in 1h 47m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4f13a49725c4c3883cc0e1d320282ed : SUCCESS in 54m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0a8b29ca254248e79e1ac83db98feb89 : SUCCESS in 49m 38s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0e3013c5a60b44e68bd3961a19e88c47 : SUCCESS in 1h 13m 11s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f64245aca9343de94c49541cd415f04 : SUCCESS in 58m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fa539a26412a411483d708541017da08 : SUCCESS in 1h 28m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/015bf94b8b994c37be90fb1f40b54776 : SUCCESS in 1h 20m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c3eb44f847f7410b954c45997f2a7213 : SUCCESS in 1h 12m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/77c77a6f9c3941b9aa871359b8811a20 : SUCCESS in 1h 35m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/daba3964462642e4ad070a7e1cb4faf0 : SUCCESS in 1h 41m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/af1d4e3140aa465fbda4f81d266efbb8 : SUCCESS in 1h 21m 46s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a0bdcd28bf0042cb9bfb247dabbd6838 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":3},{"id":"367c76a133ffbc1ebbc690374b1ef1c72f9d39ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 10:05:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e8d6b49b4263f76e226b36f3f6beb7ce3b0518d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 13:08:09.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47e1f378f8e8491a991c34a17beb6955\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e9780e71b5424904b2dabb4c7e6ac306 : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fde0e828fe9a4730a22bdc40b8138b3c : SUCCESS in 1h 39m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5654a5b7f1e4d209c67fbe9c9b893ae : SUCCESS in 1h 28m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6c67c182f0f41998ec1f9bc3a73b3a6 : FAILURE in 1h 41m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e6542834ff1f480c8b6d7add0012c489 : SUCCESS in 1h 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3e46546080487e82a63dc871fe9fb2 : SUCCESS in 12m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c68fa05732a047bb80aafc5238d38106 : SUCCESS in 21m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/117e8a321e4b49dda73c7e16d80af2f2 : SUCCESS in 27m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0851a5ce305a4e94a9ba55de7d8d612a : SUCCESS in 22m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77fa318cafb64416a3ef1124f9fd4ec1 : FAILURE in 52m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/964a927c56144f488679c3416acdc209 : TIMED_OUT in 2h 32m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/15ac52866a34423695a4251baabe589a : SUCCESS in 1h 35m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a10bdff7a570405a97444ab37b4e36da : SUCCESS in 1h 04m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b805e7ec56094a358c0ce29550e02bc8 : SUCCESS in 58m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c3ae79b028f64f648196c911545d7f5d : SUCCESS in 1h 25m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/096e50129ac1437db16b3c175b397fcb : SUCCESS in 1h 20m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/32119203fadc478f998acc3ff538fedc : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":3},{"id":"a789bb0cab65be956bc96531d70b649fcf4e9aae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-08 16:17:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"74c686a629a6953049c7c1265f9b1c3bb469124d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 16:43:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0708ed4d68442b8a95d397a27b7120d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95802ab5562d433f8aa6d44d96fc2812 : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb05f6126fb44006ab5333464fe593e0 : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8aca3df5d50f99dc307b3c62b84a6f657b671643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 18:50:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8c4cfef2fa94f96aea864a6ad8b5eb4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/96ee84e96b9447c9a82092168d20ae1c : SUCCESS in 1h 33m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9118f9c358ca464593aa3fd67e605e97 : SUCCESS in 1h 42m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd052ef551f642cd9d3481296f223ec4 : SUCCESS in 1h 40m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/98388d879d814a52bb79f030df35ee98 : SUCCESS in 1h 55m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d357313e2ff44c70a65c6b4ce1f49c20 : SUCCESS in 1h 38m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b723c8774b9b4b189738e71fb5bc149e : SUCCESS in 35m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9dd0544ea7c4ef7a4c3469313d3fd61 : SUCCESS in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24607434e77a44c6a2690a74d18175b2 : SUCCESS in 14m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b23a35b46e8f463ea9a1686f9a27e513 : SUCCESS in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcf0072a6f604693881ea962adcba141 : SUCCESS in 25m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a52b207f2a4ee2884fc3120cb0a744 : SUCCESS in 9m 57s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7c22b91dee054cc784817ccd0f6f0a2f : SUCCESS in 23m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d769ec9dd094598bd239fa9bef22c62 : SUCCESS in 59m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1158a5159a04973a8e0bceededbe574 : FAILURE in 1h 20m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d69f2e94571241c7bdc68cd152743167 : SUCCESS in 58m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e282ecc60ee7465182c10d2078307c04 : SUCCESS in 1h 05m 15s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a769dbab9806438bbb5ae3262b073e6d : SUCCESS in 1h 13m 05s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/431737c5caf942988f472bbd0299c19b : SUCCESS in 48m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c207066354e04dcfb72f566b401acbee : SUCCESS in 1h 30m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/59c813e6e42d4b1893ea6d9127064b6e : SUCCESS in 1h 32m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ac961b6b408455c97377c7a105ba6a0 : SUCCESS in 59m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c313f0a1cd5f44d186a796ca5957d032 : SUCCESS in 1h 27m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09f2206b1ef34a79bb803ee41777e658 : TIMED_OUT in 2h 16m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1bec2d423af845d2931c5b8145af9ba8 : SUCCESS in 1h 16m 22s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4206637bf46e4924834439f0ea1fc5fd : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":3},{"id":"ac2bc84ba4251fc2405d3c20c14e8eb3730adcd1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 07:59:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3131cec16cee49898a6e5b5f40b2c1ba8add45ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 08:57:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd42dde754ed4d61acb2e10236695578\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a924a14f70ae46d6a4b3960c6692f1cb : SUCCESS in 42m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c9f2bfe95db4a89b03d23b3752ee86a : SUCCESS in 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"147426d325acff3db153c3df34e5603c1b9f82fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:10:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/308f68292e4d4b909b08e6913f38e54c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fff5d5f48b5a4a399406d1c5c17d10f7 : SUCCESS in 1h 38m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15d11ae3b99e4a6694150eb5a897552b : SUCCESS in 1h 32m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/441aa9d198f5438eb5f92e027f043240 : SUCCESS in 1h 25m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/561440a170ea4f09a21dc4aba74f118f : SUCCESS in 1h 21m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4132170704ec48a1be9e046ff9778386 : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da74ffc43e4d441cb2825017e86fc93f : SUCCESS in 30m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bfa72092448442caace4ab1cff43d91 : SUCCESS in 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5aea091eca43f3a443145007729fa6 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/219fe4cc815d4637b35e4bcc8a6c3f2e : SUCCESS in 56m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6551a50f4042432e9087393799798436 : SUCCESS in 55m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e387f7f79cd4ed484c7141cd7e51df4 : SUCCESS in 10m 41s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/00b6830f13c7457b931fb7e883138b5c : SUCCESS in 51m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82c805aa90a041babee72fdffcf81d72 : SUCCESS in 1h 02m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3828461653064fa1b0e2fce7519b9d75 : FAILURE in 2h 05m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/40cadc88eb4746fd906f894459a60790 : SUCCESS in 1h 07m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/53fd2af2f3254562a0ea770a2dbd760e : SUCCESS in 57m 47s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1cdf689cb7284c3c9ab9685169f2a899 : SUCCESS in 1h 25m 31s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e76ac9f40c2e4a2880a40615b5e16fbf : SUCCESS in 59m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3eee15636b13433daa43c39c58ec0496 : SUCCESS in 1h 46m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86e3bbefe3414ef09bc637d3d9fc49ca : SUCCESS in 1h 36m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/23115940c45348de9f510aa2de5cf20b : SUCCESS in 1h 06m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/02ec40c7b67642478fab624d3d5538fc : SUCCESS in 1h 38m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/69ff9e678c2849c3af231db8f5301f41 : SUCCESS in 2h 05m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8aade636a2840fa9ee5f4d2f2d6b052 : SUCCESS in 1h 31m 27s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bbd945b27ba4487aa64b481b82881285 : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":3},{"id":"5e948515063b243ff119464c70e146e35061a909","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 10:17:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f10de826661061892b4cbc3176409c959380051","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 11:23:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f95c486ea074d66959a9a205ef22f0e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b3903960c524bcd834cdb724552e721 : SUCCESS in 59m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be30407ca4f2485bba236d9a2557532a : SUCCESS in 46m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7c4ba3f7f3f3c5f30845f472dc3141ed2931fe70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 12:19:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/820aa13112754f6596664a282015dceb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9d0244d1140e402b989c6757b0df57bd : SUCCESS in 1h 44m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8ce7b31f5f6c4c89a0f5477ef9f054fb : SUCCESS in 1h 40m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35e1f65944774d03b675276c933e4fe6 : SUCCESS in 1h 35m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/de86b6229b6543418ca08ee14a5fd159 : SUCCESS in 1h 41m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca9666c16a3547ca88396c68aed322e9 : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/644775cb735b44b78a7f2cfb66465201 : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce5f53fed5c4fde9b84fcddc6d054da : SUCCESS in 47m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac14009c9844cadb5f136ba26b0a23f : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96d10979d2614e7fabcb261d737d0e40 : SUCCESS in 23m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6e1754de0764f038403b75339172519 : SUCCESS in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43378b9c1ae4471793b02b371390e6a0 : SUCCESS in 9m 19s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9059d6372ad44c679d164703a475a8a7 : SUCCESS in 18m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b79f2b4d37f482a855e76282fad86b3 : SUCCESS in 47m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/27ec2ca17c27484da01941a2ce553e93 : SUCCESS in 2h 00m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f89874e530241c3a0c71b51c8344fff : SUCCESS in 58m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/597e5ee083464e40bf6d023fcd8187da : SUCCESS in 56m 00s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/120a8b22c583477d9544d498734ea8c2 : SUCCESS in 59m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97c1644f6eaa436a9a36d7bfacf1c1da : SUCCESS in 55m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/641de535ecf141a89719be7d8417b402 : FAILURE in 55m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a25b4eca111d4dfca7cff3ae736e019c : SUCCESS in 1h 08m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/30c68a6312d149e9821f737176b3c986 : SUCCESS in 1h 00m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1cc304c1228446518e7d906d64d6c510 : SUCCESS in 1h 19m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c0e4b4aa418b4a6eb51e3b2ea99f5f71 : SUCCESS in 1h 44m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/474c15e444aa4eaa958cd1a8736e4da5 : SUCCESS in 59m 39s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/12f1f8b429da4375a33a1d4120030f16 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":3},{"id":"170e7eae0a033e882ad831b18d24af7189fb4a55","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 13:48:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5dfe47a1d51f35176f38ef46acf9de38d102c3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:40:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e36d698515de46729703efc6318b7b66\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90d9a57beb0c44d09fe3ed5f39ad05b9 : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4bf0cc07f6d94dd4bc16efa6515da46a : SUCCESS in 49m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d78fedf5e7d52c2a9d3cc34e48c5edf326a6dd1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 16:18:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c004f19fbb9347738e11aa4eb71445c4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dae73a9f668343b2a372f7233f14fa99 : SUCCESS in 1h 21m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/64ab9e8148f2434c9e8fdf1aff6dc0c0 : SUCCESS in 1h 32m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7d80c74bcfa4c7383eecb4c78375309 : SUCCESS in 1h 08m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4028215eab6e485f883982d7870e5db8 : SUCCESS in 1h 47m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5fb621eff74745d393a4036ac90a8b5d : SUCCESS in 2h 17m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d5820c59ec2493596ec9628a5576f65 : SUCCESS in 54m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccd831545ad44074be423da1310c2c52 : SUCCESS in 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f5f231529f49b0a352723c90df79d1 : SUCCESS in 12m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd3ffdfa1ac9401f9467bfb32d4e1de0 : SUCCESS in 25m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f4988c26b2c43b8b7214b625467a58f : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8117bf327e6c4c6fa31b32856114bf78 : SUCCESS in 9m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bd9f9458f1b04280be7ef2f1ffa7526f : SUCCESS in 22m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bea8ffa79bf645278d3573757c285e3d : SUCCESS in 42m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b403ca6aefb4d3abf4ee970a583f380 : FAILURE in 1h 10m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78d3594b4d7f491bb4d6add12cb21976 : SUCCESS in 54m 00s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/81a8b7c5259a447fac7fa12009f4444a : SUCCESS in 53m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e3b29460b73c413ca24c0efaa2685c9b : SUCCESS in 1h 04m 37s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1241b0d7ac1430b8a5a98c262e28b17 : SUCCESS in 1h 01m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f2fe7ead0183480a813feda0d36dd475 : SUCCESS in 1h 52m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/84c38dfb107344118bc796589cbebb78 : SUCCESS in 1h 30m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f88965273a6c498898651d52aadad03f : SUCCESS in 1h 04m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b51dbd1762994925abed7c4a68fe07ec : SUCCESS in 1h 37m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f31801d3f8f4c689d3d5f9f23a9640a : SUCCESS in 1h 47m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a503e25f73604f7f8c1d22960459050d : SUCCESS in 1h 32m 15s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ec5cc5b9d84f4d43a2efc9c5edb5aad2 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":3},{"id":"d8a0c2aed52940fe61bc8fbad62faa78350d0bbb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-10 12:29:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a087cd423d695a0c311071dffa8e74db233852c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:22:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/357b6c2c19c644caa8fd8240aeaa8034\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c92674c986334c2f9559902bd072c0a1 : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d6f034d6764427495e776c160403f10 : SUCCESS in 50m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"694609ca352d60f925d3ed35d8834381b560ba71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 14:35:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/067f86df3b6845bda21af514ed731f34\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b8d974b7362049fea1a0fedb2c886ba0 : SUCCESS in 1h 58m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99ea61808f96405fbb5af08b0a7bf81f : SUCCESS in 1h 43m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/73908e8cd6db49919614816acf03b554 : SUCCESS in 1h 09m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c659ff78e647409e9114f4a4f934b945 : SUCCESS in 1h 15m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b9f0d422a1a4f5faf09e58cec7d8c8e : SUCCESS in 1h 53m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/677ec54a3dca4c89ba8bae9428f77434 : SUCCESS in 25m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ee26ecec7904fbfa794cb3f21acea91 : SUCCESS in 45m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb3e0e176d54a2c91b4c7448397a9c1 : SUCCESS in 13m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0201b02399b0456097307bd604508b8f : SUCCESS in 22m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75daf95fbd6e4c65b4676eb1f89c11f6 : SUCCESS in 22m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d929ace64bf641158f69fc2ed787e44b : SUCCESS in 8m 35s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a9b960262edb4c38a9cfb8b934bf888c : SUCCESS in 19m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54f261561bdb45ba98b1c7975491ed4b : SUCCESS in 43m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/390c07d72f024eefa5b48e24d7e4d0d7 : FAILURE in 1h 08m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ad0fbc41f1142b786282ed10ada2168 : SUCCESS in 54m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a91ceb742ff243ed9fad66a2ad78a00f : SUCCESS in 55m 41s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1eb3920c88e14ee691b5fc98662abab3 : SUCCESS in 1h 10m 18s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5df629317a4e430ca19ca12087446542 : SUCCESS in 55m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8be1d7e8c3c041bf8c217ef160181f30 : SUCCESS in 1h 24m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae5bec64c5b1440089cd4935f48ff033 : SUCCESS in 1h 17m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0bbb6dcd3df04382a0af5e14c7bfd255 : SUCCESS in 1h 00m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/094ffd21f38640199ff28a03e7bf3be7 : SUCCESS in 1h 40m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6c2901bfdad943f499eeaa0279199b0e : SUCCESS in 1h 43m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/53d10bfe3c784e78b1a5bf5150440359 : SUCCESS in 1h 23m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9ebde3ba190b40a983691df02b32901d : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8922d0e4f36d3aae478d16698198f8974dbf5bc2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-10 21:08:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a62fe0a9f58a4ef05566d6ac606d245433529861","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-10 22:17:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1b140e47063df74a319b611d252e4b9c52eb3be","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-11 16:27:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a6bb3e6f676230deff416565024ddd84bdf97596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 17:12:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29acb3613a6445a29c5ff151c01ab9f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/262bd687e1d341e391bd3a1877e1e551 : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07e0311f89b24e6696c34847a6d7aef5 : SUCCESS in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4cfc432d5e69430f420c2bc4d0e387a2bd9e5dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 18:36:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fa6c71121454a6a857a95cb8f8a9e23\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f14361856252485088e93b2cbbf62b8a : SUCCESS in 1h 36m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/12ed5811bcea46f795a1dcf3d742344b : SUCCESS in 1h 14m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a921eea067e491ea0ed7401b2918622 : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3c17ec4f21d2452080db79eb52e86cd8 : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aff8c9d5c16047de9621269f371fbca8 : SUCCESS in 1h 17m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ea14073a5e44ce6b3aaeb4dcfb34436 : SUCCESS in 25m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05768c912d22446693f189a96b56a1df : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f02d94d9574dc4bf93e7c568928f7e : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/615be1af5bd04aa08d140d473ad7a207 : SUCCESS in 18m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab2a33bdc4b4437b83db8a5d89c7b370 : SUCCESS in 18m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f44f833ada407e81b54355616e251f : SUCCESS in 8m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/32906dd040f544279857ad83d18a3d4f : SUCCESS in 24m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df3a5963f23d428b835508c2a4ab01cd : SUCCESS in 55m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a031977b9df4032b7246162ead8f412 : SUCCESS in 1h 28m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd9dd492d61c49058d9a13df41996b6e : SUCCESS in 44m 29s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d8cd19104a574d358fe1778bf6051ae6 : SUCCESS in 1h 12m 37s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e1adabe14dc448fa7dfb809f83dbeab : SUCCESS in 48m 09s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9d5d41d2ae3840a8a681a2f999a61215 : SUCCESS in 1h 03m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b394f59c6a24012bc1e486246f4a0af : SUCCESS in 1h 01m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6cc5d80e32104505a3713535f4cb1173 : SUCCESS in 1h 29m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2729b831358c46f28063d0d08142fa4c : SUCCESS in 46m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/015ec52c27d14cb4a6dc9234fbda71aa : SUCCESS in 1h 11m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f61b3f6cbc444b49a294ab3fe336b88 : SUCCESS in 2h 03m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5a5e2a911b1f4fbdbebda85e53a01c42 : SUCCESS in 56m 01s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/68670a398a864f4ba31325eea9c99b8e : SUCCESS in 1h 24m 13s","accounts_in_message":[],"_revision_number":3},{"id":"2174c2eedf064bee76f3104eeb39648c366680fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 18:36:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bb56cf4cbf85eb6bf09e0ad3c913a9875c063995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 20:24:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddfdc861aca4aed91ec231070eea7ad\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/992739e06a174463aca34c1b58a44418 : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5ef782277c6b4ed2a10276aa409ede6d : SUCCESS in 1h 23m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/259b66336a734f2ca487a0413dd5933f : SUCCESS in 1h 26m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b734c94cac074413882ff821ba1d75a0 : SUCCESS in 1h 04m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5cf70ba6f2514ea38b65701e9586d40d : SUCCESS in 1h 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2c928ff46f4dbfb41547fd8c6b0ffa : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47647bbd67f74a6fbfe7f995d060245d : SUCCESS in 18m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eed6d5635c94472da78ee8a7b61bff01 : SUCCESS in 18m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fca29b33085476d8bef35dc1429ec15 : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/981d6f0bbb0e4e73aedd43689912145e : SUCCESS in 51m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/020c794711c149a08adb1dc704677b6b : SUCCESS in 1h 42m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ca0b338cc46f4599b60a5dde788a84b7 : SUCCESS in 1h 24m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/34dc8d759f9f42188ff9531b766fd685 : SUCCESS in 41m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52866d84920e416f929bf1cbd9390ab0 : SUCCESS in 1h 05m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/31f52ca380fe4b7599db106fb4cfdd98 : SUCCESS in 44m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f8e526d7187c48b28f55dd50d8ab1987 : SUCCESS in 53m 21s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/008bc9f10f524e46815528d982926fe7 : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":3},{"id":"1b3e2d7397fcd3308f0576aa89c595578f1caae0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 20:24:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ef928d3806a69f7f88b4756eaa117fda41b4215c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 20:25:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350ec7580c0b4e8788b5e35a1b1f20d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14bfde98809402594dd1694de58310a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"39b65575cd8cb2f0e7ecc9e91ca4871f269b2749","revisions":{"419d385399b6bac54989d69e8818e25ebd3ec07e":{"kind":"REWORK","_number":1,"created":"2023-03-02 09:54:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/876049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/876049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/876049/1"}}},"commit":{"parents":[{"commit":"92766c85b642bdece24c629625bf80050882c797","subject":"Merge \"[sqlalchemy-20] The Connection.connect() method is considered legacy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92766c85b642bdece24c629625bf80050882c797"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 18:52:52.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 18:52:52.000000000","tz":60},"subject":"Change the release tag to use the release identification","message":"Change the release tag to use the release identification\n\nAccording to the TC resolution [1] and the release identification\ndocument [2], the release number will be used as primary identifier in\nthe development cycle. Release name will be used in marketing.\n\n[1]https://governance.openstack.org/tc/resolutions/20220524-release-identification-process.html\n[2]https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Ia3da9bf6d6113edacb0e21de9a51bd60d5e2465b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/419d385399b6bac54989d69e8818e25ebd3ec07e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/419d385399b6bac54989d69e8818e25ebd3ec07e"}]},"branch":"refs/heads/master"},"a26eb05aba4fa71711db5e911e0b747e880da929":{"kind":"REWORK","_number":2,"created":"2023-03-02 14:04:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/876049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/876049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/876049/2"}}},"commit":{"parents":[{"commit":"92766c85b642bdece24c629625bf80050882c797","subject":"Merge \"[sqlalchemy-20] The Connection.connect() method is considered legacy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92766c85b642bdece24c629625bf80050882c797"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 18:52:52.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 23:04:20.000000000","tz":60},"subject":"Change the release tag to use the release identification","message":"Change the release tag to use the release identification\n\nAccording to the TC resolution [1] and the release identification\ndocument [2], the release number will be used as primary identifier in\nthe development cycle. Release name will be used in marketing.\n\n[1]https://governance.openstack.org/tc/resolutions/20220524-release-identification-process.html\n[2]https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Ia3da9bf6d6113edacb0e21de9a51bd60d5e2465b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a26eb05aba4fa71711db5e911e0b747e880da929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a26eb05aba4fa71711db5e911e0b747e880da929"}]},"branch":"refs/heads/master"},"39b65575cd8cb2f0e7ecc9e91ca4871f269b2749":{"kind":"NO_CHANGE","_number":3,"created":"2023-03-02 14:44:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/876049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/876049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/876049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/876049/3"}}},"commit":{"parents":[{"commit":"92766c85b642bdece24c629625bf80050882c797","subject":"Merge \"[sqlalchemy-20] The Connection.connect() method is considered legacy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92766c85b642bdece24c629625bf80050882c797"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 18:52:52.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 23:41:19.000000000","tz":60},"subject":"Change the release tag to use the release identification","message":"Change the release tag to use the release identification\n\nAccording to the TC resolution [1] and the release identification\ndocument [2], the release number will be used as primary identifier in\nthe development cycle. Release name will be used in marketing.\n\n[1]https://governance.openstack.org/tc/resolutions/20220524-release-identification-process.html\n[2]https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Ia3da9bf6d6113edacb0e21de9a51bd60d5e2465b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39b65575cd8cb2f0e7ecc9e91ca4871f269b2749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39b65575cd8cb2f0e7ecc9e91ca4871f269b2749"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
