)]}'
{"id":"openstack%2Fneutron~876463","triplet_id":"openstack%2Fneutron~stable%2Fzed~I8f164c12db7c4c1aa54ba07feede3586e6a19381","project":"openstack/neutron","branch":"stable/zed","topic":"agent_bridge_test-stable/zed","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-07 09:47:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-03-09 18:04:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f164c12db7c4c1aa54ba07feede3586e6a19381","subject":"Reintroduce agent bridge resync test","status":"MERGED","created":"2023-03-06 18:24:30.000000000","updated":"2023-03-09 18:06:03.000000000","submitted":"2023-03-09 18:04:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876463-agent_bridge_test-stable/zed","cherry_pick_of_change":875586,"cherry_pick_of_patch_set":1,"meta_rev_id":"6d7b4844fc6f4021b40ee6d3b9b01a54c9758d00","_number":876463,"virtual_id_number":876463,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 18:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-09 13:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-09 13:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-09 13:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 18:25:48.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-06 19:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 08:51:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"4e7cfd2f26cb25c856c9cf02335de2b9c33a1795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-06 18:24:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"28807c92a94612d2f667584d480c2ae47aa5abd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 19:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e508768f4e242a1ad817f72aebd0be3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945dff5f6183440c9077c9fe81472f4b : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6150dba1d41e45bd8e418a835bbf044a : SUCCESS in 7m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bb4e45a33784dd29d6fa3dd9d0adf43 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7e23e18db2016db9c88aeb9e8972228fc14d8059","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-07 08:51:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2d575bf7e5ec55c65f83dab2356482a17ae5d77","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-07 09:47:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cb0e2ce7109d95ae82578bee082c0d13d8fee39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 09:48:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"215e0b1809fde3916aab1c2ae962d854d05b6712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 14:51:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c8c13abcade45f89a9d10c9f9d5564d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c937e11f85a248218b284e23de579359 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae975ddc1cb24a0d9808d66825d8622c : SUCCESS in 12m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b6381d7d3f54becb0075997999cc269 : FAILURE in 39m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d3dcf9133353b80f7227b876fbc6b422c6f174d6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 08:01:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"991173e16c870ba06691272fb8068d6f8490187d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 08:43:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b595df4ad5d94175a03add4d724bacb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f331839b1840bd8bc17123f4a0c578 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb08b00e5844b048ea7c100095d6846 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec5c7844dde34cce9137de6feda4dc8e : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3ade08c1a54359e6a01dbe34d010306b929af7f9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 09:03:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82809ea73cf6ab1b0e8eb09bd26bd925a67b7d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:06:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9f2656cf5544a19af8b6522afbb54a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d1d85e3e964e4f9093770a4f9f6309 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/813e835059bd4bb69145037a7553cabd : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/daa0ccfe7bd7419bb3e305d7bfe0ae1a : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8b2c4c15556c51f9ba8c98e17c3d0daabd9e9b3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:07:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19728e610d6689cf55d26a167d4ef0ddd5183d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 12:38:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a4c65efe9b40668b83b1dcace72894\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d50a9c58da0a4d5fbf248ea03495bcdf : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438751415af94d71b7ceeb2789054880 : SUCCESS in 9m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8b1b826531047e9afbbc8dd8f8dbe94 : FAILURE in 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c6d2231071f08b07f925366f0723269fcf8216b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 13:47:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d0d2a24836fc600f22abd64fc43b15283a97c10d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 13:47:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2ccede9fe5a79151a5fbd2dd27b978cdfe406b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:50:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b4afc0549e4298bb5ae5700e02e848\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09639664716e42c9a65210866c34d7d3 : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673f6aaef4f44113bfb10dd57e2aadd8 : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03c5241501c5445f849db9bf631a4a2c : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":2},{"id":"154a43a55458acf4ca3c8871c4c2bdfbe420b118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:51:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de931c8180b24069db928e0c7dbd1d5d3d94f085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:04:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4534b3fb51144cf8ae9e3a22b3a9e80c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f763769624a94dbbb076e1a7213a5846 : SUCCESS in 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902134524bb44e65bb083f67f3bb6422 : SUCCESS in 11m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d42534b1b4be4dbea2a3ee165b3b8389 : SUCCESS in 2h 01m 05s","accounts_in_message":[],"_revision_number":2},{"id":"59b0bc8a7a0eecf1ab48abe35964464cbfd96f99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:04:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6d7b4844fc6f4021b40ee6d3b9b01a54c9758d00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:06:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dcc40dcbbeb42fb8a9a4a6ec31025fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4408b778844e96af05672cd328a9de : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e02beeddf567e12ffb156afdcf2462980634f61b","revisions":{"664f3ba843c33ed723661a32830294b6b72c16fd":{"kind":"REWORK","_number":1,"created":"2023-03-06 18:24:30.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/63/876463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/876463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/876463/1"}}},"commit":{"parents":[{"commit":"9dd8ab4b6a61600432189f508d4996367a231aed","subject":"Make retrieval of port mac column safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9dd8ab4b6a61600432189f508d4996367a231aed"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-02-28 02:58:53.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-06 18:24:30.000000000","tz":0},"subject":"Reintroduce agent bridge resync test","message":"Reintroduce agent bridge resync test\n\nIt appears we have changed the test_agent_resync_on_non_existing_bridgetest\nmain purpose in change 709544[1]. The test is no longer testing on what is\ndescribing it will test, i.e. resync on bridge change. Testing resync on\na bridge change is still important. I feel like what the test is testing\nis already covered in _test_agent_events set of tests.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/709544\n\nChange-Id: I8f164c12db7c4c1aa54ba07feede3586e6a19381\n(cherry picked from commit 75e83602247360cf10b7b9aa4e53a0135b491cae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/664f3ba843c33ed723661a32830294b6b72c16fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/664f3ba843c33ed723661a32830294b6b72c16fd"}]},"branch":"refs/heads/stable/zed"},"e02beeddf567e12ffb156afdcf2462980634f61b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-09 13:47:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/876463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/876463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/876463/2"}}},"commit":{"parents":[{"commit":"ee28d85db943ebc6c525fad97b225236df054016","subject":"[Secure RBAC] Add shared_security_groups rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee28d85db943ebc6c525fad97b225236df054016"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-02-28 02:58:53.000000000","tz":-360},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-09 13:47:12.000000000","tz":0},"subject":"Reintroduce agent bridge resync test","message":"Reintroduce agent bridge resync test\n\nIt appears we have changed the test_agent_resync_on_non_existing_bridgetest\nmain purpose in change 709544[1]. The test is no longer testing on what is\ndescribing it will test, i.e. resync on bridge change. Testing resync on\na bridge change is still important. I feel like what the test is testing\nis already covered in _test_agent_events set of tests.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/709544\n\nChange-Id: I8f164c12db7c4c1aa54ba07feede3586e6a19381\n(cherry picked from commit 75e83602247360cf10b7b9aa4e53a0135b491cae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e02beeddf567e12ffb156afdcf2462980634f61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e02beeddf567e12ffb156afdcf2462980634f61b"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
