)]}'
{"id":"openstack%2Fneutron~876465","triplet_id":"openstack%2Fneutron~stable%2Fxena~I8f164c12db7c4c1aa54ba07feede3586e6a19381","project":"openstack/neutron","branch":"stable/xena","topic":"agent_bridge_test-stable/xena","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-07 09:47:48.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-03-07 15:17:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f164c12db7c4c1aa54ba07feede3586e6a19381","subject":"Reintroduce agent bridge resync test","status":"MERGED","created":"2023-03-06 18:25:04.000000000","updated":"2023-03-07 15:18:22.000000000","submitted":"2023-03-07 15:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876465-agent_bridge_test-stable/xena","cherry_pick_of_change":875586,"cherry_pick_of_patch_set":1,"meta_rev_id":"ecf1cf1d470eb0360b57c3087301f8dd6467709c","_number":876465,"virtual_id_number":876465,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 15:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-07 08:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-07 09:47:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-07 09:47:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 18:26:16.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-06 19:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 08:51:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"5b3cd403c392eeb53d0656a7a83a1de4c2f64edf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-06 18:25:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6ee5b79b4541402a0f56fbe16fceddaadf0b9580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 19:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8031d45a6c89472196ba296d759a769d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2239d570554fecbd3dacfa346f21d9 : SUCCESS in 14m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/148ec46c31064d1cade4c9918d676b0c : FAILURE in 26m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f92be609dcc46fa90c3996b267165cb : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c81c6b7740348e297424e6f2f6ddab8 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3640fc116cf3b259020f5aa16b92ba97695e1dac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-07 08:51:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e91b229443d851cc28624df6e3da769d3b53c4d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-07 09:47:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc204fde571c3fa83fddb31eadb369936b940c1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 09:48:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02da982fff5e6cefb58c46c2ac87a59ea7f7a221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:17:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6136a0a606a4496a8b71f1bf026832be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ae56b15acf45c2b3adea8dbe67cf67 : SUCCESS in 14m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc4cd44568e744eda8fe562d905dc060 : SUCCESS in 24m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3daeb3d443746a2ac47830f0d8b1999 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/271b38bcc6b1434c82e4a335b983d639 : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9bc5801cc1575b4623a17690272ab4bfb43b5b03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:17:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ecf1cf1d470eb0360b57c3087301f8dd6467709c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:18:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f9235424f3f4391bd3af1b88e59b0eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c05779b4ec45c68941ee0c40d436b4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee781c277b3aa47eb8f891568fb4b1b8359c2f38","revisions":{"ee781c277b3aa47eb8f891568fb4b1b8359c2f38":{"kind":"REWORK","_number":1,"created":"2023-03-06 18:25:04.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/65/876465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/876465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/876465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/876465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/876465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/876465/1"}}},"commit":{"parents":[{"commit":"69eb752cb158a33ba6965deb86a11c546dcf2171","subject":"Make retrieval of port mac column safe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69eb752cb158a33ba6965deb86a11c546dcf2171"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-02-28 02:58:53.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-06 18:25:04.000000000","tz":0},"subject":"Reintroduce agent bridge resync test","message":"Reintroduce agent bridge resync test\n\nIt appears we have changed the test_agent_resync_on_non_existing_bridgetest\nmain purpose in change 709544[1]. The test is no longer testing on what is\ndescribing it will test, i.e. resync on bridge change. Testing resync on\na bridge change is still important. I feel like what the test is testing\nis already covered in _test_agent_events set of tests.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/709544\n\nChange-Id: I8f164c12db7c4c1aa54ba07feede3586e6a19381\n(cherry picked from commit 75e83602247360cf10b7b9aa4e53a0135b491cae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee781c277b3aa47eb8f891568fb4b1b8359c2f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee781c277b3aa47eb8f891568fb4b1b8359c2f38"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
