)]}'
{"id":"openstack%2Fneutron~876563","triplet_id":"openstack%2Fneutron~master~Ib8340d9430b946a446edf80886c49fbac729073c","project":"openstack/neutron","branch":"master","topic":"bug/2009221","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-13 08:02:57.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-03-13 12:57:27.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-10 16:00:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-03-13 12:57:27.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-03-08 22:20:37.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-03-13 12:57:27.000000000","reason":"Change was submitted"},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2023-03-13 12:57:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8340d9430b946a446edf80886c49fbac729073c","subject":"[OVS] Allow custom ethertype traffic in the ingress table","status":"MERGED","created":"2023-03-06 12:22:34.000000000","updated":"2023-03-13 12:58:36.000000000","submitted":"2023-03-13 12:57:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876563-bug/2009221","meta_rev_id":"5b6a1a173abf2d55e2d3b56d7160ce73d60d458f","_number":876563,"virtual_id_number":876563,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2023-03-08 18:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 12:57:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-10 16:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-08 22:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-03-13 08:03:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-10 16:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 12:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-06 12:59:36.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 14:07:56.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2023-03-06 15:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 23:11:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-03-08 22:20:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-09 16:11:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-03-09 16:11:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-03-09 16:11:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-12 03:36:53.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"c432e9ef6e1f4bad78d64f4560e4aebd2f2e5c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-06 12:22:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3533cdb8aa7225561099bcf301ecfcb252529e3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 12:43:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aa5a9f029b84d44afb65f82c2783992\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ce08b6df0204d9ca758f83821089063 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11f45203ec1f4c0d9582355e02bfc928 : SUCCESS in 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fb43abff1b87ec90ab0fe5ad71c1cedcb59c6ee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-06 12:59:36.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/63/876563/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"49cfd989adaa188946169efda6c13da50009a1fc","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-06 14:07:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9d509735250878e69272f41e944a1a65efc92512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 15:03:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bcdbab2bdc84145952cd85b8d992f24\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/216ed6eddb91435fa7d81a4390b34e88 : SUCCESS in 1h 20m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/73effef398f441648f4d5332c58c2b0e : SUCCESS in 1h 16m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ab6f3735f0ff4c81a369d792a47221c6 : SUCCESS in 1h 24m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a838f84da86e4097bb42ab1258587043 : SUCCESS in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68bec95fc4514d909fbf72eba038d668 : SUCCESS in 12m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ca2cb91889544dbbfd4fd05c956e866 : SUCCESS in 27m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09bc5dfd60d84140a082b7e96f47c83e : SUCCESS in 25m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a45e82954644d4ca6b02a0b88920a75 : SUCCESS in 9m 32s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/aab20c91eab64747989c140c013b5f94 : SUCCESS in 23m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/68adbabae03448f8a92c4a023997b95c : SUCCESS in 50m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/378c0e4a0bcf4f62863c5270e0aee5b3 : SUCCESS in 1h 29m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/80b695ff80484d738279d3264ae60cee : SUCCESS in 54m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c8dac233683d47e3b0fc3ff994e71f8c : SUCCESS in 55m 04s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8ff481d3019444409964c7818703abac : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2cefd14cb4d246cb817fbe742695235b : SUCCESS in 53m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0e47cabef5c143b1a76847451402aaed : SUCCESS in 1h 27m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/56c6a25ba1b648dc8989ab5b01ad010a : SUCCESS in 1h 35m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/41ec17b765c5495eb04958841a9c2884 : SUCCESS in 1h 03m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/08428ff38bc7494496ff7954ca9ee045 : SUCCESS in 1h 32m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bad034123f194cba829339f1d7f7eab1 : SUCCESS in 1h 40m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a9045465eff84219877317e7b968a07f : SUCCESS in 1h 03m 21s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/dfb5aa9cfb524dbfaf2fbe181ee0f8d3 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4196550dcff26756ab3ae0076390ea25e23fcfdf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-06 23:11:39.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cd239a46107309cf5b2d004d4aad8373e9b07429","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-08 16:32:53.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f6b6edb06992ddbd3138d02492e13f28e7f4081a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-08 16:33:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b58787d773fbc459ca2fca6a61c47c9f272f18d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-08 17:51:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee5e522e3a6afe81d6be2bd2076eb027d45fc39b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 18:09:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8288e220f0f4a1fa3aef54bbaf4960c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb18b9944d364c8c911da329172441e5 : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be730093629b422c9bd15a48bda8639f : SUCCESS in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7f7ff2676050435e9ab69851250847598a3d825","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-08 18:30:41.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/63/876563/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"488d353eb1c853848abf2ae58e23724ea581ac31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 20:05:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2fd3dbb9c924d5c930b5225034d6110\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d0126c857eff49ccbf626c51506bcf74 : SUCCESS in 2h 07m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87e7bd3863254e2d82d01244cdf86373 : SUCCESS in 2h 01m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9fdca67a2ef4442fa086f8cf6dc24f62 : SUCCESS in 2h 11m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91e46ee9650c4320b473588e6a03bf69 : SUCCESS in 48m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68f6e31d0db84d43ac031f7b3826e686 : SUCCESS in 13m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fd47a88c1384336947ae8d7cfce38cd : SUCCESS in 21m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7271a7af5f1a4c39b1283dc98f5089c0 : SUCCESS in 27m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7946a9c96404a199a9ea94464d72430 : SUCCESS in 10m 02s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7db001659b164834b1f7ea03c582f874 : SUCCESS in 24m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/139064e2e39d44beac60a29ae3c6ccbc : SUCCESS in 57m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5213597e5a954ca2b96c3d698a22a911 : SUCCESS in 1h 51m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/959550da72b54105b0075c05c74accf5 : SUCCESS in 1h 20m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/21f7733402054899baf8f2248c5cc9ed : SUCCESS in 1h 00m 38s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ace65cc724224e97bc998b5f22c15710 : SUCCESS in 1h 13m 23s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b0b15fbd4b00409f96f05e462bea17a2 : SUCCESS in 1h 19m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b742b8ad17947488b42d0565367214b : SUCCESS in 1h 40m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bec1c017783b46259dc9512e02cf3639 : SUCCESS in 1h 21m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/851e83033dbb4e64b3f1de7751286888 : SUCCESS in 1h 18m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/774c723803664cc48ed6763c7b595f23 : SUCCESS in 1h 34m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/004e5ac99b0445a591317c5ad848da13 : SUCCESS in 2h 01m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6a9a6c7efe894705aea4d8513df2c668 : SUCCESS in 1h 24m 26s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/215b19f9a2f648d991532c8525699b2a : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1a7e3d861c5564bcfbe3e549dc250da9d7163c8e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-08 22:20:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"871471fcf3bea67ec76e1a52e49b5f23ad5770fa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-10 16:00:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"007b319cf7908fdb25835ea0c3b2498743d5a66c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 16:00:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dcbd49392532cc471adebb3ad8e1a016d601d1f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 19:40:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2b3cc57ea9c47908a969330955a3435\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4a2a9d297404bfe826cd315df08750a : SUCCESS in 1h 35m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61bd403d160c44eab7788e49f7793936 : SUCCESS in 1h 40m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/46f9f8852c454d998e8ad1acf545b176 : SUCCESS in 1h 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96196768814b4df785ac569945167ac2 : SUCCESS in 11m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c32491b117c43a0b1c515bd9dc6cd08 : SUCCESS in 23m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbb97f36f9534bdeb0ef727674622cbc : SUCCESS in 45m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489641e2a6af442e96b4a20129670bbe : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d1910037db844e88cf9ef694068cf17 : FAILURE in 53m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0eea01c2e85a4a9db32df9d1d19dd8e7 : SUCCESS in 1h 32m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/090b55ca35174d4489a4f93dd87d7fd1 : SUCCESS in 2h 09m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3586d1d2819049b1a277adf85c1a4949 : SUCCESS in 50m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/548f948d7d564d3ab8610994d8b51a84 : SUCCESS in 1h 06m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/74118e88efc041e4a88b49b225af5135 : SUCCESS in 1h 11m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bebeeb69b7de4acb82f5e252204b5aaa : SUCCESS in 1h 36m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/15340ee0f1a94eb4a6e94bd9005a09d1 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a3a881151b1d8ea0b56c26fb98adb12ef4dbfd08","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-10 20:38:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9dc4f5ca69d993fd2a29d30ce1698e8873fab8d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 21:29:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd1388ea589481c9ba00af426172a81\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47073b37211542e8966e9f5c54e9bc61 : SUCCESS in 49m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b769be0435304c03872a76ca8e8541bc : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"627af83b612c8360c66e2b16c807c31459ada85b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 22:53:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c4ccf482c641739102d542b325682f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f322bd99b884da3b5143d2670596560 : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e87ac020535a490da875cbdf0e0236d9 : SUCCESS in 1h 50m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d3578375c2f24c4bbc244c7ec5143ce8 : SUCCESS in 1h 29m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0289a27b730e4b6292613f2419ee0da2 : SUCCESS in 38m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c488f50ad5470caa746ef35e0e8e11 : SUCCESS in 12m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51a81f7e5fb74d6eb3d9ec7e1d477dac : SUCCESS in 27m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a976ee1d558143f9a467aee8c2841d3a : SUCCESS in 28m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776220e47aec4b67853b4bbe0eedde3f : SUCCESS in 11m 36s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5cc13534307a4801b675ca8a2d481739 : SUCCESS in 22m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c972b84046c14baba50c6362d4b292da : SUCCESS in 57m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e0177b1b25543e28daffe68cecd846e : SUCCESS in 1h 41m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/83f7d67ee57e4af581010a94ab4f17f4 : SUCCESS in 1h 02m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/68a793d20aec4155bb198a768cd3b90f : SUCCESS in 1h 02m 53s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e57c2bf1e9134cd4bbf27805828c7b5e : SUCCESS in 1h 05m 42s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ac3b22118c54d1aa284b6404d4b95fb : SUCCESS in 1h 05m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e815693c2d1f43ad80ebdb3ca0c9a1f0 : SUCCESS in 1h 41m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d6959135185a4c5f845d77a6aa671b3b : SUCCESS in 1h 36m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c60fee8b3ac14d76945b8a56e08f48c0 : SUCCESS in 1h 06m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9877bd96e03f47d0b05657720fcddf0c : SUCCESS in 1h 45m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/08d38e0d5d804e1aa5bc812b9be5f5ec : SUCCESS in 1h 41m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8287feadd1cd43f9bf1bc44f81f128f3 : SUCCESS in 1h 11m 32s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4c322d1b3c754d28a7c48d2e46c725be : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e6e75985cfb4f3ea34e09c64d0aeb30eff9e6266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 22:53:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f420c9687a054b4818f39099686799b1bd5ae258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 00:52:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c04f4b414904d3b90060922f7abcec3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c370827655b049998ccbf4abde9648bc : SUCCESS in 1h 35m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0472cdf1f62c46f988c3396baf4c2e98 : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ef891145aace478ea0206e46e1ed3cd4 : SUCCESS in 1h 34m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2524a033324cb1970f88467186e783 : SUCCESS in 13m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58447914d62342bfb0d33f95dd0077be : SUCCESS in 19m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16a58ecf27934b7eb73068eb376b1398 : SUCCESS in 26m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d628d20050b144f7ab2425905c9faa55 : SUCCESS in 10m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/07777ff23215410ca1f22630d2d54afe : SUCCESS in 58m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/34c61e6d22f646e69f3e35e8388dd79f : FAILURE in 1h 28m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6c64fb87e7db4b34bd1c42c406824474 : FAILURE in 49m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/61ef6b4397ba4d678b6b0c9d6241acc5 : SUCCESS in 1h 27m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/558599486f26448da82390b57b17a031 : SUCCESS in 49m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/092298ebb4e54556a0306a2ada1e7fe3 : SUCCESS in 1h 23m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a7af9109e137496f987e17494e0276b9 : SUCCESS in 53m 14s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5ac09fe34b09428aa0c693c927cccc53 : SUCCESS in 1h 49m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e08676b6c035602430d1e550c4294c87f6ababaf","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-03-12 03:36:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"af6b7c40f8b81ce6f5765616c3066b81dff095e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-13 08:02:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b4127613f17236525bbd00aecb8f90deeb69c34","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-13 08:03:08.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_32238\u003e\n","accounts_in_message":[{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"_revision_number":2},{"id":"f55a5051fbe97b1be2596cd674f905d06bd5da78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 08:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69f5bba026642d585db586e5b2117f4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e17a709c31f43e7b29122a27870ed8b : SUCCESS in 43m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05885aed2eda4eaba322d6498e1bafeb : SUCCESS in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32c1c834b59c591bfa8ee28b102714351111e70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:19:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c70ded778844bcba7d1d1b3f3f64fc8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/27554d55c6fd40f9bc64b339b9540d54 : SUCCESS in 2h 05m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be6ec2839b6645b0ac87d50bcf2bac6e : SUCCESS in 1h 38m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca066774e3094b259b069f37505961d7 : SUCCESS in 1h 38m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7bf7bb51cb84db5b688aaa8a4cf973b : SUCCESS in 44m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79adf6915ab549488435d75412f3a661 : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d312b25c86b94b00a12fcd58a71f0a9e : SUCCESS in 33m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54d91baace054d4b803b28f5862d4a36 : SUCCESS in 22m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a06f95640a34acab9739ffb987d1bef : SUCCESS in 11m 16s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7621bf2caaac4ec8a2e7a680245811ea : SUCCESS in 22m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/116e3a01a498473ba88ee1444811f58e : SUCCESS in 42m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/db8bdb00400244af952fb7fab64664b4 : SUCCESS in 1h 49m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b10807a74e5d4e4e897c4de60d8aa80d : SUCCESS in 1h 08m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f74fb3dbfa054deca3abf83d8aefa202 : SUCCESS in 1h 02m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d7f774ae6765463f9348bd16ab1b221f : SUCCESS in 1h 14m 44s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ade0ee7195944d89d66a5b802ab2656 : SUCCESS in 1h 06m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ff664c2055243348f404de00af4a3d3 : SUCCESS in 1h 11m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e1d1ea577c448279917cd06b87da85c : SUCCESS in 1h 26m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7c1e47200964d05be4546b44dc04f83 : SUCCESS in 1h 06m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/35f864197c1a4f4fa86574fe84584fde : SUCCESS in 1h 45m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f42dabbc6ed04e88909656b548e0849e : SUCCESS in 1h 56m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7aa0f53336984d4ebbbfc6fd6b39759d : SUCCESS in 1h 36m 13s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/eab77d4b80094e05bfddd58d31081d96 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c88c0bc65abd05270f74a48bdcd41c289ac75a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:20:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"461b65bd2dc6a8ff18058a48d073e8a030cbab59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:57:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64dd54c253194ed7a9bcb9923a57869a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5deef58ce6e3407686678e7c98367aca : SUCCESS in 1h 23m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cda0a8e4d8594246aeaa0731eabc681f : SUCCESS in 2h 03m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/51f6f169cc944bfaa9df8722a565de33 : SUCCESS in 1h 51m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be1ec8af9c14ffbb17cab53984b51ca : SUCCESS in 12m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/790c257c32bf46968fba8d0ee13bed53 : SUCCESS in 27m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22b1e098f08d4ef883e4e79d534ad97c : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1d3c54627f4bbfa0d2af197864fa7e : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/92c9f70823984ce68f87e41f65565ac6 : SUCCESS in 46m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3741c5196ce14d12938b320a2a527005 : SUCCESS in 1h 40m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b819f4c4dec8432e994946d8849d1b9c : SUCCESS in 1h 17m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/53133163f4ab4447b7510eefd85f0448 : SUCCESS in 49m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41b68251f18749d5b12d1c131c61c2dc : SUCCESS in 46m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5db8370b69794da3ac8f42ae25224cb0 : SUCCESS in 1h 04m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f5576e929a745f9b82b779dfdc32d00 : SUCCESS in 1h 18m 59s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1bfcc68687dd4d609e06c84af29132d9 : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d4d044e2788ea2f6664be593415be97f3384cf3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:57:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5b6a1a173abf2d55e2d3b56d7160ce73d60d458f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:58:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb77e27a5f6492a84ed03c968b92a1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a544429d7f469ebb8a55c76aeb1888 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"008277b8c12d99438951a308b278203fa7a7c3ef","revisions":{"55b37f6e4cb5de8089922b4752f7f8bd4ccf2212":{"kind":"REWORK","_number":1,"created":"2023-03-06 12:22:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/876563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/876563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/876563/1"}}},"commit":{"parents":[{"commit":"0a214b0437874fd7f5379ec94fd07ef5d3ff4bbe","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a214b0437874fd7f5379ec94fd07ef5d3ff4bbe"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-05 21:12:55.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-05 22:20:22.000000000","tz":60},"subject":"[OVS] Allow custom ethertype traffic in the ingress table","message":"[OVS] Allow custom ethertype traffic in the ingress table\n\nThis patch is a partial revert of [1], reinstantiating the code merged\nin [2]. This patch is the complementary to [1]: the traffic with\ncustom ethertypes is allowed in the ingress processing tables, same\nas [1] is allowing all traffic from the virtual machine ports in this\nhost to leave the node. Both, this patch and [1], are bypassing the\nOVS firewall just for the traffic with the configured allowed\nethertypes and just for/to the local ports and MAC addresses.\n\nAny other traffic not coming from a local port or with destination\na local port, will be blocked as is now.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/678021\n[2]https://review.opendev.org/c/openstack/neutron/+/668224/\n\nCloses-Bug: #2009221\nRelated-Bug: #1832758\nChange-Id: Ib8340d9430b946a446edf80886c49fbac729073c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55b37f6e4cb5de8089922b4752f7f8bd4ccf2212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55b37f6e4cb5de8089922b4752f7f8bd4ccf2212"}]},"branch":"refs/heads/master"},"008277b8c12d99438951a308b278203fa7a7c3ef":{"kind":"REWORK","_number":2,"created":"2023-03-08 17:51:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/876563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/876563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/876563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/876563/2"}}},"commit":{"parents":[{"commit":"962b4f9bb9a3cc3777b76222b69882eae7789f69","subject":"Merge \"ovn_idl_impl: fix a logic bug in get_sg_port_groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/962b4f9bb9a3cc3777b76222b69882eae7789f69"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-05 21:12:55.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-08 03:25:53.000000000","tz":60},"subject":"[OVS] Allow custom ethertype traffic in the ingress table","message":"[OVS] Allow custom ethertype traffic in the ingress table\n\nThis patch is a partial revert of [1], reinstantiating the code merged\nin [2]. This patch is the complementary to [1]: the traffic with\ncustom ethertypes is allowed in the ingress processing tables, same\nas [1] is allowing all traffic from the virtual machine ports in this\nhost to leave the node. Both, this patch and [1], are bypassing the\nOVS firewall just for the traffic with the configured allowed\nethertypes and just for/to the local ports and MAC addresses.\n\nAny other traffic not coming from a local port or with destination\na local port, will be blocked as is now.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/678021\n[2]https://review.opendev.org/c/openstack/neutron/+/668224/\n\nCloses-Bug: #2009221\nRelated-Bug: #1832758\nChange-Id: Ib8340d9430b946a446edf80886c49fbac729073c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/008277b8c12d99438951a308b278203fa7a7c3ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/008277b8c12d99438951a308b278203fa7a7c3ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
