)]}'
{"id":"openstack%2Fneutron~876872","triplet_id":"openstack%2Fneutron~master~Idbb8f98eaa429601e2f16861eb01ad72c68cffa6","project":"openstack/neutron","branch":"master","topic":"bug/2009703","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-13 12:39:07.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-03-09 03:18:30.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2023-03-13 12:39:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbb8f98eaa429601e2f16861eb01ad72c68cffa6","subject":"[OVN] Use the BW values retrieved from ``get_port_qos``","status":"MERGED","created":"2023-03-08 13:53:53.000000000","updated":"2023-03-13 12:40:30.000000000","submitted":"2023-03-13 12:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876872-bug/2009703","meta_rev_id":"2317aea67c13ba72add3ee965dde2866f0ffb872","_number":876872,"virtual_id_number":876872,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-03-08 14:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 12:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-13 08:19:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-03-10 16:01:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-09 03:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-03-13 08:03:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-03-10 16:01:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-08 14:22:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-08 14:32:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-08 14:39:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-08 16:00:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-10 16:01:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-12 02:33:38.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-03-13 08:19:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"f73b039dc1f8a863f72d7ec787c3ba7f13da9e76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-08 13:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de9faed89840f1c048816f4c3d9f23a92183e8d2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-08 14:22:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"74916edfa0c413728c14f14433337dbd20172f3b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-08 14:32:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/72/876872/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3ed3544eedb3718b937a3d1d9a1878accd792d3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 14:39:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de55eaf15a034d0694478d07de257c75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb9a10f70440409f91aa429a919155d8 : SUCCESS in 43m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2b25fd7c90c48dabe15a15d26dd0609 : SUCCESS in 42m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e3a8f90bfa4e2722bd20a14ff2f216db57dcec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 16:00:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/411419e908444b079bfccfe76409d251\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8d8ddf00abb7458284e3aa98457514e6 : SUCCESS in 1h 42m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b015eb0ebecf4510a9edab396a0a41dc : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f2bd46e15c44c1b68fda5eb1394e23 : SUCCESS in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adf1fa94cb004fa3a09055fdf9004952 : SUCCESS in 30m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7a79c1de4b84630840e10fac6cea668 : SUCCESS in 25m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfc155b3b5a43358a473cf3d9b23023 : SUCCESS in 13m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7f1250d374274c67972d9a315eb6076a : SUCCESS in 24m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/65c20150c2354dbe85932aa16572ff00 : SUCCESS in 46m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/16badad245d544f09671c1b8335cdbcc : SUCCESS in 1h 06m 39s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5dc1be8d1f94c2ea4e9410ce033d532 : SUCCESS in 1h 08m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/69a18f371c4c415092a394017fd3efc7 : SUCCESS in 1h 58m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/65547201712641b280e8740d25bd1e69 : SUCCESS in 1h 25m 47s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7a303df9abe54c2f9b78ed69cb24017e : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0f4a20d2915eb0732cc2cf9834180ae9b5da191f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-08 18:10:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb7a26d50ac3052efd1b12a053863550118c8b3f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-09 03:18:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8009a87e47f50398f85ad8360bf3070d114358c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 08:12:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b02b6aec9da1d54d8011d9b1e5fe4ca76bbfbed","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-10 16:01:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfdfb75a8c00ac8980990c02f02ef7b98d89fcf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 16:02:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f47d94c912f86455229ecc5458b441861f82ca67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 20:14:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/919634e4c4e44295b895b3b56092c422\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/360c204d057d47ae88bd680fcb19bef1 : SUCCESS in 1h 48m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb58eecba9448fc97abcf07c07d13aa : SUCCESS in 14m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68241563dc634fc99b8729e577c33281 : SUCCESS in 28m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4370c527a38c41cfa66f252b1d156cb1 : SUCCESS in 44m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e835356a2804d3886a18397300ee883 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd972c4115384e56864b9c1ddcc4e425 : SUCCESS in 55m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3ec80d26a4d04e7ea89038c50d84e9bb : SUCCESS in 1h 01m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/94df59c556ec4c98a21d7e00fd522c75 : TIMED_OUT in 2h 05m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/6905b366399d413a985f605f77884d3a : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bd62e18926f080852c53fd6c5518b0a936e77183","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-10 20:20:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e56d67bd76b203900d523f85e2ff219aadf3bc1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 21:10:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43bbcd9fc654176b33b3e82e0ebc32f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46be333e271f4bc6bf2b1055155516f1 : SUCCESS in 48m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40fb16fac49144b294f4f81c0f3608af : SUCCESS in 48m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f89d0a8c23a75696ed13bee2e56380b44828d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 22:29:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70583710ff214dca8b1575a16b1ea9fe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/19d43b7296734b01b0e716bf514d690e : SUCCESS in 1h 21m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a8b2895edd245b7bfe5577b22bb55ab : SUCCESS in 49m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38eca7bb903485aa36ec4a755c18be6 : SUCCESS in 15m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73e71581e2584b66a4dbf64b7c8fc14e : TIMED_OUT in 1h 00m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/519fa4f22eb943f6956b98334a7eda52 : SUCCESS in 46m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ef4c81d96a4733af1363162d1ffd1a : SUCCESS in 22m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e45dbb32b390405fb293dfac22372577 : SUCCESS in 26m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/791cbb45a5f647f5aa790cde69ba707a : SUCCESS in 1h 05m 23s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e5ce115523e4c48be936d102b321f3a : SUCCESS in 1h 00m 31s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3561f70e5cf94d4ba12b38ac69c1d0f4 : SUCCESS in 1h 11m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2e8b2a4badfe40a59066bec67898be69 : SUCCESS in 2h 03m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f5293869822349228fd4a3aea4c94daf : SUCCESS in 1h 22m 45s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fcc2a3102af3420fb4ab975e67c6d34d : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"302b924c790c7b77634623325ccf242881d18102","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-03-12 02:33:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1bb13158f99eebf2eda19d1cb2dd167821d388f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-13 08:03:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"590707921a2e1185492fc8acfdfc53794db9e98d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-13 08:03:54.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_32238\u003e\n","accounts_in_message":[{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"_revision_number":1},{"id":"0261745f676c6e5f16f7f3e009bdfebb6a81543a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-13 08:19:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"969d6bb375c76b707faade83281693be4c8d1852","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 08:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd30cba5e544b948e40f6642191d313\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1221dbc54894fe2804ce247d5031b6a : SUCCESS in 40m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9e7bd3c466043f0952573d5cdea15e6 : SUCCESS in 40m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0aace52145685a0df46b8f2fb76c852b9afe27d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:11:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ec1c79f512141838f15c711b7350985\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5a1de635d454e949bdc408bf479f85f : SUCCESS in 1h 07m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/833cca3072404c8192cfd166149d90c0 : SUCCESS in 33m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25b7c0ef5af400bb55b58a51509e756 : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aee845c43646485caddeafc12866893b : SUCCESS in 25m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83f13874eb2f4521ad42530a261f4fd3 : SUCCESS in 49m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fbfa127eab5420996cecde752c433fa : SUCCESS in 10m 05s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f960fb1a31ee484ebe46069b623109b2 : SUCCESS in 46m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4cddbe1e16344e59fdf37d3e1b31855 : SUCCESS in 1h 02m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3de07667398c4ffba62846d8569947dd : SUCCESS in 1h 21m 48s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/70cf088204c6427dbed6bd0c93d6d489 : SUCCESS in 1h 23m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4176f8331f854f04a9e3c9501e1339cf : SUCCESS in 1h 59m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/007a527141df40c1b315e8193391a1ab : SUCCESS in 1h 32m 35s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f3ef5763ce1d438a957e04a404d11bd2 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"525a7790fefbb2b24816175a51936b99cd724d53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:12:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"018e23f46af81feb16e6b313ff5d899b3d1ba368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:39:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/206b46b290674f4e9aded939d2e12139\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/10b28462825b41c9823c44694767e517 : SUCCESS in 1h 44m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8949b5998f1841b5a26bdbdec182ec68 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32976f1ddbd542ba99d8d4bf3af0dc8f : SUCCESS in 25m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a3d7e6e3f9446ea9faac75852715408 : SUCCESS in 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe75bb6f18db4b2eb4360d326765a63a : SUCCESS in 9m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1801a14ce8946378e826d829a92650d : SUCCESS in 48m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fe115024907f4020924fd07eb9becfc2 : SUCCESS in 1h 00m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/552676a370a24e839119425467b8c511 : SUCCESS in 1h 07m 52s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f52939b26a3241538f8b0baaddb729b3 : SUCCESS in 1h 44m 38s","accounts_in_message":[],"_revision_number":1},{"id":"77172cb06f96f80b3c2159428daeef516068eb8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:39:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2317aea67c13ba72add3ee965dde2866f0ffb872","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a6670118f1444a9fe837edc855fb85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55e838791c140ffb7d001b2984c7421 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9567033fc23e9288bf555eff306145ac66f260a","revisions":{"b9567033fc23e9288bf555eff306145ac66f260a":{"kind":"REWORK","_number":1,"created":"2023-03-08 13:53:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/876872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/876872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/876872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/876872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/876872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/876872/1"}}},"commit":{"parents":[{"commit":"962b4f9bb9a3cc3777b76222b69882eae7789f69","subject":"Merge \"ovn_idl_impl: fix a logic bug in get_sg_port_groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/962b4f9bb9a3cc3777b76222b69882eae7789f69"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-07 22:31:33.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-07 23:28:43.000000000","tz":60},"subject":"[OVN] Use the BW values retrieved from ``get_port_qos``","message":"[OVN] Use the BW values retrieved from ``get_port_qos``\n\nIn ``QoSMinimumBandwidthEvent`` and ``QoSBandwidthLimitEvent``, the\nbandwidth value is provided by the method ``get_port_qos``. This method\nhandles when the value to be retrieved is not present in the\nrequired register and returns 0.\n\nCloses-Bug: #2009703\nChange-Id: Idbb8f98eaa429601e2f16861eb01ad72c68cffa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9567033fc23e9288bf555eff306145ac66f260a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9567033fc23e9288bf555eff306145ac66f260a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
