)]}'
{"id":"openstack%2Fneutron~876945","triplet_id":"openstack%2Fneutron~master~I03ea78fcc5825e35c07dda1aa790769aa918dd8a","project":"openstack/neutron","branch":"master","topic":"bug/2009804","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-10 17:49:41.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-03-10 17:49:41.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-10 15:58:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-03-10 17:49:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03ea78fcc5825e35c07dda1aa790769aa918dd8a","subject":"[OVN] Method ``get_port_qos`` should always return 2 values","status":"MERGED","created":"2023-03-09 08:31:12.000000000","updated":"2023-03-10 17:50:54.000000000","submitted":"2023-03-10 17:49:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876945-bug/2009804","meta_rev_id":"b59e6df5fd8bac880e5703b18b1663249f72bdb2","_number":876945,"virtual_id_number":876945,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-03-09 09:10:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-10 17:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-10 15:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-09 14:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-10 15:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 09:10:15.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 09:31:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-09 10:15:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 14:42:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-09 16:08:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-10 12:29:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-03-10 12:29:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"a0d59b7ff7704bb5341641959aceaec86b0e0aec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 08:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8e2ff067cdd7b0a1764132e070bf188dab58b11","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-09 09:10:15.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/45/876945/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ea912832110d579a94395c2e3074f43343f94039","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 09:31:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da09f516a9e847529f762bb883b69a57\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b66c60c769d043ddb887d21f020f3ee3 : SUCCESS in 58m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47243c7c942b4c52875f154f0d85f48c : SUCCESS in 57m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33654bfa585957efdcf01c99037d3197257bde1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:15:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aff8c94a3884462cb9ca1deb68673ecb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/731ea47620d84882b2e1e9ebfaf33835 : SUCCESS in 1h 34m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25911d44734c4726a86ca32d437fd09a : SUCCESS in 39m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6e14bc8a84408f8cf389892afb248b : SUCCESS in 13m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff957dafb97440649f58b976498d0469 : SUCCESS in 32m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08661378cf8e42cf8cc0d2824fa7abaf : SUCCESS in 30m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d112e7f8a4419b9e520b4e5f86a8e8 : SUCCESS in 9m 39s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/2b457e2251a841f0afb5b9c09c9f1076 : SUCCESS in 19m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/529aebe4bd66478abbdf9a1cba395428 : SUCCESS in 1h 02m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b3c9e65f7886403aa93843a283ae6237 : SUCCESS in 50m 33s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/05b503f85236495d82adeb411a162168 : SUCCESS in 1h 10m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b6a0ba3ecb94231a0c508b3f63522a3 : SUCCESS in 1h 42m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5fbca1ee30dd4fe0895181d3001f4274 : FAILURE in 27m 27s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/be4f007e317c4a158f94f47ce656141a : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c9700e529be8eddfb6b6b7498dadfcbe0d51c40f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-09 14:42:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a47ec7aedca4fc06a54306cea092faff2b283e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 15:29:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb18e8ab73894a51ae5ff4e8eb5ba389\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be25a9b51fd84af4bdbe656e84380a1e : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3060c7b39907435081477b566479bc14 : SUCCESS in 44m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3877e273b4edca96acf83f6dbbe3a1085f4def3e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-09 16:07:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc9c10f5349173ce49ce2ccaa00c99ecffb4c968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 16:43:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a454a868ae247c69065ebdc641a6c6d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba6138f3e44d4cbc811850675da12ed9 : SUCCESS in 1h 04m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8c67ed203c24f3eb88696dd955ab938 : SUCCESS in 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d7b8a22f3ae4c3daf8e09e0cd4033a0 : SUCCESS in 12m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/688ab0a396dd4680bbd92de0dad7cf6b : SUCCESS in 26m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/428b6f45a07c4ae3848fa1d8a7135351 : SUCCESS in 23m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585a2d1cd11b42db816ab8ead2e43318 : SUCCESS in 11m 07s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e087084aeec2438db8d56ccfade46c71 : SUCCESS in 25m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96d646ac5594478fbe912401723ae2b4 : SUCCESS in 1h 02m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/87d411e829d949f09359c4d12ee6b1df : SUCCESS in 59m 52s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6c8eadac085e45ec892b1fbb501fd20f : SUCCESS in 51m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a71e7d2bffa44528db6a13227694780 : SUCCESS in 1h 54m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3e5a95b396934fc0931d77e39d0071f8 : SUCCESS in 1h 18m 17s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/272d847fd9094fba9ade6885a0eb2c59 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b7b7c6833c0a2d574acdb8fa831b4d8481b7f024","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 16:58:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45eea0bb66c412ea22de7a0bf3ca52f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/417372d1a902498bb48937986987ae9c : SUCCESS in 48m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b83c9b1f9612409d9b411d99af42eb25 : SUCCESS in 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a073cc0c557d9e385f78f71bf5ddac0737909c4e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-10 15:58:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9e7e3c9d81591abc3d5bd60de1add39ececb743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 15:58:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94e6c0e1643d980cd9274db8e4cb703d16983069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 17:49:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ff58300247438399257addba81ac31\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8c80141868b45bd9ff448f93bf02b32 : SUCCESS in 1h 07m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19c411b716f4041a3e69c8b0b201f36 : SUCCESS in 14m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e45b291ffbc24e0d8658049743278909 : SUCCESS in 32m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2c3719f7b9e4496a116a281bd9fbb53 : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3968bf0e28d405685cfa555054c3054 : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4a393a84717405c99bc07a3919b7140 : SUCCESS in 1h 02m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea1ffce08efc431b8080da336167d9e6 : SUCCESS in 1h 13m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e0252a3d634c4891b9ed1e2398471d39 : SUCCESS in 1h 26m 30s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3aabd24d3d904a35ac391a56314ef46b : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"48b9b2bac2a63eac76ead9478fce51de0b0faf8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 17:49:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b59e6df5fd8bac880e5703b18b1663249f72bdb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 17:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c64d59ae1e874064aa7a082b99bbd508\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2943dfd37ee4a269cadb0bf4444f8dc : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bffa642b351f7fc3bb5cfb7a1bb2a3499c4b4506","revisions":{"bffa642b351f7fc3bb5cfb7a1bb2a3499c4b4506":{"kind":"REWORK","_number":1,"created":"2023-03-09 08:31:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/876945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/876945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/876945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/876945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/876945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/876945/1"}}},"commit":{"parents":[{"commit":"50dc9dd24377ddb04bcb7fc9be362894b6ed1490","subject":"Merge \"[sqlalchemy-20] The Session.begin.subtransactions flag is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50dc9dd24377ddb04bcb7fc9be362894b6ed1490"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-08 08:08:33.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-08 08:12:10.000000000","tz":60},"subject":"[OVN] Method ``get_port_qos`` should always return 2 values","message":"[OVN] Method ``get_port_qos`` should always return 2 values\n\nThe method ``neutron.agent.ovn.agent.ovsdb.get_port_qos`` should\nalways return 2 values, even when the LSP is not present. This\npatch:\n* fixes the output, returning 2 values\n* catches the exception raised when the LSP is not present in the\n  OVN NB database\n* returns the default max_kbps and min_kbps values (0, 0).\n\nCloses-Bug: #2009804\nChange-Id: I03ea78fcc5825e35c07dda1aa790769aa918dd8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bffa642b351f7fc3bb5cfb7a1bb2a3499c4b4506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bffa642b351f7fc3bb5cfb7a1bb2a3499c4b4506"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
