)]}'
{"id":"openstack%2Fneutron~877049","triplet_id":"openstack%2Fneutron~master~I3888d457d1c069c24eb8943fb54b9a9dc9f14819","project":"openstack/neutron","branch":"master","topic":"bug/2006949","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-19 20:02:18.000000000","reason":"Change was abandoned"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-04-19 20:02:18.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-03 14:21:59.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-04-05 14:52:29.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-04-19 20:02:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3888d457d1c069c24eb8943fb54b9a9dc9f14819","subject":"ovn: allow RA and NA for stateless SGs by default","status":"ABANDONED","created":"2023-03-09 23:43:19.000000000","updated":"2023-04-19 20:02:18.000000000","total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"a4f0fad60463c3d8d5ec773c21b01141c95cd447","_number":877049,"virtual_id_number":877049,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-30 09:26:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-29 23:00:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"disliked":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-04-03 20:58:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-04-03 14:21:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":-1,"date":"2023-04-03 18:52:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2023-04-05 14:52:29.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 23:51:22.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-09 23:51:22.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-09 23:51:22.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-10 00:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-10 01:46:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-10 02:10:02.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 20:58:43.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"bfd54d579c3a8d2debe19c3c153bc6750facb808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-09 23:43:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4750ab9715a20a89de03746fb41da162ba934c99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 00:34:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aaecd87e46c408492d4e9e9f7be5291\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04333681a1554a4cb0c65c64e40bc28c : SUCCESS in 48m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ef01a77868f47f3a410f4a0702c63f4 : SUCCESS in 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"089799c32d63fefa344f25d7e06d2091d2f0545b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 01:46:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643d17a3548242169f949b7ff7b068f2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21ee8538ab884614b453e65ef6479570 : SUCCESS in 1h 12m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9df77f8f87940ab8c375ffd7b0a47a8 : SUCCESS in 27m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b50405d8a62849bea806a6f4fcba3a77 : SUCCESS in 45m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e972eca3f94627b6b3074654d0cb5f : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce4cc7c3530746b8a11bdf44dc378308 : SUCCESS in 40m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6976cdb35fc44e93b450566eb91a74d4 : SUCCESS in 40m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca24ac270a344dc0aa3b9ca32e9d80dc : SUCCESS in 10m 01s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c8a2fc74e077495eb6b84f4e0a13fdad : SUCCESS in 37m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5166ff2e545a41f98a400d85e496c3c8 : SUCCESS in 59m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cacdce4e9c5c4f9eadd3503f699ccf46 : SUCCESS in 1h 22m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8566e4675d9d41e9b3dbe86c3cbfa885 : SUCCESS in 58m 43s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83fb99f0669e41e8abc5700f35c2f554 : SUCCESS in 1h 03m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ef3332c81894a4d8e3918d69e14d6ad : SUCCESS in 1h 40m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ad165fda8d44f86ba93a62326ebc9c1 : SUCCESS in 1h 03m 23s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a406720f18e04f108d9f92ab704688cd : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4541ebc996b452aa8e7bfad3fa1d049e74e86520","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-10 02:10:02.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/877049/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"47a9aaace87f9a3c51c0149ded1288cea5892350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-10 02:38:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4338fc5c991d66490e4ebd3c2ac49c84cd588ed0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-10 03:26:22.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/877049/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7a9d2323334cd152c6b70ef77081e5c4c82ccfca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 03:52:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c901208dbba4166be0def7ce1a74932\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1bd912c97a04c528fe4c0f334e370b8 : SUCCESS in 44m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb676776270e4b7d9ffce6d58fdc8db2 : SUCCESS in 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"064b94c153e046463e41b3039b5c120d9e4fc0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 04:55:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdd47a4716aa4d449d90c25dd10fd506\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/710fbf3810e343038604efac69cf4c4d : SUCCESS in 1h 21m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47a8e1476ec74a61bde0e6114de4941a : SUCCESS in 25m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/021c1355642c4e48a4757f0ac4d1f3a5 : SUCCESS in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b4c15383204451810441d8576504ad : SUCCESS in 12m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d5398fc5bf940f19dc0ebe1e568bfeb : SUCCESS in 26m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8870bfe300704014bfa84a554f7ae2c5 : SUCCESS in 31m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc1889a21bd4afbbbbb8334bcbfebf7 : SUCCESS in 12m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/248d995fc090461eb8683f6778c081c3 : SUCCESS in 23m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f50b0cbbb964a3f887dd53363022b8a : FAILURE in 54m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2234fb5182bf41a5b942a9678c1bbea7 : SUCCESS in 1h 31m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4b98e1bb57d94b0b809862d5d206140d : SUCCESS in 1h 12m 19s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/76096a51bbff4a19bd8fd06dbabdc45c : SUCCESS in 1h 09m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf4fa791a6174db88bf5529ac98ed60d : SUCCESS in 1h 51m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/343de3a151c5410ea9e4055a6e100aa2 : SUCCESS in 1h 09m 17s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/5eec6d1e032843b18f12c9a53041eb34 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b693c72bcaf86861a4f43fcc948657d5ad241ec7","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-14 15:41:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95c0e05e7d95c5dd1d4bcb0547542f8e12096ae4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 16:35:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ef0a3ec6994e1ca72c9ab6df16e2dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dade460ec3c84abfada3012e4200f51e : SUCCESS in 43m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7f15abb6058453d9b33d1750540d2de : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"926fa6d628968fbb12c897c4f6a4d5411f5c3eba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 18:01:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80bdde9f5c1e4a4ba7521906516ee609\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20e752f9bc3e4f6c93b394af09365a6d : SUCCESS in 1h 33m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a206f8214e044a2c9b9c3a1fc2a55b1f : SUCCESS in 37m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adc7538eb56d4539acd52ab3cd0293f3 : SUCCESS in 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aece75d181846d398dbd139ca2c687b : SUCCESS in 13m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0b981b285ff4295a4cba4433b409299 : SUCCESS in 25m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bac1c20324d428cabd836cb626e22c6 : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1216fff906ac4152a00a03dded3f048d : SUCCESS in 10m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/52301b99a3294b01824609d004cade9e : SUCCESS in 22m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b6317cd4819407a945546e26d14cab4 : SUCCESS in 54m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5321266beeb4e769735d59e8b2df91a : SUCCESS in 1h 44m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/baec60029170442ea0a3557559f86985 : SUCCESS in 57m 03s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f7a09486a21f4aa2967b57b4cbd308c7 : SUCCESS in 1h 04m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e29ca43f6875421bb423bd496adc4a2d : TIMED_OUT in 2h 13m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/876e4f9ac5454b0d94ea8a45c38f789e : SUCCESS in 1h 03m 44s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/21e5a78ca153493b8651496f60b77559 : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":2},{"id":"72aeee5b127289ece3e4def7fab75a1691add41d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-14 21:11:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"639253a5081d039dac393a01affef189d3027179","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 21:55:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f55a5772d8e42818ff2f3cd805b0b2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb12fbc6062d4146bf2a5f10490a1de9 : SUCCESS in 40m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f4c1367c6af42e1803ad806fc02a8c4 : SUCCESS in 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f72da6702aedfd04c18d09787ac08c6138374079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 23:18:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9299a294f34916a25d34878941a45a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49215937422f42c2ad90f056e6646933 : SUCCESS in 1h 06m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8f13d0f376a40e880e69f667f2d4020 : SUCCESS in 27m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c31145585904583a22af9fbde01a571 : SUCCESS in 52m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec22953a4b543f6bc5cacf4e9a4a427 : SUCCESS in 13m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8daac6c436e46f5975aeee83601de9d : SUCCESS in 27m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe0266b591c9434e9c2f7aaf31c813d8 : SUCCESS in 30m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b41131fbb184271ae0b6a90e9736f3b : SUCCESS in 11m 42s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8e5a0e7d8f3b4ab0b1e0463a2a70b610 : SUCCESS in 24m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7b3b59fa2ae4da2a98972261a02fbc2 : SUCCESS in 1h 01m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b41874d828347b898af7e825e7bf9d0 : SUCCESS in 1h 45m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b7177c729a1e4c43871352f712d8a516 : SUCCESS in 1h 24m 22s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/547fc9b94651452cb0b782b513753b4c : SUCCESS in 1h 10m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79bb78a9411e4fe3bfa9c1011a96ebf4 : SUCCESS in 1h 58m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f389b88f899a4ac0b1cc272faa488c9b : SUCCESS in 1h 19m 54s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fe10f3ccf2354b9bba30274bbf5dea62 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b3db42b0878cf37524bdd70a531e34d4fb5c320d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-15 15:09:46.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68e18331cdda6fb43ffd3d97c86655022191bfe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-29 20:39:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87803420a56a7df6b5d55280339c4494aa2dc5be","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-29 20:41:34.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fdfc76e8119c4e8724e572878134727532f4246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-03-29 20:58:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d404df0fefa6fd3fc27a63fb916ecf1631b8a08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 21:47:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84edf28882d240b5befd7c2bdbb81b39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/688010c0d2854e51b3c0f5c9820e3160 : SUCCESS in 45m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2da8ec29f3f4a648b54c0d3726c1df4 : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59b7d2ee7ba0463880c655c864e32a9a : SUCCESS in 46m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b232b2bfcaf3164542f93cc3e4c8918625b491a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 23:00:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95607a2781d741e99d649ca372602ee2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4033232d49d43358792653bdf418f92 : SUCCESS in 1h 12m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54abc1ef40bd4685b21cd324e3a383ff : SUCCESS in 36m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c074cff4bf7f4b82887a04610e6b07fc : SUCCESS in 36m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf184fafe7b74101b269c43b8ebc4a5e : SUCCESS in 14m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d84cbe0112ed4d9b8fae626f62bfa2d6 : SUCCESS in 25m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/412c7a990e3b44a2923b26ece607f798 : SUCCESS in 29m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd514647a66642a4a38a5a911d05603b : SUCCESS in 23m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb5a9ccc106455c96d9efa0de3c5516 : SUCCESS in 10m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c114e3c616314bcc874b6c5f1374d55d : SUCCESS in 28m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f21c7ee0714747b3bae95c77795b42cb : SUCCESS in 40m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96929498bf634f4ea6a77407fd252c28 : SUCCESS in 1h 45m 31s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c749a4817329426980439616dfa26506 : FAILURE in 19m 37s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/018fbf0a45fa4653b16f0fbd0577c2b8 : SUCCESS in 44m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7104f85dae26409fb96c3bfd5a8ae884 : SUCCESS in 1h 40m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e91dc0539cd44229a3ef91461565eeea : SUCCESS in 1h 06m 04s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e4fc799fff084be9a3c96de967d121c3 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":4},{"id":"d1a8bffd3760c4d57c105687aebe682ecefbf7cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-30 08:06:17.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/877049/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"854bf4b9db2a79137ea39305dcb2db2a12d6ea40","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-30 09:26:11.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/877049/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"55fe91e7c902c6aca0331e3f9135ed7c53920c84","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-03 14:21:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ed93f0f5be6a976e029af9f8907ee643475a8fbe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-03 15:29:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a6ba66ceb3f0b01cfb1bfcb41a44f22718076a2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 15:53:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a8a385d4ec4ab4be3e6994f91e971e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6b2b79730784a95bac5968c4a0ef0a3 : SUCCESS in 21m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c477ad881898446aae69147ed589527b : SUCCESS in 22m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8df9e4c5de2245889ff5f225ce432ce8 : SUCCESS in 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a3f8f0903293102a4319b053387e40a6c589198b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:44:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a4d13d39a1419d87085bb09540ee2d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca07af53ec5d422c926ba5939104fdc5 : SUCCESS in 1h 17m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9830817c84c0428c9510d0a1db4aa868 : SUCCESS in 28m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc8bdd97550146acbbf0a2c477140689 : SUCCESS in 44m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebcdb8501d3f4875ab32048962c219b3 : SUCCESS in 14m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1589b4fb44984673acac6a406b6ba165 : SUCCESS in 27m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40bebbcc94d546b5913621db49560565 : SUCCESS in 25m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cb000f793cd4616b5f0fc3ed3128def : SUCCESS in 29m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eeb53a74b084562bee3f0d67ce826f2 : SUCCESS in 10m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f32b32c3fd9b4083bef5b903a2cbfa8c : SUCCESS in 24m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95ba8378da0b4a019866478027496035 : SUCCESS in 41m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/685f0fb9c1314b73809855b16b474104 : SUCCESS in 2h 12m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/731fd82e7fa343c9a6da9ef44d986555 : FAILURE in 22m 28s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e3d59f39041e47119068a25966b283a8 : SUCCESS in 1h 01m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33fc1dd86e7b405a93f065460f41a55c : SUCCESS in 2h 04m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/774f831c7c444dbb895db2d93ec19e63 : SUCCESS in 1h 20m 23s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c7a2080c6a314001a254ae90a506c387 : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":4},{"id":"67b0e73456a8ec3ccd1d00a793e215b37ce3c26e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-03 18:52:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"173f7c0d9261d5ff03ca1710f84c7ac8575c0873","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-04-03 20:58:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3d4d928e0ca8fcbb9f08709e94d65ab9ce30ee91","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-04-05 14:52:29.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ad38763d2b51ace9bebb1dcd365a0e87a07fb421","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-04-05 14:55:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a4f0fad60463c3d8d5ec773c21b01141c95cd447","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-04-19 20:02:18.000000000","message":"Abandoned\n\nThis patch is not needed because we\u0027ll enable the protocols in core OVN controller code, not neutron.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8f8f32f0c5276ee4cd419f00aa512e68da1a760b","revisions":{"0504fc935c0ccc483d35f81b4b9ab2ddf1cdbbdd":{"kind":"REWORK","_number":1,"created":"2023-03-09 23:43:19.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/877049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/877049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/877049/1"}}},"commit":{"parents":[{"commit":"4c80527d8292141a2b4b8f28f270fad07d487f62","subject":"ovn: allow returning metadata packets for stateless security groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c80527d8292141a2b4b8f28f270fad07d487f62"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-09 23:24:42.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-09 23:36:52.000000000","tz":0},"subject":"ovn: allow RA and NA for stateless SGs by default","message":"ovn: allow RA and NA for stateless SGs by default\n\nWith this patch, stateless SGs will allow service ipv6 traffic by\ndefault. This is needed to allow VMs to configure their SLAAC and DHCPv6\naddresses.\n\nCloses-Bug: #2006949\nChange-Id: I3888d457d1c069c24eb8943fb54b9a9dc9f14819\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0504fc935c0ccc483d35f81b4b9ab2ddf1cdbbdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0504fc935c0ccc483d35f81b4b9ab2ddf1cdbbdd"}]},"branch":"refs/heads/master"},"dcd69797eb8808bb895f5eac9c3b82b6bf4cd193":{"kind":"REWORK","_number":2,"created":"2023-03-10 02:38:15.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/877049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/877049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/877049/2"}}},"commit":{"parents":[{"commit":"4c80527d8292141a2b4b8f28f270fad07d487f62","subject":"ovn: allow returning metadata packets for stateless security groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c80527d8292141a2b4b8f28f270fad07d487f62"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-09 23:24:42.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-10 02:38:04.000000000","tz":0},"subject":"ovn: allow RA and NA for stateless SGs by default","message":"ovn: allow RA and NA for stateless SGs by default\n\nWith this patch, stateless SGs will allow service ipv6 traffic by\ndefault. This is needed to allow VMs to configure their SLAAC and DHCPv6\naddresses.\n\nCloses-Bug: #2006949\nChange-Id: I3888d457d1c069c24eb8943fb54b9a9dc9f14819\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcd69797eb8808bb895f5eac9c3b82b6bf4cd193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcd69797eb8808bb895f5eac9c3b82b6bf4cd193"}]},"branch":"refs/heads/master"},"44efd4a14d2cd9fe7f5aab294fbdfe6e1f54d3a2":{"kind":"REWORK","_number":3,"created":"2023-03-29 20:39:28.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/877049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/877049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/877049/3"}}},"commit":{"parents":[{"commit":"e245cb6b9d6d51b19518ba987c9b359e44c72ede","subject":"ovn: store neutron:revision_number for Port_Groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e245cb6b9d6d51b19518ba987c9b359e44c72ede"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-29 20:38:49.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-29 20:39:03.000000000","tz":0},"subject":"ovn: allow RA and NA for stateless SGs by default","message":"ovn: allow RA and NA for stateless SGs by default\n\nWith this patch, stateless SGs will allow service ipv6 traffic by\ndefault. This is needed to allow VMs to configure their SLAAC and DHCPv6\naddresses.\n\nCloses-Bug: #2006949\nChange-Id: I3888d457d1c069c24eb8943fb54b9a9dc9f14819\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44efd4a14d2cd9fe7f5aab294fbdfe6e1f54d3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44efd4a14d2cd9fe7f5aab294fbdfe6e1f54d3a2"}]},"branch":"refs/heads/master"},"8f8f32f0c5276ee4cd419f00aa512e68da1a760b":{"kind":"REWORK","_number":4,"created":"2023-03-29 20:58:31.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/49/877049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/877049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/877049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/877049/4"}}},"commit":{"parents":[{"commit":"d103d9d983bb03ce44fa3da7458d46bffdcc49ea","subject":"ovn: store neutron:revision_number for Port_Groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d103d9d983bb03ce44fa3da7458d46bffdcc49ea"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-29 20:38:49.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-03-29 20:53:19.000000000","tz":0},"subject":"ovn: allow RA and NA for stateless SGs by default","message":"ovn: allow RA and NA for stateless SGs by default\n\nWith this patch, stateless SGs will allow service ipv6 traffic by\ndefault. This is needed to allow VMs to configure their SLAAC and DHCPv6\naddresses.\n\nCloses-Bug: #2006949\nChange-Id: I3888d457d1c069c24eb8943fb54b9a9dc9f14819\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f8f32f0c5276ee4cd419f00aa512e68da1a760b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f8f32f0c5276ee4cd419f00aa512e68da1a760b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
