)]}'
{"id":"openstack%2Fneutron~877133","triplet_id":"openstack%2Fneutron~master~I0e55763e873670b5c008ba1f55d140c810a16df1","project":"openstack/neutron","branch":"master","topic":"dnm-test-fullstack","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-03-17 21:13:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0e55763e873670b5c008ba1f55d140c810a16df1","subject":"DNM - TESTING PATCH: functional and fullstack CI jobs","status":"ABANDONED","created":"2023-03-10 21:09:31.000000000","updated":"2023-03-17 21:13:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"12b5fe28baf86478692db3e5bfdec56f99d77f06","_number":877133,"virtual_id_number":877133,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-10 23:14:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 23:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02b18376e6d317696a13cf160dbf03c02f22a676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-10 21:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7113edd08ccfc56597b6d18fc2757f98bcab2a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 23:14:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/356b558a52354619a46426f5d181d407\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/866cc8a6ddf84b38874d3a676873816d : SUCCESS in 1h 35m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/994f20fa9bcd4c9080f492822bfbfb57 : SUCCESS in 1h 46m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33d1446eac024c4f8752494a69a3db10 : SUCCESS in 1h 43m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a3b8b53f2af2435790b44c641378fd5c : SUCCESS in 1h 29m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72c8c2d1e3d9416fb00b5f90134e3e9c : SUCCESS in 1h 32m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdc0aad144ec4ddd8de9e463940218e0 : SUCCESS in 29m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5a26e9fa7b41aeb330abe35f26ada0 : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c5639c94a8459b8b30c811865874cd : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4e75782bf644c68894b6cb61f67f483 : SUCCESS in 55m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8df5c03cce840b2ba360de7f9dfc45c : SUCCESS in 2h 03m 22s\n- neutron-functional-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/78f38e117fd5477fa5643598c707df19 : SUCCESS in 53m 40s\n- neutron-functional-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/a7f620fb40124568b1447e05e67a13f8 : FAILURE in 38m 44s\n- neutron-functional-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/a358c9fcaa19426ca63e6810692bca63 : SUCCESS in 1h 02m 28s\n- neutron-functional-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/1bcf2ea2617a4af7b799027ff52e4f7e : FAILURE in 51m 45s\n- neutron-functional-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/90231cfbfdf747db966a7a641e9b7564 : SUCCESS in 44m 20s\n- neutron-functional-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/40d1c920ab0c4f258064c04bfa23b2c1 : SUCCESS in 45m 35s\n- neutron-functional-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/8752696e1c8c41f0943bd69f9d43c773 : SUCCESS in 56m 22s\n- neutron-functional-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/336e13e4b86540128a1b44a785846902 : SUCCESS in 37m 00s\n- neutron-functional-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/49913cb4877f4bd2bab63ff71e58b356 : SUCCESS in 46m 54s\n- neutron-functional-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/13c51746805f4163a20688df12d17592 : SUCCESS in 43m 41s\n- neutron-fullstack-with-uwsgi-1 https://zuul.opendev.org/t/openstack/build/fe4ec6704e7a44e8814c8b0ab5dc28fc : FAILURE in 1h 31m 24s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/19cbd962317a4ae1a11ffee54749b2eb : SUCCESS in 1h 30m 10s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/21ee4d89b9de4717a75a097f5ad73a09 : SUCCESS in 1h 45m 35s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/89e243964f2043e798424e576bcec567 : FAILURE in 1h 26m 33s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/3ef7f0d1082d4fa49e28eaea1c083be0 : SUCCESS in 1h 45m 19s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/15ff724190c8425fb6661dfe1bf59103 : SUCCESS in 1h 40m 45s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/78c0929da58a4655b90bfcce23c8b30d : SUCCESS in 1h 44m 00s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/eac853fd864f4439a1a5b88317563c0a : SUCCESS in 1h 44m 59s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/8820d59bcd764b30957e93ed4fb367ad : SUCCESS in 1h 52m 44s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/3c5714ee23574a80b96f2d921be8fe2d : FAILURE in 1h 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"12b5fe28baf86478692db3e5bfdec56f99d77f06","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-17 21:13:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"105772c9c2c1290e41a64fb11e33b5b74d03249b","revisions":{"105772c9c2c1290e41a64fb11e33b5b74d03249b":{"kind":"REWORK","_number":1,"created":"2023-03-10 21:09:31.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/877133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877133/1"}}},"commit":{"parents":[{"commit":"fff1f8c20f82b53b49b83693a0f031c645f7b9f2","subject":"Suppress IPv6 metadata DAD failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fff1f8c20f82b53b49b83693a0f031c645f7b9f2"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-03-10 21:08:32.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-03-10 21:08:32.000000000","tz":-300},"subject":"DNM - TESTING PATCH: functional and fullstack CI jobs","message":"DNM - TESTING PATCH: functional and fullstack CI jobs\n\nPatch tested:\n  https://review.opendev.org/c/openstack/neutron/+/876566\n\nChange-Id: I0e55763e873670b5c008ba1f55d140c810a16df1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/105772c9c2c1290e41a64fb11e33b5b74d03249b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/105772c9c2c1290e41a64fb11e33b5b74d03249b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
