)]}'
{"id":"openstack%2Fneutron~877253","triplet_id":"openstack%2Fneutron~stable%2Fzed~Ia05416df88904e864d4fc9760ffcdc97a4651f9f","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-13 18:17:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-14 12:09:32.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-03-14 08:30:06.000000000","reason":"\u003cGERRIT_ACCOUNT_23567\u003e replied on the change","reason_account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}}},"hashtags":[],"change_id":"Ia05416df88904e864d4fc9760ffcdc97a4651f9f","subject":"Ensure vlan network traffic is not centralized","status":"MERGED","created":"2023-03-13 13:08:35.000000000","updated":"2023-03-24 15:50:15.000000000","submitted":"2023-03-14 12:09:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877253","cherry_pick_of_change":875644,"cherry_pick_of_patch_set":2,"meta_rev_id":"6c6ecd4d222d8e9c30a563782667dd2df0a235b4","_number":877253,"virtual_id_number":877253,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-14 12:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-13 18:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-13 21:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-13 21:35:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 14:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-13 14:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 14:49:35.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-13 14:49:35.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-13 21:35:05.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"50f5880a2b9d66132c398788b611894e995f4601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-13 13:08:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c878cf3d42961561a42ca89cc850ee897d7ec07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 14:27:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e04e01e16f4ee4915dc43420d20a1f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c43abd2b12874a9a94705a3bdb1697ba : FAILURE in 1h 13m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7a4ee912a5b49029c70aa942705dd04 : FAILURE in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08ffe903592283636f0d9e1d326c62981d089375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 14:44:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779d78c1657648719f583be19a4473f8\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/13610410711f4e619634b9914a077870 : SUCCESS in 1h 34m 15s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/eff557efa7e447a1baeead4f8a5b7322 : SUCCESS in 31m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5060783066ab4184a4065d10c3c8f861 : SUCCESS in 33m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8de0d87fcc0488d9ae4aa868a4364a7 : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/391a84f0f7fc470caa326c3341d098a9 : SUCCESS in 23m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/490b0ed6753a4ab49ac3877ce8f2c605 : SUCCESS in 29m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d67a57b9fa1e4d80bcd39370f706cbdf : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737be254f6c6441ca84367582676b644 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec94d595336c4a98a88f09bba9b479c5 : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7667954111e4628ab13d235a575e9cd : SUCCESS in 39m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d7527603b20450e945b5b616cd83d45 : SUCCESS in 1h 25m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6faf1775e7b4198b375f9ba1542120b : SUCCESS in 1h 00m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e5c774f2fb54e38b1e99f004f0da457 : SUCCESS in 58m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f638796d34b7412f68e1ae96d302e8c63cec104","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-13 18:17:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8101b37bb72f41ba4138435bdd05a7a847b17dfc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-13 21:35:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"588a39eb92bc36c1f40e8504117434bd55442afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 21:35:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85a715e999beac9fa501daeaeec3574292c4b186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 23:09:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a23e2d8383014960b17aba5ce082ae81\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6ea18fe80c0b4f42bf59a4e6451be404 : FAILURE in 1h 25m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace8b04ad06d4b90ae94d462fc01815e : SUCCESS in 12m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2837e44cc0a4ff2a8f1bc6bd75b0584 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f7e7e211b784259bcabd013ad16f5db : SUCCESS in 26m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bbe12e91b74326bf24a765593829ab : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a6d70a51ae429094450d8d1f3645af : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c61297192ecf4653a146c7564395fc18 : SUCCESS in 52m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f5a1b3e1a9c4624bfda07cba42395c2 : SUCCESS in 1h 32m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b640213d8f894226bc48d7fd727625ee : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"20a2b18c711da6c7288058d8543473878a70b354","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-14 06:19:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7a10d29043d792dd78b481e7df99ce4a727247e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 07:12:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15badfb9748b434d8113755216cc894b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6456ce59c7594ed69658fe5c05094926 : SUCCESS in 50m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d14a663bdd084648bbd57b6867553b34 : SUCCESS in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d84a6f3155c7386b59add3c19ee64d1ed669d368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 08:06:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b33a61cc8684fb0ac0064e822fb6af1\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/72b2ce8c4189433eb0887adda98d6a5c : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/946337dda6a0483b94326db8f15fb5ae : SUCCESS in 28m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36feba86dc1460384c3784624523ef4 : SUCCESS in 45m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cb236431b674da2bea05f9f0ee75be4 : SUCCESS in 12m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c2d9969341a48339b06b5566db4c2ef : SUCCESS in 26m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45f84d61f4334383afa3c6879b7b5028 : SUCCESS in 25m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd1b65d85d564a8793aedc7332281511 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c2c3ba746b4f4b9afe22be3d65e387 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb60bb95ce59413ab48eefa49d30019a : SUCCESS in 8m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b391fae476044bcb837f588dc9e6444 : FAILURE in 41m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7feffd46ac784ed3b3e544b2a4553833 : SUCCESS in 1h 45m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cb5db64deb046908ad56aa7936095ef : FAILURE in 56m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fdf4219482654b979e0fe6bb3c961f4e : SUCCESS in 1h 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cad59d1005ef2087e3a821dab7ce9d403a17767","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-14 08:30:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"813d5de54dbb3ad18b083eb27e4cd14f12affbca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 09:16:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fdce29e9710421ca00304ca364461c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d876a104ee094d8bb1f907722b2330a0 : SUCCESS in 44m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95161ebb5a564f3590e1c1e7edeff545 : SUCCESS in 44m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e55564f3da0a014957a8e03bdfeac73174b7bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 10:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43ce9159abc465c956e55df4138cf31\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c5d1deef50204d639032a90bf436b8fc : SUCCESS in 1h 51m 24s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/5c5b8d7e019244c7a250d171b989864f : SUCCESS in 25m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0ca1c32b7f347d18916e923f54cb84e : SUCCESS in 26m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2405a212a41a48e18e3b660756804696 : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da0fb14ac694a73b40ad29d6481bbfd : SUCCESS in 22m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64bf9a9c218e423790072094b6685651 : SUCCESS in 19m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa9ed2d558ae489db5ceceb3e0ec5ed5 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffe5e0ed23646068d34e93825cac012 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c49b98dd8472445c857d7e31be67bb3c : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2abab3e2908421689c6953233bc1247 : SUCCESS in 40m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/390470c986ec469e98bf0d686af87673 : SUCCESS in 1h 57m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/663de7fa197d41e4a35a98b1e106be23 : SUCCESS in 49m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd8c7f4c0c2e4eaa8a9ec4e544bb60d1 : SUCCESS in 1h 08m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7abd94f37bdc9c402d1c66805eba2788c94ef11a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 10:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89f7c052fbb368aa732d62726f7dba0e4405e1ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 12:09:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991b1608e36f4a5988cfd9eb1f93d9b7\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/32b2127c80614984941676a8c5f49afc : SUCCESS in 1h 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f52aca48d74736b448783c5abd7574 : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2abc8f59726645b9b2ef708cd9b509ac : SUCCESS in 28m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48f26157ebd441a8b43776c46ae675a1 : SUCCESS in 29m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b54f66d09a0450b80d5524ed262775b : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c3947ce73f94aed9274b262cac86b1c : SUCCESS in 11m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3e11dacef36471ba92509171bbf6791 : SUCCESS in 43m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3785df6bf9c4441ab1945ba753b2ce40 : SUCCESS in 1h 35m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b18fde9d7c2c4b20b747bec9d74bfa40 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":1},{"id":"779ee1c3f3de83a658f9689d492da0a2a26b2266","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 12:09:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fca307d8f865e3c3e867b6c6678b67fde569db2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 12:10:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62f4fa401df5406e8cfce6a438c7b64b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec0ca35fe3be4209aed086cb4bf9518d : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c68e892e9444f24b454d061fe89ea19 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"6c6ecd4d222d8e9c30a563782667dd2df0a235b4","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-24 15:50:15.000000000","message":"Created a revert of this change as I9daa7fcc0c1edc9c3d162228153be187dea8ae52","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a415bb4c98fa0a7b30f36cf9d2811576181fadf5","revisions":{"a415bb4c98fa0a7b30f36cf9d2811576181fadf5":{"kind":"REWORK","_number":1,"created":"2023-03-13 13:08:35.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/53/877253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/877253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/877253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/877253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/877253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/877253/1"}}},"commit":{"parents":[{"commit":"db948928337a1b165022f73a73c1905c09562d7c","subject":"Merge \"Rename setup-mtu-t1 argument because it is misleading\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db948928337a1b165022f73a73c1905c09562d7c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-02 10:04:11.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-13 13:08:35.000000000","tz":0},"subject":"Ensure vlan network traffic is not centralized","message":"Ensure vlan network traffic is not centralized\n\nThis patch partly reverts the workaround introduced at [1].\n\nIn patch [1] the reside-on-redirect-chassis was forced for vlan provider\nnetworks to force centralized but not tunneled traffic for those\nnetwork. In this patch we are making use of the \"redirect-type\" flag\ninstead so that the traffic can be distributed and still not tunneled.\nThis flag needs to be set on the router gateway port (port connecting\nthe router to the external network) unlike the previous one that was set\non the router interface port (port connecting the (vlan) internal\nnetwork to the router). In this patch we are setting it on all ovn\ngateway ports if DVR is enabled, as:\n- It is needed for vlan (provider) network to have their traffic\n  distributed instead of tunneled to the controller where the cr-lrp is\n  associated\n- It is not having any effect on the geneve tenant networks as it only\n  applies to network that has a localnet port associated to them.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/871252\n\nCloses-Bug: #2003455\nChange-Id: Ia05416df88904e864d4fc9760ffcdc97a4651f9f\n(cherry picked from commit 8e3bddbf8bc5296143357d3df7fd27aeb9bb9bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a415bb4c98fa0a7b30f36cf9d2811576181fadf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a415bb4c98fa0a7b30f36cf9d2811576181fadf5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
