)]}'
{"id":"openstack%2Fneutron~877255","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ia05416df88904e864d4fc9760ffcdc97a4651f9f","project":"openstack/neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-13 18:16:30.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-14 00:53:19.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-03-14 00:53:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia05416df88904e864d4fc9760ffcdc97a4651f9f","subject":"Ensure vlan network traffic is not centralized","status":"MERGED","created":"2023-03-13 13:09:08.000000000","updated":"2023-03-24 15:51:28.000000000","submitted":"2023-03-14 00:53:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877255","cherry_pick_of_change":875644,"cherry_pick_of_patch_set":2,"meta_rev_id":"624d412399f90a69969e1b42a6354ec9d4755221","_number":877255,"virtual_id_number":877255,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-14 00:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-13 18:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-13 21:35:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-13 21:35:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 13:56:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-13 14:50:00.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-13 14:50:00.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-13 14:55:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 21:35:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"f224771b5609bf9fc30c2e51894a4aa215167bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-13 13:09:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cf0f1361e10251746a4d177456476ded590b4766","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 13:56:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0272001bad74266aff4d34543c855de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56e8f28e49a04bbf8facf4d4ee552246 : SUCCESS in 45m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b79dde32e6994233ad5e90714252ab2c : SUCCESS in 45m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e404627a631d2ec24e4bcad2a107f0d4e00c8d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 14:55:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc8fb4d84b24ed6bfef9310b1a7f774\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/a38ad3ec8f464204826b7b61b1c400bd : SUCCESS in 42m 42s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/ef4dfdf49d45407b9f8ac80b426658b6 : SUCCESS in 1h 13m 00s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/f772e7e596fa45b9a2fe2e9b644a23b1 : SUCCESS in 38m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afd3bee4fc97423ea8a1447a9f0a04fc : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2763fc161c143ad943371cf0b6839b1 : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7540c3df1d064ee69ae1146dd93bc779 : SUCCESS in 34m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1298da6d0594fda817f6a4c0c5c15fa : SUCCESS in 29m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10fc6630b9444e3286bbd1ba1468e07c : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c898e7da93954d9f853b2f23b5e59d1b : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c67fe2899674a6191ffd5ac81829a98 : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aeaa9020a5524d8b88d2e0f838b68e07 : SUCCESS in 47m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ee9ac063d9845339b4efbcfd3021722 : SUCCESS in 1h 35m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/808e7b15927e4cb7b5bb75b60bb914a9 : SUCCESS in 59m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab004282acb949239059eb5923365ce0 : SUCCESS in 1h 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5706ea2e83584626e5da20fbcad0f3691bed119b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-13 18:16:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27eb2f2f66566e1ba3e7b7973ff2a185ab035b06","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-13 21:35:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0567b5bb4287cdce7dd366ea381c2fe4f07b9b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 21:35:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9bf75a6b801bc1fee32d0866b429719c96cc5c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 00:53:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca97dd8194c4159923cb9d7832d0127\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/3421958a92a8433980209631c6002507 : SUCCESS in 53m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1c71664a1048a6b3b215ea98672fd4 : SUCCESS in 15m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b5eb58a58b4dd7b7ae990f37f86538 : SUCCESS in 29m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b541a97e9dc04b10a2fe47fd614af490 : SUCCESS in 25m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7869952b6924c5596636539ac61ea54 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c93c734f1b48288510d634cd824e6c : SUCCESS in 11m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7589e79c5c1548e4b87bcda8e0651e4d : SUCCESS in 49m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3b86751e233474492b63fb6b10b1b99 : SUCCESS in 1h 04m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03f8d64a97db48e18798d0765453f9d3 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b5cbd2fddd2f2a90ecc084eab9083d060cb4a1c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 00:53:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d8ebea24ee5a8820733bfc8f26723b013e8a2ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 00:54:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02521b3ab2a1470e90fa23f959724f3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bf4d9b5cfc4e24bafea7340b9e9796 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d35030983baf49a88184b0f07eee9c6c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"624d412399f90a69969e1b42a6354ec9d4755221","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-24 15:51:28.000000000","message":"Created a revert of this change as I60f495747f87d27d9c5668d9c91f64ecc159bd69","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6","revisions":{"fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6":{"kind":"REWORK","_number":1,"created":"2023-03-13 13:09:08.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/55/877255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/877255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/877255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/877255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/877255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/877255/1"}}},"commit":{"parents":[{"commit":"93fddc453cd79b2cf09b9e9b853e62d655b9bc5b","subject":"Rename setup-mtu-t1 argument because it is misleading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93fddc453cd79b2cf09b9e9b853e62d655b9bc5b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-02 10:04:11.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-13 13:09:08.000000000","tz":0},"subject":"Ensure vlan network traffic is not centralized","message":"Ensure vlan network traffic is not centralized\n\nThis patch partly reverts the workaround introduced at [1].\n\nIn patch [1] the reside-on-redirect-chassis was forced for vlan provider\nnetworks to force centralized but not tunneled traffic for those\nnetwork. In this patch we are making use of the \"redirect-type\" flag\ninstead so that the traffic can be distributed and still not tunneled.\nThis flag needs to be set on the router gateway port (port connecting\nthe router to the external network) unlike the previous one that was set\non the router interface port (port connecting the (vlan) internal\nnetwork to the router). In this patch we are setting it on all ovn\ngateway ports if DVR is enabled, as:\n- It is needed for vlan (provider) network to have their traffic\n  distributed instead of tunneled to the controller where the cr-lrp is\n  associated\n- It is not having any effect on the geneve tenant networks as it only\n  applies to network that has a localnet port associated to them.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/871252\n\nCloses-Bug: #2003455\nChange-Id: Ia05416df88904e864d4fc9760ffcdc97a4651f9f\n(cherry picked from commit 8e3bddbf8bc5296143357d3df7fd27aeb9bb9bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
