)]}'
{"id":"openstack%2Fneutron~877484","triplet_id":"openstack%2Fneutron~stable%2Fzed~I12fa0764cd0ff509f1859b61060d64cc5a54a7b9","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2000078","attention_set":{},"removed_from_attention_set":{"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-03-16 18:06:09.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-03-16 18:06:09.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-03-16 18:06:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12fa0764cd0ff509f1859b61060d64cc5a54a7b9","subject":"Delete the PB level registers when deleting the duplicated PB","status":"MERGED","created":"2023-03-15 14:39:37.000000000","updated":"2023-03-16 18:07:22.000000000","submitted":"2023-03-16 18:06:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877484-bug/2000078","meta_rev_id":"902efa42ca621de7e8515dad8c359eb07f4a280d","_number":877484,"virtual_id_number":877484,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 18:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-15 18:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-16 16:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-16 16:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 14:39:37.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-03-15 15:22:40.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-03-15 15:22:40.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-03-15 15:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 16:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-15 18:19:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-16 16:20:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a29dd8f2c69e8a4ce5710835b36ed3e781ff1b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-03-15 14:39:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9317137959246329bf04834fa065676f13a56704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 15:30:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6168821fdaf4c4085503948ae4a6e7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/671e1b7e121344f88bbd14a9a7b0f68b : SUCCESS in 46m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2521e6c7f1a649d4881f4941de6ef2a5 : SUCCESS in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2c6e9dacdf30ab63f8277a9472a4787b539d084","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 15:33:15.000000000","message":"Topic set to bug/2000078","accounts_in_message":[],"_revision_number":1},{"id":"15db047d0f8b40d5a9a780d87b6ceda859eceaff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 16:57:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15657475b8984e2b9c86dbf8c9772a86\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/61fd6c6c40f64c7a9eda0ef4a8a41805 : SUCCESS in 1h 49m 25s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/58e5a6cbd8194d058cfbee88a3a004e4 : FAILURE in 2h 15m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/9a1cc4f6016d4a88975fd8a1bbf6f0fc : SUCCESS in 1h 46m 20s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c5b7c25a32684db4bf2dd47ab501a081 : SUCCESS in 2h 02m 59s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/9da4b7a5f573473ba0214870fa7bba85 : SUCCESS in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/880c2e3870304e108344cffdb38b2cb0 : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca7818776d004215b2c5ac067e5b9f10 : SUCCESS in 12m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58e1179d974145438e525a769de2b92b : SUCCESS in 24m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/614856343e714a95bf3307f8b58ab3af : SUCCESS in 23m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc9f796602fd4349bf3555f2f2586aa9 : SUCCESS in 24m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafe8b13a4724f398dbff962c541712f : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/68de95b9f2804f2abd5ea0d9760c3912 : SUCCESS in 57m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b6b01b2fee4477a8893de4359c97070 : SUCCESS in 1h 31m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d93a0e7e59794216ad7a9f0906be1296 : SUCCESS in 1h 35m 24s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7869ca53894d479bbbf01f64e1f6e5ee : SUCCESS in 57m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/353cd50fdfd54fb380c965879c3fcb2a : SUCCESS in 59m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5a859e8ac25f4037a314cfff9dfe7f23 : SUCCESS in 1h 18m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cf8f09535801480a9c5f9ad0b2517138 : SUCCESS in 1h 16m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b8e6c85b1bd49d5bc68e5622981f2a6 : SUCCESS in 1h 04m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c5a678ac6b741d48d09cc467960d6b2 : SUCCESS in 1h 08m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d60cbf687805e8494f7b7ab921ad7c33f2cd091","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 18:19:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed309582d11a9d571040bae78a0920cc0c77ecf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 18:44:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7d307a35ff4a70b023130970c6a7b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0df4aea7126d43089be9d3eab8deec35 : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2b234398a834e8897627f47851ee34c : SUCCESS in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7676556871d5663d0843a05ed02833ca0156298f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 20:07:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bded6191793454280cf339d262e11d7\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/76b30b6bf18d4ad7b4b545c284dbbbe6 : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/989a814266cb48648a3a0c69db965905 : SUCCESS in 1h 34m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/f139ffd9f9684c608f8022345ac2344c : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/2c6a4b902bb6452cbde6c2917e2550c3 : SUCCESS in 1h 45m 54s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/1e5373fb5486459dac37337f34656b90 : SUCCESS in 31m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9eaaa6850e249288b4c9fac68fbf98c : SUCCESS in 39m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4580b8b77753449fbc5458c2b805e1a4 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0a30cccf87240f3942c9e19383020bc : SUCCESS in 21m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0df6ad8c0b54491a925126d5690d9e31 : SUCCESS in 23m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de2a1a57fc47424b92cc692b6ac79207 : SUCCESS in 23m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2ee17a56fe4a1a918ba92ec534fd94 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8078f54795c54ed5a5547671d4575960 : SUCCESS in 50m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d60f330a80ff48258c954d92691f44af : SUCCESS in 1h 37m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ed490845e6074312add48d0631ce4d51 : SUCCESS in 1h 21m 25s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/63fd853c841d49a785851372cbe6edb4 : SUCCESS in 1h 01m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b14a7b7a59d4b32aed290882106d1b0 : SUCCESS in 1h 14m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/61ffd50d4fd54e90b7b62d6c7dd3f090 : SUCCESS in 1h 30m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cc8b3a7b080b42a68fb0d43db17a42c4 : SUCCESS in 1h 15m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2900b1fb4b50414687744824a22d210d : SUCCESS in 1h 14m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c6f017d08554c4daefdd8b580d5749b : SUCCESS in 1h 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3acdb23a0875774cbc604b7a2b25ba5e784040ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-16 16:20:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dd58392dc18028eb736aa87ded19885c62adb6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 16:20:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adb5c8eaa4a4f0fa753dce7f1a7914af454f8731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 18:06:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/211e6a3cfd364260ace59a6454e6f83d\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/41f18957d9ae4a648b6c816fee3d6596 : SUCCESS in 1h 22m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58918a3cbdb241868c5409f7d70b305c : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbcdb6ae47bc46cb9990af1d2ba259a3 : SUCCESS in 26m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4906a3401c6c4b0088a047585bae8489 : SUCCESS in 27m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420ba2a5b329420ea6b1a1faa22b77bf : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/895930d74d42461ca412aca63692c5aa : SUCCESS in 52m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f6a09fc89e04c29824eda125e309202 : SUCCESS in 1h 40m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f137038501a44a3cace3095df0eb166b : SUCCESS in 1h 27m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a51090101fd3486da0674a09e6feea72 : SUCCESS in 1h 04m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8af15082c5824dada0227c6f807648c8 : SUCCESS in 1h 00m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acfc3ccb61bf417fb503a9bfd58940b4 : SUCCESS in 1h 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dc0b1e080bf502f570316098e704bedb6c697aa2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 18:06:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"902efa42ca621de7e8515dad8c359eb07f4a280d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 18:07:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4cc75c5f3174d1e9eeb7a27570e4e3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142eb9d635224b529294016faaba26e3 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07855a68266484ae187c76708a727b8a4db1e2e2","revisions":{"07855a68266484ae187c76708a727b8a4db1e2e2":{"kind":"REWORK","_number":1,"created":"2023-03-15 14:39:37.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/84/877484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/877484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/877484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/877484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/877484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/877484/1"}}},"commit":{"parents":[{"commit":"a415bb4c98fa0a7b30f36cf9d2811576181fadf5","subject":"Ensure vlan network traffic is not centralized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a415bb4c98fa0a7b30f36cf9d2811576181fadf5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-12-12 04:10:01.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2023-03-15 14:39:20.000000000","tz":60},"subject":"Delete the PB level registers when deleting the duplicated PB","message":"Delete the PB level registers when deleting the duplicated PB\n\nThe ``neutron-remove-duplicated-port-bindings`` script removes the\nduplicated port binding registers (\"ml2_port_bindings\" table) that\nhave status\u003dINACTIVE.\n\nThis patch also removes the corresponding port binding levels\n(\"ml2_port_binding_levels\" table) associated to those inactive port\nbindings.\n\nCloses-Bug: #2000078\n\nChange-Id: I12fa0764cd0ff509f1859b61060d64cc5a54a7b9\n(cherry picked from commit 572185e1fd252772d8baaec099bf8bedfc837013)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07855a68266484ae187c76708a727b8a4db1e2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07855a68266484ae187c76708a727b8a4db1e2e2"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
