)]}'
{"id":"openstack%2Fneutron~877528","triplet_id":"openstack%2Fneutron~master~If362fab20ac03f8b62471b60c031f9349171ce93","project":"openstack/neutron","branch":"master","topic":"bug/2011724","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-20 11:18:29.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-03-17 13:38:16.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"If362fab20ac03f8b62471b60c031f9349171ce93","subject":"[OVN] Explicitly define the fixed IPs for the metadata port","status":"MERGED","created":"2023-03-15 16:48:14.000000000","updated":"2023-03-20 11:20:03.000000000","submitted":"2023-03-20 11:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"877528-bug/2011724","meta_rev_id":"4321d3c820b078f7e8eb935dec0152a217d79efa","_number":877528,"virtual_id_number":877528,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-03-16 12:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 11:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-20 09:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-17 13:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-20 09:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 17:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 17:28:06.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-15 17:47:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-03-15 18:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 13:38:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-20 09:00:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a65faf9b1555689f294073a71a0e39fd97efadca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 16:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b3b6b9ab5ea5acd1cbddf8b82f7599694e1a620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 16:51:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"116d4db8e8e8cdb7031f1cd8bdf7d61bd51ef341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 17:13:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a3a256872734513890d2218e5edda49\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8a57c089f41413191b9ac9dad488d65 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/822ff241ec4c4a8493d75eca725aa99d : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67041146a2aad00fd3b5479df0b9803995855b27","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-15 17:28:06.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/28/877528/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"26ff0adc738587efbb2fb3374d592c05a4b6eb57","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-15 17:47:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5559a5a9a5229fad277a982117e4596acec0f174","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-15 18:07:41.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/28/877528/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"313c64166b4dfc535b7ca6c478900c8f17853d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 18:23:31.000000000","message":"Patch Set 2: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/817356f51c714cf89a927f5d9788b7a3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6dfd518ae39c4bea811316240d6af41a : SUCCESS in 1h 29m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03a3f2eb01044deab176ae75f125748b : SUCCESS in 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03dd066182a043a7bef94d2b9004884f : FAILURE in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9be9a099005849a3a85b8002dce341f5 : SUCCESS in 24m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbc7b7811cf64f889e0663b4605ec087 : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b05a4531f964cf7a672cdc6ce54b82a : SUCCESS in 10m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/79067ab7545640bea6d74e23412fe40d : SUCCESS in 20m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/daf58b7edac84bd595dca522a5552145 : SUCCESS in 46m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/802a1554aa6b41859468f9a97576b35e : SUCCESS in 1h 04m 35s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a03b8fde1524c688c4ab5fed353f9e0 : SUCCESS in 55m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/203a6c17d7a14b8f99f878aa7e71f772 : SUCCESS in 1h 25m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ded8b289c2dc4c8f89a38edfa258fc24 : SUCCESS in 1h 10m 13s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4c081832f8ab4cda959cb13b794ec95a : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":2},{"id":"480aabfe007c36640ee4efdf9c8d27a9dafb5972","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 09:02:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a76c2327059670842e1211103422a90fcab2c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 11:33:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63b35f405434403f8c63e5fcafdb60d1e55aa532","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-16 12:33:48.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/28/877528/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"db11d91da92083f6c495cb9dd0c2832430490e4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 12:42:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/167026dea34d4e2d8e80d5b8ff18678c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dd5470255bf45cc8dc202f0b193c8ca : SUCCESS in 1h 07m 13s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44fd66234a604ac68b88fc52d2a30d1f : SUCCESS in 1h 01m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/354a440b40f04c1fb6e2b6dda5865831 : SUCCESS in 1h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d6d8f45f2fbe6848ed650d4b049d6d58e1d3817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 13:14:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216f046a61b64e44af1ef40c68339374\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d7d48db56abe4d6c882bdaf90f524295 : SUCCESS in 1h 17m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e869b8c323841678d620c36a7bfc157 : SUCCESS in 32m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f75cd95ec57f4d0d962c8766c45fad70 : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07e9ef9bfdf84115a63b3c2f2c4052a7 : SUCCESS in 23m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c76ba24a578d4930a487a08e7c8ef6f1 : SUCCESS in 24m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfb63207c6ac4a3dbe6726f87a6d63bd : SUCCESS in 25m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98da290a0942499e9ee131756117be1e : SUCCESS in 10m 07s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/24156c54850e468eb7851d6a5db4378b : SUCCESS in 19m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b18add7c0ccf411c8c79a67c1756e1ff : SUCCESS in 57m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7a6ab97e12044ba1a7b8fdf5c2fd8283 : SUCCESS in 1h 01m 54s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dfb4069f19e3452d9776ab71f31e6a5c : SUCCESS in 1h 02m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b1c426a3a1b4405a4357742e27812a5 : SUCCESS in 1h 38m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/486b9abbc6974fa7922afda8726d1fae : SUCCESS in 1h 25m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/06d6abb399b9499d9d0d65e6cab86a1c : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f6f5a921aab08935eeed096d120263e2d8ac6019","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-17 13:38:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f6852a9c94ce350064f7be4301250078165657fd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-20 09:00:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9f7f9e8a18cca40aac65691726523b93d0e20a17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:01:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"53d68d9b4d0dcb49a1595af3b462d0669b5b31fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:18:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9194467229740f5b3b6b99e74ccbeec\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/07d8ea490d6b43d691f224aad4c2ff65 : SUCCESS in 1h 50m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b6719ce9684424ac9e81806b9f037e : SUCCESS in 14m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/838dcd37bcc1490c87ef8dce74087576 : SUCCESS in 26m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3f5709244546cc88ea76e3030a562c : SUCCESS in 26m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a53971e46e9a4f298beab83d39450fb4 : SUCCESS in 30m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7337a10ca4e24aa2aedb234cd74b2e8f : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bee4b8d029b74024959ed84e441326b7 : SUCCESS in 58m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/151293d6030f49d79abf940baeb36ca8 : SUCCESS in 1h 14m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/87874e3f8f5e4a249af706dca6bb4531 : SUCCESS in 1h 21m 18s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a087ba79899742be975a79006d434191 : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":3},{"id":"a29ee1da08468ec61437bdf2527393a159540620","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:18:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4321d3c820b078f7e8eb935dec0152a217d79efa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 11:20:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59be5c312aa440f78c42ce38942ac70d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf66a8adc1448fda4863c54cd63b6f8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9704dca84ea3ad21ecf9730eea03692daeddf382","revisions":{"f4a1ed86f7c3cbb089208a4c6e840f410d98d974":{"kind":"REWORK","_number":1,"created":"2023-03-15 16:48:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/877528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/877528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/877528/1"}}},"commit":{"parents":[{"commit":"df0d3dcfa72d91dace75df1bb00acabef495e024","subject":"Merge \"Use neutron-lib policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df0d3dcfa72d91dace75df1bb00acabef495e024"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:29:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:29:39.000000000","tz":60},"subject":"[OVN] Explicitly define the fixed IPs for the metadata port","message":"[OVN] Explicitly define the fixed IPs for the metadata port\n\nThe metadata port fixed IPs depend on the subnets \"enabled_dhcp\" flag.\nIf the subnet has this flag disabled, the metadata port doesn\u0027t receive\nan IP on the subnet CIDR.\n\nThe method ``create_metadata_port`` should explicitly define what fixed\nIPs should request the metadata port during the creating depending on\nthe subnets \"enabled_dhcp\" flag.\n\nCloses-Bug: #2011724\nChange-Id: If362fab20ac03f8b62471b60c031f9349171ce93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4a1ed86f7c3cbb089208a4c6e840f410d98d974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4a1ed86f7c3cbb089208a4c6e840f410d98d974"}]},"branch":"refs/heads/master"},"38922346119e786265b9824b3c9dc9a679229401":{"kind":"REWORK","_number":2,"created":"2023-03-15 16:51:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/877528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/877528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/877528/2"}}},"commit":{"parents":[{"commit":"df0d3dcfa72d91dace75df1bb00acabef495e024","subject":"Merge \"Use neutron-lib policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df0d3dcfa72d91dace75df1bb00acabef495e024"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:29:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:34:58.000000000","tz":60},"subject":"[OVN] Explicitly define the fixed IPs for the metadata port","message":"[OVN] Explicitly define the fixed IPs for the metadata port\n\nThe metadata port fixed IPs depend on the subnets \"enabled_dhcp\" flag.\nIf the subnet has this flag disabled, the metadata port doesn\u0027t receive\nan IP on the subnet CIDR.\n\nThe method ``create_metadata_port`` should explicitly define what fixed\nIPs should request the metadata port during the creating depending on\nthe subnets \"enabled_dhcp\" flag.\n\nCloses-Bug: #2011724\nChange-Id: If362fab20ac03f8b62471b60c031f9349171ce93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38922346119e786265b9824b3c9dc9a679229401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38922346119e786265b9824b3c9dc9a679229401"}]},"branch":"refs/heads/master"},"9704dca84ea3ad21ecf9730eea03692daeddf382":{"kind":"REWORK","_number":3,"created":"2023-03-16 11:33:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/877528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/877528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/877528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/877528/3"}}},"commit":{"parents":[{"commit":"df0d3dcfa72d91dace75df1bb00acabef495e024","subject":"Merge \"Use neutron-lib policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df0d3dcfa72d91dace75df1bb00acabef495e024"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:29:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-16 11:32:56.000000000","tz":0},"subject":"[OVN] Explicitly define the fixed IPs for the metadata port","message":"[OVN] Explicitly define the fixed IPs for the metadata port\n\nThe metadata port fixed IPs depend on the subnets \"enabled_dhcp\" flag.\nIf the subnet has this flag disabled, the metadata port doesn\u0027t receive\nan IP on the subnet CIDR.\n\nThe method ``create_metadata_port`` should explicitly define what fixed\nIPs should request the metadata port during the creating depending on\nthe subnets \"enabled_dhcp\" flag.\n\nCloses-Bug: #2011724\nChange-Id: If362fab20ac03f8b62471b60c031f9349171ce93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9704dca84ea3ad21ecf9730eea03692daeddf382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9704dca84ea3ad21ecf9730eea03692daeddf382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
