)]}'
{"id":"openstack%2Fneutron~877533","triplet_id":"openstack%2Fneutron~master~I5e1887ddf1dbbd8332663e085084852591ae75cd","project":"openstack/neutron","branch":"master","topic":"bp/keystone-v3","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-04 11:03:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5e1887ddf1dbbd8332663e085084852591ae75cd","subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","status":"ABANDONED","created":"2023-03-15 17:08:58.000000000","updated":"2023-12-04 11:03:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"469250366628695ad4fa582879200bae6329dc80","_number":877533,"virtual_id_number":877533,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-05 13:28:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 17:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 18:03:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-03-15 18:41:31.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 19:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 09:40:06.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"CC"},{"updated":"2023-06-14 18:59:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"7b3de2bed1685d0c762f54fdff01c6a33cc02119","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-15 17:08:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c14467b120361be447e16e6eed6fded0a9c268f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 17:35:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c100cd2301d445e88d9379829923d503\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74ba5889fae649e1a08e1f578589e824 : FAILURE in 21m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a95ecc0e71d4bd6a5df166737b59265 : FAILURE in 22m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db94643c03aeb7827b46f883aca0dcbffe7adf84","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-15 18:41:31.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/33/877533/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ddc3575285aee834d769912ee700cb2b844327cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 19:15:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7282f200637847cdb58854a3e03b31d0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5a1e1d08bd5b4972ac3af80a59c7479e : SUCCESS in 1h 51m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0c604d72d98f435fa1953844dd8d10e1 : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da2595f0586b4106a43e73c5c8b6be91 : SUCCESS in 1h 21m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b9d88b8345c4764bed1449cb9c76834 : SUCCESS in 1h 22m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/191b7f8c51b6415eb8159b2d89a46e48 : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/399a1b0d80304ccfb06f0806a6890a44 : SUCCESS in 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1eae634fe3ec4ab1b735cfa599e03eb9 : FAILURE in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3e8babd031649818bd45f5bc5368709 : SUCCESS in 14m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f00dd63b96614ac3b04989c15aefee15 : FAILURE in 27m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63bbac046e134d8e821b0fed4d5baa9f : FAILURE in 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00733deea9148dbbba2e32e5cced8d8 : SUCCESS in 10m 58s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/908a111eed8f4d4a90c31c7901ca7f57 : FAILURE in 25m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b4aa922d31744ed9aecbc0d194840bd : FAILURE in 56m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3aa36e4a77b7404b95c919ad04527178 : FAILURE in 1h 32m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/49b2e83cbeed481fbb63615a2f06806a : SUCCESS in 1h 06m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bd00893ef95b47b79e5b790d2ada7f3a : SUCCESS in 1h 07m 53s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7812f7e05ed34d9490da04596cddd6ba : SUCCESS in 1h 00m 39s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/af70e7e4651f4da9b9c91cad12f91976 : FAILURE in 1h 10m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fcf93425c0bd4de9abad6ce73ec5f5af : SUCCESS in 1h 39m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d05e01fdc4354baa85105d0970ceb8f5 : SUCCESS in 1h 21m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c4dfc8908a4478d8fed509cc7dbc79e : SUCCESS in 1h 13m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/11eef9859f944ab79785a2ff2b03104f : SUCCESS in 1h 53m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5807c8ff277a4f0cb8ac78bdb7799eff : SUCCESS in 1h 57m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a8046d422c4b4114b3599a4ea5ac9a1c : SUCCESS in 1h 16m 24s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9d2fdd68a7c74fb787961c412d01c81f : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8586d5dd4ac4e0dfeb92511518a461860bcc4999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 11:29:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa449579e25acc0c2678ba808e79a975a3488df5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-16 11:55:32.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/33/877533/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"53274f8bbb612252f03d5dbc29ec398565137e8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 12:33:03.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da0952dbf8874141aebddf56a685a7f9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d639e21cc944023af859ebe4c1d8d6c : FAILURE in 51m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82f80f4daeda45709ec3ba48fd1d3a80 : FAILURE in 1h 01m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d5480f7d93d4992946868a1e01861fa : FAILURE in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17da137cabf0b7cfc46e73d2b41ca8217c180154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 13:10:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbe741694de84fd094513967a0badccd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/419580430cdb491ab8005023ad4dd9ac : FAILURE in 20m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f47e5d6fb3d54cbd986eac16c79f28a6 : FAILURE in 19m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/07334e19edfd4faa8a280523c05b920d : FAILURE in 18m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b052271fa034ed6933575e32dca5cac : FAILURE in 17m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5ecc7a72928e44199509a67e91ba8156 : FAILURE in 20m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a24829bb90174c7ba07e3631b55e08a1 : FAILURE in 20m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d75f085bbd4189b11007ec2d3ac9ad : FAILURE in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2242014c545b44be9fa1d685d64f08b0 : SUCCESS in 14m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7abac9dca7044ea997e03101e363989d : FAILURE in 28m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12d68f8dd3e64af0821475fc70eba4d4 : FAILURE in 23m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/449ae93dd96349eb96a572c2a0fb0834 : FAILURE in 28m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f505817ba83d4fc78810ffeeddab488a : SUCCESS in 10m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/458b66beafd7418eb158203420124ac7 : FAILURE in 23m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0423079c9f4249fda5288e7fad6e87ce : FAILURE in 56m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1dd3393317a74385aa77a92b0517ee01 : FAILURE in 1h 32m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7d13f88eb3e4bae9dfaa8394f86eb53 : SUCCESS in 1h 08m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/971352ded7d94f73a1fe48d6beb9ff10 : SUCCESS in 1h 07m 10s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4556dc416c7f41cf97ac0d45b485381a : SUCCESS in 1h 21m 17s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6866fc72f94a4bd08cbeae01b28d6481 : FAILURE in 36m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb30cefe5c6f4e41b0a5530147871017 : FAILURE in 29m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fda57894178f4925a5b5d614f5bcb626 : FAILURE in 30m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/50bc6bc6359a4d3783fa44f222175b73 : FAILURE in 25m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/622ab4c2082e47a9af0d7166321e042c : FAILURE in 30m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a4efe2b6afe046c1830991f17f3074b7 : FAILURE in 17m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc516e1432e64adf91eca14c0bdd56d5 : FAILURE in 27m 20s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ca7ddb0918b34251bc140331938f87f0 : FAILURE in 25m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e3eeaf3d2e2fc35849031bc1b44a5a0bc69b49e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-24 14:17:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e3ee68b55f48103142bf9926cb7ee62841ada73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 15:22:07.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19616cdc1984acba6822b638b2f2e4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d93ffd27fb34375ada828b170e901d1 : FAILURE in 43m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45bdad587f3c417883674b2dfabb81c7 : TIMED_OUT in 1h 00m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67208f62d00e47e7aa611ad4ef081e4c : FAILURE in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3a260520d2264b00360f6eef3723d9ee00706b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:37:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa7f70fc4a7348a5bf4e9bc1763bf48a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1d87fc77159745d08acdefc17e6e0f35 : SUCCESS in 1h 20m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7cc487a56d404fa5b28c8df7ff93a0b8 : SUCCESS in 1h 14m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/edd1f3bbe6134f499fae8e5343968f7f : SUCCESS in 1h 07m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dd44a087afea4b68a6284ae8e021be70 : SUCCESS in 1h 20m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4112a0a208fc4bc3a09d61b3630ae734 : SUCCESS in 1h 38m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8fc1f1dddcfa4179b38c93ff75571570 : SUCCESS in 25m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fec35fb2595c496d8f593a4ac9f90cb8 : FAILURE in 40m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640ca26dae4447deb15cd78fc02fe222 : SUCCESS in 13m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ec45c810484b0db0e68ea55bed2438 : FAILURE in 30m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c51e61e7c474423bbbb8ca59091cfe6 : FAILURE in 30m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6489acbdf19946749b57517052f4884f : FAILURE in 27m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfa22e4e4494135a35dba1309f3f6eb : SUCCESS in 10m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/09557bffcb7443d0836664ae5cc1ce16 : FAILURE in 24m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/408a8b585bbe467894303b0ac5a562e9 : FAILURE in 42m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3072da5b34024b63a0bb3bdaf4a5c54a : SUCCESS in 1h 35m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9108e35468e4bb7be360cc5e7b7c07f : SUCCESS in 51m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6b95dfdd018b405ba535478fef8a2b82 : FAILURE in 21m 11s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a39c555c0a894c83b6c10283610ad61e : FAILURE in 15m 13s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a4333e35749548be9f3c88976bcfcfa4 : SUCCESS in 1h 01m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6243c73f327400a99c8c0722197fb01 : SUCCESS in 1h 19m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ab72ec5c51f6464cbf964af0eb7fc3bf : SUCCESS in 1h 31m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/34af10b5e83548a1a87314b7ae2b8416 : SUCCESS in 59m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c7d0bd6e57874baf831d138e21151372 : SUCCESS in 1h 48m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0478d02d279e4ab9880b0e2323b05302 : SUCCESS in 2h 11m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5fc4e16075e24d6ca1812acaa20a2e62 : SUCCESS in 1h 11m 46s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f67a97af9f0244aab2b57d175ac7461b : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4b0521ea4957b64cee5cbcfbb23d21de679b3c9a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-24 21:30:56.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/33/877533/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"b736cefb16757173d4d4f8a5d01fb1391a27ddba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-25 16:11:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bf5fc300017f4a130988a988d7e5f40895515ef1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-25 16:43:52.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/33/877533/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"cda2d417f17e0e6e6fbc8a239ade54bdb0ff4023","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 16:48:19.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8697283e08f34490be05da9ed0af0e4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7d059c2e6ce432e8b08ef25239170c1 : FAILURE in 35m 29s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed2438306568409e87b1ae43cf9f40b1 : FAILURE in 18m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0b563f565c84e92821eee628ad6f861 : FAILURE in 18m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"86f39058709068b69affde362ec5db3dcc5b9776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 18:17:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaa23b1449a24e8c828d774b5c5a4e89\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3b70b4ff23e144cd9192977ab529ac5d : SUCCESS in 1h 14m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9d1f2d73840043d79a0bc78910022717 : SUCCESS in 1h 13m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6bb58feac0054977a62c8383ab89ea9f : SUCCESS in 1h 15m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a4791f233c6247f4b4f607ce023ade39 : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f011e2a1ba84c98b121b2d1e4ece067 : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7466f10e7eb4c788d22a4608eff8e45 : SUCCESS in 27m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d3e2221885c48d5a15e6703d7c314f3 : FAILURE in 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ccffd2d19e44d87b28e00039c4e1613 : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fe2f38984624def8817b74959312f3d : FAILURE in 34m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2896efd5641c4d8fbaf84767b5c237d9 : FAILURE in 27m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23c72579444b44f9b5c52dfbb09b1460 : FAILURE in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780a83e6d0094ec1b30c075294e1a76f : SUCCESS in 11m 01s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/751c97b4d7d7420997649d5931eefa71 : FAILURE in 31m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8cdd35f432eb4fc5b934a024ca0bfb22 : FAILURE in 49m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7541e67d8f564233a198edbdeeecdbe7 : FAILURE in 1h 35m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b989aabefcc44728cabf4c47ceb7467 : FAILURE in 55m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/721f8f7ccbdd46f4a0305a33c378de78 : FAILURE in 20m 01s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9775481a59d54f418bf760d7ef6f9a3e : FAILURE in 22m 45s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fe86b8044d1344a09a751679c0dad2d4 : SUCCESS in 1h 05m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4164ba8959354f4bb7cdb5393c2b5397 : SUCCESS in 1h 27m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/14f58ab76c174aac8b7c3d0c5ea420cf : SUCCESS in 1h 32m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6934a519cb324886a8ac2be5da611495 : SUCCESS in 1h 18m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/905fa54925be49b8879c17535b31332e : SUCCESS in 1h 59m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0dfb872a0a12481cb217a76ee1150ddf : SUCCESS in 1h 52m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9c43705b78734b2c92097644ba7b383c : SUCCESS in 1h 23m 32s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2d5b445e74364fc2af119b0032d271ab : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":4},{"id":"22699db893ede348e9212016e176d8831018ad45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-28 17:09:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"39e9f9d63cc9ba12dba4fbb55ccb1f11ed7d9644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:33:52.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/684e435853444d1aa43417470c33fff2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66c41aa4b9c84176af2fe50b22dca367 : FAILURE in 22m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55b480453c43449ca9c409a8852fdc3a : FAILURE in 22m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db69af09361147939048f27eef8015a6 : FAILURE in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"86cf307d521606a3c82a112a20a72ef51dd8341b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 19:07:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af0200d3d6f84a9893a031a1cd0fead6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a0ce64086f9442759baf0c0a2e0ab003 : SUCCESS in 1h 49m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e8e19872cfa14d66b93bc4b092fc530f : SUCCESS in 1h 39m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11dc0829d23549a1980d04e992c9b4ec : SUCCESS in 1h 19m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cf98ff52cbe0441a86e0462f18707f7e : SUCCESS in 1h 31m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fba3b4184d084d96a07049f779197dc3 : SUCCESS in 1h 52m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edb3bce37c28408094bccbf09e5338ae : SUCCESS in 27m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b98cedea784586bfeb91f38324e654 : FAILURE in 33m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d438e79b3c44a4a89c91643ef5718f5 : SUCCESS in 13m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/915da0eaa6854a4a9cecb13a360be84a : FAILURE in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1179557ceb5f434183e2352af280a8e3 : FAILURE in 26m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe6c6d0ed802414fb0f7a238485d852b : FAILURE in 30m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae1ba50c0c1140b38b52a636df6ed747 : SUCCESS in 9m 38s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d6bf0fc9d6f64db6a684fa6e41aefca7 : FAILURE in 23m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/775d939afb5d40c3a83b66797780d6c5 : FAILURE in 42m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12c954be75c94de687cd10e0ef0b1a55 : SUCCESS in 1h 42m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36c91b3ed84d4cf29ef18d58bd45b343 : FAILURE in 36m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/48540ea0a6db4928913b96e9a3873600 : FAILURE in 19m 19s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/df42441bf08e49bd8d83143b5a2b396d : FAILURE in 11m 36s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4fef6c9997234d1f8592e6398363fb5d : SUCCESS in 59m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d66c914b7bc45c58d63eafabbd7f00b : SUCCESS in 1h 03m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ddb3fc3aefa249588449a869fc3d6c4b : SUCCESS in 1h 30m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e39d65af27ff462cbcfa77e0c76bba44 : SUCCESS in 1h 06m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/291a5d793bda444db1fa47195d9080be : SUCCESS in 1h 18m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6db606bfa9ca404aa6d6e3d47875c965 : SUCCESS in 1h 49m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/79862cf0d33f41769f2fcef813fbd857 : SUCCESS in 1h 22m 53s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4f45caa9f2d04e81be883eaa238cfc0c : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":5},{"id":"85f7286326a18030fa0f04c4c40d2095cdef4619","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-28 21:17:06.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/33/877533/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"20277218e003883fd6304920146569bf018c818b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 16:12:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c154c03ae3c97a83b017bb018d6521f82d949d1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:23:29.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf8efbd29b08445c9cff311f42317f08\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0e0a85a204a4653a14453e60645aa04 : FAILURE in 1h 08m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94ec42ffa69842e591b2d78f89519a34 : TIMED_OUT in 1h 01m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/161e01ac5493440a916e7cc242fde933 : FAILURE in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"00c8cf07b07d0571b4d732ccd1056e9476e8531d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:35:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3576dcc21f534b3982d932985365713c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/119e0e7bc3c24872a82853dd32c74032 : FAILURE in 1h 45m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02acd51fd98b49d1aff416c8d20b7a6f : FAILURE in 1h 20m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9de04c3e082b4a0aa26d4a582a01416a : FAILURE in 1h 43m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c77b7917500c4cca8c2c03e8974c1a98 : FAILURE in 1h 32m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c92c8e7173444655bcedb4a51cdf28c5 : FAILURE in 1h 57m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca33e50baaa04d91998ccff7332523cc : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/742297da2c154490a8b03e8c0e2e52df : FAILURE in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d1210ae9cf4fe5ad142e49e98c86db : SUCCESS in 16m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bc08bef78834d59bdbceab42ccec471 : FAILURE in 30m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84612731a0574a87a529a5d6e4547202 : TIMED_OUT in 40m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2130e03f57045d89d27f74123bd1978 : FAILURE in 26m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4605db4990c049ec933dcf158f99c4cd : SUCCESS in 11m 53s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/051ecee515b74fe997c21c8af8ecc26b : FAILURE in 25m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bee81ed44a0649eaa5d175ff1f85598f : FAILURE in 58m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e1070e5f44d4ea88ecebae02c6f52db : SUCCESS in 2h 10m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32242268106440c3b3d448e718a764c7 : SUCCESS in 1h 28m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/68530feebf584a3f9d5209a604ca2227 : SUCCESS in 1h 29m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a26e813f826545b7b9d2ea713f2ea632 : SUCCESS in 1h 38m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a766293b3fb4455bb6cad811b1fd5307 : SUCCESS in 1h 40m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6c427cfa81934b2a8444f28f99f48b18 : SUCCESS in 1h 13m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc316b6ad25d4e179c3c35b9a259ced7 : SUCCESS in 2h 03m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f84215feee9c4ad2a4838af6235fff38 : SUCCESS in 1h 47m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9cfe219d96cd4b44858b3e850c8fcf65 : SUCCESS in 1h 22m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9e62d175e6b14deda23a851f5e6ec724 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":6},{"id":"a4803c9a9ded21a35d15f8d562c0c0a41c17a12a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-17 22:26:18.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/33/877533/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"962312a7081ef5f25b9ff5dc67ce2514854d10e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-09 10:47:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d60984c38faf658962d4198ce098731d8dc958af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 11:44:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d4e1ad941c84e6abe94817322a5a90a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b666a384a3f46fab58fc3471538c42e : FAILURE in 46m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c39c1884cc214f949493e7d3a5b2d6b9 : FAILURE in 54m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73efbb08538b44ccac1752503cdc7f68 : FAILURE in 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"349743d06a173fa83d72d2fb14761574b18f4a96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-09 12:04:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"39d6ace629f55a8afbec5d1af7da629a574e1c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:52:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80dc7d358ed5490a92cfbf3078323531\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8e6b7b0e3ea428b82d3881b4e763bd3 : SUCCESS in 46m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b829dfbbc4b489495ee0818b6d99efb : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76f4d8b60eda468fb5b4e118e8f45e24 : SUCCESS in 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"69da3f6d5d1b95b5a0f2f4dc38abe96649e3d40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 14:17:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b237f50eb804133af6e6c7cd3b35eb7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/722b7adf448f4559ad84b5b899f70e58 : SUCCESS in 1h 50m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b66ad831779148d1a6b5654b63b2b0ac : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/967977a142794581bcc2f856e2ae8405 : SUCCESS in 2h 11m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4e29815b1dcb4e199f793758cbdb12a7 : SUCCESS in 1h 36m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c437707b4be2430fba5c7935da794a7b : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3209a61a9b7e4bd98aa27f719776c601 : SUCCESS in 45m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87e5d6a531ae40e0beb209bd1588e489 : SUCCESS in 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d638fe1741ad44299d0eed33665e4b99 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/863f1153e4f74b3cac4ae69850a32ba7 : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7b1646e665945bcabcc1fe535bff246 : SUCCESS in 25m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8da83395c6ce492fba8902d5c051b6fe : SUCCESS in 28m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16765ce1b7984ad0875f0b3ab6b01124 : SUCCESS in 9m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6ccc2ab3854d4d5881106ee90f35429d : SUCCESS in 24m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/626c8a83a4174ba89728b2f6881f8b13 : FAILURE in 56m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c2aa58513064b12895d4ab998153ad6 : SUCCESS in 36m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d36eaa78dc06491cb4dbe90df31df34b : SUCCESS in 51m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a64a5ba718954f479519769e38fda01f : SUCCESS in 1h 16m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/736ccd4bbc9843b296ce0ef872deb150 : SUCCESS in 1h 40m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8d343fcb1078488eac09d184e8a277ef : SUCCESS in 1h 22m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/34218b9f74b046fe8f7bbaa2fb2fb668 : SUCCESS in 1h 04m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e5e90628fab45109b7282299f802b1d : SUCCESS in 1h 30m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/63daa011a4204cafb6bae975a87a6ba3 : SUCCESS in 2h 03m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eb0258273f154119b6e2fefaaeeac1d7 : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":8},{"id":"84cf166336472c400f1aa8d944563e45588e4ded","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-09 17:19:29.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/33/877533/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"f32785a8f75fb206ef0437e6d7615022440a79f3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-09 21:51:10.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/33/877533/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"7c0379e5e52368f0aca694d09a62073126302d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 09:28:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"70390c9ea38b81e541095a8a83a476507c97d518","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 09:38:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87311dc5ae714a8c9905aaa35a93a1db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58ae4b3a1a2840ce84bbe3e5383bcbf9 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9f9489f77a74f91978b5fa475f34f7c : FAILURE in 8m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3675a159ff93457a80c326fd4a79b1d0 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f08a79dbff46c7f32d87cf822149d50edda75b19","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-10 10:02:06.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/33/877533/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"02f38110f3a025d033bc63a0685a16d567892c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 11:31:34.000000000","message":"Patch Set 9: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d814b970b994a9fb6850daed9f4b7db\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3b036a8e805641d3a952b5b3bcde9af0 : SUCCESS in 2h 01m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c0f33738979445ca97fe0d755e1a34c5 : SUCCESS in 1h 28m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c2eadafc0f84bbd9d385e2a43085f01 : SUCCESS in 1h 48m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55f0d471c5e1430a87ddc07079b004fc : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cee544686f904a7b951df840f6ef63ef : SUCCESS in 1h 50m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/94169913b373430c8320b97180490533 : SUCCESS in 37m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c79ff2bdd413428d9dc07e6916f52391 : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30e55bd05c54b50906924c2213bbf5d : FAILURE in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91c37e08bcc049ee96377f272e8f01f1 : FAILURE in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/540dcd7b622d4f90af8e018d6f33e1e8 : FAILURE in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f124dee9decf4d5f93261337fe1694e9 : FAILURE in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10f96542d664db783fbf621836da078 : SUCCESS in 9m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a2346089608048baab7a364d47437df7 : FAILURE in 4m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/862477145bb04d4e8d1a28320c586197 : FAILURE in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e786c4d32f343db82b5be0d0a16d16d : SUCCESS in 36m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9053f73e4e0c4482a969108f60978dab : SUCCESS in 1h 00m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24920924e9824a039c28ac103a4a805e : SUCCESS in 49m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee04402e998a4a408190197929cab8aa : SUCCESS in 1h 37m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e7fc7acfb3b42ddbd43dea66d22c63e : SUCCESS in 1h 11m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eda3c1d00dd94a978af46827bd753221 : SUCCESS in 1h 08m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/936ca51e45024418935db382a78be4d1 : SUCCESS in 1h 19m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a14c51356d3f4c0db81cb48bbc366295 : SUCCESS in 1h 30m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c8e5e3b436643008e858abed290a0da : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":9},{"id":"68292b4f35a29295fbdfd8666b8998735f9ded79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-14 18:59:46.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"bae21de5fbd0b9312e4fb8f84b7e845e57bed80c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 19:11:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db60c67f15474a7eb50cd455aa134f98\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aedadf83606a466d9ccf2f7b94ca9ec4 : FAILURE in 8m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6399ff78f8144819b942481431c3c32 : FAILURE in 10m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38907871afdb4ff08ae2329cfb919265 : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"02969a916f69449a203f28567b8093e1d4674f58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-14 19:35:21.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/33/877533/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"0e39feb9bcbe09fe76b11300459b32e25214742a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 21:17:56.000000000","message":"Patch Set 10: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7f9ebde896447b8943d0943e80daed5\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3ec74651fcf04b77a16f0f90538236fa : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aa98b99709fc490e98e93915433198b7 : SUCCESS in 1h 43m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/193f9470089f4c3c83d719cd6844425a : SUCCESS in 1h 50m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ad05c98d4a5545319b245d70c7e53669 : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f49af792ac204a0eaaa6cd0a1d0f5a59 : SUCCESS in 1h 32m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a4b5c4e73b94f6b8939c38c6a93c30a : SUCCESS in 32m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdd19131abd240ef8d9df3264501a12c : FAILURE in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907a44f00c644872805f4b09cdd368de : FAILURE in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71575d4744a7407aa6dd852de9e19983 : FAILURE in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/203fa16698ae4eca90e12c7f480f4dee : FAILURE in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/007be725795546f59339ee017e70b9f0 : FAILURE in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ed962fea22475d8a6d74f050f7db9a : SUCCESS in 9m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/4369e9b181b447c682bcc55c71aeb42a : FAILURE in 5m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8aefcb9421074d97b608410f5815a081 : FAILURE in 41m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/737974f9ee75418981cbb56c850987ff : SUCCESS in 1h 59m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/455c8c9e6de4484195a36f3e9226eb49 : FAILURE in 49m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1af35bdba1f645f59ccd2032c89b36e9 : SUCCESS in 1h 00m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b1610d7f8eb049dc85fe9ea588b860b1 : SUCCESS in 1h 29m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/665d62236c1847048931e42fa542cf93 : SUCCESS in 1h 20m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8b66858e597c4b11b3d27fec5f50e6bf : SUCCESS in 52m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/252d90e9d7c04fda932e37900f2ffa87 : SUCCESS in 1h 31m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7cc00bca4e744a84ba79130b0d6fc249 : POST_FAILURE in 2h 12m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/033d59f7747c49c3a32291edc259dc99 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":10},{"id":"d46619b8a7b2db01d7b8073691763f4b99dbe2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-26 13:33:39.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"2560a52b3978879eaf6be4a32a533e664d5046b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 14:22:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd65d07e3e9947fbac416a74ddb45716\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c03b12ce1dc4b5fb7265b7e52c0b327 : FAILURE in 20m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/437d79bd69944717ae9d2a99e307aa2d : FAILURE in 34m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ceccdd465a6e4622a5a6c5c338e3c133 : FAILURE in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b87c32330e70867112e5e3cff9fbd995f16ef595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 17:44:27.000000000","message":"Patch Set 11: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/895d0ad99a714f2c80d192f5ffb63539\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/902696dac838410fa04a37d737b2844d : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3809ce9df50c4535a67482b6c3a3a56b : SUCCESS in 2h 40m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2afb360d452e4545bebc7d577ca27783 : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6180d6e1027e4849bcc069ed6f29863a : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/51bd88277dda46999958e2333d8b0e58 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0249b54b5be447a79916a82ea65c3d5d : SUCCESS in 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5cf3fe6097d4e1aaeb269a126ef3a06 : FAILURE in 40m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9b56b92e824908aee7cff13c04a27a : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fd69eb5d0034cec9f98638a316fbc21 : FAILURE in 29m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27730e5d69024ddaa2cc4d3f18bb207a : FAILURE in 29m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f65962807f044a484b9a8af564b2259 : FAILURE in 29m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb0fa8980e94052b4cdab5cda877324 : SUCCESS in 12m 59s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/520ed9e6d7b644e798790e518560d800 : FAILURE in 31m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18caa7005b1a422cbd072a08b8216bda : FAILURE in 1h 03m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/95fd9d95f4094f0bbea487179bd65386 : SUCCESS in 2h 03m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ba60d389fe743d59071e1c970155158 : SUCCESS in 1h 23m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ee51be343a3c488a966ae00805055690 : SUCCESS in 1h 09m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1b7ca8b45c204478a91d41a9431d2698 : FAILURE in 1h 49m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7de8ea80b65d4103bb402d63ea3b91e2 : SUCCESS in 1h 46m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/149cdfb5aaaf4ca5931ed515d8abc34e : SUCCESS in 1h 20m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/57449cbbc7104a1c8038345a914ea58c : SUCCESS in 2h 20m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad61b307ce5e43859246169caf6ccb38 : SUCCESS in 1h 57m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2d03f7fffc8482d824d542004649260 : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":11},{"id":"7d2f0022a47364dec46cb452a6c8adc909375501","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-26 18:07:14.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/33/877533/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"1cb6c2d96a2b92336b9e8ff2eb45b8c99edbb8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-27 09:18:33.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"32fca8ed22922f220ac47c958e6e5c1e36e92a2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 09:43:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db6967b2eaed470eb56c24257f5fd09d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b94bd6f37654561ad2504d5ef8d365a : FAILURE in 21m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f070d794a0a4568a6728c5e2bb90152 : FAILURE in 22m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91c32c2457154ad9a484312b3c91e1c5 : FAILURE in 23m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"89eb8191f4271a90551a87cfadba1240cb9a2b9e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-27 10:57:07.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/33/877533/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"71c04a171d4b8018c2a9c077eff32c93f97b52e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 12:13:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a910f56e3b144c7f918a11d5300cc396\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e518103f82014de88fc8f869df797d11 : SUCCESS in 2h 14m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b087bb9a091e4b7e9599831bafc43354 : SUCCESS in 1h 23m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea095b7fff9d456a885d8bc73665e164 : SUCCESS in 1h 37m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/45b50a9aa7334c0a97986b628ca21a79 : SUCCESS in 1h 17m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3388ce5afdf34918a13496a709733374 : FAILURE in 1h 58m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e1bf2591fc84e4bb40dcd13b69d2257 : SUCCESS in 37m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58867ba596dc4b43839b7ba34231bd08 : FAILURE in 42m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be021df60a9d4ee9a2de8ad77e80d45d : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fb00f8c13d64da6926ae0d71e75e427 : FAILURE in 28m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47b55238610b48299835308f6af60f20 : FAILURE in 27m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47df5b473cf847f593fddb6e0ab7f055 : FAILURE in 32m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5183e73344245a1bad4c2efba1656e0 : SUCCESS in 11m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/08907775a95749a497f2f8b13b9d9c61 : FAILURE in 22m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4d4d5b54b654cb8b9f53e13b66e41a2 : FAILURE in 47m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9f2c3345a32442a9a87fe224d232057 : SUCCESS in 1h 58m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/267e625c06c649e9b9688ab79343285b : SUCCESS in 55m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/913238755a20409eb85f1fe8f73439df : SUCCESS in 1h 12m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ef29cc3d2be4ef09ccebc239c0878ee : SUCCESS in 1h 32m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/acc533f7ffe149f3bbbf6367d99527be : SUCCESS in 1h 21m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4feee3b4fe314c2c892b43c7f4fea8f2 : SUCCESS in 1h 28m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f6954489ced4f45963840e6cfc73ab2 : SUCCESS in 2h 16m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b1e4a25457384b138f6f917d55b3229c : SUCCESS in 2h 31m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c71bc0a487ef4d4989ff0d191d6df55a : SUCCESS in 1h 41m 41s","accounts_in_message":[],"_revision_number":12},{"id":"a513fe4fa499c2f461780439cf12c34e039cc7d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-16 14:54:38.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":12},{"id":"f667d9a81d54a39b8fed359499c6af0b5c763bc9","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-16 15:35:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":12},{"id":"ffd25c422291f3cef0e7f99014e19ecf97067edb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-16 15:35:24.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"bb9d46a698b421120c461b5cdc872610a35964ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:08:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/460025aa18c14a26882c9075cb4472e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/892c7c1f8ac247adbf2cfbe33069f590 : FAILURE in 32m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4936443fc11a4e04a9f667a4d72dfa4a : FAILURE in 26m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fece6483759445afab277c9616509402 : FAILURE in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"51680cc45933e54220c2326d9b17a3686d60006c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 18:29:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ba8c47ceeeb44af94708713d0f58648\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3179d92c6b974244b6f611dabf01fa98 : FAILURE in 2h 51m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc75965fa79641a19a59803950d9adfe : TIMED_OUT in 2h 51m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/04d661f9efa040cf94e8aaa68eeef81b : FAILURE in 2h 35m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/987df02c86564fe68dc6eef00416c1f4 : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c207486117c643b391426f4c43aa5f62 : SUCCESS in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bffdf9ce3f442969c79ef361929f491 : FAILURE in 39m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506e4af929f2485e98e5be75b409906a : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac14d6b6a2ad450cb9182289bec2000f : FAILURE in 24m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8713cea5de414850b76002a5c4f69df8 : FAILURE in 31m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/579adec22c3040d7a190e781c254ef3e : FAILURE in 26m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efda2e2ddbb41f48d728efbdb2179b3 : SUCCESS in 12m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/126cbe824a7744528f009b57ac28e479 : FAILURE in 29m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4881a1ad15614c919f288238fcbb4cc1 : FAILURE in 58m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdcd9b3f68384e589bc55dc9e098b425 : SUCCESS in 1h 51m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2c813a3429242858765d14a30162281 : SUCCESS in 47m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/80de707f9c5d4889b98fde899970e760 : SUCCESS in 46m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0703f027db2c4991b7e44e63da25af85 : SUCCESS in 1h 48m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eeb8c4e3e34e4048bc77abecc33249d0 : SUCCESS in 1h 18m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7f206555a26542d4839d328c80aa83fc : SUCCESS in 1h 22m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4692adc9624d4d549d634af85c85c1ea : SUCCESS in 2h 06m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e97dc135d55d499a8e18a4251d226057 : SUCCESS in 1h 49m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c11db92d462b4a979134e4658d3cd94f : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":13},{"id":"d0d0b86dd12acbcaae7c44f12c056d9bdb5e89d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 05:22:28.000000000","message":"Patch Set 13:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/33/877533/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"eb559720fdcb0f70172abfb66bad8c61c8134e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-21 15:01:14.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"2493d77263aa4c749c0eb5479e5ecea47661f89b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-21 16:21:50.000000000","message":"Patch Set 14:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/877533/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"872aaee8e04cd27ed61b9b024cdb1561b8131287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 17:30:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f888fce5e9574a13b2405cd1bcf2811c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ff3c5b25682470faee0a795e5a293e1 : SUCCESS in 1h 39m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ca02590c69045f699e28ee3f70b2f0c : SUCCESS in 1h 40m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f612f1ab50954c428dcab4e378b86b3a : SUCCESS in 1h 21m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bb881e59ab764e25af70f6bbf84314e6 : SUCCESS in 1h 54m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbcad6fc5c514628b55b10105e27a8f7 : SUCCESS in 33m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8ee6dd39a9b4c1397382422c08043d7 : FAILURE in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f989b76c88c549f9be53bf42dcf3ec98 : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a449c56a1290444aab288ecbc4aa32db : FAILURE in 24m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edbbbfd703c14ddeaf1d5b9a0d5ed00c : FAILURE in 30m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb3fbcf9bf9b43c9996fcfce28a781d6 : FAILURE in 26m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db3297315e3e4a3c8a1f6f61e5c89865 : FAILURE in 25m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901f78885349446ea56cb20f2a93412d : SUCCESS in 11m 11s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/221cccdf605a4b5d91e0904999d74404 : FAILURE in 31m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b57c63c6834840479d7f25d03cedca3f : FAILURE in 49m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0bc9a091546453c98259f1c878cfd7b : SUCCESS in 2h 14m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7693c3a4433241329f4ad67d70f8b911 : SUCCESS in 1h 05m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/faedb71f9ffd436f9946e54b9da98b59 : SUCCESS in 46m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6134ba8084f4f58b8f40030fee021d3 : SUCCESS in 1h 31m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/db0ad6974c464b26bd95a9845e056c30 : SUCCESS in 1h 16m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/98eb20a958584efc919f4179d60fecc4 : SUCCESS in 1h 14m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/91ac871997584ba786370aafb9ba0bb6 : SUCCESS in 2h 21m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/de7c391c0f8043bd97d67ed464d73340 : SUCCESS in 2h 01m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b9c64778f754f85a6ea94d858b04c76 : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":14},{"id":"cb4cbc6231561f9e11b7333404aa4450c247640f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 19:11:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf8e47c1bb74058ba64521bfaa38038\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8e19655dc39492aa5da270edd9a1ae0 : FAILURE in 46m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4543d79649f4d9b939d74919ad2b3e6 : FAILURE in 1h 03m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14ff7259db684e3bb181d1c95f7df551 : FAILURE in 1h 07m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3136f6c082884b7089a2f8be5b2760bc : FAILURE in 59m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9d4b84822ab2f77a23a67ce3acb510458a504fb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 13:45:26.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"9215369fbff9618ae8199c7ba422c3f91e8e50f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:24:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d5a11a67bb847ce8970faa3359eeee8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6c69015c9bd84d448fc9dd8821c1f6fd : FAILURE in 1h 50m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d5af9275df6429998d3720cbfc3e79d : FAILURE in 1h 21m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9a2d4b2d4124429aab6d9ad5d5feeb2c : FAILURE in 2h 06m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c189f9ee68504990836e44a440d2c6f9 : SUCCESS in 1h 51m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e27bf3476fa4c93a08af9e2046ff535 : SUCCESS in 32m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fe1566578364e46a17438b185fda3a8 : FAILURE in 36m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a6ba16b6d05455199ed54a11a905af1 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d72b414eb20c4178baf40cf380cbbef6 : FAILURE in 26m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3347a24d91a846638ec89328f9b407fd : FAILURE in 24m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/200773891b7946fea837318a41c53690 : FAILURE in 26m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d093d7d1c384c65aab90bfff9062f8b : FAILURE in 28m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a795cdf6dab3469babc117155c50d6b5 : SUCCESS in 11m 55s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4a0446e5b1a74550b87fb2a3e0f39423 : FAILURE in 29m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/517c4a671f6c4543a267cb77d016eb6e : FAILURE in 54m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82f0e2ad07374794b91b0e6ecfb35dfe : SUCCESS in 2h 33m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/13c476c1e0124c9b82c956dcbf7f36c6 : SUCCESS in 57m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32796bcb86fa4b0cafe426498a75e604 : SUCCESS in 53m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2c972b058f974943a2bd33256497edfd : SUCCESS in 1h 27m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/51ea7477529843b0a0ca991fc5aed615 : SUCCESS in 1h 20m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/82451814f01e4d61957ef31c06419150 : SUCCESS in 56m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e51819fb750048e6a66a3286c9237d47 : SUCCESS in 1h 35m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4fdb92c53914279883a640606b231a1 : SUCCESS in 1h 48m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a5ca9507a10427fa5cb89c768336225 : SUCCESS in 1h 17m 11s","accounts_in_message":[],"_revision_number":15},{"id":"2443f44237ec98b05c2e4ce4f2e4837e821d169b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:01:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd431275a584af296907f816e924567\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d9b6fcc772745149fa3797f607f32d0 : FAILURE in 44m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5826c4e0d9fe46e8a6f59d602cf36d0f : FAILURE in 1h 09m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3562339a16f64059a9652a9cc096d2ea : FAILURE in 40m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48497f6eeb3a4177a8e9aa34be66bf21 : FAILURE in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f0e10e11b4dfa95481ce0bab41d9bd28eb2f21ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 17:03:07.000000000","message":"Patch Set 15:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/877533/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"67a1fdbb5510734a35d2e30b03e6cd541f978a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 10:37:56.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"d139f18df1c83793872eaf70d8d708e7c0738879","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 12:28:01.000000000","message":"Patch Set 16:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/877533/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"02e021f46bf3abb17f941640b063b3f04f7f320d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 13:26:35.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3b59ee47d74babbbfbf3ea238859d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a3367c190c64c7996927c4f4582d43d : SUCCESS in 43m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6494f8d3f0e14ba689f5390d795f71df : SUCCESS in 40m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57a8b5e9b09b4dcaa23a82d212599d42 : SUCCESS in 46m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8141b87269f44e598e4e118be752a461 : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"76022d58fb0cdf0696577067eafe992f83cd634f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 13:28:49.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/233094a37fd54b9abd789a810adecb73\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aab616089f644b4bb9a9f3d394f330c9 : SUCCESS in 1h 41m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33166742a2654eacb1ae87c205466d16 : SUCCESS in 2h 03m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bb47729ed8f2429a9d76fedb83977302 : TIMED_OUT in 2h 49m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ff6b44755204a57a8ff2443cbb6d1ed : SUCCESS in 1h 26m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/451e4bfdf9a443b1b12079d5e164d67a : SUCCESS in 33m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4967fb89ab5842579a860e732cfcee83 : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c8695b600c410e94b9ed9b7cf2262b : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a7273aa815449bea297b1eccacd3472 : SUCCESS in 29m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b3de794ec1429fbe54add5291dc545 : SUCCESS in 29m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9726fa847404af7b44f04e9c0cea91d : SUCCESS in 22m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3521c16eded941078a38f6b8b4649729 : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76570f4ac6ae42eaae4024154e9a1264 : SUCCESS in 12m 24s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f8bce5df6f404a8e8839b107646d23cf : SUCCESS in 28m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5accb5e2c944d90a792894f8bd9660c : FAILURE in 47m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/18965eb9985c4cbdb0f8d154aad3bd31 : FAILURE in 2h 02m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f001208fe034facbf37a1874a1dcfff : SUCCESS in 58m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f1f7f10fdd4142f89aef5180661d5be7 : SUCCESS in 58m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/782f3510c3f94d3a944f061692816151 : SUCCESS in 1h 24m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2a84668996cc419b98ec6fd6d278f87f : SUCCESS in 1h 10m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/da3ad6cbb7ad4353bd10bb03a5e86886 : SUCCESS in 57m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad5e001beb5b41aebc1caf6c2d766fe3 : SUCCESS in 1h 53m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74573d1c85174111ac26d7627df096eb : SUCCESS in 1h 52m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77bb8824e4034e9e85f76c83d28e82e7 : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":16},{"id":"469250366628695ad4fa582879200bae6329dc80","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-04 11:03:55.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"1ad35a72127ba4bfd774a6624a8d13a93567e654","revisions":{"e04f8a225f6693a79dde1e2e9cdabc0263471106":{"kind":"REWORK","_number":1,"created":"2023-03-15 17:08:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/1"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e04f8a225f6693a79dde1e2e9cdabc0263471106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e04f8a225f6693a79dde1e2e9cdabc0263471106"}]},"branch":"refs/heads/master"},"849088a7ee3512c0fd07f579470b2d98ba2a2149":{"kind":"REWORK","_number":2,"created":"2023-03-16 11:29:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/2"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-16 11:28:58.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849088a7ee3512c0fd07f579470b2d98ba2a2149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849088a7ee3512c0fd07f579470b2d98ba2a2149"}]},"branch":"refs/heads/master"},"bab4062dfc13e78086ef505ffef07197f58f27f6":{"kind":"REWORK","_number":3,"created":"2023-03-24 14:17:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/3"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-23 08:14:49.000000000","tz":60},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bab4062dfc13e78086ef505ffef07197f58f27f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bab4062dfc13e78086ef505ffef07197f58f27f6"}]},"branch":"refs/heads/master"},"358a153918864222cabc9322b59aa2b4d663c83e":{"kind":"REWORK","_number":4,"created":"2023-03-25 16:11:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/4"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-24 10:27:08.000000000","tz":60},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/358a153918864222cabc9322b59aa2b4d663c83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/358a153918864222cabc9322b59aa2b4d663c83e"}]},"branch":"refs/heads/master"},"d1eae6de4d2fd5eb96ed7de584d290095726b91e":{"kind":"REWORK","_number":5,"created":"2023-03-28 17:09:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/5"}}},"commit":{"parents":[{"commit":"b73399fa746d951a99fdf29950a1c0a801e941a2","subject":"Merge \"Revert \"Ensure vlan network traffic is not centralized\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73399fa746d951a99fdf29950a1c0a801e941a2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-28 17:09:00.000000000","tz":120},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1eae6de4d2fd5eb96ed7de584d290095726b91e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1eae6de4d2fd5eb96ed7de584d290095726b91e"}]},"branch":"refs/heads/master"},"c9daa594e4dabc966b37a81a613542739fc54440":{"kind":"REWORK","_number":6,"created":"2023-04-17 16:12:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/6"}}},"commit":{"parents":[{"commit":"18bc5b82a1d313a27411a9a88488f1cc306f0fee","subject":"Merge \"OVN: Always try and create a metadata port on subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18bc5b82a1d313a27411a9a88488f1cc306f0fee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-15 02:38:06.000000000","tz":120},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9daa594e4dabc966b37a81a613542739fc54440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9daa594e4dabc966b37a81a613542739fc54440"}]},"branch":"refs/heads/master"},"a5e2b908a7ff6c71515fa3420c1f8e1fe0543257":{"kind":"REWORK","_number":7,"created":"2023-05-09 10:47:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/7"}}},"commit":{"parents":[{"commit":"9319ba00a986e693eeb4707055aa53111601e7c8","subject":"Merge \"rbacs: clean-up to use defined constants ACCESS_*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9319ba00a986e693eeb4707055aa53111601e7c8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 10:47:31.000000000","tz":120},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5e2b908a7ff6c71515fa3420c1f8e1fe0543257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5e2b908a7ff6c71515fa3420c1f8e1fe0543257"}]},"branch":"refs/heads/master"},"eeb3cceb470f7ee2514d941debc8a70ccebcac05":{"kind":"REWORK","_number":8,"created":"2023-05-09 12:04:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/8"}}},"commit":{"parents":[{"commit":"9319ba00a986e693eeb4707055aa53111601e7c8","subject":"Merge \"rbacs: clean-up to use defined constants ACCESS_*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9319ba00a986e693eeb4707055aa53111601e7c8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 12:04:13.000000000","tz":120},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eeb3cceb470f7ee2514d941debc8a70ccebcac05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eeb3cceb470f7ee2514d941debc8a70ccebcac05"}]},"branch":"refs/heads/master"},"c86cbbaeace4f27807e0f6406ef31f5282615550":{"kind":"REWORK","_number":9,"created":"2023-05-10 09:28:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/9"}}},"commit":{"parents":[{"commit":"9319ba00a986e693eeb4707055aa53111601e7c8","subject":"Merge \"rbacs: clean-up to use defined constants ACCESS_*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9319ba00a986e693eeb4707055aa53111601e7c8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-09 19:41:10.000000000","tz":120},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c86cbbaeace4f27807e0f6406ef31f5282615550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c86cbbaeace4f27807e0f6406ef31f5282615550"}]},"branch":"refs/heads/master"},"d6fb10c89ead366a5d6fd15f9aaccf2df7df0448":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-06-14 18:59:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/10"}}},"commit":{"parents":[{"commit":"6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e","subject":"Merge \"Allow Multiple External Gateways\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-06-14 18:59:46.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6fb10c89ead366a5d6fd15f9aaccf2df7df0448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6fb10c89ead366a5d6fd15f9aaccf2df7df0448"}]},"branch":"refs/heads/master","description":"Rebase"},"47d2666d823cbd77835e49594f215ed0421dfc32":{"kind":"REWORK","_number":11,"created":"2023-06-26 13:33:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/11"}}},"commit":{"parents":[{"commit":"b86ca713f7d422489f2526c1a94ffc8393b3ed75","subject":"Merge \"Explicitly define the subnet creating a new port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b86ca713f7d422489f2526c1a94ffc8393b3ed75"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-26 13:33:19.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47d2666d823cbd77835e49594f215ed0421dfc32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47d2666d823cbd77835e49594f215ed0421dfc32"}]},"branch":"refs/heads/master"},"3728b51b4ffee36c41024bc57d959dd277747adc":{"kind":"REWORK","_number":12,"created":"2023-06-27 09:18:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/12"}}},"commit":{"parents":[{"commit":"0c5d4b872899497437d1399c845be756103a46d3","subject":"Merge \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c5d4b872899497437d1399c845be756103a46d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-26 17:56:34.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3728b51b4ffee36c41024bc57d959dd277747adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3728b51b4ffee36c41024bc57d959dd277747adc"}]},"branch":"refs/heads/master"},"86d99a8d7fd966783df583ad7b2eda0bd71f08e7":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-08-16 15:35:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/13"}}},"commit":{"parents":[{"commit":"186e87e3894e8ac1151425f93d51b819002c02cb","subject":"Merge \"[OVN] ovn-db-sync check for router port differences\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/186e87e3894e8ac1151425f93d51b819002c02cb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-08-16 15:35:24.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86d99a8d7fd966783df583ad7b2eda0bd71f08e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86d99a8d7fd966783df583ad7b2eda0bd71f08e7"}]},"branch":"refs/heads/master","description":"Rebase"},"2dcdbd7893ef006ae982ee4da200962a69a46786":{"kind":"REWORK","_number":14,"created":"2023-09-21 15:01:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/14"}}},"commit":{"parents":[{"commit":"dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177","subject":"Merge \"Open the 2024.1 (Caracal) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-12 01:40:39.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2dcdbd7893ef006ae982ee4da200962a69a46786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2dcdbd7893ef006ae982ee4da200962a69a46786"}]},"branch":"refs/heads/master"},"1d1815d277000b69fa98a25d88437002c9260248":{"kind":"REWORK","_number":15,"created":"2023-10-04 13:45:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/15"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-03 22:01:26.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d1815d277000b69fa98a25d88437002c9260248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d1815d277000b69fa98a25d88437002c9260248"}]},"branch":"refs/heads/master"},"1ad35a72127ba4bfd774a6624a8d13a93567e654":{"kind":"REWORK","_number":16,"created":"2023-10-05 10:37:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/877533/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/877533/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/877533/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/877533/16"}}},"commit":{"parents":[{"commit":"a76aa019772cc35b07a404c213b327169c188243","subject":"[OVN] Add the default condition check in ``PortBindingChassisEvent``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a76aa019772cc35b07a404c213b327169c188243"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:51:56.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 10:37:36.000000000","tz":0},"subject":"Replace \"tenant_id\" with \"project_id\" in IPAM engine","message":"Replace \"tenant_id\" with \"project_id\" in IPAM engine\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\nChange-Id: I5e1887ddf1dbbd8332663e085084852591ae75cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ad35a72127ba4bfd774a6624a8d13a93567e654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ad35a72127ba4bfd774a6624a8d13a93567e654"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
