)]}'
{"id":"openstack%2Fneutron~877585","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ib8340d9430b946a446edf80886c49fbac729073c","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2009221","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-16 18:15:43.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-03-16 16:24:37.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-16 16:17:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Ib8340d9430b946a446edf80886c49fbac729073c","subject":"[OVS] Allow custom ethertype traffic in the ingress table","status":"MERGED","created":"2023-03-16 09:17:17.000000000","updated":"2023-03-16 18:16:53.000000000","submitted":"2023-03-16 18:15:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877585-bug/2009221","meta_rev_id":"7fb39877e44424d4709dbf9019708967d4179042","_number":877585,"virtual_id_number":877585,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 18:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-16 16:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-03-16 16:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-03-16 16:24:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-16 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 09:54:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-16 11:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 16:08:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-16 16:08:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"9a86ad0cdb53aa2dacd6aedc2c415692d7258fb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 09:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"298353f876289debe0b791755269d4aa57eea984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 09:54:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b0c343e03814f6bb6810478eb3231cf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd58a03c83ae485abe2f5fa70772697b : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5b8e8410db04e488c02ded211956502 : SUCCESS in 22m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2964966f12f3ce45c09d2594fe6e1c6a25d30f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:10:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b1736161ff4b38a1d6e6af8dbfca4f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/faabfa3f58764d41baefdf2f12909781 : SUCCESS in 1h 10m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/366278bbc908436c9cab76d2a78466d5 : SUCCESS in 1h 10m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/405276c0680c455ca2dd1414b9e0b632 : SUCCESS in 1h 22m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de12430d7cc142b2b714536999d79cfd : SUCCESS in 33m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad29590fb69d4bc9abbd5dacf1f657c7 : SUCCESS in 12m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6558794c956343cb8465f8da9f228df6 : SUCCESS in 22m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a556e056fc114b12b50ab2cdcebde45b : SUCCESS in 23m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8575b5ebd6478ba32bd8350f12c285 : SUCCESS in 9m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/335c58a6dead4a06b6698695bbd2936c : SUCCESS in 22m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/25f33585f0de4f089e33b60b5cdefd8f : SUCCESS in 47m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7edde25e08534499b77e90496af0c6ed : SUCCESS in 1h 33m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/df2a1cf496574a9ca0f960f5413df331 : SUCCESS in 1h 21m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/87f62826a1594c998b1aa21bf63a842b : SUCCESS in 1h 25m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83a18c3988cc4190b6a68ef504815cef : SUCCESS in 1h 06m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/52d2276af4444a638f79e8cf181b0d88 : SUCCESS in 1h 48m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3b1c9e93939144f0a94ca8d0a3be4149 : SUCCESS in 1h 45m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/345a68fd2a0d4efd96a463b277fe352f : SUCCESS in 1h 20m 40s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3253567feae044a28a41aff4f6c70f21 : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":1},{"id":"daf4666958c2c5b213ed21e97ac45d753ab38ec1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-16 16:07:58.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"567f0e1eb57a9ad4d235f7e28e0e44db3ee5432c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-16 16:17:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff0f6e4757975cd3f64647b4fcd73a0243d42942","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-16 16:24:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"551f77e64440d147a7288552bf0e6f1e872d0e9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 16:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3b6c51931c72eec3abc44786d245e29e282b968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 18:15:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63717851217946db9ec4e435d3d18019\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8546163ae326484abe48d6c12bcf62de : SUCCESS in 1h 12m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cbf5ac94f74140c399e4e1739ed88fec : SUCCESS in 1h 11m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1be06fc0bda745cf958ce27baae242c5 : SUCCESS in 1h 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7687f36a6043728953bf8a51c1b7b5 : SUCCESS in 13m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cafd1ac4ed2a49f483cf3948388b1775 : SUCCESS in 27m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5d52c683c7942ed896a47737daba619 : SUCCESS in 29m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7c249ff8564081a3f35adcb4eae0ac : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef26aa526c04461498fb74dcbd910146 : SUCCESS in 52m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/145505af4365484689b3508b2a20a673 : SUCCESS in 1h 45m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d85bc1ab721498e83ee9b71de2d19b8 : SUCCESS in 1h 12m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/64301906a0f849dd88cfb0a39ee620c5 : SUCCESS in 1h 03m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c923ef398ec48e789b28810673b4cc4 : SUCCESS in 1h 24m 08s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c4f03918df8347dc8a51e1b3e84e981c : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2915968c7176eaf53d0a7a6148b3f9b6f97fa748","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 18:15:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7fb39877e44424d4709dbf9019708967d4179042","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 18:16:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/114785b91c0042a6a67a754c8712ae86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8ebd205dbf40b6bf1fa994ec266148 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17faa288cedcdb7a365be3105f281cfd957a1464","revisions":{"17faa288cedcdb7a365be3105f281cfd957a1464":{"kind":"REWORK","_number":1,"created":"2023-03-16 09:17:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/877585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/877585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/877585/1"}}},"commit":{"parents":[{"commit":"8c89c3e5d0157fdf2ad49b18b1a4be043e29d30a","subject":"Only create a frozen  Row on matching events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c89c3e5d0157fdf2ad49b18b1a4be043e29d30a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-05 21:12:55.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-16 09:16:43.000000000","tz":60},"subject":"[OVS] Allow custom ethertype traffic in the ingress table","message":"[OVS] Allow custom ethertype traffic in the ingress table\n\nThis patch is a partial revert of [1], reinstantiating the code merged\nin [2]. This patch is the complementary to [1]: the traffic with\ncustom ethertypes is allowed in the ingress processing tables, same\nas [1] is allowing all traffic from the virtual machine ports in this\nhost to leave the node. Both, this patch and [1], are bypassing the\nOVS firewall just for the traffic with the configured allowed\nethertypes and just for/to the local ports and MAC addresses.\n\nAny other traffic not coming from a local port or with destination\na local port, will be blocked as is now.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/678021\n[2]https://review.opendev.org/c/openstack/neutron/+/668224/\n\nCloses-Bug: #2009221\nRelated-Bug: #1832758\nChange-Id: Ib8340d9430b946a446edf80886c49fbac729073c\n(cherry picked from commit 008277b8c12d99438951a308b278203fa7a7c3ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17faa288cedcdb7a365be3105f281cfd957a1464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17faa288cedcdb7a365be3105f281cfd957a1464"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
