)]}'
{"id":"openstack%2Fneutron~877601","triplet_id":"openstack%2Fneutron~master~Id61031d7e707d0ba7b007bae0c9e0f59b8b40f8b","project":"openstack/neutron","branch":"master","topic":"bug/2011687","attention_set":{},"removed_from_attention_set":{"34564":{"account":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"last_update":"2023-03-20 10:49:24.000000000","reason":"Change was submitted"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2023-03-20 10:49:24.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-03-19 19:17:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id61031d7e707d0ba7b007bae0c9e0f59b8b40f8b","subject":"doc: state that O flag can be 0 in dhcpv6-stateful","status":"MERGED","created":"2023-03-16 09:33:09.000000000","updated":"2023-03-20 10:51:47.000000000","submitted":"2023-03-20 10:49:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"877601-bug/2011687","meta_rev_id":"b565706a4ddd0c486859cd110d3da86fdeb9beba","_number":877601,"virtual_id_number":877601,"owner":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 10:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-03-20 08:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-03-20 08:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-03-19 19:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2023-03-20 02:34:37.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-03-20 08:52:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-03-20 08:52:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 10:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 17:57:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-03-20 02:34:37.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2023-03-20 08:52:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-20 08:52:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"28f3a5d948504cf95c08854d4dbe99d99cef615c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-16 09:33:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e9e07c7b01cd48e4e3cb636696752b4af8e897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 10:00:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5520a0c7f48454fa13f0d589a3ba220\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b86826d245a467d8e61c7d2d3f2ef3f : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259df2b1b8294a21806b782ad7690e51 : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e93cd685eb8d262a8ff235cb997f1e4c74cc09ff","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-16 17:57:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"205d4039760c4fbad6fb64a09546bcebeff5519d","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-17 08:41:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72b44a45b000f046a6a5725561d70ccf37934940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-17 10:16:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90c149b2a629a180a216dabc99829cd4a36847a4","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-17 10:26:40.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4df1d0bbcb634ec105f9adbfddc5bf0197d42f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 10:34:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d8156aedff4c36bff42f2e90b2c837\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7b46159fdb4b4985de91f624125d14 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d4d7a3c40e4775bd86101c179c1ebe : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f46b1ff7b984362208d0864052aebc52750b5b27","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-17 15:56:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd73e3c241e098a0bdcdb43644e428750e2d882f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-19 11:18:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df33ac091e5fbab4df7f730c00e8ad05ed3da58c","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-19 11:27:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dec5b456750d13c83eb3423be4e2e7340cace0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-19 11:28:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7261d73ee24821836c36098507555d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874cb5a97d324b568c95dfc991d0fb18 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d00f91c4279d48559683dd5ae16bcf5d : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":3},{"id":"524bf59fe851751499c52220a2fe52639a3054c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-19 19:17:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a5a5642cad7fadc0b7fbcf7c5e9174c3657c39a7","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-03-20 02:34:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"71ddeeae6d125b32240c97090673aae1866b3b0e","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-03-20 03:48:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a9c9ad7d128e5683a318e37d2cc8a0969fb0c0b2","author":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"date":"2023-03-20 04:16:02.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"936f1d1044413c3fcce4730a29d1002c2a337967","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-20 08:52:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db58a4e15af901480f7ed09db8bec22c6b627467","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-20 08:52:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7dbce8a6978ec89f363767587243594d76a2d03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 08:53:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"51449041ba016c5b2aad64b0a566f9217964899e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:49:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0641f41ab16a4a60aa267c344a9b5f40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f95fca68ce48d492aa3e3de6c79b98 : SUCCESS in 15m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70029ec84d2c4000892a605ef8ff763d : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":3},{"id":"49efcdba5eba6e31d0ff8ff42be7b0562fda1cc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:49:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b565706a4ddd0c486859cd110d3da86fdeb9beba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:51:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faaab8bcb4d04853a949664eb1c6bfba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36043b9ea110422ebfe866ec0a56564a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c97dcfd03fccd8bfc011a26a2cbc62b4e6fa52ad","revisions":{"a9e687d818194176a48ec2a2a32e4e6399554a16":{"kind":"REWORK","_number":1,"created":"2023-03-16 09:33:09.000000000","uploader":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"ref":"refs/changes/01/877601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/877601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/877601/1"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Yamato Tanaka","email":"yatanaka@redhat.com","date":"2023-03-16 09:15:34.000000000","tz":540},"committer":{"name":"Yamato Tanaka","email":"yatanaka@redhat.com","date":"2023-03-16 09:30:53.000000000","tz":540},"subject":"doc: state that O flag can be 0 in dhcpv6-stateful","message":"doc: state that O flag can be 0 in dhcpv6-stateful\n\nThe current document states that O flag will be 1 when ipv6_ra_mode is\ndhcpv6_stateful, but the actual implementations of both ml2/OVS and\nml2/OVN set O flag to 0 according to the following implementations:\n\nML2/OVS\nhttps://github.com/openstack/neutron/blob/f545c002dc699f849026ccd1bad403b2933373aa/neutron/agent/linux/ra.py#L53-L55\n\nML2/OVN\nhttps://github.com/ovn-org/ovn/blob/a5238e6234d17e119dca952d9e43c36dce057d5e/controller/pinctrl.c#L3733-L3734\nhttps://github.com/ovn-org/ovn/blob/a5238e6234d17e119dca952d9e43c36dce057d5e/lib/actions.c#L3349-L3350\n\nThis actual behavior looks correct because O flag can be either 1 or 0\nwhen M flag is 1, according to the following statement of RFC 4861:\n\n  https://www.rfc-editor.org/rfc/rfc4861#section-4.2\n  If the M flag is set, the O flag is redundant and can be ignored\n  because DHCPv6 will return all available configuration information.\n\nTo make consistency between the documet and actually behavior, this\ncommit changes the document to state that O flag can be either 1 or 0\nwhen ipv6_ra_mode is dhcpv6_stateful.\n\nCloses-Bug: #2011687\nChange-Id: Id61031d7e707d0ba7b007bae0c9e0f59b8b40f8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9e687d818194176a48ec2a2a32e4e6399554a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9e687d818194176a48ec2a2a32e4e6399554a16"}]},"branch":"refs/heads/master"},"70bae028fed07e22c6df9f1b268f41d587ec1993":{"kind":"REWORK","_number":2,"created":"2023-03-17 10:16:22.000000000","uploader":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"ref":"refs/changes/01/877601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/877601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/877601/2"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Yamato Tanaka","email":"yatanaka@redhat.com","date":"2023-03-16 09:15:34.000000000","tz":540},"committer":{"name":"Yamato Tanaka","email":"yatanaka@redhat.com","date":"2023-03-17 10:16:12.000000000","tz":540},"subject":"doc: state that O flag can be 0 in dhcpv6-stateful","message":"doc: state that O flag can be 0 in dhcpv6-stateful\n\nThe current document states that O flag will be 1 when ipv6_ra_mode is\ndhcpv6_stateful, but the actual implementations of both ml2/OVS and\nml2/OVN set O flag to 0 according to the following implementations:\n\nML2/OVS\nhttps://github.com/openstack/neutron/blob/f545c002dc699f849026ccd1bad403b2933373aa/neutron/agent/linux/ra.py#L53-L55\n\nML2/OVN\nhttps://github.com/ovn-org/ovn/blob/a5238e6234d17e119dca952d9e43c36dce057d5e/controller/pinctrl.c#L3733-L3734\nhttps://github.com/ovn-org/ovn/blob/a5238e6234d17e119dca952d9e43c36dce057d5e/lib/actions.c#L3349-L3350\n\nThis actual behavior looks correct because O flag can be either 1 or 0\nwhen M flag is 1, according to the following statement of RFC 4861:\n\n  https://www.rfc-editor.org/rfc/rfc4861#section-4.2\n  If the M flag is set, the O flag is redundant and can be ignored\n  because DHCPv6 will return all available configuration information.\n\nTo make consistency between the documet and actually behavior, this\ncommit changes the document to state that O flag can be either 1 or 0\nwhen ipv6_ra_mode is dhcpv6_stateful.\n\nCloses-Bug: #2011687\nChange-Id: Id61031d7e707d0ba7b007bae0c9e0f59b8b40f8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70bae028fed07e22c6df9f1b268f41d587ec1993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70bae028fed07e22c6df9f1b268f41d587ec1993"}]},"branch":"refs/heads/master"},"c97dcfd03fccd8bfc011a26a2cbc62b4e6fa52ad":{"kind":"REWORK","_number":3,"created":"2023-03-19 11:18:04.000000000","uploader":{"_account_id":34564,"name":"Yamato Tanaka","email":"yatanaka@redhat.com","username":"yatanaka-1007"},"ref":"refs/changes/01/877601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/877601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/877601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/877601/3"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Yamato Tanaka","email":"yatanaka@redhat.com","date":"2023-03-16 09:15:34.000000000","tz":540},"committer":{"name":"Yamato Tanaka","email":"yatanaka@redhat.com","date":"2023-03-19 11:17:55.000000000","tz":540},"subject":"doc: state that O flag can be 0 in dhcpv6-stateful","message":"doc: state that O flag can be 0 in dhcpv6-stateful\n\nThe current document states that O flag will be 1 when ipv6_ra_mode is\ndhcpv6_stateful, but the actual implementations of both ml2/OVS and\nml2/OVN set O flag to 0 according to the following implementations:\n\nML2/OVS\nhttps://github.com/openstack/neutron/blob/f545c002dc699f849026ccd1bad403b2933373aa/neutron/agent/linux/ra.py#L53-L55\n\nML2/OVN\nhttps://github.com/ovn-org/ovn/blob/a5238e6234d17e119dca952d9e43c36dce057d5e/controller/pinctrl.c#L3733-L3734\nhttps://github.com/ovn-org/ovn/blob/a5238e6234d17e119dca952d9e43c36dce057d5e/lib/actions.c#L3349-L3350\n\nThis actual behavior looks correct because O flag can be either 1 or 0\nwhen M flag is 1, according to the following statement of RFC 4861:\n\n  https://www.rfc-editor.org/rfc/rfc4861#section-4.2\n  If the M flag is set, the O flag is redundant and can be ignored\n  because DHCPv6 will return all available configuration information.\n\nTo make consistency between the documet and actually behavior, this\ncommit changes the document to state that O flag can be either 1 or 0\nwhen ipv6_ra_mode is dhcpv6_stateful.\n\nCloses-Bug: #2011687\nChange-Id: Id61031d7e707d0ba7b007bae0c9e0f59b8b40f8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c97dcfd03fccd8bfc011a26a2cbc62b4e6fa52ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c97dcfd03fccd8bfc011a26a2cbc62b4e6fa52ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
