)]}'
{"id":"openstack%2Fneutron~877675","triplet_id":"openstack%2Fneutron~master~I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-21 08:23:36.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-23 15:14:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2023-04-28 17:36:33.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2023-04-28 17:36:33.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-04-28 17:36:33.000000000","reason":"Change was submitted"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2023-04-28 17:36:33.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-04-28 17:36:33.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-04-28 17:36:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a","subject":"Add support for localnet_learn_fdb OVN option","status":"MERGED","created":"2023-03-17 15:01:01.000000000","updated":"2023-04-28 17:38:02.000000000","submitted":"2023-04-28 17:36:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":56,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877675","meta_rev_id":"89619ed9b8474dabfc153f3b11fc0d69ec70e864","_number":877675,"virtual_id_number":877675,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-13 10:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-28 17:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-21 08:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-21 07:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-24 09:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-04-21 08:23:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 15:02:55.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-17 15:02:55.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2023-03-17 15:26:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 16:06:17.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-03-17 16:31:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-03-17 16:56:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 18:44:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 07:19:15.000000000","updated_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"REVIEWER"},{"updated":"2023-03-21 17:46:26.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-22 10:08:46.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-04-24 09:39:34.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"6c3ce7287851966bfdf7b83141ebc8c8bf65bf71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 15:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08a407291c7d711a2fcbe4fe30d6bc4c352700d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 15:26:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8508ad6b3546eda07d4ff924f296d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52e86fa9c0b8447ab509ea51dae8fc45 : FAILURE in 23m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e80560c7ad2f4c9995a29a7e8f0964cc : FAILURE in 24m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c8401468414743d8b817abce27bc3154 : FAILURE in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"702aacc6487c6f61d1d1f8db5916004baf1b3c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-17 15:34:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49247a63fc87864160863584c7327b21411eb1cc","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-03-17 15:48:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6757b3cefa1033f727343f6298a8534f6d8c4a62","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-03-17 16:06:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"65bb686272ee37bc7e67470f480b8532b052b07e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 16:07:49.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fcedb19b50f4a22ba314bc160717f42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e777383d645e450d9edd65a5861b636a : FAILURE in 27m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03f7cf3825b7458e8689fad475ab50f9 : FAILURE in 31m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ade73a16c7041e3967ccb73b5852865 : FAILURE in 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"478f1fb9313aed8599e478fc0d62ee7c4e1d8de0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-17 16:31:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a967cd6c1bfa21035cf2cfa292f9c0b3a1ef43ad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-17 16:56:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/75/877675/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d333648deae6673e6605022e92460a77ba2b5ae8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-17 17:35:40.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/75/877675/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"81b2a625179b0a3652024b1e9e88b488fc905cfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 18:44:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/215bded939fb469d87e6739ed8eb810d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/43740b88a6ff4263954aca9f47584907 : SUCCESS in 1h 37m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/776975343bbe4aaa9453961cb69d9d2a : SUCCESS in 1h 38m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba2132a17cce497ab9a359de60687bac : SUCCESS in 1h 32m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e51448f339748e09d8a36a2531050b9 : SUCCESS in 1h 39m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27d2d68af5cf4074adfcc9284e38c3e4 : TIMED_OUT in 3h 03m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a02642a52c5442f7b58e1ece33849245 : SUCCESS in 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86b28674841b4d278ad476fc87159010 : FAILURE in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/441c8b3272c847b1a6f05f481d9e74ab : SUCCESS in 15m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9feea8fb22394495afbdf3b6e45f5f78 : FAILURE in 22m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fcb26940d0e47f4b4bb7b3ddf0f55b8 : FAILURE in 25m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6eeb07e9db7743c69988c742649922cb : FAILURE in 24m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29377a284cf74bf0be5708e8cdacada1 : SUCCESS in 13m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bbeaa5e880114d9087af55a633413a2d : FAILURE in 22m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e590063b18848ada6408ccc5612c937 : FAILURE in 45m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcf00695180f4067ae70e459e137409f : SUCCESS in 2h 25m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c0c95f2702e4a8099058d72b4c83bf6 : SUCCESS in 50m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ddb6b7054a774fd88b08bed4e49f4fa5 : SUCCESS in 59m 09s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/42aa54f229014ca2b0e497e478c91a8a : SUCCESS in 1h 12m 13s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a1619bd77bb49c6930a6cbc344856cb : SUCCESS in 54m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d127ffbcba71460f81ee5a6ee86e159b : SUCCESS in 1h 27m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c45c01caf10040b1b5dd03a219762944 : SUCCESS in 1h 10m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d73a450b84f94ad6bdb387e42bccdb6c : FAILURE in 1h 12m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/190dcb6f7c8045178b7b681d28114ffa : SUCCESS in 1h 34m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac7471d5d4274d91819e94db89e4343d : SUCCESS in 1h 30m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac5fd39089f24fb19e8cb400b25ef0c5 : FAILURE in 1h 20m 34s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/568abf3c3d9740c597b10283d8e624fa : FAILURE in 1h 25m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e18cad0b7ea0fb120f4f9e8d80fc59b20f3852de","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2023-03-21 07:19:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fd1c25d40c85372362359b24ed431185970a385","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-21 09:11:46.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"81d58e98433624e1f1d72b8d1896e8e26eb35048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-21 13:57:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"69aa1c6da1bc8a3a7b2c343fd28cc4650e26f6e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 14:42:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09a0bedecf52426e8e67d72eff316e96\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/248f0157b8074e02b5534be04c7e9176 : SUCCESS in 44m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fc08c082c0f450ba67bbff8e8a0710d : SUCCESS in 41m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/725d2bde2a864c8c87e9d7e1b7782b1f : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"824210a27fc832ee701feafb8c5999d5da267c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 16:10:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6901932290094e8b965a7eb0c5c0aed0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/db8cb5f488a34c8ebe5e9e77979df6c4 : SUCCESS in 1h 53m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4d3b55601404283afc5f6aac9428ba5 : SUCCESS in 1h 44m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d99d52f4e6dd4166bbc866a27c3e3009 : SUCCESS in 1h 33m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8998bd1e79d44be9b4878fd69ed01e3a : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/16e4bda709f24bb8a17ac5ad27045e60 : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82bf65d27a574458ab261ba7334ccfa4 : SUCCESS in 35m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0bada16026b434691ea0004af5ab308 : SUCCESS in 36m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a4f18a2878c485cb0d813e0c5091599 : SUCCESS in 14m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65131d239c494219a43dc10e59b2e743 : SUCCESS in 30m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46da53aab5ea410db208dca0a5ba90b7 : SUCCESS in 30m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9642613fa6814be6b0e30d839b421045 : SUCCESS in 30m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542c59affb13449a9c1151a45fedb32f : SUCCESS in 12m 01s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c4eba292466c4c67a10aa2727d892659 : SUCCESS in 26m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/865179055d2d48dea03ae6543f298712 : SUCCESS in 42m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00426a63f95a454f81395668f194a1a1 : SUCCESS in 2h 05m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef23a731b6d14bc787fc9097706b2b4f : SUCCESS in 55m 19s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8831b84e91d4435fbd5326c1a276f4d9 : SUCCESS in 50m 36s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a2fa1a228f8441d388932e30b146a87e : SUCCESS in 1h 06m 12s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52387dc3d9cb4138ab3597f9b095efb6 : SUCCESS in 1h 07m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9997df3c468d4ddd828403e41fdb5564 : SUCCESS in 1h 35m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d148ebfdd9947c1af312cb36acd1f1e : SUCCESS in 1h 12m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3177a35b37844921b8f657f15bb634b0 : SUCCESS in 57m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/07dc366d7124428c92c137b5d31b613b : SUCCESS in 2h 00m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ea87f7989b043bea782b3a2cf1f3223 : SUCCESS in 1h 57m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3488ee40f05b44c5b7c209afc6eddab0 : SUCCESS in 1h 14m 15s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4ea1c598d69f4a6f81ea836e609452d2 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":3},{"id":"055256f28fb2c36e719c1f43d1371f4e9eb1681e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-21 17:01:39.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/75/877675/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fff756d9bc00117d8530ce844172e5c0459d8d74","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-21 17:46:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff397ed1bec663493fa4cba91d2171d7d15036cb","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-03-22 10:08:46.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1d400c54d43a3c798c5525d6e3eec480570c3037","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-22 10:37:35.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"15895093de70d7060e6ada71b796b3a5b8589129","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-03-22 11:04:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f2b0dfaa1c0c7f2fe8a696a1cad01a85b5e66fc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-03-23 15:14:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70497a3198fc589e25bb6081f3e365046cdd27e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-24 07:08:20.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"64e25cc86de5da94c318ae13ff759f822319e79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-24 07:08:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0b9e8b28979ac684267d471ae04c0c9fa04a272b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 07:48:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5ad1322463042e39c9fbba77ef277fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d9ce6df9ff041d8bbb8188a993ba76c : SUCCESS in 18m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/964d7caebb4445ec93a793744936b8d5 : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad03a758ed24438289b0b43f379aa88b : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"57cd897673d34503df798b627924cb21a77463e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-24 08:06:12.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/75/877675/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"3bcaaeb9684a6d4c6d2876160f0e1e9bbd9b33cc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-24 08:44:17.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/75/877675/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"ad73f8257a1360a09b94ff75fbe8baf2d2895b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 09:17:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c926005017a6412da5a8fde6557397ae\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/725acc8556b044dd9bbbdec951784fa5 : SUCCESS in 1h 30m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/61ddeb6a9084436cbfc57a3f47ac89d2 : SUCCESS in 1h 21m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eec3968694a04ee88523a07366c09ac4 : SUCCESS in 1h 24m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3aed0a890394479cb0fd1aff86411123 : SUCCESS in 1h 11m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4605b38d1db48c5bbc083f682e9df98 : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/69bdd2805d7f411f879e12faeb44de8d : SUCCESS in 29m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5673cf929ff412a82506cd75fedfe46 : SUCCESS in 29m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4beb4b3be134e29aa60ce9ce1cb175d : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d95e7378c854b0599386b6fbf238db0 : SUCCESS in 26m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b7fbd0aa893433c98c5fec7a04c81b5 : SUCCESS in 29m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c04a47c1a0744e8b558b019a617cf5d : SUCCESS in 25m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805e039e16994fc0a23db2214e7b0b76 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2152536822514150954ce0ec635cadb1 : SUCCESS in 9m 03s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/de5fcb08075646cab55d08d4e344df47 : SUCCESS in 24m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7d5372d7b4a4c08b33a0efdbe0f152d : FAILURE in 58m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/15fcecda85864339b68bd415ab1880a1 : SUCCESS in 1h 40m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cb3aaae7cbb4baca4b4b910693b8a60 : SUCCESS in 58m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/734329aa4e4042d4abcedf2ac39c0377 : FAILURE in 20m 34s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b0b610e2ada46639f542362c129d227 : FAILURE in 11m 54s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1d7e5c9efdb740bca95d5697aba406af : FAILURE in 45m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6620d54d6b940149292549d4199c26d : SUCCESS in 1h 28m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8bdfe6191a0a43f1aec24f4f89fec94f : SUCCESS in 1h 19m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/365f851003884272b4fea06fdb38adb1 : SUCCESS in 1h 05m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a76d40a916f4d609a5e3832183d1a48 : SUCCESS in 1h 32m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91cfa3a783ab4b468b1250de4f883f69 : SUCCESS in 2h 02m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e8928020012a4581b6b050157425f5e5 : SUCCESS in 57m 54s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ad75c2456ce947efb75a4c32674e8b18 : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":5},{"id":"12b24f475545c887651cdb98c28861923a4ae153","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-24 09:19:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d121c7426db600de16f4e4491969b0eaf1f062d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 09:40:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bedafc2afc01482f86150a826d2b3c3d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c01bfa7f87194f01811d7fb45cad0f1b : SUCCESS in 19m 08s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b20873b76c34690bdff3ce9f20eb690 : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f70782812414449babafcba1012fe6a8 : SUCCESS in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0fbeba9ac4662ee36b54fdc2dbd2bea8c078bdcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 11:28:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ddce679f45d4e71a9925368924a4c80\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/287f21c0ec1345a78e16581af47fd033 : SUCCESS in 1h 24m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9b70d131cea549c9a473fe5b39938760 : SUCCESS in 1h 48m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50c13d9dec5c4418859a565a80386170 : SUCCESS in 1h 21m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/92a0259d429744a48536226253109857 : SUCCESS in 1h 58m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1718d5747d2b4a97bf07869d847d871e : SUCCESS in 1h 15m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6098f344e9824686ab57bd00717e5833 : SUCCESS in 40m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c75250a3142466ea625301a56269ccb : SUCCESS in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec2f89f48064b1dbfd589547d9e8a0d : SUCCESS in 14m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8298330c4e549b589c529d4c95b5910 : SUCCESS in 28m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/799ab63b36ed418eac31c7d7ae1578b7 : SUCCESS in 24m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d44dca78c4cd47eb92eea946ee43bd39 : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a5af7babde41f286d7483ee4b7c2d7 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39417a3c9fdd435da05c26bde5bc43d6 : SUCCESS in 9m 05s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0bef4dbd3e284e07a4938589a0d9280e : SUCCESS in 24m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6a527e64c0e41908ec0ba7bd8c5161d : SUCCESS in 56m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/516611fabd694f459b9cf777a973fb47 : SUCCESS in 2h 07m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b90fa34d26542d9a1835d0569e6e711 : SUCCESS in 56m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3f0d22269d95457a9d1959cead4cd15d : FAILURE in 21m 16s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e4eb5005a69f4460834276dcbdf01e24 : FAILURE in 12m 37s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bad7c2af541479183069b0ef0d06447 : SUCCESS in 59m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9ccd49daf2e84e1a8fc08c89250e59c6 : SUCCESS in 1h 34m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7631ba29b9cf42c7877d3010bfe35ab3 : SUCCESS in 1h 24m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5531f507d584e01bd7903fa54ce767e : SUCCESS in 1h 13m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9054d856fa804cb0b510abd323f55b6f : SUCCESS in 1h 52m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa7fac3ffcb7470aa3e26765b16e50b6 : SUCCESS in 1h 50m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/418910f471fe41cb9ec15fc5e5732f10 : SUCCESS in 1h 22m 43s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/76ccd07b53214da59ea03091eca50469 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":5},{"id":"84480b34aefc98309db81e5610afada1d44cc0a6","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-03-24 13:41:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1ce26958d7343675b3613a40dea0a7425cdee921","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-31 05:29:22.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4113d49b933a6015b2c9a96988937e5700044840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-05 05:59:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1e7f0bc8fe9d224b4ce409fb935c66032dcd8156","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 06:45:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd7e456adf54748bda9450c6fdff979\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d37e795e153f41c59fa1dd1605d2b9b3 : SUCCESS in 35m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99e5319c97464e8ea114191df452a695 : SUCCESS in 44m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d5bf219c620455689264f4fd7b0473d : SUCCESS in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"be961b2ae8078fad1b15c7950eab2aed3a1fe398","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-05 06:56:30.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/75/877675/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"3a364132bbdf0c5c8c3410580bdc01eecfa8fe98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 08:03:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86fb38c3d262435aa4ae5f241e52f266\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/12bbba546a4b4ed597628b1136d36b98 : SUCCESS in 1h 34m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/41e23c50b826401094c6efcaaafc5556 : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1635f774cd7d4f98b8e99693167d26ce : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3655cea88db478ebc6aa10152b351c5 : SUCCESS in 1h 18m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5dfa3b422af0464db2070e8997a0eb12 : FAILURE in 1h 53m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/839187037e834df7bd972e40d9caeca1 : SUCCESS in 31m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99db1b9e77bc4db39ef642ec411de6de : SUCCESS in 39m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f891c68927ee4d2385b68cc049e56769 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a60853e21e143079ea31661130006d1 : SUCCESS in 32m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d63f7efafd0e485c8748f625ed05ff8a : SUCCESS in 23m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/038e18aa71d5424994736b02b70d80bd : SUCCESS in 22m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ad74e1470d411eb31bab7dc6ef69ab : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/421bd343d5494557ba47438449176c97 : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8d3ab45bd9ff400591e0750da7583974 : SUCCESS in 24m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dec616d5d8f441e6ae541cd2fdfe122a : SUCCESS in 1h 33m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b9bc19aa5784064aa15acc90ae50a4a : SUCCESS in 1h 25m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d01e02fcb5d4dd6b30c543833976588 : SUCCESS in 55m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0d957c74191b412793698802f8153a92 : FAILURE in 20m 27s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/99fa94e4cabd4264be1a4cd68a954cbb : FAILURE in 21m 56s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf3872a366b343b2b2e9452e48e992eb : SUCCESS in 48m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/41fa5b4dc568423280280f783b861b7e : SUCCESS in 2h 02m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/df829144ef7149de8f952bd94d35a42b : SUCCESS in 1h 15m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9fabd6624fd34379bddcdc601957f12d : SUCCESS in 1h 09m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/782610e64da540fab45bb85b1b4c30ad : SUCCESS in 1h 37m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a47882672d14cdb85b23b9e41b79988 : SUCCESS in 1h 22m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0407a198af9a43b6825d773c9d2968f7 : SUCCESS in 1h 25m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/dc30183d76644c52b3843122f3f6910b : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":6},{"id":"f1f185ba4215a21795551611e5e92ec7109b3111","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-05 08:50:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e18885514684db2b94d0b3c7977b45791fb2e758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 09:17:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11e24c29593644788d90699930c67563\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d6d03229fc542cfbd1a31e1fadca79e : SUCCESS in 26m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b1f4024db7749d8a0373a8eb45fbf40 : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2157cd9128ce47d9afc4487deab3a400 : SUCCESS in 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4eba0c2313c96d5c9a8aefb6293cf9af84aca249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 11:10:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a57ec4b141114068b35c443aa82e2800\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6f8f4b2a7ad445d9882a5aa10e398981 : SUCCESS in 1h 47m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eb8c1356bffc4ed4a2af6f6675ad7230 : SUCCESS in 1h 23m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f492db3482da4140960446ebf76defa2 : SUCCESS in 1h 59m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c7ee0e7e8ce948d0b81708c58b24be82 : SUCCESS in 1h 05m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/df46a1220bcb42db8b5801427c344ee0 : SUCCESS in 1h 38m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99dfe54d449642fb857c419987d583e6 : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5b2dfc64b3248759bb6ea154b6711d3 : SUCCESS in 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676a4dfcd409447ebc895ca5ed97cd78 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca3871b66bed438ba4c1c540f2f61169 : SUCCESS in 27m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58c59ad575d847ab96704accb571ffd7 : SUCCESS in 27m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1f6966dd9de45138a9840dc743c74e9 : SUCCESS in 26m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382c6ef78efe4bb3932677a51f5561b0 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/252e178955f24f84af3ffa442409303f : SUCCESS in 10m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c55e089ff6514b60ad2abacd352d5434 : SUCCESS in 23m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4633211b52045f49948c40ad0052c63 : SUCCESS in 1h 04m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d524b15474724982aa9e29e681a44867 : SUCCESS in 2h 18m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d4a0dc86d1a4c45b7fd8806d0185518 : SUCCESS in 55m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e87264b33b5e4b118bbb7312023ee637 : FAILURE in 10m 32s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d8b4888b56a949deb27fd690e3548ef5 : FAILURE in 11m 54s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/137d5e408a31430bb2dd462274ef6aab : SUCCESS in 56m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc05bb2620254d069779cbfd9dbf3ed8 : SUCCESS in 1h 34m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e36ccdefbd64e9aa9c5df5a14db64ca : SUCCESS in 1h 19m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e4e2b8de3652436ab99a52700e4a86f3 : SUCCESS in 1h 12m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aaac007ebe584963b0f4da4d480bdc6d : SUCCESS in 1h 33m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea936817f7334133abebbea920f5c67d : SUCCESS in 2h 05m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0380b447208e4b099e643fc1870fc063 : SUCCESS in 1h 28m 02s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/630b3ac6a9b14e7a930cde45c8fc96a7 : SUCCESS in 1h 20m 32s","accounts_in_message":[],"_revision_number":6},{"id":"23e1707f121d840940522763fc0939dd604631ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 14:33:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2bcb3982c12a9a552dc390ce65795a199f5ad3fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 14:35:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"36ea53d1ae4497517fda9d28c256b339e74667c9","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2023-04-13 02:51:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f84918ef0412502793412c743daf997af86fbb4b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-13 09:37:09.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"50f49a48e0eecf88ccffdcba8c8610b1fdedc395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-13 10:01:29.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"903271eba6a888de87f92b5d0b8070b5c25d897f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 10:30:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bfe0dd8ef0841488a946ffcdc227dfa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f95de7cdc43149d2b3d8f37ecc20f6b7 : SUCCESS in 23m 10s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72d00f2f52784d72b7a05cfe3e2a266d : SUCCESS in 27m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3094d476f5914a64adb0e93833ff7d4e : SUCCESS in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a324acb176c0e56af4df00b6db40a70457420295","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-13 10:38:29.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/75/877675/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e554b094f6325e2980b634195da5d049b6b0eaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 12:02:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3c6e52c40584d1e89c551613a8da524\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7cf3e6d9687a45f3a4f4eea08b2e9e8b : SUCCESS in 1h 26m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/927efdc851994d7f91e8a13b518878c8 : SUCCESS in 1h 30m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d963260f69714741baabe116f99a3f02 : SUCCESS in 1h 26m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a17cda22facc46498059a346cc4dcba6 : SUCCESS in 1h 11m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f3b22def72534339bbeb1b0a94220206 : SUCCESS in 1h 18m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f7795e4fc6454e6e8e8dcd93360330cf : SUCCESS in 30m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211c0a9de4f341839290b73074a0020e : SUCCESS in 37m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa1daeddcfb41b5a93ec18a86d3bab0 : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01e2f18fefb3433ca2d7f18a36fa9837 : SUCCESS in 24m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e82b2af4b9f4ff0bb488366592cc713 : SUCCESS in 29m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11c41666f29c4350afc1972de4866075 : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc38dad98f204f128c4bfbf42fac1944 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8372d24e96441b869b3881315d951d : SUCCESS in 12m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5bab6a54fe7b4a88b049459aab8462fe : SUCCESS in 22m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b11ab38e09e41db816bd4fb71106deb : SUCCESS in 58m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/33882b4a93eb423999807a6ea68ee8a6 : SUCCESS in 1h 52m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/62e1e933d88644ff93047bfd3c9834aa : SUCCESS in 1h 01m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca55457152f040e19ba6a08777cadab2 : FAILURE in 46m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4fd62cc9c1ae4963ab4a5eda0d184e4d : SUCCESS in 1h 38m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d57d02d93144b109579a01a0df8d76e : SUCCESS in 1h 34m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0e1493b45f474943b7f46b5e2b8c2bf7 : SUCCESS in 1h 11m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3b05d8c9c2034ffbb4d538eb58048e0d : SUCCESS in 1h 43m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a946a4822be14c979a61491b81df46e8 : SUCCESS in 1h 42m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/26e11aafd6a741fcb60ed30c782c299a : SUCCESS in 1h 09m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fc30c33e96154139966bea1f2d04ac4d : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":7},{"id":"f4148c828933fb77c088f97fa4f4945f50787b61","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-13 12:07:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3282a87c735e79dde390c2ec7924cc75ea590951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 12:41:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f07dbf159664a72b8db842c55e40297\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93d38251667249f4a60a189d0dc46ec8 : SUCCESS in 32m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebc4555e95f844c0a95b2da60b9793df : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dc8a4db94a54171a94fc01805294371 : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6bf2dfdb75196f42e29c38c6d963ffc017716f22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:08:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25f224b8c8654581a9629f6b8b03ff4c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a2f34a366d5b4b0dba9d8cc2cb997d7d : SUCCESS in 1h 44m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20cd00db0e154cdb8c168989612d40f5 : SUCCESS in 1h 18m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/96d37e6c5b89445294aa390fe01a163e : SUCCESS in 1h 27m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5be2e0257e17455ba77cb7fef5dc76bd : SUCCESS in 1h 15m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed8804340531478887cc69d790c37bbc : SUCCESS in 1h 22m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e8252bb864243099244b33cdfc1d171 : SUCCESS in 33m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52500dc7a1034fe1a3048057046654d2 : SUCCESS in 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ebf5f1f43564ab5b8a8799cb30d7f7a : SUCCESS in 14m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab7e2e006fb84e09b98468b89613f107 : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a148558e119498f844d402f303d9f87 : SUCCESS in 26m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea9f1abf34e6416986626c49c443afd7 : SUCCESS in 24m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4469a9e1a44f108741151851705132 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/561ef85e170a411e945fc7ab2a1f1892 : SUCCESS in 10m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fe685ca595944beebefaef11fdcfa64f : SUCCESS in 21m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a602a8231f54c33895da84d0c6663fa : FAILURE in 51m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a37674ddc4b4d6c9768d7886731b1ad : SUCCESS in 1h 28m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/24b8e3c27f96466b9ca37c3d53f75c25 : SUCCESS in 1h 05m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/608b94dff74c406c980c3305dd4dfc4e : FAILURE in 46m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/edd022124fe84b7d88fdeff2ef695987 : SUCCESS in 1h 30m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8836d268ac2f4c718a3a7fbd0cd9765b : SUCCESS in 1h 28m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0d0919eda764dce875388c6815b0e9a : SUCCESS in 1h 32m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3460cf440380451b853ec07faddc6aaf : SUCCESS in 1h 50m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d319c01ea484c42afae1991de516720 : SUCCESS in 1h 31m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c2807006f1c2495ca3e6344f1128c476 : SUCCESS in 1h 34m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/4ee6ba45a7284975988b5e024205dc13 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":7},{"id":"e2c2a3324afe79a9d49a4973ec8935c8094aed8c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-13 14:21:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4df5c60055f776dd554f3544a1f6dc5926ffe07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 15:13:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a306ac55ef1b49e78ce49abbceaeda5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a9a84e4d3a743f1b026a2a4f3fb42b0 : SUCCESS in 49m 01s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f5497c5e96248ffa474584ee23db870 : SUCCESS in 48m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86ee7d21946648e9bb0b84e60e54b18e : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79459f004f4efd04e38ae0b9c5a95efcd459d842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:59:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33153427990545d7bddf09946e006d34\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9c69dbfb10274ec7bcc298231898248f : SUCCESS in 1h 41m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eff20e9e00de4e5aa0a6db8b3f206130 : SUCCESS in 1h 19m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fb7645df17645e8ae898d624f6b82c0 : SUCCESS in 1h 26m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f04d64999de4f8fa1ac98da489d64b5 : SUCCESS in 1h 24m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f742297acd754d88846de3f517a5d520 : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3e1bf7fa0fd4a7a9d9aa5106b48e1b2 : SUCCESS in 41m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91c049621d6442269187d98fcbbd3ff5 : SUCCESS in 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fde459306194e58bfd883874ca4996f : SUCCESS in 14m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d8c2ade563e4b2b9668836027359ad4 : SUCCESS in 29m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c106e198f6b84d80869785077f7c2d18 : SUCCESS in 25m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/981f730eb5b144eab325949f2d619e22 : SUCCESS in 30m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b36d7db0544bc1abbd8c86213afec1 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca50772c4ae946a9bc549957f658d37b : SUCCESS in 12m 03s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0ab74255c0a84f90800da4d20bb2e418 : SUCCESS in 28m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/731949d604df4af0bdf7ece85991c677 : SUCCESS in 1h 03m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b89d349def746fba3ae79d4236a3163 : SUCCESS in 1h 39m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/21e1519ee1fe4f138fabc1c54c95bc76 : SUCCESS in 1h 01m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f3649b68c55f4be792efd9a70bfc65bb : SUCCESS in 44m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a545ef570d14ecfa676c103ab85e995 : SUCCESS in 1h 32m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/406f3faea8dc447390bb0f841e730495 : SUCCESS in 1h 13m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/296306f0c458452b9dedf00789e506d5 : SUCCESS in 1h 10m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7048718824cc4dfaa85fca3a4d16b39c : SUCCESS in 1h 50m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2db46c46a07a49e8856f36f3419f6531 : SUCCESS in 2h 00m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7200b09eb2b24e20a1b94694e57396ef : SUCCESS in 1h 19m 02s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/159be7b42c8c4de3bad41a2be0a91259 : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":7},{"id":"6c0a05c1766bc9fe9ee5cdc9dc9907b0053aea90","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-21 07:14:01.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"28cc1073aa3f16f1ff86b8c550d5fdd57d298f4f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-21 08:23:36.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e3bf40d666e1b0b05b7d69a407c40f660c5d31f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 08:24:13.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a160a4213f8c3db84122ca9b309b772c68e426fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:30:15.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/278b8cb16a7b4717929fe58ec67b20e0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/52124586d223413cbabb8539ce962b75 : FAILURE in 1h 58m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dcade57761d14bbc9224f16bb77e5c49 : SUCCESS in 1h 33m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f082959b03064804928d1656f346d185 : FAILURE in 5m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9168d1e3c2034be69b66a83ddb687317 : FAILURE in 16m 52s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ce14657894d6457598dee6e31feca6d4 : FAILURE in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc404ef13304304aad4a98da5a37c78 : SUCCESS in 13m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/692a4cb2bce447e78d5bbb216ad4ce71 : SUCCESS in 25m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ed3b85899414911893828fb141de4b6 : SUCCESS in 25m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/084c4c0826d94d7aa09e26b8592be783 : SUCCESS in 26m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6596bb530f8a4e56a9dd04bb5069a9e2 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb882536f76d4581b3b6b9c5517abbdb : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03b7c3404a4142b29e26023737de536b : SUCCESS in 58m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b580060523646449abccd462851deef : TIMED_OUT in 2h 32m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b66bd407edd04d5983ed476a87f88ca3 : SUCCESS in 1h 28m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84a7376b65e74cc1a6d8c319befb1a3b : SUCCESS in 43m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b69b538c5fc4abcaa6e2cf85b7ef3b6 : SUCCESS in 1h 00m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ceedf48889f44176b8446ca72f4570f4 : SUCCESS in 1h 11m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9499bc558ec34ab2a412e5169c44d4a7 : SUCCESS in 1h 14m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/0ad3d6728323445caba611ff40f059b7 : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":7},{"id":"71c856235ed7e7ec9f301c1da9c148776dce54c8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-24 05:17:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a8210091fa8b038dd2f3001a856228422993f165","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 06:03:25.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7cbec90677744129db99a59996b5353\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/355332cc7f7a4456972e8f9a9d77d13d : FAILURE in 3m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3e8166e03b342c595b7a7aae4a991f7 : SUCCESS in 37m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aabdaff9634b463c92a7bad6f8a2aa8c : SUCCESS in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"647238dbc4c0f77e80ac8a0165039f662f4a968e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 07:37:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3b24b706ffa4137a9ed28000f6ee8a9\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f7ccac51eee7441893a7a43cc7255296 : FAILURE in 10m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1c4bfca5974c4c0c8aad819824544a7c : FAILURE in 10m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c79f8111beef4e37995dbfc6f2ad49a3 : FAILURE in 10m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fdfaff9bd05a4322be881db312cf9a52 : FAILURE in 10m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/606c93064e9b45a0b181583ae4fc027f : FAILURE in 10m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26731213fe5c41e793872a8d3bb1772c : FAILURE in 10m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6957a541e709419f9acadaee2c77ca52 : SUCCESS in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188568df43e54f0b9012974b80fcd8e9 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b0d8c195c4d4221b58e8b0bde120ed0 : FAILURE in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6299ff7697d34000a7231457ebc34830 : SUCCESS in 22m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/484c4415e71c4500999826a073124641 : SUCCESS in 26m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad70cfb7e4954f6d840411c14022143b : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b3dd6b7c8c745aba4b0afe1766b2a4d : SUCCESS in 10m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/49227b9c577d4b14ae237d8da7104106 : SUCCESS in 23m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87e4802f49524041b96b0a4b12b9a9ff : SUCCESS in 54m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/caff20dfe4a7479bb8b9c8706f279557 : SUCCESS in 1h 50m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/16e7dec0958e4ef486927c76f0f33b85 : SUCCESS in 47m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c149654aecd14baebe4e6bd37bad8ba2 : SUCCESS in 1h 07m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7714bb4fbf45495b821578a8616cae46 : SUCCESS in 1h 28m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6086dc6509f3415291ff35871c133b87 : SUCCESS in 1h 21m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5e89d3d04504eb3abc3c49e7bd2350e : SUCCESS in 1h 10m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f7b17c6507b1413ea4796fcd52221465 : SUCCESS in 2h 08m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/846561638f4c496b88cf11a6fb761cf9 : TIMED_OUT in 2h 13m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c8f2b5728dc48b696193c297f91be5e : SUCCESS in 1h 17m 44s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a44a5860c81e4a3b980e7ceaf54a4ef3 : FAILURE in 23m 16s","accounts_in_message":[],"_revision_number":7},{"id":"204a11baaf1b3da1f2cd86f155922d6c3d55da9f","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-04-24 09:39:34.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b1e909dd51eb6bb4502511f1b87c24357115ca52","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-25 05:29:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d5eb40151954401dd8bbee6992d9909ae153e15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 05:52:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a02f47a4544ffca2f9fdb6f0282ecb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2f6949f0c864890ad7f41a3cd3cc59c : SUCCESS in 22m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3af9b3881ae472891ad62c3b81d1dc9 : SUCCESS in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"88769299d9b0c30f6bb9ab0a7c6f32fe7d7f1cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 08:03:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59ea449e6fe44c7f906a7cd5e7ac3d5e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/126307f5d5dc43b4ada6c3a19322bc58 : FAILURE in 10m 06s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/46d57d04a340495bad522aa81d232287 : FAILURE in 6m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/02abb372d5014ddfb29818fd8d0539a0 : FAILURE in 10m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/699ec76fb6f94c179dba02b8fe89ef7f : FAILURE in 10m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c3946a174414203b013305686503c3f : FAILURE in 10m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1223d0cf795843c4a07e3b3d28eea3fc : FAILURE in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d10c1ae30aa2436881f2ec449ae5e4c2 : SUCCESS in 45m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a68bc7e32b4d36bba706e4e898ced5 : SUCCESS in 13m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f80b0518da742ec80e1388a285c4b0f : SUCCESS in 25m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f050f6125fdf412aa572fa21841d54de : SUCCESS in 27m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8041b641836e49dcad81c043b1ef4e87 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beea5ca422eb477dbd4db80d57da8de6 : SUCCESS in 11m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7ec8e8dfaaf2422793defad2a5efcb4d : SUCCESS in 25m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c3ef2b990dc462bb676206fcc1030e8 : SUCCESS in 59m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fee121bc2bb4b00b0102dd0708ffccb : TIMED_OUT in 2h 32m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a84eafd76bc742da89b6d83194052061 : SUCCESS in 58m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e21045e222024de2916c6eacfcd01aaf : SUCCESS in 53m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/05609941507a4cd599ac3cc8f608353e : SUCCESS in 1h 00m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/952f82527b3d46aab046622ac1d8839b : SUCCESS in 1h 32m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/337a2684c62f4e99b5655f17d73f21d2 : SUCCESS in 58m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/88325892ce234e60a1af3a0a535c13a0 : SUCCESS in 1h 31m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9ef25e383d74564b7d910112ed3bf29 : SUCCESS in 1h 50m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd4681ed1d134bd8ab0dddf8f3970841 : SUCCESS in 1h 15m 17s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f9253cb7a8d949a09ce9e29dd5383ff5 : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":7},{"id":"e6a8aec6e216a209049bcdfacbdeab20d44715b5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-25 09:22:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"aba150b0ef4e02c6f93e485aae76ae5da6526d46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 10:00:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1106ce83b58242f8bc96a67516fceeec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7e1667969cb4546bef725af61d51e00 : SUCCESS in 23m 48s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df0def08332242549e77b6d26917b30b : NODE_FAILURE Node request 300-0021054802 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/63251309e7b74a859d6ca1c94c63a938 : SUCCESS in 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7c9f603066c97e5b37a0182669f97d91cd5babc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 11:22:50.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74bfa5dc3cfd46f2bbc172cd0a910fb2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f9c672c9694e41918aee323b434c4cbf : FAILURE in 11m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b73c44bbba744e33a3b17df2b3d181b1 : FAILURE in 10m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2c72ba40c34436cbb944b5e41fefa8b : FAILURE in 10m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f283fc41f72a4025bedf3621072556fa : FAILURE in 10m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ccbb6c538f5f4090afc8c286adb3b8c1 : FAILURE in 11m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4cf91a6dd914780b6df0e0576dc0247 : FAILURE in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46d6de22e45a4680b534a4a5c0d8e480 : SUCCESS in 38m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e759484d8f49609a45775fbb07f3df : SUCCESS in 12m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10b67083a83e4caf9234c47c61d884e4 : SUCCESS in 29m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cee99188eb7a4d988cfc26e5f21afedb : SUCCESS in 24m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0efe4812c3b4cd7bb72a3d27deb0cdf : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12aff404508c4c3dad7806a765791374 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/307ba07f4288497b8a9ec0c1a2f7a800 : SUCCESS in 12m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d55e3e1cab704570a6be892d476a5797 : SUCCESS in 25m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a646042c64ce4c33b9aedb105b19b445 : FAILURE in 52m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c87c679ec8bb4c8f81a703ba5025cc71 : SUCCESS in 1h 38m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/35707c7041624b18b0e701e51149bacb : SUCCESS in 57m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b365c6e31a54f54b5c40020eee4e63a : SUCCESS in 1h 03m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1dd59ee7b5a240e2b64d58dcf98574a0 : SUCCESS in 1h 19m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5890fabd25724f80b3668a544ef4f946 : SUCCESS in 1h 23m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfe03247e1bd4e738e9002fa2e3d89a7 : SUCCESS in 1h 12m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/994738c777704390b28fba41ac01753b : SUCCESS in 1h 52m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4bbbc0f0f15c4f858e2aeb8a7754a84f : SUCCESS in 1h 54m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/57de992ce84f49898ff2f50589a8198e : SUCCESS in 1h 19m 23s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ce6b0574e9ed49f48027dde22ccc77d9 : FAILURE in 8m 46s","accounts_in_message":[],"_revision_number":7},{"id":"418faee0286c0fa96c52d437845cc3f8e392cecc","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-26 05:56:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"29f96565f9f4d0cd69a77e5529da1e2e486a9f02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 06:44:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9598181ea8d14813a4007fbc21e622dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de6f93664ff948e08ed916c79f06d5a6 : SUCCESS in 18m 14s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09bdd95227d74e5694adfc92b3e150ee : SUCCESS in 44m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5c5e252844f45f2b7eccb030ba3df20 : SUCCESS in 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e2b0239f7d7c7a1496aa91f35258fb4b4a91e2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 08:19:12.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84fc7c5a3d354e048b9f9503c0a8311d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0bb63d142e2b417087697c9510b21300 : SUCCESS in 1h 55m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5e24259f88bf4103b9efd4b9c66f19ba : SUCCESS in 2h 09m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/882e16a982ab40568f8d4446bdbc4add : SUCCESS in 2h 19m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/315ace1ba42e4c4c8575918cf7394318 : SUCCESS in 1h 56m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8508ed97039a43bcbc1be2e358ec5f45 : SUCCESS in 1h 47m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a34265ec2684a598e38cd57c3b3e6ea : SUCCESS in 35m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa15fbd309564641ae9185b490f2d087 : SUCCESS in 29m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853d9cd2fad84e4ba5c95556e01333de : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69eb56c5507145bc802bbe8f4dce281d : SUCCESS in 24m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f5e8ff3debb43f18aab4897117790bf : SUCCESS in 24m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43ee9c8283da45c6942d8221055ddd9d : SUCCESS in 28m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0d5c464ff442a6a41165835946b661 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68beb7e82cba477486d919229fe6a16a : SUCCESS in 10m 34s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/47e8808aaf9046d2a1a182db0256af0d : SUCCESS in 18m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed4587f1ba27477388be641b40cdafcc : FAILURE in 56m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/342dfd59a28a4ab99e81317a8cd7a33f : SUCCESS in 1h 43m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e562ba5ecc984b3fb684af5449b9c1d5 : SUCCESS in 1h 02m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6349e37859644c9a80b14c36cab9087 : SUCCESS in 1h 04m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/211b12a4a00b4d41a8e104e19f2a9738 : SUCCESS in 1h 36m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c4da2b6869114b34ae7c43cd0811c0b7 : SUCCESS in 1h 28m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4740618f420a4471aa96cb3da07d757e : SUCCESS in 49m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8dcedaa6c2da413e9612341cb9aed98e : SUCCESS in 1h 51m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a72093c39a034bdebab05bf62644cdef : SUCCESS in 1h 52m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5a51a64e97324c77b575dff5cb76bd57 : SUCCESS in 1h 19m 47s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/80168a43d78c401cad4d739bf09ce82b : SUCCESS in 1h 30m 27s","accounts_in_message":[],"_revision_number":7},{"id":"4abaa391d3128f99ec1687b2d80111d662c0c350","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-26 08:32:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"73a4081ca6b4f0a394ef450ce31d37a6ea4dc294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 09:26:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cdc7cd8308c4c8cb83e65fd343b413d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4ae04859713452ab497b860275de893 : SUCCESS in 48m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96729f82517b410fa70e16a473f051f2 : SUCCESS in 48m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b86cc2765c44f2787b7acb935af1313 : SUCCESS in 52m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"39f8336e896c3cb48168873fb2c5fb4fc6869532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 10:59:45.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/494d9123e5ef40769a1ec8ac9ff583db\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/37750813f1bd40ea8909ebc22545af58 : SUCCESS in 2h 25m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8b6668d628294d919ea4075bcc4ee3a0 : SUCCESS in 1h 41m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/68f010c5429645c0b9d7f63282f865f6 : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/63c7532c3fe04b6fb97e217058d3031c : FAILURE in 7m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2da01f86e38045b4937e1e63fd05bcc2 : FAILURE in 6m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c026f4923fe448d1bb1fde9d4368522e : FAILURE in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/126dfa7367e5422d83093e1a885ceb16 : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f0fd7edf564fd2bea31b2e676e242e : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c895f51b1f94175be12c0ae400ef14c : SUCCESS in 25m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/249e21ba91c445e4bf63eaa788165da0 : SUCCESS in 20m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd77e4e66e364db28e4a32c353340105 : SUCCESS in 29m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168a04cf94ca4b1595a3a65f548e9c85 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a7d5ce2c50d44ccad1c45d67bdb00a1 : SUCCESS in 9m 59s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0e347cf9461c4bffb55f492985557a7f : SUCCESS in 21m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/69769a892eb44325bacd4fe4237b6a4c : SUCCESS in 1h 01m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09fe15f8a0c14317bf518c92aa56a342 : SUCCESS in 1h 53m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9e15120e4004d72aaab490beb93c3b7 : SUCCESS in 1h 10m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e1ea395edc245679f749cb7458488e7 : SUCCESS in 52m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/19bdbda4acf6403f923a43df33c7137d : SUCCESS in 1h 48m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5533e8ed4551403ab7863e6b3750acdf : SUCCESS in 1h 27m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b4f885a78c4d45098704c8cd0d66b0cb : SUCCESS in 57m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3283b59a272746e2b7888a9ddb35cd37 : SUCCESS in 1h 39m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53d443fac2bc4338b0feeef304187306 : SUCCESS in 1h 29m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2beeac4a3553408aa0797d50e16e129a : SUCCESS in 1h 14m 59s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1b799182abf04bf1b69cbe31c9b6aeed : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":7},{"id":"15044c279f650c92a49e995952fba3c7f391d979","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-26 12:23:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d5c2593de8c0874e4abcd0f13dbd6779f6e2fa2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 13:15:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1605063aa8244d039044fb219239dfc4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fd3b0bfc8e543bfbb55d2fd262dad48 : SUCCESS in 50m 08s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96239983f036491fb74a79e7af49919f : SUCCESS in 49m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f6d9fc09a9a4b2abc4a971a842bbf80 : SUCCESS in 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"62105e9a548ab49317ec2559223b12e2cd262858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 14:21:48.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3ee74a8f2194e76934306734acbff96\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5ab011692a3540e689b4813c40c9fd54 : SUCCESS in 1h 45m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc3ba40e52fa4354938fc968dfdd18b5 : SUCCESS in 1h 40m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/009d9a1499a94a71b13764aa6e0dcbb0 : SUCCESS in 1h 49m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0beb55efa65e4a9d8ebce2aad899981a : SUCCESS in 1h 45m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9b03dfbae5b44c1889a21151bcf7301 : SUCCESS in 1h 51m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d1687727e61444cb5e6cf3cbbb5201b : SUCCESS in 44m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4673421ba63d4758966474eab23e5ac5 : SUCCESS in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72368313d3d04d038593ac7173f7efa0 : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c9e42ab91e9461ba3b66227760f16b2 : SUCCESS in 28m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0212013b70754e2e85fb37c086e8f9b4 : SUCCESS in 32m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a10702e5c90478cb21c51e581841eb4 : SUCCESS in 27m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb8b95b5253438d8e9760bceafd07af : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d1984f41a1a4d0c8d4d56d511ad0261 : SUCCESS in 11m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c473bcd074af4305874255b21eec09f9 : SUCCESS in 22m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c552c32f11044cc99be1ef6d9d470b4c : SUCCESS in 1h 11m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f96cbfa78fff466ba3e7324f1f44881e : SUCCESS in 1h 32m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ac29cbe78e144b9a999a23718f51ab2 : SUCCESS in 1h 00m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a2267f73c24e4b32a57a57c2352b6196 : SUCCESS in 1h 09m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0da874e2dd61470d8b4a248a050ebb05 : SUCCESS in 1h 46m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a71701af75714929ac6076193a1130b9 : POST_FAILURE in 1h 23m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83c2398a3e96499eb7772e3a9935c415 : SUCCESS in 1h 06m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5af70ce4c3af406da53445d3d96f002a : SUCCESS in 1h 40m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/631a4e9fc9984ce5ac164de51c9e65c5 : SUCCESS in 1h 57m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9357ccd5f220419298dd79ed216259d3 : SUCCESS in 1h 14m 21s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f1a2150f7017437ab138a66fbfb9458c : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":7},{"id":"a3c75c0470fb30c816e2d14f54dd8d3321aa298a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-26 14:22:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9d7e59928160bdf25642349ce05a6ef154be2d50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 14:52:06.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e148a1eba57d46c59b92d71058e3c5a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/326b26f5b5f64066862955b1660da1f6 : NODE_FAILURE Node request 300-0021067701 failed in 0s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88efc260358b439ba89a374cd6e1152a : SUCCESS in 27m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b509329def54ec3b36ebf3d44b15862 : SUCCESS in 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"36056b08dbf586ec0d51645e74d6cbf65ba90076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 17:08:50.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5338b6d7214040da994c2ced71546bbb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/40ec1abf8d494e0e825fce76ec106b95 : SUCCESS in 1h 30m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c927ef398d154d40a49841e71b02f2de : FAILURE in 1h 38m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5028e5bb7d545ffb40723d38e87dbca : SUCCESS in 2h 42m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/27d022be1b504819b26044eda2ea0aca : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a80d415fa9e4325bedb8d5c618b7f70 : SUCCESS in 1h 38m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba37ddbffe4b4a46893d890304160fb0 : SUCCESS in 39m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b9a53da4b6545a896e5d5f626a5f2e7 : SUCCESS in 36m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00274c79b86b4d029c94364eb3c15814 : SUCCESS in 14m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fbb92f92fa3412faaf46451402ef9ff : SUCCESS in 37m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c60df4de532b4b83b32aa0009b80f4d0 : SUCCESS in 28m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb5f8f823ab74bbc821217d8f3e90a00 : SUCCESS in 27m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d64f7c446ff446b835716b31c43c2c4 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6afb7acdaf414756adf741fc03d3bf11 : SUCCESS in 12m 09s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/66479182317e4be591b5d700128d6b7c : SUCCESS in 24m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7bf1e55f2444c32aa4b6ae81912a956 : FAILURE in 42m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4de099b89484abea9176a6b55fa71af : SUCCESS in 1h 37m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d97c05751c6749658b5804d15a4cde9e : SUCCESS in 1h 00m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f6d6274e5e954482b0b2b49525663a4c : SUCCESS in 55m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cceb272812bb429c98ef2b8bc7193dce : SUCCESS in 1h 49m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/37a4220bba3a4e20ae961d180cee745b : SUCCESS in 1h 22m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/11b0c293138d42049bb1dc6ceb2167a5 : SUCCESS in 1h 18m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0ae181614103460eb73c333954e0236f : SUCCESS in 2h 05m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad2d6c0c11b0427ea36e80c0fea089c5 : SUCCESS in 1h 53m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/908a85ba71484354b95442d33ef0416f : SUCCESS in 1h 12m 10s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/16bd71c901364ef1b5ede692ad3d81df : SUCCESS in 1h 20m 09s","accounts_in_message":[],"_revision_number":7},{"id":"6228343c5f346e2173580dcb5279cd677c24d248","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-27 05:33:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"78c0501bc31e22931e337a6199b5654477f71a12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 06:15:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b98f5dc98f4e06bcd72aa007f70315\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75bf9959a1b64983be4736b349760cef : SUCCESS in 24m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0828c7704b4e48dbb705658ec0eeda41 : SUCCESS in 41m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d70860570fb74333aa99f1832748a503 : SUCCESS in 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b7601253a5bd42cd72f14e32e45073b53473d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:44:14.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bf3e79ad30a4547a0769ee81b72e6c9\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0f9369dc8ec54c7f8aa5a8d23dd7d1b2 : SUCCESS in 1h 57m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5dff272b973140f1a892b17941b7a0b8 : FAILURE in 6m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b1cd11cd36048a8991cde601ea51c66 : SUCCESS in 1h 39m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41b63549b01c466796c8ed58cb4ffda6 : SUCCESS in 1h 34m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/62f1114b50bf40beaa40d15e20c752bf : FAILURE in 6m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30b2390f0e0246c4a76eb630241c7106 : TIMED_OUT in 1h 02m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b81d5e2c9b81438fbe7760c2de1750cf : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d85c6427e144d2bb881f586da36f52 : SUCCESS in 13m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7418b1619df945de99364d54fd2be5c4 : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc9bf0e0c827486fae14d4434e5fad4b : SUCCESS in 31m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb3bbf7755a5497697b43e0feaba8aaf : SUCCESS in 20m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f354da8ce95b48a99886cf637acee029 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72064fb8fa53467faebe165139e9d3f5 : SUCCESS in 9m 01s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e8a6a70a049b4ce1b2f4aa6b8a502e2f : SUCCESS in 19m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff4ee37aceca4bd48ffc91175bdd731f : SUCCESS in 55m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f48421a4fd3e4addb8025fd65289c6b1 : FAILURE in 2h 00m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8490cc4b83f94d5891a32616fdd1b647 : SUCCESS in 45m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11bf35cff3bc401aae7bc3bfe7ba9baf : SUCCESS in 56m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/65c8b175cbb0482eab288bf1e1ab5c87 : SUCCESS in 1h 30m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/db7d2fe2422740c1b8778ff130c83c73 : SUCCESS in 1h 30m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/638a24df0037475b876bee001e74eb6b : SUCCESS in 1h 13m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e86a5943cab4ca196ebba3aab7a4921 : SUCCESS in 1h 45m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3d91dcbe44fd4fb1bcb996bcaa394b55 : SUCCESS in 2h 02m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/934d0cff3320423f9ea645331408278c : SUCCESS in 53m 58s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/9063de01c130452da8845a63759a8f41 : SUCCESS in 1h 21m 00s","accounts_in_message":[],"_revision_number":7},{"id":"379855ba489af5a6e7e60150d203bcb26a6f760c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-27 07:46:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"93a581045dda6b129b2c9b23a0d4de3081b20d8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 08:29:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b615d224736843febbd3f75ce2908d81\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4eeab8808fb414aafa3cd1fcdef48f2 : SUCCESS in 41m 27s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8713cb809ef426b8020292dc2bc94af : SUCCESS in 41m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43779c80f50a473cbe0d127b14a3ddaa : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c0ed841a86e053d0ae91264e8ef0c549f3955a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:14:29.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79385440f60f4b28bffec5600091138a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/98deb3ff6d484d2c80e4ae81b5c35f88 : SUCCESS in 2h 25m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ac41f6ca59154dd48517163f0e2c166a : SUCCESS in 1h 37m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db8b3104e67448c4af3543fc80c9815d : SUCCESS in 2h 06m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cf141a9e033c46ffb051c57fedb8bbb7 : SUCCESS in 1h 45m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b1f4592e81d4f95b086f696a8211e55 : SUCCESS in 2h 03m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ec2c076d71f48b99b276d44855f296b : SUCCESS in 41m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a030f509143a47c4b633e6d9fe8fb1c8 : SUCCESS in 40m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c050c8bbe22c4bbeac9f828ae795c070 : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d6568ce6014fd28cb918243aec9cf9 : SUCCESS in 27m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/713e056fe9c64226bec4d077fc0b58a7 : SUCCESS in 27m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e199a060bb04f8fb892c87e4c870fb4 : SUCCESS in 21m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ce5d73c6f24199bb75471cb42482ef : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0c7013eebcd4683a9c69bed1abfbf6b : SUCCESS in 9m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/06c6daf7e8744d8f80353323f8cfc5cf : SUCCESS in 21m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/133c6763219e49c58cf8c1ab99b0ed2f : SUCCESS in 51m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ce49cedcfc24556afef289821a35cdf : SUCCESS in 1h 38m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/465d0a0ad76846d1a2faa33a3c8c2f57 : FAILURE in 41m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b165cc69c8bd42af810d9900c16dfff1 : SUCCESS in 1h 00m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5c43a08db3984f6fb7a0c5f48173f51d : SUCCESS in 1h 45m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d9108509b5345d3b0985434dcb2e6e8 : SUCCESS in 1h 21m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d1ea9fb323854724a1d86f0cefe5af49 : SUCCESS in 58m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fa25b144795a4a75b1830225bed20c03 : SUCCESS in 1h 30m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/463eefcc33de4e9ea985e59940862d9f : SUCCESS in 2h 00m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/62f01d75687f40899d139320d51606f3 : SUCCESS in 1h 26m 56s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/6af05ca119314cdca925b4f5afb196f7 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":7},{"id":"413978521981afdf3cfa04b13374d2e9660e798f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-27 10:20:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"582300258f205465e86f92c138c799aae1db4e28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:43:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1756850cd4649fc86205707081a6a93\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d38c4504c4514ded98bd7eedef7d8fce : SUCCESS in 19m 01s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab67d8ef8b6a4ca7bbc361499dc200bf : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/00c0a122d31246a5b9d9e9fb8603d62b : SUCCESS in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e98d1614f21a780d09dd86dbc513777cacf3f871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:47:16.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2706bdf32ee494ba5ebac6d1fc19e22\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/95cfff421edd493e94b482e956a4422d : SUCCESS in 2h 14m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d330ae67db1c4fa693c01066fbb25f08 : FAILURE in 6m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5b18449e1ee4fb4aa66a1b8a4bee94f : SUCCESS in 1h 33m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b045a622e2624bf19204da0806866e54 : FAILURE in 4m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b71c1b9905674aa78c484334be22af96 : FAILURE in 32m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d8187e3189048ef98161db2266c3f7c : TIMED_OUT in 1h 02m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3192c430039d4a3baf156aa89c8f98a8 : SUCCESS in 45m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fd96a56594422a9400de3377be36f2 : SUCCESS in 13m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f008c4e8fd3144f1924a3462a30390a2 : SUCCESS in 28m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6e0362e3439428d96f4a35179484fc9 : SUCCESS in 26m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11e0f8ae364941a2a562e6c9a428fb32 : SUCCESS in 32m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c9c6c1e93c46898bbfda8ccaac4f20 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07daafa975cb42179a88102f22dc7325 : SUCCESS in 12m 19s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fbf1cc8fc4784ae99777b5d4c5eb5bcf : SUCCESS in 21m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4151b38325264d7db9fefab6481b524e : SUCCESS in 58m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/56342c5cb5e54c7aa289214bc949e433 : SUCCESS in 1h 31m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaa77c4f8bf04b39b7b3c6f84cc00e4e : SUCCESS in 54m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1778dfd3abe444929aa7f9c6e52a7b36 : SUCCESS in 42m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/86b6d00d6eed41ca995a453e229fa359 : SUCCESS in 1h 23m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f504da8b45648f79449618a010738e9 : SUCCESS in 1h 23m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/13ef173a3428409eb9639776dd6c8871 : SUCCESS in 1h 19m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f234df724c8a4748965d655a0f3c1a23 : SUCCESS in 1h 43m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/36b9f37d9f7842129136383741f576f9 : SUCCESS in 2h 13m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3fed7442aa594c46a14c5158a05ea98c : SUCCESS in 1h 07m 29s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d76a439cbf624da2a3274b43d12c2ee7 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":7},{"id":"62490f65ac3bfad80cc4d9d4ae17c5dbc7b2b313","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-27 17:08:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"89da6882446aab10ef5642c9e5b4ddfdddacb27c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:37:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41cdbcf859914030a3620da93b001bd0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8388f925ab943e1a4e89ea81012c900 : SUCCESS in 28m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92fca036d12c4a1cb32cb54f6881f751 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4482917c58a242e19b425a3b4c8dc5c2 : SUCCESS in 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"117a32423b9deff6c392cd82392f8b98250e97e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:26:47.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2255a8e71c254501890bfd4886307ca7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8299cf8f7f634b76a0d9646b2b6cf68d : FAILURE in 1h 46m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/76e25537313d4f8f8545bd4f401f9824 : SUCCESS in 1h 26m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bf0c894401134910861d94e1a56778f4 : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd1e4b07ab1c41cd9c560fc15b274ddb : SUCCESS in 1h 27m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f75ae2ef94b7420d88406cc8a9fc61f0 : SUCCESS in 2h 17m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/691dcfcfd6d842549619a3b2073171f1 : SUCCESS in 44m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fde42c1bd3d14ff89e1f4f1f08087e27 : SUCCESS in 41m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84910d75829f419196af639da29a4afc : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3adc0146dbd5410d99d5898a58418c0d : SUCCESS in 27m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9656b853e4d442938616010e922b5876 : SUCCESS in 23m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32a7ac4c450746bc9db4a4492bab952b : SUCCESS in 23m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f16f0fe494e47409bdff5aa9f4f75f5 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07de07c7f6a24d15bd44b457440bc859 : SUCCESS in 12m 13s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b51cdb3e524243c5be8c2c6fbca418bb : SUCCESS in 20m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b76ce019fab441895745138babebb03 : SUCCESS in 1h 04m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e33937fb58244ab964140ec50a92813 : SUCCESS in 1h 41m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7a1ebe6ffe04828bbde4e131445ce61 : FAILURE in 39m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2ecdf200f2934a74866f98dede4920b1 : SUCCESS in 47m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4509a3a85a3247a2911a7f9967241cc3 : SUCCESS in 1h 38m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a87c207f5c8a41348c03c9de1b8d8134 : SUCCESS in 1h 32m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/08e1bfc59dba40d186bdc13e43ff3ba0 : SUCCESS in 1h 11m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/53137ee75dcd470da8186ee162304723 : SUCCESS in 2h 02m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a967f2d059844d2ab3948401bbb03a2 : SUCCESS in 2h 02m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/94943d3684f04a6992c0fb888df072f3 : SUCCESS in 1h 00m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d2733f78948e48ca8d87f51337792f46 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":7},{"id":"56f64177892efa5efcbe46dfbe98a9f738c13a37","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-28 07:28:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"121932f4149f9817480ec2c03d0f5f9ec5f83424","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 08:17:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3b8715101641eb9dc496dbffea91ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47e41a5f1522495f9bc5da8ffe353ac1 : SUCCESS in 18m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4b9d756fc974cd09f2358f73a650fc0 : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0228e2a82a1a45f3bd11859a98029c66 : SUCCESS in 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1ba638ddba77840aadb441db4108ff4f269b966f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 10:16:59.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff64428805354de384b21e3309c10213\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/130d3459e3f1417fb565c4b579d04776 : SUCCESS in 1h 35m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bf9ab153dcd64a7fa7cda078d0084ddc : FAILURE in 6m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d70cc45e26bf4cc8ae4c37534aed6bc5 : FAILURE in 5m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/32b6b59fdfd3444b84714e0a094d3ad1 : SUCCESS in 1h 36m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61c0cb506d764ff4ad30ec1c5109edd8 : SUCCESS in 1h 33m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9767efa43fcf4918843048f2fd7c3200 : FAILURE in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2539fb0736a9448fa3e8c881c83eba03 : SUCCESS in 39m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb40bd90995e47a2810048affa7a31af : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/003c7bb9096d43e1a1cfc7fd9abe03de : SUCCESS in 28m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b68a32c446ee47bd9f4b815ea23855c6 : SUCCESS in 27m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0c22e30a0d741b68c34f228d08ec831 : SUCCESS in 26m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc38b7f7a704a2b81e51f9097306eff : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e07d8cd2894418c8c453f1277f97bca : SUCCESS in 13m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f8c47f0371de435cb42978065ccb3206 : SUCCESS in 24m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/70263fe5c30e4cdcabbfa7c19779692a : SUCCESS in 1h 01m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b37d155388584bf9bd7602ac3b0f52b9 : SUCCESS in 1h 44m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c253556bd184319a7317f89fb6d5d9a : SUCCESS in 1h 01m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4cd209e3e2f84b299e9c258771d4a806 : SUCCESS in 1h 01m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2fefcf506b84399a6f9c75bdc32cdc9 : SUCCESS in 1h 31m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da016fbcf9e3489bb20fade82d3320d9 : SUCCESS in 1h 32m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cdc271fe4efb4dc7bb130205ddcd93eb : SUCCESS in 1h 07m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/071e8d7c7bf446fb97f8ccdbd122dfea : SUCCESS in 1h 41m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ebb2aeb82b0843919d4e19f5ff7aa4b7 : SUCCESS in 1h 56m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52efdf56a3544ad6bfc6621787d6ff16 : SUCCESS in 1h 04m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/108c0c1473d542c2ae26c17910f62472 : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":7},{"id":"316597174c206cb293212ac8bfa6c969f7b649d2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-28 10:18:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c4c831076d0729214971c9ef29f4d40b5cdf02aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 10:40:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c39a95058b2142b99fc60622db58b6b0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d90bb97eccbf4a4cabbc9fa8531e5578 : SUCCESS in 20m 33s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f9f25a83e4e4876a8513a0b69f0ed11 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/033026733bf648fa9ef86f717662ed04 : SUCCESS in 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bfb75cc50904b72e55dc74a9d7cad24314a20681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 12:23:41.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfd25d68195c4dd489cf8db7cb5a18f7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0aec0e84746146b683865e506d1af724 : SUCCESS in 1h 50m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/299c7bc9c8824e7b899609a91d4977b7 : FAILURE in 4m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2de01ab74004ad4ba7ab234e85e071c : FAILURE in 7m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7bbf7446f94940df818ae3a59195ea49 : FAILURE in 4m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95b5528600b24283ad54746642663566 : SUCCESS in 2h 01m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df2fc005a2fa472eb1728beae614437e : SUCCESS in 35m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79f4c0ffe02041dcbaa0fabf5442de1f : SUCCESS in 37m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87cd796796fd48c7b70b648899b7b71b : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5f1d70d30a24cdaa93573610df21d19 : SUCCESS in 24m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7db77d3a2694ba793e922d813a650fb : SUCCESS in 26m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ed592fa12f422da74c1ed727065c4f : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2feaf95c6d04fbc8c0eab0bdcd86bbd : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f32865c96a6a482988f497922c46a6dc : SUCCESS in 12m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8520f4cf01c54b1e9274f57e71baf8ba : SUCCESS in 22m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/08a827998087465aa6f620da1b88b848 : SUCCESS in 57m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b555705bb8149e681ee3693781b29cf : SUCCESS in 1h 31m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74db5dd51ca5420e9465408630004c2c : SUCCESS in 50m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/09cbc1e405294822acc5fd9d9a40fea7 : FAILURE in 37m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b2959fb38074ee292182d7951c30ba2 : SUCCESS in 1h 29m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/131631e3a64346988ad7b06f1288cd51 : SUCCESS in 1h 27m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/844060d4505b4b9a98df52df2cc8ee97 : SUCCESS in 1h 09m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fcf22abcada34cef8e8d6384578805d9 : SUCCESS in 1h 43m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/357e7bc0a134443f88b61e0984e90006 : SUCCESS in 1h 39m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba8b10feab4e4cdc810235712a296895 : SUCCESS in 1h 20m 57s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b65f7e3592164b03905e9a6191c03efa : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":7},{"id":"6af484f6d5a8784eaea7b3de60cfef09c3e63a90","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-28 13:10:28.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a0cc12e151ad96dcd952eb63cd3bf868207f176a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 14:07:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59e9e462249342c591ab8654ba17c50f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f27fe1897ec84bdabe0f55b06356295e : SUCCESS in 24m 39s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19b39f77c1ae43f88c212ba5f5d6349b : SUCCESS in 24m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca7b32a0eaec4ac6a4068c038b5fdb90 : SUCCESS in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0e9d74a3ab7af4ee84021755489ed738c0c3b546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:19:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11cbdb1f382140489cff2752e48d982f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e98f9dace69f43aeb296f7623ef8b99f : SUCCESS in 1h 57m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c45a19ed29194866a7b894bfccbb06a7 : SUCCESS in 1h 52m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c0ef86b68964bd8aa500838104c5c30 : SUCCESS in 1h 24m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7465369ed7e34e21b04af89d6df97956 : SUCCESS in 1h 12m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/31688c84b5034191916677672001cc8b : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d367cf1464a34d9c9dbc15ed8097e752 : SUCCESS in 29m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa0365ba899040bb99192d4b85e90b1b : SUCCESS in 36m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7175414e5a2b4b96a16576ca99cd9843 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/564ced30718f4b71a612d0b56c7ffe17 : SUCCESS in 25m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c068f818fa14067ad9237aa3b067b63 : SUCCESS in 29m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5241af506de4a5897b23e0466c95545 : SUCCESS in 28m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab5e81e75714e0892109a9390171542 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/970150fdffa646d1aa229ba2f7c579f0 : SUCCESS in 9m 24s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a1325065e480415aaa508deeb7327a54 : SUCCESS in 24m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67d47215e22842d4bc8462e999fe877e : SUCCESS in 1h 00m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbff494bcd3847baabe743ff44eced05 : SUCCESS in 1h 41m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb314b87229e4ee1a101f26cc43f6840 : SUCCESS in 1h 00m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e6529a7b97314b46a722008022f9b07e : SUCCESS in 58m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/51addcfd61eb4e69b51e8b46f75c64ae : SUCCESS in 1h 44m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c1b3b8e92e14b8f9498be268249174f : SUCCESS in 1h 40m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b30ecc1325ae4625a4cca1bf3f6c8b7c : SUCCESS in 1h 15m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0e17c2292c646c7ac1d8092abf7c770 : SUCCESS in 1h 49m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61c9427a9d8c4f7496f5af1d2c9d38ca : SUCCESS in 2h 01m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/38fa4d37c19d4139bcb0c771e27009c5 : SUCCESS in 1h 18m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1463d1f95a8b4039acac195d11fd5ff3 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":7},{"id":"0a83407d28cc79f7c9d23cbd682c5dab5d083089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:20:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7ebfd9f255f06eae38b227d7c5e8b4b070c05d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:36:33.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58db981439b24a00aaddd2d551cd1ed0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b5dd1e8fc27a40bd92369559c5259cd8 : SUCCESS in 2h 13m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/947a18cdf65d4ef18513c32548ed6859 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/639716dd50034e338086078a45a21f0e : SUCCESS in 2h 02m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/118c8bb5a3784cf48bb46f92474cd809 : SUCCESS in 1h 49m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b03c9aff98384200ae9d66d7539a5eb9 : SUCCESS in 1h 35m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874aa1ea80f74010affa4e25c212f53b : SUCCESS in 14m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f489e30c61644d49974f2108854b18f : SUCCESS in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6fcd406e0640b7af6545d9c405403a : SUCCESS in 26m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c934141bc0d48b0bb096afc5e00905c : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9192d15e700429ba9449f5183c635c3 : SUCCESS in 14m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93985ba598e54a728f92e11692a922cf : SUCCESS in 14m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca962e0886f5448d87c14862867e3506 : SUCCESS in 54m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57a00f26f51c4a8390fab072ec87ea1e : SUCCESS in 1h 47m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/028da04bb88c49f9989e3b326acc1e38 : SUCCESS in 1h 45m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20f9140489e648b381d0ea031ff18fe3 : SUCCESS in 56m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d580593c285f4adb889732d6760d197d : SUCCESS in 54m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b3a2522232d74a4284afe894955fe618 : SUCCESS in 1h 13m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1c7250cc18046ee99bb3b78bf2f39d9 : SUCCESS in 1h 14m 49s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d508623c7e3b416fa01c0716e3f1763f : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":7},{"id":"14056d734505acb5f1409894710cb6e39a5b9d86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:36:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"89619ed9b8474dabfc153f3b11fc0d69ec70e864","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:38:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7819bd7d2f46969d18d266abb08f38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f987f657b6694415b889f5672a0ec375 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e5b666ac3404157b448b89ed6074a60 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7dfbdf65a71b7da2865d475cd91988728f734652","revisions":{"6970c0d6b1c617da7253e5dc7c9920ce563f6041":{"kind":"REWORK","_number":1,"created":"2023-03-17 15:01:01.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/1"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6970c0d6b1c617da7253e5dc7c9920ce563f6041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6970c0d6b1c617da7253e5dc7c9920ce563f6041"}]},"branch":"refs/heads/master"},"8c0e78bf8c399498fc9430bfc402e16e0ce1ad05":{"kind":"REWORK","_number":2,"created":"2023-03-17 15:34:39.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/2"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 15:34:26.000000000","tz":60},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c0e78bf8c399498fc9430bfc402e16e0ce1ad05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c0e78bf8c399498fc9430bfc402e16e0ce1ad05"}]},"branch":"refs/heads/master"},"3499232e2d5a448dae38da23818ba4245e275dce":{"kind":"REWORK","_number":3,"created":"2023-03-21 13:57:44.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/3"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-21 13:56:41.000000000","tz":60},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3499232e2d5a448dae38da23818ba4245e275dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3499232e2d5a448dae38da23818ba4245e275dce"}]},"branch":"refs/heads/master"},"e8c7ad6465e6548c1aaf7700b76f756774db3462":{"kind":"NO_CHANGE","_number":4,"created":"2023-03-24 07:08:20.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/4"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-24 07:08:07.000000000","tz":60},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8c7ad6465e6548c1aaf7700b76f756774db3462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8c7ad6465e6548c1aaf7700b76f756774db3462"}]},"branch":"refs/heads/master"},"9c82dd684996cc005ef135b0c5e67e74be22b471":{"kind":"REWORK","_number":5,"created":"2023-03-24 07:08:46.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/5"}}},"commit":{"parents":[{"commit":"85b82d4452ed3199c7f1f7c2455d2a75faaa2991","subject":"Merge \"Only create a frozen  Row on matching events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85b82d4452ed3199c7f1f7c2455d2a75faaa2991"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-24 07:08:39.000000000","tz":60},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c82dd684996cc005ef135b0c5e67e74be22b471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c82dd684996cc005ef135b0c5e67e74be22b471"}]},"branch":"refs/heads/master"},"d97849532cc0bc5da35099a10921225afc3e5510":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-05 05:59:21.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/6"}}},"commit":{"parents":[{"commit":"b1ca854791fe391ccbb7ac3a019dc9b9d46a3995","subject":"Merge \"Mark \"test_multiple_agents_for_network\" as unstable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1ca854791fe391ccbb7ac3a019dc9b9d46a3995"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-04-05 05:59:21.000000000","tz":0},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d97849532cc0bc5da35099a10921225afc3e5510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d97849532cc0bc5da35099a10921225afc3e5510"}]},"branch":"refs/heads/master","description":"Rebase"},"7dfbdf65a71b7da2865d475cd91988728f734652":{"kind":"REWORK","_number":7,"created":"2023-04-13 10:01:29.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/75/877675/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/877675/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/877675/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/877675/7"}}},"commit":{"parents":[{"commit":"b1ca854791fe391ccbb7ac3a019dc9b9d46a3995","subject":"Merge \"Mark \"test_multiple_agents_for_network\" as unstable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1ca854791fe391ccbb7ac3a019dc9b9d46a3995"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-17 14:59:05.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-04-13 10:01:18.000000000","tz":120},"subject":"Add support for localnet_learn_fdb OVN option","message":"Add support for localnet_learn_fdb OVN option\n\nIn OVN 22.09, the option \"localnet_learn_fdb\" was added so that\nlocalnet ports can learn MAC addresses and store them in the FDB\ntable. This avoids flooding issues for VMs on provider networks\nwhen port security is disabled\n\nCloses-Bug: #2012069\nChange-Id: I93574b4fe9a79b649bfe755cf7e0697ccc7eb83a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfbdf65a71b7da2865d475cd91988728f734652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7dfbdf65a71b7da2865d475cd91988728f734652"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
